)]}'
{"id":"openstack%2Fkayobe~825832","triplet_id":"openstack%2Fkayobe~master~If604adb0815f583c3fee4772f90ad53b84180372","project":"openstack/kayobe","branch":"master","topic":"network-mapping","attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-02-07 15:31:13.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-04-12 13:08:30.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"If604adb0815f583c3fee4772f90ad53b84180372","subject":"Rework network interface to host mapping","status":"NEW","created":"2022-01-21 16:02:45.000000000","updated":"2026-09-28 13:41:50.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"6942b431f8a9ee997e41c733d0e99c5290f03698","_number":825832,"virtual_id_number":825832,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-28 13:41:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"reviewer_updates":[{"updated":"2022-01-21 17:41:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-07 15:31:13.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-10-29 14:48:43.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"}],"messages":[{"id":"c5fc439a1830c7bb9276adc5408d70772bdf860f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-21 16:02:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50cde9b239690875048e1687940983d659d9dcf7","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-21 16:07:05.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"c8060682e524bfda76e9427b29d40216f71b8726","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 17:41:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a697c7d74478472185a6497edcdcfb9a : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2fd1cf07fa4440da7593aa76c61e22a : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66acf76b4786452a9474076f4586eec7 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0eb6d7aca3240d3b0bb734b5bdc2ad8 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a6355f6b0bd4fd6b798c9a6ec15c3ea : SUCCESS in 10m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/afe0edba984b441bad9aa2ff3c84908c : SUCCESS in 6m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0ccc97252ac34459b518ffadcee0557f : SUCCESS in 13m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f30772716e8c4a729b6ab22bfef61b46 : SUCCESS in 34m 06s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7b2ceb7173364aa7a1943e6e570e39ca : SUCCESS in 54m 31s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16855313fada4e8abfe45a2b2bf0198e : SUCCESS in 48m 57s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c1357f0ec425463c948da0ea15321c72 : SUCCESS in 1h 13m 24s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6853077046054924af8528e92d6c1182 : SUCCESS in 21m 06s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17567a55d70c40038e9de3d1b27db186 : SUCCESS in 14m 58s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ad81423801554bd5a1bbf5642521e11f : SUCCESS in 1h 35m 52s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd81703aa567471d9b55752347e9ccae : SUCCESS in 1h 04m 49s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e6f43766ddc742b8b44dd4bf667b070e : SUCCESS in 47m 42s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71431e177cbd4b95a9210e28f3c98125 : SUCCESS in 37m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0d8fd4868a1b468784ddb830d480058f : SUCCESS in 26m 15s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f13f8c216a8941d9b58ccf911251fc7d : SUCCESS in 25m 40s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9f2c2b201501459e9e46292b2438fc12 : SUCCESS in 18m 12s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fa0992cc8ec4b2486eb0de7e0307ae9 : SUCCESS in 12m 20s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/d4d3db5262b346fbaae52ae0fcb3a3d2 : SUCCESS in 16m 00s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dba33a42c95b48d49c905444bb9df6ef : SUCCESS in 15m 04s","accounts_in_message":[],"_revision_number":1},{"id":"3e3251842faf9312122b6ec3a770f356d086b1f2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-29 15:55:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ede72763367f8f1b6e62e1258b94dca267ed77fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 15:56:15.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kayobe for 825832,2","accounts_in_message":[],"_revision_number":2},{"id":"e00b17ee7c63511da54811b5841c4b8a7646db8e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-29 15:58:33.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"9c6d3d3c717a96e0a52711b307df888fb6e7481a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 18:10:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24527ab8127441749e1b075eff870552 : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0501358cc21c4a3ea56631919ef2ee88 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aba857da7f6448db8e82b970d4e5c2b0 : SUCCESS in 7m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f671c403af504591bc8f8e24cd706bb1 : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1ea7260e94f437abdb55a7d56003056 : SUCCESS in 10m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/529e8c3b07674fc79d199a19c31881d7 : SUCCESS in 7m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fc747743adca493f8efa47c1142f686d : SUCCESS in 15m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ea029c2cd6f84004a68baaa000100841 : SUCCESS in 30m 51s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9943e155be8b4d9b99cb6865147d4d3e : SUCCESS in 1h 10m 24s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/3fe9f46dd9114c71abb1e659f4f7d1d6 : SUCCESS in 1h 07m 02s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8790902aa4d45e4b3d0a38c6351e271 : FAILURE in 10m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/af4819bc19ce4c71b9d250543cd3be7d : SUCCESS in 1h 05m 48s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ce3a98e5840c404f939343fded7b1f77 : SUCCESS in 25m 51s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/97d63b4f7ef34a48a16b229c925385ac : SUCCESS in 25m 47s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1b0a41b8b6e4c368d58d3ec24e2d93d : FAILURE in 10m 24s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1ee16f55d346436488f33b6cfcb1d304 : SUCCESS in 1h 55m 59s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4758fa45f5c4e57b0de4490ed2854bf : FAILURE in 57m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/443c1c2bb2b74fad82a553c97d81c5d3 : SUCCESS in 46m 59s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/d38d6da5d5214d99bb1a560d122046df : SUCCESS in 40m 55s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d3943a1544c4db1a3852931dcf4ec0a : FAILURE in 10m 24s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9317bffd194d4d2cb78af8021564c96c : SUCCESS in 37m 09s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2556b5f2bbd440b7b9de7370f929f1f5 : FAILURE in 26m 01s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7d913c0ee9e644f9b21a27ff0ea60669 : SUCCESS in 16m 39s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6dea427397e2404bb4c112f3ff9b051e : SUCCESS in 16m 16s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd32869e3c2241d487c89c25238f276d : FAILURE in 11m 13s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e39ff3863f1a4cf289c08b493adecd79 : SUCCESS in 19m 05s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e1c37cd3625546efa8fc8000a058dd1d : SUCCESS in 19m 49s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3980399beabf4576aba6f264811e8474 : FAILURE in 10m 06s","accounts_in_message":[],"_revision_number":3},{"id":"604b2f01880b860759ab8a0933d320e774e5ed93","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-31 14:54:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8efc06d464133935cfde4472d3848675bd7229ec","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-31 16:06:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"788d312bead9e3874b469b796c4d65bd23e4373e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 17:18:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/692fd15ba84a4948a8191cc6d278a5d1 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5e318d23bc7416aaa7dbe06a666dfa4 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81e2c76cdc084412b6f09d363b1dbcd3 : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8df446d7385b47cd91de392f1e632a80 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85a268c849db4a9190c34f27872172df : SUCCESS in 9m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/070b13030e9e4034bd22a885d70b302a : SUCCESS in 5m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/789775c7a157486795bb44b28e2c9d14 : SUCCESS in 11m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8a9e63c265f64cd794b5b2cd322b22b3 : SUCCESS in 25m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/eafe9ac8394e4dd38a5b1e44c14031ea : FAILURE in 8m 02s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/15636921eb614c6f96675871b8e0b3cb : FAILURE in 10m 12s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3eef2b1812314cbd8e52a178458d1fef : FAILURE in 9m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/53c3d65b003344208619f6b289b6c614 : FAILURE in 8m 34s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b3cfb18732b846259000cfabb85b3eb2 : FAILURE in 10m 25s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/7f81c87aa78e4accb82717cac12c959a : FAILURE in 9m 35s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f4daffcdcf3422785145cdcda3bb429 : FAILURE in 7m 31s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d65f130251d542d4b4a624026f7ffe4f : FAILURE in 1h 00m 50s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17c382ac9daa49648e66703799d35cfa : FAILURE in 42m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/31bde0b22e0647b782ad76f990b34692 : FAILURE in 11m 38s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/f2e9531727fc4082915c4ffb7790ca94 : FAILURE in 13m 04s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e7bdc9d1b9542eda87b8b09561033e3 : FAILURE in 9m 19s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/65a12a7335844067a4bd3bed2e7ef197 : FAILURE in 30m 42s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec9a75b19a0d4c5abc1bd958ef34097d : FAILURE in 22m 50s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/098a643d7220437a8f69e7cd6d8bf7b5 : FAILURE in 13m 00s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c3c64a42b71d47869b1c7e1981ca4085 : FAILURE in 12m 47s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/891aaf72c67340b18aacffcea36dfe20 : FAILURE in 9m 37s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/8e69fcfaadf24256bcde4fe8a85813ba : SUCCESS in 15m 53s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/bbf6081623f94f33b53361b828e273db : SUCCESS in 20m 23s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df3fd9f750904194bd8b812ba6b0ada2 : SUCCESS in 19m 01s","accounts_in_message":[],"_revision_number":5},{"id":"c8e6960bfd8c29dfc4c3f0c53379b16656fb54fc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-11 14:35:20.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7340bbbdd2a45a85b07424fd25dbabc0ba29e63c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 16:21:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d919d3fd15d444ab141e27cc6e6627e : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0ff53a49f854d7685c17247a416126f : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89a2d2f9654f4608a23e1d41faf02bcb : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4e5295de1f14b468e9d1d0cd4877e7c : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d186cf33f8f434996293324b56360f5 : SUCCESS in 11m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c33f11b43b01482b8fb48c97422b1a0d : SUCCESS in 7m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ef07a7a669c34fb39a8378397e8ef64b : SUCCESS in 12m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bf2778a829754781a2868fe46f1dc3c5 : SUCCESS in 29m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ef44518b5c144d88afec0f3a29396a5e : FAILURE in 11m 26s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/907b6ae92ba443b385e91ae110e3e1ea : FAILURE in 10m 13s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4eb18b2928c54a7f9ac1cc4ea69e6398 : FAILURE in 8m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f8591fbcb04c43b2b9607d4fac9f13b1 : FAILURE in 10m 40s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f57b4d96399a4ec9a90346ac31a977fd : FAILURE in 11m 40s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/22f9fea8242f4ec4be8b08fac973400c : FAILURE in 12m 18s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78dca82f8dbe4b2289cb0e2d09965c8d : FAILURE in 9m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2b35b8f067be49d8aba860fce144c2b3 : SUCCESS in 1h 40m 40s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3956dd097ebd4c628d177e06e8bb3cfd : SUCCESS in 1h 03m 02s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8be6913e09bc437eb334b94d59b09eb4 : SUCCESS in 19m 00s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/5a6e9118e13c498d89848e447b964859 : SUCCESS in 18m 32s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afdbb837078f4d8ab1f47b4cba268eca : SUCCESS in 13m 52s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/fabe5767fa0a4488a7372b87a9d25e6f : SUCCESS in 44m 17s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b1bdc3f4702e46c1ba729322e73583d6 : SUCCESS in 43m 23s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/408c6612ae7c4287802d9a6edb13733e : SUCCESS in 26m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a0b4a3c28dc541dfb5f36cb5d73b9948 : SUCCESS in 15m 32s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/84fd04a109ec4f72b1367310fea644d6 : SUCCESS in 13m 48s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f61fbbe68264a9fa423729273d9e589 : SUCCESS in 13m 52s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/69d81cba847d4edeb6e41f4876503f46 : SUCCESS in 16m 56s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f406e65db9fb4f52866ebbb43982c210 : SUCCESS in 18m 23s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34766bf6471b49fe86f3f9fdf90a4ff5 : SUCCESS in 12m 12s","accounts_in_message":[],"_revision_number":6},{"id":"5765e3fbc9f54c072a31e8f8b6d7ecf2dfa5ea1b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-12 10:28:37.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e07951aaec39aeb47b9670d40a4f3073d6ac6015","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-12 10:58:59.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"9839038b7fc2802fc446d361152fdd213b8f2f16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 13:08:30.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b69a60efcf2c4b88963c85bd2efccebc : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1758ccdc69334ad480f163a56a705246 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b40c1c911de540e2ba67213be2863392 : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18d9ff20e78e4da394f9596ba0bc6b1c : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6571bc5c7b6c42918fcc519b4a21b14a : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d705a8a957c451da88e5d20e931ef08 : SUCCESS in 4m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7f5a180023a34f13b4fae6df42116b02 : SUCCESS in 6m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/428c39794a7144eb95a04138108383f2 : SUCCESS in 13m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3ecb762312f740ed87b00a8d2d8c627d : SUCCESS in 21m 44s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3681a550546d44bf84ae2a2489534ea4 : FAILURE in 9m 39s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/cb4bc220981a488381a538c2b1d887fc : FAILURE in 11m 25s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ac0043841fb428bb986f466abec9778 : FAILURE in 8m 42s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8f5491a0181347dcb852034375b13fa8 : FAILURE in 12m 02s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/34027e827bc84f649bcc4efcfe6e22f7 : FAILURE in 12m 55s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/cb7264c114644895a4747327358f2cbd : FAILURE in 12m 15s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eaa0ed5f20674481a10a100eb46c74b8 : FAILURE in 10m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/86dad0b44ce844aab324595e4357f8f9 : SUCCESS in 1h 43m 52s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4c4ddd91ea842829e75018a7582cde5 : SUCCESS in 1h 22m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/48403c21836a44968e4728a644764025 : SUCCESS in 23m 13s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/343d25c022614eb3b81c8c0d4b05d0c5 : SUCCESS in 22m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99cc16d6f62e49f6b49f2cfa8291768f : SUCCESS in 20m 35s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/3c52bb0cc13f4362a068c457e89ed342 : SUCCESS in 53m 57s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4cf05eee17c549078ad566a9e840edc2 : SUCCESS in 43m 41s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd4179b47aa3417d8df19ab97cb0871e : SUCCESS in 33m 53s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/33ea44f3a5214335b9904987926d721d : SUCCESS in 18m 41s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f8360b5573c84c7fbb90114789b40daa : SUCCESS in 14m 41s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4d22208997e4b79a199ba346fecf099 : SUCCESS in 11m 00s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/da1efadfedee41998373894c2bd41757 : SUCCESS in 20m 13s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/cde290262f904794ad9d0c40ef9af5a9 : SUCCESS in 17m 22s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea1f79836a6a40d29443cc43bebd8a37 : SUCCESS in 13m 23s","accounts_in_message":[],"_revision_number":7},{"id":"962e9f1d78682154118784a5b0168aedd5308943","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2026-09-28 12:15:40.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"6942b431f8a9ee997e41c733d0e99c5290f03698","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 13:41:50.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/89abfa05801b45f594999b4b4553f44b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86e447a6400645f78d5a637cb45649c4 : SUCCESS in 10m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3a36f80393d44d8fa1a611a1363c48b4 : SUCCESS in 8m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d018050b2c3e45a686c114061f94117e : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fadb47c7523a4a48838f317c79b767ac : FAILURE in 3m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/915d13a65a82404bba966a944b4678ac : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36fe0613a58d4119b67564e9a5a25f50 : SUCCESS in 5m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3721acc3be874689b1a2d33c4132449d : SUCCESS in 9m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d560e81d59b04b839ad437eabbdfc078 : SUCCESS in 18m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3bfb9bb1a040419e8527182ba40ed0df : SUCCESS in 9m 23s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/87a83a07ce6a4298ae30ae310c16b5fb : FAILURE in 27m 49s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/b03978ecf2b6491b9d6575d906193750 : FAILURE in 32m 18s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/deabdb99feeb422e846bea62642f78da : FAILURE in 35m 02s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3e1374f3eff145c78d978e58f9e60bde : FAILURE in 37m 44s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/55d5f5d84f2c4d68b4dbdcb2dbb431f4 : FAILURE in 27m 49s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87cb6043b52c43ea9a619d284615062a : FAILURE in 31m 08s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f754a0fe81934c5697e5c56cb634f716 : FAILURE in 30m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ace22e276fa847dca585de8ff5c2719a : FAILURE in 38m 07s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9004544cca7249e495b1294541c5a8d2 : FAILURE in 39m 08s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ec598faf68c24acc8155863dde73b6db : FAILURE in 33m 49s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8237cb6848594e3ea271c961c2745d6d : FAILURE in 30m 41s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7cc6712ecc7c4f91848004448b79aad5 : FAILURE in 33m 24s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0326b900b4f24b609ee5ba979f77c1a6 : FAILURE in 32m 56s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cc5b3c6da44849cfa3e046463574cf73 : FAILURE in 1h 25m 07s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/24668ac86b2e46ada97f1312af70ab29 : FAILURE in 28m 06s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/65be8e878a8d4743851710ef40ba2013 : FAILURE in 26m 52s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce3e69bdd6c1474a9d04f84628f847e0 : FAILURE in 34m 33s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f05fbc4b9f15454a93ac8eb7bbc3803c : FAILURE in 40m 51s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/490b7ee09cc145f9b58f9f108c415dfa : FAILURE in 33m 36s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/959c6d8c6c9244aeb716b6e36d7e2a46 : FAILURE in 23m 54s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/917cd906f2934404a23d4c5eb87d48f0 : FAILURE in 26m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67a90df826f54b548ec2555fc3a89b6f : FAILURE in 36m 55s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"f79112f3eb90e87d81473585756580b9f32489b8","revisions":{"0c9fce2cc7cb3e85effd74320e2209da4a852075":{"kind":"REWORK","_number":1,"created":"2022-01-21 16:02:45.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/825832/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/825832/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/825832/1"}}},"commit":{"parents":[{"commit":"1b49f86e86e971e326fb4dd6f493f5a9be1cab83","subject":"Allow more fine-grained groups for ironic, loadbalancer \u0026 network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b49f86e86e971e326fb4dd6f493f5a9be1cab83"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-21 15:11:45.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-21 16:01:34.000000000","tz":0},"subject":"Rework network interface to host mapping","message":"Rework network interface to host mapping\n\nChange-Id: If604adb0815f583c3fee4772f90ad53b84180372\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c9fce2cc7cb3e85effd74320e2209da4a852075"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c9fce2cc7cb3e85effd74320e2209da4a852075"}]},"branch":"refs/heads/master"},"9c7166f7d73cf9aabc89a054414601a0df52a805":{"kind":"REWORK","_number":2,"created":"2022-03-29 15:55:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/825832/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/825832/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/825832/2"}}},"commit":{"parents":[{"commit":"1b49f86e86e971e326fb4dd6f493f5a9be1cab83","subject":"Allow more fine-grained groups for ironic, loadbalancer \u0026 network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b49f86e86e971e326fb4dd6f493f5a9be1cab83"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-21 15:11:45.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 15:55:26.000000000","tz":60},"subject":"Rework network interface to host mapping","message":"Rework network interface to host mapping\n\nChange-Id: If604adb0815f583c3fee4772f90ad53b84180372\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c7166f7d73cf9aabc89a054414601a0df52a805"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c7166f7d73cf9aabc89a054414601a0df52a805"}]},"branch":"refs/heads/master"},"a335dcc377a66dca814c1ce37064cd0559659b0f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-03-29 15:58:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/825832/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/825832/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/825832/3"}}},"commit":{"parents":[{"commit":"b3fed016f3fecc58cfb8d6643369bdaf114a703d","subject":"Allow more fine-grained groups for ironic, loadbalancer \u0026 network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b3fed016f3fecc58cfb8d6643369bdaf114a703d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-21 15:11:45.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 15:58:20.000000000","tz":60},"subject":"Rework network interface to host mapping","message":"Rework network interface to host mapping\n\nChange-Id: If604adb0815f583c3fee4772f90ad53b84180372\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a335dcc377a66dca814c1ce37064cd0559659b0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a335dcc377a66dca814c1ce37064cd0559659b0f"}]},"branch":"refs/heads/master"},"72eecd0b4b107478b278d3691e2367ed197eeaee":{"kind":"REWORK","_number":4,"created":"2022-03-31 14:54:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/825832/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/825832/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/825832/4"}}},"commit":{"parents":[{"commit":"cc9181f5c52b2716c290ff75e593f6913a646b6b","subject":"Allow more fine-grained groups for ironic, loadbalancer \u0026 network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc9181f5c52b2716c290ff75e593f6913a646b6b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-21 15:11:45.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-31 14:50:43.000000000","tz":60},"subject":"Rework network interface to host mapping","message":"Rework network interface to host mapping\n\nChange-Id: If604adb0815f583c3fee4772f90ad53b84180372\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/72eecd0b4b107478b278d3691e2367ed197eeaee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/72eecd0b4b107478b278d3691e2367ed197eeaee"}]},"branch":"refs/heads/master"},"f0662cecb487fa6d787b1a33b06a9826e78d8b5c":{"kind":"REWORK","_number":5,"created":"2022-03-31 16:06:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/825832/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/825832/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/825832/5"}}},"commit":{"parents":[{"commit":"cc9181f5c52b2716c290ff75e593f6913a646b6b","subject":"Allow more fine-grained groups for ironic, loadbalancer \u0026 network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc9181f5c52b2716c290ff75e593f6913a646b6b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-21 15:11:45.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-31 16:06:36.000000000","tz":60},"subject":"Rework network interface to host mapping","message":"Rework network interface to host mapping\n\nChange-Id: If604adb0815f583c3fee4772f90ad53b84180372\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f0662cecb487fa6d787b1a33b06a9826e78d8b5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f0662cecb487fa6d787b1a33b06a9826e78d8b5c"}]},"branch":"refs/heads/master"},"db3186ee4e752b85fc0dbf49eefbb65db3fd8221":{"kind":"REWORK","_number":6,"created":"2022-04-11 14:35:20.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/825832/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/825832/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/825832/6"}}},"commit":{"parents":[{"commit":"cc9181f5c52b2716c290ff75e593f6913a646b6b","subject":"Allow more fine-grained groups for ironic, loadbalancer \u0026 network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc9181f5c52b2716c290ff75e593f6913a646b6b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-21 15:11:45.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-11 14:27:54.000000000","tz":60},"subject":"Rework network interface to host mapping","message":"Rework network interface to host mapping\n\nChange-Id: If604adb0815f583c3fee4772f90ad53b84180372\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/db3186ee4e752b85fc0dbf49eefbb65db3fd8221"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/db3186ee4e752b85fc0dbf49eefbb65db3fd8221"}]},"branch":"refs/heads/master"},"5e89012eac66014d4e054d0f8519f3d202009a54":{"kind":"REWORK","_number":7,"created":"2022-04-12 10:58:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/825832/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/825832/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/825832/7"}}},"commit":{"parents":[{"commit":"715366b527932b09190f9b61d90759ea4d033ced","subject":"Allow more fine-grained groups for ironic, loadbalancer \u0026 network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/715366b527932b09190f9b61d90759ea4d033ced"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-21 15:11:45.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-12 10:58:16.000000000","tz":60},"subject":"Rework network interface to host mapping","message":"Rework network interface to host mapping\n\nChange-Id: If604adb0815f583c3fee4772f90ad53b84180372\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e89012eac66014d4e054d0f8519f3d202009a54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e89012eac66014d4e054d0f8519f3d202009a54"}]},"branch":"refs/heads/master"},"f79112f3eb90e87d81473585756580b9f32489b8":{"kind":"REWORK","_number":8,"created":"2026-09-28 12:15:40.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/32/825832/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/825832/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/825832/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/825832/8"}}},"commit":{"parents":[{"commit":"838618cb7354b61f6174fa28824f13609243be95","subject":"Merge \"CI: Add idempotency and diff check tests for host configure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/838618cb7354b61f6174fa28824f13609243be95"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-21 15:11:45.000000000","tz":0},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2026-09-28 12:15:11.000000000","tz":120},"subject":"Rework network interface to host mapping","message":"Rework network interface to host mapping\n\nSigned-off-by: Rafal Lewandowski \u003crafal@stackhpc.com\u003e\nChange-Id: If604adb0815f583c3fee4772f90ad53b84180372\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f79112f3eb90e87d81473585756580b9f32489b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f79112f3eb90e87d81473585756580b9f32489b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
