)]}'
{"id":"openstack%2Fkayobe~826745","triplet_id":"openstack%2Fkayobe~master~If4477e7fa4c022ead6cda4a34a72b342f115936f","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-02 17:34:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If4477e7fa4c022ead6cda4a34a72b342f115936f","subject":"Bump up Ansible supported versions","status":"MERGED","created":"2022-01-27 17:25:57.000000000","updated":"2022-03-02 17:36:09.000000000","submitted":"2022-03-02 17:34:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826745","meta_rev_id":"f1667e18fa8e93907385fbaf383c9fa092fcd7b3","_number":826745,"virtual_id_number":826745,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-02 17:34:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-03-02 15:41:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-10 07:57:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-03-02 15:41:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-27 19:25:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-10 07:57:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-03-02 15:41:42.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"9c80c6d7602633b0e6b6b923306d7a31c9d5e010","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-27 17:25:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65ac9709f8731f1adec68e6b05c62ef534130c44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 19:25:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/818e1d3265c746deb121f1e0f505c4dd : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81a5c90864c04a1c83adc86639e4fcfe : SUCCESS in 8m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6479cb64893412a94e037e5fdf19882 : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41c7c41a7c9f4f79bbb36470636ea922 : SUCCESS in 8m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc9fba2a8d884cbb8d7ae643445409c9 : SUCCESS in 11m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/871d71b524f149afac5eb7a1c9400bee : SUCCESS in 4m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/94848728232a42838b4f9404ae6bcbc1 : SUCCESS in 8m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b073b3a6f2e74eebabf636cfb1521fa7 : FAILURE in 13m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/facdcb88683347a2b1fed65e348193b4 : SUCCESS in 34m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/82283caa971d4650b65e8d400e51ec88 : SUCCESS in 1h 11m 22s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90de8f3da1fc494ca394c872ed0f32c8 : SUCCESS in 50m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0ec72a7009a94d1eaf2a69dd75d52494 : SUCCESS in 1h 08m 32s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6c2d4472d71e4cec95e94b34b934d87a : SUCCESS in 24m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e02461ec4d6647d6a5490984a54a9e44 : SUCCESS in 15m 53s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fb8c1c24d64e4846b53bd6f53c3fbfc2 : SUCCESS in 1h 50m 03s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8015bdbf2bf149e3a4bbda2b5031a714 : SUCCESS in 1h 15m 03s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/32f5843728054bc2958b9d8088b02217 : SUCCESS in 44m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab15d01a22734ccc9bb65ee54d6f032a : SUCCESS in 39m 18s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/357d3f986af4476ea14f2304986cb7d1 : SUCCESS in 28m 19s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6818451daa8f4762b3a664881abdbfba : SUCCESS in 25m 36s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9cac4bbf41c54773bd17459bc9f45679 : SUCCESS in 19m 57s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/972d91612e1a49f38e01d9e44a1df0da : SUCCESS in 13m 07s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/4a981e834d6a481087bcfef00de46962 : SUCCESS in 19m 10s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0498edc31ab14823a13b841d25d0a308 : SUCCESS in 12m 40s","accounts_in_message":[],"_revision_number":1},{"id":"97d3edec710e4994486a0d78582019d1f5e2d34b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-28 10:49:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4dcece07fa773fbb4d3449e2365f5c70c025214c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 13:21:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/215e63c868b14618bdd7b0279f8eac7c : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/319298367a0d46229763f60193845893 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ff73f2943f44f2d981f6eae8fb29107 : SUCCESS in 7m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d6d1e415d8b4dd695269dd8af129bae : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80d41a362f754e639ac9731bce0a3c5c : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20dc914a7f414b498d2fa6467c6c6596 : SUCCESS in 4m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e0e90a96fb9343acac3e4b53e8b42181 : SUCCESS in 6m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dd9c4e93108e47ef8408513825bc479c : SUCCESS in 13m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4c866aa927514f0d8bc58f5c0dd389c8 : SUCCESS in 33m 54s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5ddd1206074f4a729ff9209e2a72f7bf : FAILURE in 58m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf900a4a84e44e31b177bd90955e6ab0 : SUCCESS in 46m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1a9460840ac04c828e256feea8160311 : SUCCESS in 1h 16m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b6fe07cef2a1411496417959e253fbb6 : SUCCESS in 17m 20s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c588a01494b4633a905b73b39494387 : SUCCESS in 15m 45s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/834a2578589945209e0ffea327b476f8 : SUCCESS in 2h 28m 25s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fedd32a79744a148c33f7ab05040df8 : SUCCESS in 1h 03m 57s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/11a2416e31954c08b3455958cdf55ef2 : SUCCESS in 38m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/390170d3a97c4e6d92a74005803865c2 : SUCCESS in 36m 14s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fadd5c065c06461e87083f652fcecf4a : SUCCESS in 33m 53s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7475ebf5ff146ae946a136bb24ffc2d : SUCCESS in 21m 29s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/46e27c5e7b3f48739ed9801012e2c4a8 : SUCCESS in 21m 42s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d890d961d574359b21157892de985ad : SUCCESS in 13m 58s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/941fcba7830a45b58356b3b6fcdca707 : SUCCESS in 20m 51s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b4a0ad1c7d44445b2c678bc205a8f8a : SUCCESS in 12m 01s","accounts_in_message":[],"_revision_number":2},{"id":"e53937b595a783dc7ae6885c84be15757fb913fd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-28 14:15:35.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"e975ca43922af870c078389f80c2de0b160d94db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 15:52:27.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d659d8e076942ab8fb943f78b62e6d3 : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fd79538423d4dc5a4a968466099a334 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d59c5ffcb4f46b8938b39c6a4dc7435 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4eaa51701ff457daee7535fb72050c2 : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/119fe1a358624943b584b1cf316917f0 : SUCCESS in 10m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2be87d023e934ab68b9157b17092d876 : SUCCESS in 5m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f567fe78d3a04d96abc3ef6812ca5dcb : SUCCESS in 7m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/37a77dcca1804274b0ed8d5b4d5c1cf0 : SUCCESS in 14m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3a4d6b9cc2664eefaa852b3407c092bc : SUCCESS in 33m 18s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/bd8e7bbdf2634723918f40c006e9d1a6 : SUCCESS in 1h 00m 30s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/870c6f37f0214596bcd8338af697abd2 : SUCCESS in 43m 32s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a65469d567764f2cbb1b0b58bf4654d4 : SUCCESS in 1h 02m 35s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/22f73fc5c78b433fac0e3ea6a5986dea : SUCCESS in 20m 15s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5474346c4de443669e9c66ea758c710b : SUCCESS in 14m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f2099486291d4fd0aa84733c9a7d6db0 : SUCCESS in 1h 32m 27s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/484f35f4e495450199347d19f430a679 : SUCCESS in 1h 07m 01s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0f10611735eb47d7aae7bcb3ae492272 : FAILURE in 23m 04s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94a7c72be8c846b9b117fdf634f04454 : FAILURE in 18m 55s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/31e51239a0b049d4b5c30e64823c0b84 : SUCCESS in 27m 13s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44542a8c32f54cb59a3b7d8f83319205 : SUCCESS in 22m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/32cfa01caacf4e86bb0574151ad777a7 : SUCCESS in 16m 50s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a75714bd292d478b9998375b88af8279 : SUCCESS in 11m 17s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/d5c58c4773b743639ed336e7e6131aa1 : SUCCESS in 16m 35s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a928bc80f17f49d093f88730c0cac8a1 : SUCCESS in 13m 16s","accounts_in_message":[],"_revision_number":2},{"id":"50f6a24fe666e9d218b20cdcce8f16017bd78316","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-09 10:13:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f7c51107014631d5cfae4b42c3aca482a5bcb12c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 11:55:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a304741fbd24899aaaec22d2eed3c03 : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/881603542a1f4713b246a6d374990aab : SUCCESS in 8m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94771bcffa6f45a489d2b6f3750a087e : SUCCESS in 7m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0f32ec0b72144d1931cd06b5f68f10d : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a514c7f84142468ba2c5979972eb6519 : SUCCESS in 8m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29b42db2ef654fff8ba1350bbd57e3c7 : SUCCESS in 3m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d768ca94d4b44d4285e152a1e81579c5 : SUCCESS in 6m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9521e686533f40ea8e6d16fe8f43c7a3 : SUCCESS in 11m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3b0f9eea68a4423e95463ddf9500bc78 : SUCCESS in 27m 49s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e1b34e9141e4414ba5e6b980f14f24c6 : SUCCESS in 57m 33s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/313c2a45a2da4efe8d8fc89e6a8da61a : SUCCESS in 41m 02s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f15e5a6876424a81bd425deed90c6b00 : SUCCESS in 1h 04m 40s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2259c84cdb244e2bae478477b792dc67 : SUCCESS in 20m 32s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/384eee498b53411c8c248686eaf02221 : SUCCESS in 13m 54s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d18477be02ee400cae5d9fe55e168dbc : SUCCESS in 1h 37m 24s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78f919a652fd4efcb107803dd05b3bbe : SUCCESS in 57m 50s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5126ab749ac54a449260c262ea6a29d6 : SUCCESS in 40m 35s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecb9051b042046679aa9615199176a93 : SUCCESS in 30m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fb5e47c4ad1147a49fd8ffc0f970f97d : SUCCESS in 29m 06s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ab9507ab3b94f498b6fd8a5a8d8a328 : SUCCESS in 25m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/96e0df9972564a2da55b1a73b4440946 : SUCCESS in 17m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31ac1f9ac3b64a1c889f8dd3c414041a : SUCCESS in 13m 18s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/631160c2ab94474fb54a07c30b298127 : SUCCESS in 18m 30s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6df0a9a2ba634456a0acb5acee458c0c : SUCCESS in 13m 19s","accounts_in_message":[],"_revision_number":2},{"id":"d2d20899e9d7333afd39a7964f993ec6afb32b79","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-10 07:57:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f709a1f561626e868fb8861d10640cbcdc652a9b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-02 15:41:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dcdc6b8f1114a1f3666c2daf5d3804733403d897","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 15:42:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"49f861ae756556ba9e38c261cbc3f07fc4e2f661","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 17:34:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbc6f440bcbe4b7f984fb86d17c99456 : SUCCESS in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cc098799e484591a0fd4126a0246d04 : SUCCESS in 10m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/610ed208d8d74b208d17f3fc7c2786d8 : SUCCESS in 8m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4f2e8c6bbff4707858fca1d248f62c2 : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b69b00df911483b98d4eb147bd46a00 : SUCCESS in 5m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d67ec1bb69034aceb9b1385e83c2618d : SUCCESS in 8m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/68b8b18820ba4b1d92e858be5447a55c : SUCCESS in 16m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/19379e8db0774dd18fa032aa015f56bb : SUCCESS in 35m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7bba18ba29044397ab23faedfcd59109 : SUCCESS in 1h 16m 21s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76e1cdc7b8fd4a8daf82b07059d71d41 : SUCCESS in 59m 19s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c78ca0428f9c402a88a329c77a6f2dbd : SUCCESS in 1h 15m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7de48ef445e94b3fb21d78f7d826e145 : SUCCESS in 22m 53s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9853e8674d2f4a1997535258fdd672c8 : SUCCESS in 15m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/22b8f7187d28429e9f80f4d90081ca47 : SUCCESS in 1h 48m 29s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cebfaf5afe7645ea9de21a8ad16e9a0b : SUCCESS in 1h 19m 37s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ff78d2dc84134a29821e197130c30327 : SUCCESS in 47m 31s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0dcb94fab78b4c30af2ed50d2e4be129 : SUCCESS in 43m 59s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1c5e45ecb1b843ed957ab96b61fdf452 : SUCCESS in 29m 51s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbf5867cd388490798e79d5826a2f7eb : SUCCESS in 26m 26s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/46fa2dc1b72346bd9002dea0bddeea1c : SUCCESS in 17m 56s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5da0ab67400541b29fcbfb10ada9730b : SUCCESS in 14m 03s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ee7c865a201c4d9ca6d7215818ff6e50 : SUCCESS in 18m 51s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3323a8bb34de4ed2b483f306fd58994e : SUCCESS in 15m 31s","accounts_in_message":[],"_revision_number":2},{"id":"0a4a6a0c394dd7aeba82c4ed19a2672e6eb1ded7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 17:34:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f1667e18fa8e93907385fbaf383c9fa092fcd7b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 17:36:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46bded972e5d42c0a21aa222168449e7 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/822c0c2e7cc94d6bb0f91e6da81948f6 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fb5649bd2f303332ed1280baa9ab2f4db9c40ab3","revisions":{"60503704d72904821223ad290006cd28a0b51bf0":{"kind":"REWORK","_number":1,"created":"2022-01-27 17:25:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/45/826745/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/826745/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/826745/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/826745/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/826745/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/826745/1"}}},"commit":{"parents":[{"commit":"abb2edd8a0de089f219b845ddef8a2c24d0fafb3","subject":"Merge \"Remove chrony cleanup from overcloud host configure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abb2edd8a0de089f219b845ddef8a2c24d0fafb3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-27 17:20:44.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-27 17:25:45.000000000","tz":0},"subject":"Bump up Ansible supported versions","message":"Bump up Ansible supported versions\n\nThis change bumps up the maximum supported Ansible version to 5.x\n(ansible-core 2.12.x) and minimum to 4.x. This synchronises Kayobe with\nKolla Ansible (see change Ia373f9cc3bb69eba0288bbb3e497e8cadb7cc4d3).\n\nansible 5 / ansible-core 2.12 raises the minimum version of Python to\n3.8, so CentOS Stream 8 environments will typically get ansible 4 /\nansible-core 2.11.\n\nChange-Id: If4477e7fa4c022ead6cda4a34a72b342f115936f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/60503704d72904821223ad290006cd28a0b51bf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/60503704d72904821223ad290006cd28a0b51bf0"}]},"branch":"refs/heads/master"},"fb5649bd2f303332ed1280baa9ab2f4db9c40ab3":{"kind":"REWORK","_number":2,"created":"2022-01-28 10:49:08.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/45/826745/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/826745/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/826745/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/826745/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/826745/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/826745/2"}}},"commit":{"parents":[{"commit":"abb2edd8a0de089f219b845ddef8a2c24d0fafb3","subject":"Merge \"Remove chrony cleanup from overcloud host configure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abb2edd8a0de089f219b845ddef8a2c24d0fafb3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-27 17:20:44.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-01-28 10:44:32.000000000","tz":0},"subject":"Bump up Ansible supported versions","message":"Bump up Ansible supported versions\n\nThis change bumps up the maximum supported Ansible version to 5.x\n(ansible-core 2.12.x) and minimum to 4.x. This synchronises Kayobe with\nKolla Ansible (see change Ia373f9cc3bb69eba0288bbb3e497e8cadb7cc4d3).\n\nansible 5 / ansible-core 2.12 raises the minimum version of Python to\n3.8, so CentOS Stream 8 environments will typically get ansible 4 /\nansible-core 2.11.\n\nChange-Id: If4477e7fa4c022ead6cda4a34a72b342f115936f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb5649bd2f303332ed1280baa9ab2f4db9c40ab3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb5649bd2f303332ed1280baa9ab2f4db9c40ab3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
