)]}'
{"id":"openstack%2Fkayobe~827403","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I4b8a42704f2b0a145ee9dec433d91df67628cd9d","project":"openstack/kayobe","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-03 19:33:25.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-02-03 19:33:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4b8a42704f2b0a145ee9dec433d91df67628cd9d","subject":"ntp: Fix service mask when service doesn\u0027t exist","status":"MERGED","created":"2022-02-02 16:45:19.000000000","updated":"2022-02-03 20:52:10.000000000","submitted":"2022-02-03 19:33:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827403","cherry_pick_of_change":827486,"cherry_pick_of_patch_set":2,"meta_rev_id":"6a8b398ea70e484b87c93f98ad2958fc2c35c10f","_number":827403,"virtual_id_number":827403,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-03 19:33:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2022-02-03 17:51:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-02-03 18:15:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-03 17:41:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2022-02-03 17:51:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2022-02-03 20:52:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-02 17:09:32.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2022-02-02 18:07:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-02 21:02:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-02-03 17:51:00.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-02-03 18:15:10.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"e817000b7cf8f0bf793f2f8103d7a170650e017c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-02 16:45:19.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"532bd8d40be030c67ab02fe78c6fae3e85e8b3fa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-02 16:46:37.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"5380a9b5dad0dfc1d784c035e69a60b05396a786","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-02-02 17:09:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3c0e0d8d4301b67c511575f286e97b828529e9d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 18:07:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ae9eaf3c5ae46ccabd8fd80a837cc14 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1d27611138844a2af8aeb23da3140f0 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49e26c04676444bdb531e5d61060966c : SUCCESS in 7m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73814eb8269c47b58fdb3e56f00beee0 : SUCCESS in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7177774a5f4480ca2bf4a25120b30cb : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/435d9bda250843beaf82bb9d4b93de6f : SUCCESS in 8m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2447b6aeb17747af9f95821b4079e1d1 : SUCCESS in 4m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/862d8801cf5247649e9ca773ffb1be6b : SUCCESS in 6m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b335cd513c1f4a41ae7e78da27f17331 : SUCCESS in 10m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e76af5ed78494631ade2bfa855d6fdbc : SUCCESS in 17m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d8b71e2634ce4b229c3dd7196729c102 : SUCCESS in 57m 20s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b08391f3845c407298c238b58bea25f9 : SUCCESS in 40m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2ddabf2097db43e78854d0e0ae05a13d : SUCCESS in 56m 46s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/48d926dd84ec4d4a9cd88b6e2cde48b1 : SUCCESS in 23m 19s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8cf04c621c24e92aff18269f3265c82 : SUCCESS in 16m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7333654817914ff083c154d30fa48a90 : FAILURE in 1h 14m 49s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/629c5833690f4377b989c5f91207075d : SUCCESS in 33m 40s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b992c57b15ba4a67abfc381afaf9f27c : SUCCESS in 38m 20s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f55016a37e3e415ebd36a693f6a8fd0c : SUCCESS in 26m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/dff6574e2a42447aba85ff2a95ebef5c : SUCCESS in 19m 51s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2677b35cdfad4c4181fc7890dab80667 : SUCCESS in 12m 21s","accounts_in_message":[],"_revision_number":1},{"id":"f3d1a1604eef7a08d2e0eb5b1b36f95838b3523a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-02 20:45:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d88620e1693fea284a7ce14be75173bff834a27","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-02 21:02:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"74f2c021eefe5e63b77b53c374661a65d95485a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 22:24:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3e5fec2ed3549d9bdca0dc134df5753 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b208a23dc13a42c7a8687fe5b3b8e9c0 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0052911283704840a5f188ebaae187fe : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95503883efdc4873bf5b14fd85ec6cb1 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c694d8f635ab482f966655d972cb054c : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0948d308e43447ac96c60ea3343c3766 : SUCCESS in 8m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13f2644755e144da8189eb0a003634b1 : SUCCESS in 4m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cb04db2ab7774d9ba7179aab6b6f1a99 : SUCCESS in 5m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6dc7b24dad3440659e66f678fac1a6d3 : SUCCESS in 12m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/054d8127e5374059b468a864ab680388 : SUCCESS in 17m 32s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6c063b1042944ff981b4a0311b0062e4 : SUCCESS in 1h 10m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a41acc3dfd9a4182964036e6b37cd57e : SUCCESS in 54m 45s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c10e746d4e18488f840b081f8d8444b6 : SUCCESS in 1h 07m 18s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0502fe0774b44069b9896a5b971446c3 : SUCCESS in 21m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5ac0ac856a448868d57e585724967b8 : SUCCESS in 16m 18s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b6d6e9a1ef004827b030d906c9edf907 : FAILURE in 1h 33m 38s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/39d95ae170cc4a4b8fc08b7124ea98b0 : SUCCESS in 35m 08s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/283469fd229e412ca4763680b5ed3155 : SUCCESS in 30m 59s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/40f9e5f3a728445ea0c556f3e25f6671 : SUCCESS in 25m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7cece3856b43492b98284a3acafbfa26 : SUCCESS in 15m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/268174b80e384886b56e917fd76feebf : SUCCESS in 12m 22s","accounts_in_message":[],"_revision_number":1},{"id":"65cafbb9b03313a1511c86a4501fbc0b7291c9d2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-02 22:29:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec5dc399ec01d23b3af1a080a0af2306cb3f59d0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-02 22:48:57.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e7f2b0eba1eed9abd0c5fb50ce56f9f8fe2a6f7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 00:08:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3ed77db680e440b91a87b65bcd6531f : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d3b3cc019044f4d8f3fac3690345930 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c280ce3938a46169b2c002216ffe5e6 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/229f72d2a3934a2cb51e958c6f188471 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/336350b0ce2b44d4aac5308282748263 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84b726650eb94bfc9f575328cf060d68 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea53bebd13ea46a8ac6fcc8021050f2a : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d5188be23bd04b3c84f632a91eaa1246 : SUCCESS in 5m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fede8108ebcd4353bdea305042aa61e4 : SUCCESS in 11m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ffb094a11f5a40eab6d420d60ca157fb : SUCCESS in 16m 42s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3857917ef1f9414cb9e32dccf93f8b3f : SUCCESS in 52m 52s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39e81f22d55549b783cc3f3433853dac : SUCCESS in 38m 49s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/28a27034e7bd461590af28607e848464 : SUCCESS in 55m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/255c2827473b4305bb645a443945c6f8 : SUCCESS in 16m 35s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd62253008be49968673c8234e5a4664 : SUCCESS in 12m 28s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a3ce0c6b8cf64d539df95954c75f732f : SUCCESS in 1h 36m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9328dbcbb4324e5fa14db7ba4c1ac42d : SUCCESS in 27m 45s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dad6c0c2b0f64196a7c9547f0914f20c : SUCCESS in 31m 28s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c75024c5946e4e0099ea1681402fc65c : SUCCESS in 22m 59s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/21718a8804514c75af81fbe6ca74f5c6 : SUCCESS in 13m 04s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4047d8c4dbe847f991b37699e051b40e : SUCCESS in 9m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ed1b98012b5f8cc3ca00ef962315da02d3d82b74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 00:09:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22bfacee24802aaaf2f3e19dc8d1065136ddce4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 01:25:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d25a208f7ffd4de890985832668cb883 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89042ff8a14b4e33b943c76152381897 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/869c55f4f4d345549660c4cbf63f4d33 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21b13b43b98b488abda11a6e58f7817a : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7158cf9fd2d478983563935c48989a6 : SUCCESS in 4m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0d134e495c4e49d2b2df62b3d007308d : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/250261f857324bb1aa6b5a97d4c8fd33 : SUCCESS in 10m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/db4d871a5655486aa32bd37658d76ef4 : SUCCESS in 17m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0e8ef908df754960ab78e42b96fc4a21 : SUCCESS in 1h 09m 57s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1aa3dccf233c41adb9073a40534f8051 : SUCCESS in 43m 07s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e02299f385374833839b69d39cbba5d0 : SUCCESS in 1h 12m 30s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/49d9042fcf5a48d8bd62a858ccc49476 : SUCCESS in 19m 15s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de6aedc3b9344d0183e4eaf71e6a2e89 : SUCCESS in 16m 32s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/58df142cba3847e58f91b35a2d979787 : FAILURE in 1h 12m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f842430c45aa4753a45f4701fb2f7ec7 : SUCCESS in 35m 22s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75d62bbbda24435d84fe9eb95042fc3f : SUCCESS in 28m 27s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e5d8ccfd3a564936a83e15f84a742c98 : SUCCESS in 24m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/957d1d98aee84e3887d026cadd2df886 : SUCCESS in 18m 35s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00cb14ab518b45e99f7dd36547262cbe : SUCCESS in 11m 03s","accounts_in_message":[],"_revision_number":1},{"id":"37b87ba71b640b9011bbea4856077e8934f351a9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-03 07:58:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e10a32e0d18e61a3b5beada93ba9a9b406f47c56","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-03 07:58:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d6c844bec16f60aa85ebbd6292e6ce7d27ca7a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 09:28:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1754125e0bb94bbc9c1c329e847f2970 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cffe9b9268764ea2817886aa9a81d4eb : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9985901e0ff04b9d9c2364029df0f1c0 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d64cab030a14e7eac7a455290e5658b : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e654f8a00b5d4ac5815d8363b7552aa3 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2e01bebe2404a6a9dd454344232a484 : SUCCESS in 7m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58fd3d35e0af466aa978dac9fd69634b : SUCCESS in 3m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7067170af8544837a07cdb5035ec7b5b : SUCCESS in 5m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/81d68a38ea8c4d62ae2d80c8b43c73f3 : SUCCESS in 10m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2bf0cbb153704584847ec920a4673dd2 : SUCCESS in 21m 11s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/485568926e8841ffb966f6b07f5bcb73 : SUCCESS in 1h 07m 58s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50058b97f7d549459dc5f8a1e8b56e47 : SUCCESS in 46m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c11d4e63cb3f4e19a95e840101d03324 : SUCCESS in 1h 07m 38s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9fad3b771ad444969efa10735b575b95 : SUCCESS in 20m 19s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ee6e43e2055401780d90ff71ac58cae : SUCCESS in 13m 54s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ed29522e4e264972b0e970d1e23017d1 : FAILURE in 1h 24m 45s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/569a6d7507dc484091de24d331b0fabb : SUCCESS in 33m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4c5d186f37f424da04e689e6fdcd163 : SUCCESS in 28m 19s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/289e26c62f3447f1bf9a1d1ba26e6150 : SUCCESS in 25m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/66807e567b114325a9d29699b034396c : SUCCESS in 14m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3d1708e961d427db5c6dbb8f8dbc7f2 : SUCCESS in 11m 02s","accounts_in_message":[],"_revision_number":1},{"id":"2702d28a451ab4ba75de3c684c4cf9c29e8d9ce9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-03 09:32:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e9a8b1f01eddcbae49d44e0a06dc10ae6ff3960f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 10:55:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6658c6c0fa2e4eaf87107498148ad724 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a711b8f04a1b490199744b4452ec235c : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f457cd0fcf2e4422bfd550caf37995e8 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a442c967b6f452b901e500c6289432d : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb5916afb1bb4c91bdc1a743a6fee805 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7875584a6b9944a8a3ad5ca70105a048 : SUCCESS in 9m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db332a1f3eb54a77b3ee882e09fa3faa : SUCCESS in 3m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8a1f5e43ad6b48598b66b578a7d839f4 : SUCCESS in 6m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/07fd89a9c97c45f8a6bc04da581b4e86 : SUCCESS in 9m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f546db7d4f97418e8cd424545441f35a : SUCCESS in 14m 35s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8fb36dea204848b28aa33d8722420953 : SUCCESS in 57m 52s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b3900b330d04f7da5e8b3f5ce8d5802 : SUCCESS in 37m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d28961e2504847c795e7d57069f434be : SUCCESS in 54m 34s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/01058561c997405ba9637681c5fb4875 : SUCCESS in 17m 49s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/443d8ce58d714bb7849e7334da515635 : SUCCESS in 13m 13s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/baab8c87125449689e934349a4c5535b : FAILURE in 1h 21m 56s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1eba8487f7834cb182a9ada30923d6f7 : SUCCESS in 34m 50s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ed62b60cd6840eea45e169e76d786eb : SUCCESS in 22m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8fffabe558764ef3a69c1411d7c59eb8 : SUCCESS in 26m 59s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/78aab9e2c8d5415ebeca153a7bbf3ed6 : SUCCESS in 19m 09s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9390ce8607bc435cb82185edf1263b52 : SUCCESS in 9m 47s","accounts_in_message":[],"_revision_number":2},{"id":"1cef72ed4fe59a5f53290de70d236a3a3bed7959","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-03 11:38:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9511552181c97e028b50ed2d318eb75afbf17a61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 13:29:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8867e52c81cc4f3598dad9c42ae5f7a6 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ef3c8c05c5848f2ba529b04d81ee093 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3f37fb9969b470fbe81d0d0289ea5dd : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6197dbdc58442c8a034f759a04149a8 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd0c7b89e46d42198a62602bd61d8479 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59cbd489713444308936616a7957453c : SUCCESS in 8m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02fad2b6368e49c69f757f435912e753 : SUCCESS in 3m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d9898fd74bd14baab069d4b327e7718f : SUCCESS in 5m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d22008ca5d054394bcbf5fb16172ce60 : SUCCESS in 10m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/89421f62acc1488fa5edddfca01907eb : SUCCESS in 20m 19s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1334e7efdc9b4dfb8278601cc8c899d2 : SUCCESS in 55m 56s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10255bc1986f421282e5f7938f74f76c : SUCCESS in 43m 56s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8031509adcdd4dad81c8534bb1f16b68 : SUCCESS in 1h 06m 36s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/50a3cd4a0b054a8786afc95b5e7a5620 : SUCCESS in 18m 21s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6543b9affcb4791ab87eb298e5fe21a : SUCCESS in 14m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/becfd0191fb74319878b9c98d0206ee7 : SUCCESS in 1h 34m 23s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/293f35da09f94d889fbb986fc88a95fb : SUCCESS in 36m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46ee28c5af5d41c2971835a06b86352d : SUCCESS in 23m 29s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e7b71608be1d4f68a39ac45d87c1e0b3 : SUCCESS in 25m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/469f0982bb434d1aa1e905fad3d5e7f7 : SUCCESS in 14m 52s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48b345fbcca546f8b043b87b5da2f076 : SUCCESS in 10m 54s","accounts_in_message":[],"_revision_number":3},{"id":"c83c7d45e1165ecab16c154f0c0327a1a0eff60d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-03 14:00:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2cd30dc4d7d29fd1ffdbbd314fcd8f1bf0f1a36b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 16:03:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f121e1f0cf94397862ae128a10499f8 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/103ee260676e4257ae71e9c917ad1c10 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a67aaec77bdf45fab617a29c63456f26 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35572d003abf477fba1345ddd2a8e924 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9f1db94bae148cda7abd74b8aece8ef : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0e7b6a99a2a4348873d7b90aa822c01 : SUCCESS in 10m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffcebffea5464d8ab1abee155c9ec627 : SUCCESS in 3m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b83675cb742245809f43d1b921d8b998 : SUCCESS in 7m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2afbf088ff804881b712bef8221d2a0d : SUCCESS in 10m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6c718aeb94284e54ac37cd0f20794952 : SUCCESS in 16m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/88e9c2ce0a234c8cbb3a046155f74481 : SUCCESS in 1h 14m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa8e85b7a7de48ac9dd71018c866d66b : SUCCESS in 40m 36s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c7f44420105d44da992b65a17b7a94bf : SUCCESS in 1h 17m 12s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a60088e5ecf24fc28b404cc367343e8e : SUCCESS in 18m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5be5f65d7896456e98ca18cec9c10db3 : SUCCESS in 18m 43s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6c108e077b3140748cd2138c6029db83 : SUCCESS in 1h 59m 04s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a91dc0e6d97c49999653d88df9939c76 : SUCCESS in 36m 20s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6640febd94cb483fa0db7b22a59b2f9a : SUCCESS in 23m 56s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d723d8584c14410b83acdc5d3f0f35ae : SUCCESS in 30m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c7e85f69bf3a4c9e915777b3e2f6c2fe : SUCCESS in 14m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c92696ef9f8f4e779c7b1a3bfdd0efc3 : SUCCESS in 9m 26s","accounts_in_message":[],"_revision_number":4},{"id":"e82c90d81d65779f38e4ebcd4373a4013cb5b739","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-03 17:41:06.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"be25546ccf2561eb75829fb20d32946a6e780ee5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-03 17:51:00.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"da837e6c81609901a681b95719931f213e9b4637","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 17:51:46.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0a8f6600f1f6b490957de2e17947ca17b724a64d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-03 18:15:10.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"3480003a70d22f600d2fb03a800fdf607c4df816","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-03 18:29:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7053536e058cca349338d9de4c2b50040995465a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 19:33:25.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33b5ca4a286e4c5b85a237a6aade7489 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f1c92217ba94607b0590d09b51a4d63 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0459ef5e29634dbdbfab91dae399695d : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a1e7815af9e4bc6973e5d08f89a9b83 : SUCCESS in 9m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27bef22783a445a59eae8832c3f8994c : SUCCESS in 6m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/25245b2c81504f1896632d930f9c56d1 : SUCCESS in 9m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/639e1e22e58945c1b0ad905ca1efd302 : SUCCESS in 14m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8f9b9f048b91400abaa9599de1d0331c : SUCCESS in 22m 11s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f1229f6ccb85463da6317137e8f4d469 : SUCCESS in 1h 06m 09s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88efc7add57440c4bbcaf05e95c4893b : SUCCESS in 36m 22s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/dc809578bb0c4476bf41c5486c63de3b : SUCCESS in 1h 09m 39s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bad93b0b744e4d92af10c2cd4ae4d0ed : SUCCESS in 21m 12s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80b3aae7c5c34e9db7b2812b761b0178 : SUCCESS in 13m 31s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1c766c80e6634580a1f705b68c5586eb : SUCCESS in 1h 37m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/30e1118d17644a5bbeddae0b51893334 : SUCCESS in 38m 21s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2678aeac47b1492dac995f8800ed0e48 : SUCCESS in 31m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4d72a0b71ac540369250f7354ee0b8c6 : SUCCESS in 33m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f1722b72203743758360689db256d9bc : SUCCESS in 16m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bf9cbec71e4467587a17fa60f17a900 : SUCCESS in 13m 23s","accounts_in_message":[],"_revision_number":4},{"id":"0aed31d29774bac042a95323cdc0608f8191dc4f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 19:33:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"4dfa0b3dff8d652412e99559077d9a4b8888cd5e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 19:35:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db0d5e6b030a41859343903ca7a6dfa3 : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbf1d164536740739099a4e3c09695f0 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":4},{"id":"633f8afaeb9693e172d05e074a5553e34ca9a44b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-03 20:32:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"366567ce53ee921958645f57115c4eab03f17f8b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-03 20:52:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"180fd4ba38061365e790699fff0e87c5ce4a5b98","revisions":{"dceb5b9524f44cbe4e593913118f628541c45216":{"kind":"REWORK","_number":1,"created":"2022-02-02 16:45:19.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/03/827403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/827403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/827403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/827403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/827403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/827403/1"}}},"commit":{"parents":[{"commit":"11e8ffc69b529445dcdbbc48280cd7edf5286a4b","subject":"Merge \"Fix Sphinx syntax typo\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/11e8ffc69b529445dcdbbc48280cd7edf5286a4b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-02 12:04:04.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-02 16:45:19.000000000","tz":0},"subject":"ntp: Fix service mask when service doesn\u0027t exist","message":"ntp: Fix service mask when service doesn\u0027t exist\n\nThe ntp role attempts to mask NTP services that may conflict with\nchrony. Currently we are seeing the following failure:\n\n    TASK ntp : Mask alternative NTP clients to prevent conflicts\n    Could not find the requested service systemd-timesyncd.service: host\n\nThe service_facts module shows that the service is not found:\n\n    \"systemd-timesyncd.service\": {\n        \"name\": \"systemd-timesyncd.service\",\n        \"source\": \"systemd\",\n        \"state\": \"stopped\",\n        \"status\": \"not-found\"\n    },\n\nAccording to the Internet, this can happen if there are After/Before\ndependencies in one service on another that does not exist.\n\nThis change fixes the handling of these not-found services to avoid the\nerror.\n\nStory: 2009821\nTask: 44401\n\nChange-Id: I4b8a42704f2b0a145ee9dec433d91df67628cd9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dceb5b9524f44cbe4e593913118f628541c45216"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dceb5b9524f44cbe4e593913118f628541c45216"}]},"branch":"refs/heads/stable/wallaby"},"ab3897c30e9e3515d03c23bf827e2641e64b8726":{"kind":"REWORK","_number":2,"created":"2022-02-03 09:32:25.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/03/827403/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/827403/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/827403/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/827403/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/827403/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/827403/2"}}},"commit":{"parents":[{"commit":"11e8ffc69b529445dcdbbc48280cd7edf5286a4b","subject":"Merge \"Fix Sphinx syntax typo\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/11e8ffc69b529445dcdbbc48280cd7edf5286a4b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-02 12:04:04.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-02-03 09:31:44.000000000","tz":60},"subject":"ntp: Fix service mask when service doesn\u0027t exist","message":"ntp: Fix service mask when service doesn\u0027t exist\n\nThe ntp role attempts to mask NTP services that may conflict with\nchrony. Currently we are seeing the following failure:\n\n    TASK ntp : Mask alternative NTP clients to prevent conflicts\n    Could not find the requested service systemd-timesyncd.service: host\n\nThe service_facts module shows that the service is not found:\n\n    \"systemd-timesyncd.service\": {\n        \"name\": \"systemd-timesyncd.service\",\n        \"source\": \"systemd\",\n        \"state\": \"stopped\",\n        \"status\": \"not-found\"\n    },\n\nAccording to the Internet, this can happen if there are After/Before\ndependencies in one service on another that does not exist.\n\nThis change fixes the handling of these not-found services to avoid the\nerror.\n\nNote this patch also disables Horizon deployment to work around disk\nspace issues in CI.\n\nStory: 2009821\nTask: 44401\n\nChange-Id: I4b8a42704f2b0a145ee9dec433d91df67628cd9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab3897c30e9e3515d03c23bf827e2641e64b8726"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab3897c30e9e3515d03c23bf827e2641e64b8726"}]},"branch":"refs/heads/stable/wallaby"},"588d838a6b8b836b3c6069274417ea432411a8f8":{"kind":"REWORK","_number":3,"created":"2022-02-03 11:38:19.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/03/827403/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/827403/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/827403/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/827403/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/827403/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/827403/3"}}},"commit":{"parents":[{"commit":"11e8ffc69b529445dcdbbc48280cd7edf5286a4b","subject":"Merge \"Fix Sphinx syntax typo\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/11e8ffc69b529445dcdbbc48280cd7edf5286a4b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-02 12:04:04.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-02-03 11:37:57.000000000","tz":60},"subject":"ntp: Fix service mask when service doesn\u0027t exist","message":"ntp: Fix service mask when service doesn\u0027t exist\n\nThe ntp role attempts to mask NTP services that may conflict with\nchrony. Currently we are seeing the following failure:\n\n    TASK ntp : Mask alternative NTP clients to prevent conflicts\n    Could not find the requested service systemd-timesyncd.service: host\n\nThe service_facts module shows that the service is not found:\n\n    \"systemd-timesyncd.service\": {\n        \"name\": \"systemd-timesyncd.service\",\n        \"source\": \"systemd\",\n        \"state\": \"stopped\",\n        \"status\": \"not-found\"\n    },\n\nAccording to the Internet, this can happen if there are After/Before\ndependencies in one service on another that does not exist.\n\nThis change fixes the handling of these not-found services to avoid the\nerror.\n\nNote this patch also disables Horizon deployment and reduces reserved\nblocks on the root file system to work around disk space issues in CI.\n\nStory: 2009821\nTask: 44401\n\nChange-Id: I4b8a42704f2b0a145ee9dec433d91df67628cd9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/588d838a6b8b836b3c6069274417ea432411a8f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/588d838a6b8b836b3c6069274417ea432411a8f8"}]},"branch":"refs/heads/stable/wallaby"},"180fd4ba38061365e790699fff0e87c5ce4a5b98":{"kind":"REWORK","_number":4,"created":"2022-02-03 14:00:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/03/827403/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/827403/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/827403/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/827403/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/827403/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/827403/4"}}},"commit":{"parents":[{"commit":"11e8ffc69b529445dcdbbc48280cd7edf5286a4b","subject":"Merge \"Fix Sphinx syntax typo\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/11e8ffc69b529445dcdbbc48280cd7edf5286a4b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-02-03 13:59:45.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-02-03 13:59:45.000000000","tz":60},"subject":"ntp: Fix service mask when service doesn\u0027t exist","message":"ntp: Fix service mask when service doesn\u0027t exist\n\nThe ntp role attempts to mask NTP services that may conflict with\nchrony. Currently we are seeing the following failure:\n\n    TASK ntp : Mask alternative NTP clients to prevent conflicts\n    Could not find the requested service systemd-timesyncd.service: host\n\nThe service_facts module shows that the service is not found:\n\n    \"systemd-timesyncd.service\": {\n        \"name\": \"systemd-timesyncd.service\",\n        \"source\": \"systemd\",\n        \"state\": \"stopped\",\n        \"status\": \"not-found\"\n    },\n\nAccording to the Internet, this can happen if there are After/Before\ndependencies in one service on another that does not exist.\n\nThis change fixes the handling of these not-found services to avoid the\nerror.\n\nNote this patch also reduces reserved blocks on the root file system to\nwork around disk space issues in CI.\n\nStory: 2009821\nTask: 44401\n\nChange-Id: I4b8a42704f2b0a145ee9dec433d91df67628cd9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/180fd4ba38061365e790699fff0e87c5ce4a5b98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/180fd4ba38061365e790699fff0e87c5ce4a5b98"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
