)]}'
{"id":"openstack%2Fkayobe~827404","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I4b8a42704f2b0a145ee9dec433d91df67628cd9d","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-02-03 20:54:24.000000000","reason":"Maksim Malchuk replied on the change"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-04 11:51:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4b8a42704f2b0a145ee9dec433d91df67628cd9d","subject":"ntp: Fix service mask when service doesn\u0027t exist","status":"MERGED","created":"2022-02-02 16:45:44.000000000","updated":"2022-02-04 11:55:28.000000000","submitted":"2022-02-04 11:51:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827404","cherry_pick_of_change":827486,"cherry_pick_of_patch_set":2,"meta_rev_id":"28163de84f7b55efe6a8f1a963afd92658c90397","_number":827404,"virtual_id_number":827404,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-04 11:51:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2022-02-02 17:09:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2022-02-03 18:13:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-03 20:32:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-03 20:32:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"date":"2022-02-03 20:54:24.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2022-02-02 16:46:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-02 17:09:51.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2022-02-02 18:49:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-03 18:13:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2022-02-03 20:32:22.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"0e427c3147e3d08c0a3b3158c21aaf1e2dff90ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-02 16:45:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c1bacdf710be3a9e4e78b6b73bcda6f4f6d2318b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-02 16:46:29.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"2b9be6e9255f34bbf0443529854001f047ec6c4c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-02-02 17:09:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ee5371b7f8df82fe30dfb8910def88ff5ce9786e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 18:49:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9558e9a281474570a62a0e34d1f25645 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3ba6838ecad4019be7b3337e2f9389c : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85b6a9bfa98e43ee810f590a38870ace : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f03a4bc52bf4e39bcadf6e3867cd960 : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8ac5f85b6d747798fbce76a19290c87 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d287e0d8d2e144b88c1fab27d13d44b7 : SUCCESS in 9m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38c602b27d2045e8a406fbf4664b44d1 : SUCCESS in 4m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7929cbe8162e4355ae3abaece4e2de9c : SUCCESS in 7m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/93b16971e50b472c9628ac03ea6e1e38 : SUCCESS in 13m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5a9fe7093da84e9fa607f264e8084555 : SUCCESS in 32m 45s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2c0534e13ba043c3a418b02d0ce2813f : SUCCESS in 1h 05m 51s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cb6accb1dc1474e9e84dbed55fa8a75 : SUCCESS in 1h 03m 28s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/15dfff1441854da89714960043ea97f3 : SUCCESS in 1h 09m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/58a35809ac65461cb68cdcf04ffbe427 : SUCCESS in 23m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2407195e60524d7697ad5feb699dea94 : SUCCESS in 15m 21s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f2a5ec72406b491b9070ac127f15c0d6 : SUCCESS in 1h 59m 51s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ec2bf7ea9874f23a856cbc2c172309b : SUCCESS in 1h 13m 29s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4286e420aad946669b1058b776fab6ec : SUCCESS in 33m 56s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7016524880d142a1b01e4aaa21dca8f8 : SUCCESS in 30m 26s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/61fb7bef376a42498d90271f6fa6851f : SUCCESS in 33m 09s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b179cdb4a1e24a9db94e9d14c9d18d40 : SUCCESS in 23m 38s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c3ae3aa1bd2347298f8610fa2aebaa41 : SUCCESS in 17m 12s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/002564541ecd4aba89784ae7359165cf : SUCCESS in 12m 58s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/64ad947063df42eeb8307f3265ce1ff8 : SUCCESS in 20m 53s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c1947a6b2f644df96048d79996d0d29 : SUCCESS in 14m 20s","accounts_in_message":[],"_revision_number":1},{"id":"e4ffe3a843904a2482ef6f222a77401422babc0c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-03 18:13:21.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"4743d7d8f779808d613138ab8a9903d8140b7492","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-03 18:13:43.000000000","message":"Patch Set 1: -Backport-Candidate Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"daedb09b873c3806555ff33ac0ddcc7c5a938fbb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-03 18:26:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7eabed18db6693111317306d82ecda12cc88f93a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-03 18:28:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c18836f8a89c0ae732b25d7fce12188e4085549a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-03 20:32:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3769b4cabd9bded8c991630796ac538a2b368483","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 20:33:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fdfde5d258773c9c11461f302f3dd54ed9716b8d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-03 20:50:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d7a8b898051a92cebe8daebec1c1eb23695292e1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-03 20:54:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4193e5cce60810eed233a63c222ea845d01cc0a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 22:13:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/757b85544d7d4fada8f720fa96bd7172 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b4a56060e774597bd5f8808baacc0b3 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f93e0d49c12240a083b5798352cf5674 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d1e78e0f42d4a62adca646615ca4c69 : SUCCESS in 10m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6beea0e5797426aa0f7e6e5ca725be7 : SUCCESS in 5m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f1587ee3ad0f4598a8fb8a4445e19e37 : SUCCESS in 7m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/827c93e9af1a42b3a73cde5929ada22c : SUCCESS in 11m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/06fcc23aec154949b244ba9eb00f4d62 : SUCCESS in 30m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6c6b01a975124fb19dd389d6f79295ce : SUCCESS in 58m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39795fe780b04c8daa9a2d9c3de49a9a : FAILURE in 43m 41s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/840f7d8b307544e28c7cdb2890fa9ca9 : SUCCESS in 1h 02m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d4a9bab7a4bc422d944bf1bdc808110c : SUCCESS in 19m 31s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b589e8056784967800c6e2b21e5d564 : SUCCESS in 12m 50s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e4f183155f1f495c83c95f577dba3465 : SUCCESS in 1h 32m 52s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64fb567c80ac4b538eebff84e00707ee : SUCCESS in 1h 04m 33s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e52d1f19afce41f8a600bc00326039f2 : SUCCESS in 31m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5e53f0f84624f828a96f53565f0234d : SUCCESS in 32m 54s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ba110a3f15074463bc982779496e7e84 : SUCCESS in 24m 53s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/219af0a636ed49a7b7e663395686f642 : SUCCESS in 19m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c6c6b6c2619e4269bfdea2d2bea4678e : SUCCESS in 13m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8042e3773974e5e993a1da117b8b368 : SUCCESS in 10m 07s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/171e9121ab464f2c841adf7e9330f336 : SUCCESS in 14m 24s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dfd920811db42f98a42d00129228a5e : SUCCESS in 10m 57s","accounts_in_message":[],"_revision_number":1},{"id":"593c9314a148698b57280f4c47a1fceeb69423ae","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-04 08:03:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c94429f51d773d8c0afb6a1084de1180b7f0a933","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 09:57:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7539d220dc5d448385013be81c3ed2fa : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99f213fd4d91464a926090bc14711770 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18bca344295b443abc5fef0344083a56 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fdff62474db476492a30a7201d7a3d2 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce58068963644a32b5922b169c8c232a : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/136ee06adecb4de9b1eaff3f907e0b78 : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7d2d002217c42ed9e87984f68011f59 : SUCCESS in 3m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/67c70b48d59247cf9eb25617744d3d51 : SUCCESS in 5m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dd400e2aa8044f9b938896155811b672 : SUCCESS in 13m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/063b4828af474bc69e28f232449a633f : SUCCESS in 31m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ce03c701ca13473dba55fe6fb3193ce0 : SUCCESS in 1h 05m 39s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17129661265549fca07a1b37387e4459 : SUCCESS in 45m 12s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/64e42b8ca1074db9a89fb67ddae70cc5 : SUCCESS in 1h 07m 11s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8e503fac70db47eeaef2375caaa71447 : SUCCESS in 22m 47s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ba1f269c7eb491a9a75966a2e89ff8f : SUCCESS in 14m 38s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/400742f9003f4c0086e62eb0794f71f3 : SUCCESS in 1h 49m 51s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d67f037f772243bfad853fb0003b67ad : SUCCESS in 1h 46m 06s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2161b799d9764d19a933f4be097e2d8b : SUCCESS in 33m 22s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cc1113828d6458db87a1958d3870ea7 : SUCCESS in 27m 41s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b85c2d73365f42c9bd00f95549dec494 : SUCCESS in 31m 30s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab13e53c57ad45d19f9e752c0484ee5f : SUCCESS in 22m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/92a0a948b80445b58ef3a16be0ac5f20 : SUCCESS in 17m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87ef4a48884148a9be19b2b3d3bef08a : SUCCESS in 11m 53s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/6164d509e43d4150bced07830bb3c12b : SUCCESS in 17m 44s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e187068d1744c99901593275655712d : SUCCESS in 13m 48s","accounts_in_message":[],"_revision_number":1},{"id":"67a0918939158433a3e6c021b3376c2100bc95bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 09:58:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d62c624e7c3035114d98f7b61ed113f990d4a62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 11:51:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62e035cfac594c9a88b3f3ed3c490a68 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f17e11492eb49e9a885d8c16d781760 : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc399900c47041e083d4e0155f97a3cf : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df8de5d978c0458b87a5e5fbb164f52e : SUCCESS in 9m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb23afdc0df0414c927c6b7b515d4307 : SUCCESS in 4m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/78fcf6e9496c4cf2b6799a9925f196ca : SUCCESS in 7m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/93c6680d821e4e499a6ecf19514c3bdd : SUCCESS in 14m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4096a0573c9a4355ab5782ca11f7a051 : SUCCESS in 32m 27s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/48d4f47e00cc4cd4865fa32b3ff8885a : SUCCESS in 1h 08m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a657f99d1f447d0aef28d185760d0c5 : SUCCESS in 45m 51s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/69f4e425ab074091bb655447529853b4 : SUCCESS in 1h 10m 02s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8d730f1fe9624edd8252ce53b50b4ac1 : SUCCESS in 24m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b68d450d6b73462da95009e78ac91f5b : SUCCESS in 16m 10s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e25834129d3549989a6dc6ac34888311 : SUCCESS in 1h 48m 30s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e2be4a8184749f4840d5f6dbf3332c6 : SUCCESS in 1h 32m 45s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0420dd82ec6542f19089613b51bc635b : SUCCESS in 39m 45s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67e4d6596274419b84e9524c57b21d46 : SUCCESS in 31m 01s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/604cfbfc54d948808bdbb922f9d931d6 : SUCCESS in 36m 17s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/160b1925be414f4b95c512dce7f858ec : SUCCESS in 28m 59s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ba21013ec351475b90013da6d3814760 : SUCCESS in 19m 56s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3306eb38a94f466c8bc74a6c3014912f : SUCCESS in 14m 45s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/31a317ca118541e682f4e1efb07c28b3 : SUCCESS in 21m 06s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60560c89d3954463a8187378a17ba3a2 : SUCCESS in 15m 47s","accounts_in_message":[],"_revision_number":1},{"id":"ba1d10f433853cffcadcd32171a638258572b316","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 11:51:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"28163de84f7b55efe6a8f1a963afd92658c90397","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 11:55:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a336b79076134da3b8d32327087c872c : SUCCESS in 1m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/365d45888e4f45199ac00166d2d42d79 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"816456972eb48adc03135b455928fb3b3ea58fdd","revisions":{"816456972eb48adc03135b455928fb3b3ea58fdd":{"kind":"REWORK","_number":1,"created":"2022-02-02 16:45:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/827404/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/827404/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/827404/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/827404/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/827404/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/827404/1"}}},"commit":{"parents":[{"commit":"82cb85ceb812c8d28003c9422521885260893ad6","subject":"Merge \"Fix Sphinx syntax typo\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/82cb85ceb812c8d28003c9422521885260893ad6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-02 12:04:04.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-02 16:45:44.000000000","tz":0},"subject":"ntp: Fix service mask when service doesn\u0027t exist","message":"ntp: Fix service mask when service doesn\u0027t exist\n\nThe ntp role attempts to mask NTP services that may conflict with\nchrony. Currently we are seeing the following failure:\n\n    TASK ntp : Mask alternative NTP clients to prevent conflicts\n    Could not find the requested service systemd-timesyncd.service: host\n\nThe service_facts module shows that the service is not found:\n\n    \"systemd-timesyncd.service\": {\n        \"name\": \"systemd-timesyncd.service\",\n        \"source\": \"systemd\",\n        \"state\": \"stopped\",\n        \"status\": \"not-found\"\n    },\n\nAccording to the Internet, this can happen if there are After/Before\ndependencies in one service on another that does not exist.\n\nThis change fixes the handling of these not-found services to avoid the\nerror.\n\nStory: 2009821\nTask: 44401\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/827403\n\nChange-Id: I4b8a42704f2b0a145ee9dec433d91df67628cd9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/816456972eb48adc03135b455928fb3b3ea58fdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/816456972eb48adc03135b455928fb3b3ea58fdd"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
