)]}'
{"id":"openstack%2Fkayobe~827486","triplet_id":"openstack%2Fkayobe~master~I4b8a42704f2b0a145ee9dec433d91df67628cd9d","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-04 10:21:45.000000000","reason":"Mark Goddard replied on the change"}},"hashtags":[],"change_id":"I4b8a42704f2b0a145ee9dec433d91df67628cd9d","subject":"ntp: Fix service mask when service doesn\u0027t exist","status":"MERGED","created":"2022-02-02 12:12:19.000000000","updated":"2022-02-04 13:39:50.000000000","submitted":"2022-02-04 13:37:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"827486","meta_rev_id":"30c509baa567f05cacb3371b8ec69bc031cff49c","_number":827486,"virtual_id_number":827486,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-04 13:37:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2022-02-02 17:10:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2022-02-03 10:44:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-02-03 15:14:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2022-02-03 10:44:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2022-02-02 16:46:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-02 13:18:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-02 17:10:03.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2022-02-03 10:44:03.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-02-03 15:14:59.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"d5c68222ff46e18a07c4a1a9f93be03b114f667f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-02 12:12:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7affe4d6ac71a981ca28927ac856bce54d000c3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 13:18:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1d80baffaff4a95ae58249719a17f1a : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42d1d42c1e9a4d4c845c696de574fc84 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d53eb574a6e4e50ab5c8ab731409204 : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e52f32abc1244766a856f03952245c6f : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1934b94a39fe48febb15bfce120f7ea6 : SUCCESS in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6e6265a5b71428aa4e480780de64f2b : SUCCESS in 4m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/972de53482f84e6e82c554385acba7b8 : SUCCESS in 6m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0a9d297a8025415baeda8c8e18cf7302 : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2fcd2d15a64e43c2870bb898cb6e5035 : SUCCESS in 34m 26s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4a6690ddd51f480a91a3925090d11319 : FAILURE in 14m 38s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/430a521256ff49ce8800d262271bed91 : SUCCESS in 48m 38s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f9c7fb9fde2f42b09eaec525ff8f6918 : FAILURE in 10m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1402aee7196140e2afa689f34cb5d2d9 : FAILURE in 15m 04s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff2d9dbb855349cf8a9cfb87da1a7b70 : FAILURE in 13m 31s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/14e68fc4a7ff4367a5581b2443eb5ee6 : FAILURE in 14m 35s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6801b13b0f2c4c61a34055f5e771900f : SUCCESS in 1h 01m 18s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/649bd995a501467999ffbb7ea6a3f29d : FAILURE in 12m 56s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b434a966baa4a1697fc19ac43f2573a : FAILURE in 26m 58s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c3a00ea1171347a193d095ce1694d946 : FAILURE in 13m 47s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef901fb16da74957a1f493cc6b127cee : SUCCESS in 18m 57s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/862733283f6a47f3a5dc64e72de1a7f4 : FAILURE in 12m 32s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d48d79161ba644739c77780e3dc0cd5e : SUCCESS in 10m 24s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/a2c6c43fb05d4fb2acbd0834e00a325a : FAILURE in 12m 26s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6906436352854c9392cef4f8aab7014c : SUCCESS in 12m 05s","accounts_in_message":[],"_revision_number":1},{"id":"7b3b11518eff0a5a8f3c3e5b276c004e72aa4c66","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-02 13:42:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1b811b3c4f380569bef0838a5764b9916a27bf36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-02 13:57:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"32d3f5a60ae0033277bbd302a146b4128aa8fd19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 15:40:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f94010554994dcaa83e7be44710f485 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18a3c8d41d694275b287312a039d56b4 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3152bca29034c68a5793eedaa7b8b29 : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a5a520afb5d4b1faa353b722a049504 : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b01dfe9c963b4880a7b8aee86c401fc6 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/262f672ad26a46438f8921ba8054449a : SUCCESS in 5m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/924dcfff10f340c98ac9805a6670916d : SUCCESS in 6m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f574757e4b30430780b8141c0f6507c9 : SUCCESS in 13m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aa519bd70c4b4a729637e15ee0e7b27d : SUCCESS in 34m 42s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3b20ac1919824cd0804aa2203aba7730 : SUCCESS in 1h 15m 17s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8db4aba107c24f4ea0f42200112e9d12 : SUCCESS in 49m 57s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/df9c230e525c46f892a20b3fa64812eb : SUCCESS in 59m 22s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/adbb29e923b34cce92e01d0aa09f3836 : SUCCESS in 22m 03s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03f90b2953e24c17b11f6524fb786d04 : SUCCESS in 19m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3df5a63a53814ed7913657a648617a19 : FAILURE in 14m 24s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/407703636f454f5c934ecf92511a9a30 : SUCCESS in 59m 09s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/73beb103d6c340669156be5b0163dee3 : SUCCESS in 44m 32s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1dc0de19b7654c5daf42619472fd1fd5 : TIMED_OUT in 1h 35m 27s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/eeba3300df994d87a967350e6a3939a0 : FAILURE in 12m 58s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6d1344cf9154cb38053338143a424ea : SUCCESS in 24m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9e66fee070e84113b7668e40e605d610 : SUCCESS in 13m 16s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/644fcf3cf0f94064afa3af334a751c70 : SUCCESS in 13m 18s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/f84d9a9fc9be4fcf99d79233c6e38aec : SUCCESS in 21m 20s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b77e856588ff4e0589a01196e58d609a : SUCCESS in 15m 26s","accounts_in_message":[],"_revision_number":2},{"id":"ec3be293bb3954feae1d22c1bedbe961fe47541b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-02 16:46:03.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"cdf902d1a9cbc7c094ad1467cfb953d04a84a521","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-02 16:46:14.000000000","message":"Patch Set 3: Review-Priority+2","accounts_in_message":[],"_revision_number":3},{"id":"9293c039c7931f6bf46434ce4674ec133ca9085c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-02-02 17:10:03.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8f07915e44cf854156561430042bb65064f3b88e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 18:46:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fc20190563947d9b3fb0bf5c4c31128 : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a039c1b0de44de3bead3f1dacbda6a3 : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32a8375328e0461f860ff83a56cb1295 : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12457a46e9ca4edcac4755a87e5f61c7 : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffa0bc0ae6b747d194f615f617053cda : SUCCESS in 11m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f746c133217540a9a885401f390287e5 : SUCCESS in 4m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0235a398a5fe4a2e94cb9126cfd5a129 : SUCCESS in 6m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8d3c8ae29e274b1bbc95683a6bc0b4ba : SUCCESS in 14m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7b5b04f2f9dd4649812b86edb2d24fb7 : SUCCESS in 32m 28s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0fc1fc72835f48708a1fe384825a5505 : SUCCESS in 1h 25m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2bd163803484418918b3caa5403cd52 : SUCCESS in 53m 07s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/83b312dafd57464db6ac0e9304554c0b : SUCCESS in 1h 09m 39s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e90081d380db4f94a887400e10276329 : SUCCESS in 23m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53a5b782a8c241adb6757d00e680a058 : SUCCESS in 18m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/df9b2fb921d145af80d2718b29e30b10 : SUCCESS in 1h 57m 13s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c620718adc0f450183985d4748d98a57 : SUCCESS in 1h 19m 35s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/03ddba73c0c846a7b9db3a133c591ba3 : SUCCESS in 53m 49s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96811678c9324665baa3dfce45787623 : SUCCESS in 47m 03s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f2727ca1a3d44a9598c5b3a97ba84f22 : SUCCESS in 30m 45s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79f62d8b358b47a49831baa2e8498ca0 : SUCCESS in 24m 03s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c3e75e3c60c04b4eaff9aa95e9ce338b : SUCCESS in 17m 32s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14eff8c04a6843fe8175ac4f583c597e : SUCCESS in 13m 56s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/6cce23d6e8844085a1e0ed60e3e0d7e6 : SUCCESS in 20m 37s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4daeb5dc54c40b08c3a0f7a5519d2ad : SUCCESS in 15m 23s","accounts_in_message":[],"_revision_number":3},{"id":"5cc59b6cc5fc3c87c1f23199bdc1cfc90f62b068","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-03 10:44:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"09563d38f25a1558692f0ed2276d4107fbdc76d5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-03 15:14:59.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"988a96566a47f9c1ab33574926702b489d06e13b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-03 18:25:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0c481c98a57ea5a01b05152157b589bc1680d558","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-03 20:02:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"89a676163e6ea3a0cb5ece9e3457ff0013297c55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 20:33:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"52ecde6e7d7b069af4ffaa83b8098817c072a88e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 22:13:07.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"6cdc7042e84ad20337da051af8c161a8a3a4d171","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-04 10:21:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2f7c4c4d4c05f30b0edbf811448f25c9f23acfaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 12:04:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48b6ced977404c90b05468d6b087db37 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0c50e64e3484e64b9688384319e680e : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37bf6ea9ccc64ad1ad3b04b1bee96a97 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6333f2903fe45358b927fb32489fd18 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60c073074dd4403eb0d0b06b87d6ac18 : SUCCESS in 8m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a3a4614d1ba4d99a85ef608482fb998 : SUCCESS in 4m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9b93f71f7f2948c3843033750cc773c1 : SUCCESS in 6m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c1a30498ede14655b340e1a860fecdf0 : SUCCESS in 13m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5812512762de4bad91594b8cfa9b5775 : SUCCESS in 32m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/36ba07ae31494e7ba5f05babcdd9cb41 : SUCCESS in 59m 43s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b20a894f83764d1d91fe3ed58e806566 : SUCCESS in 48m 10s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/98cc823e5b644de0baa3839ba65f4146 : SUCCESS in 1h 10m 09s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9e9326f7264043a787647a335d0ce33a : SUCCESS in 23m 46s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a1d2fce4a1347919e7148070f1410c2 : SUCCESS in 15m 33s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bc2c1687bb2f47c3a14fa768a5d0e5d4 : SUCCESS in 1h 39m 27s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97d1e06247004894af7c5947a35b618e : SUCCESS in 1h 09m 48s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3c343335b38349bfa9f0f30d4031ffb9 : SUCCESS in 41m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0799480e466947c6aba25ebf5cf7e5c7 : SUCCESS in 37m 04s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2a5d378223df4146b384406cb8e45743 : SUCCESS in 30m 50s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a7f97c5607241f0b4bc24297df833d9 : SUCCESS in 22m 26s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/36315f0cc59f4863a02d93fc44cf4eb7 : SUCCESS in 17m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/505de053c0e44817982d211bce0236ad : SUCCESS in 13m 24s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/3497b9fbf426409688a4bcf5ac8040c4 : SUCCESS in 19m 53s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba7af97f2a2947eb97ce0addf994287b : SUCCESS in 14m 08s","accounts_in_message":[],"_revision_number":3},{"id":"76dcb40705846844a0ba0de8ecd40fc9a3d2f93e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 12:09:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1703d2b93eb0fe2b6a036b57dcbc6acc54d95594","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 13:37:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1987e27c8a5346c89512095b5f79e551 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aeb95175a41c4c15835e4f929a812b92 : SUCCESS in 7m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/199d9be4f1b441b2887462482dd18bf6 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebc0c0be17384cbe8bee8258fd1824c2 : SUCCESS in 9m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be8068101e0f4282b3b7b3a8d6ed81df : SUCCESS in 4m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/63c42d844f9a48bca154c24be87ecef2 : SUCCESS in 7m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/34c2a1470ac24583a483006f85870bbb : SUCCESS in 13m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/19b72a9670f14cd68c4b535810b94670 : SUCCESS in 32m 11s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f9aa5b75a2c74dc6a44bb705660cda49 : SUCCESS in 52m 29s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/119d97a9520d41c3b96f760a5dd11968 : SUCCESS in 46m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/266bf4765e424f1099926ab9611de21f : SUCCESS in 1h 23m 14s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/29aeae75d93f4caf9a60d82369a72c55 : SUCCESS in 28m 21s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/250ee38020754597bd26b18d54eac91d : SUCCESS in 17m 19s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2f10883369c34afd849ba8e47f5aab40 : SUCCESS in 1h 27m 48s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95725e344fd84a679e4a810b266f28e6 : SUCCESS in 1h 19m 29s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5221d4ce64c04cc88d8603a21dc03109 : SUCCESS in 40m 03s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34909de119804fff8afbc7adfa497890 : SUCCESS in 39m 58s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/828c211fe0504eda82076929abf3b6ce : SUCCESS in 28m 12s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87c3c6c8a1e9449498503e7a93c1d861 : SUCCESS in 23m 53s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c95ba1c84b804d468c339d1f551bd7d0 : SUCCESS in 17m 32s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb514e8e71194251a3b5847e2953fe9d : SUCCESS in 15m 14s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/78a851cbb04342c181624b68e76d1fa2 : SUCCESS in 20m 11s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99cab64dde0b4d83ad254ebe6c9efdbb : SUCCESS in 14m 55s","accounts_in_message":[],"_revision_number":3},{"id":"11132a401048379bd066030ae50209d93ba27523","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 13:37:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"30c509baa567f05cacb3371b8ec69bc031cff49c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 13:39:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/041120eed21f446696f7eb770fe7a882 : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bcc10cc8db84548bd79e9210cf49597 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7173a9f73126eca1cc36bd6997acf10ac75832ff","revisions":{"e4df5b09f6f9a9d7331b2fa81bb8bbfe94a62226":{"kind":"REWORK","_number":1,"created":"2022-02-02 12:12:19.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/86/827486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/827486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/827486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/827486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/827486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/827486/1"}}},"commit":{"parents":[{"commit":"d6e836586762e41211940532b736967bad631cc3","subject":"Merge \"Support for untemplated dirs in kolla config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d6e836586762e41211940532b736967bad631cc3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-02 12:04:04.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-02 12:11:19.000000000","tz":0},"subject":"ntp: Fix service mask when service doesn\u0027t exist","message":"ntp: Fix service mask when service doesn\u0027t exist\n\nThe ntp role attempts to mask NTP services that may conflict with\nchrony. It appears that CentOS stream 8 has removed the\nsystemd-timesyncd service, which leads to the following failure:\n\nTASK ntp : Mask alternative NTP clients to prevent conflicts\nCould not find the requested service systemd-timesyncd.service: host\n\nThe ntp role should handle services that do not exist. This change fixes\nthe handling of missing services to avoid the error.\n\nStory: 2009821\nTask: 44401\n\nChange-Id: I4b8a42704f2b0a145ee9dec433d91df67628cd9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4df5b09f6f9a9d7331b2fa81bb8bbfe94a62226"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4df5b09f6f9a9d7331b2fa81bb8bbfe94a62226"}]},"branch":"refs/heads/master"},"7aa0207cb074ec75d9461ac910987b97b8358c1b":{"kind":"REWORK","_number":2,"created":"2022-02-02 13:57:41.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/86/827486/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/827486/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/827486/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/827486/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/827486/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/827486/2"}}},"commit":{"parents":[{"commit":"d6e836586762e41211940532b736967bad631cc3","subject":"Merge \"Support for untemplated dirs in kolla config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d6e836586762e41211940532b736967bad631cc3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-02 12:04:04.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-02 13:57:15.000000000","tz":0},"subject":"ntp: Fix service mask when service doesn\u0027t exist","message":"ntp: Fix service mask when service doesn\u0027t exist\n\nThe ntp role attempts to mask NTP services that may conflict with\nchrony. Currently we are seeing the following failure:\n\n    TASK ntp : Mask alternative NTP clients to prevent conflicts\n    Could not find the requested service systemd-timesyncd.service: host\n\nThe service_facts module shows that the service is not found:\n\n    \"systemd-timesyncd.service\": {\n        \"name\": \"systemd-timesyncd.service\",\n        \"source\": \"systemd\",\n        \"state\": \"stopped\",\n        \"status\": \"not-found\"\n    },\n\nAccording to the Internet, this can happen if there are After/Before\ndependencies in one service on another that does not exist.\n\nThis change fixes the handling of these not-found services to avoid the\nerror.\n\nStory: 2009821\nTask: 44401\n\nChange-Id: I4b8a42704f2b0a145ee9dec433d91df67628cd9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7aa0207cb074ec75d9461ac910987b97b8358c1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7aa0207cb074ec75d9461ac910987b97b8358c1b"}]},"branch":"refs/heads/master"},"7173a9f73126eca1cc36bd6997acf10ac75832ff":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-02-02 16:46:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/86/827486/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/827486/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/827486/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/827486/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/827486/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/827486/3"}}},"commit":{"parents":[{"commit":"d6e836586762e41211940532b736967bad631cc3","subject":"Merge \"Support for untemplated dirs in kolla config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d6e836586762e41211940532b736967bad631cc3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-02 12:04:04.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-02 16:46:03.000000000","tz":0},"subject":"ntp: Fix service mask when service doesn\u0027t exist","message":"ntp: Fix service mask when service doesn\u0027t exist\n\nThe ntp role attempts to mask NTP services that may conflict with\nchrony. Currently we are seeing the following failure:\n\n    TASK ntp : Mask alternative NTP clients to prevent conflicts\n    Could not find the requested service systemd-timesyncd.service: host\n\nThe service_facts module shows that the service is not found:\n\n    \"systemd-timesyncd.service\": {\n        \"name\": \"systemd-timesyncd.service\",\n        \"source\": \"systemd\",\n        \"state\": \"stopped\",\n        \"status\": \"not-found\"\n    },\n\nAccording to the Internet, this can happen if there are After/Before\ndependencies in one service on another that does not exist.\n\nThis change fixes the handling of these not-found services to avoid the\nerror.\n\nStory: 2009821\nTask: 44401\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/827404\n\nChange-Id: I4b8a42704f2b0a145ee9dec433d91df67628cd9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7173a9f73126eca1cc36bd6997acf10ac75832ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7173a9f73126eca1cc36bd6997acf10ac75832ff"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
