)]}'
{"id":"openstack%2Fkayobe~827487","triplet_id":"openstack%2Fkayobe~master~I86fcc75e844eb922f62c90c45a105519845cc1a7","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-24 16:30:17.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-02-24 13:02:39.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"I86fcc75e844eb922f62c90c45a105519845cc1a7","subject":"ntp: Remove removal of chrony container","status":"MERGED","created":"2022-02-02 12:12:19.000000000","updated":"2022-02-24 16:32:41.000000000","submitted":"2022-02-24 16:30:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827487","meta_rev_id":"dcc9d08269677855e29b53717293f0bbab023bab","_number":827487,"virtual_id_number":827487,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-24 16:30:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-02-24 16:32:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-23 22:36:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-23 22:36:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-02 13:55:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-10 07:54:40.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-02-21 11:56:42.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"4d6f8ea78c3db3c3a9d27028e274760e809de9b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-02 12:12:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4fd78f259e0316d219fa6f0aa7c9c2fc45a013f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 13:55:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84c6171736064c8c93c0d0b6b2f4eab2 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9b3f092e829479cb756c2c27ec6c0a6 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37aa5b84d70c43bca00fde36e8441704 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0a443e88cad488090f4b5397418fdce : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cae4017a5ee4f2690aad3da77999b77 : SUCCESS in 9m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8ea91553632e4baaa7d07c7a058de829 : SUCCESS in 6m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4c0ec8fbc3484030939df0a2313e6f4d : SUCCESS in 13m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f3031e6eb65f41c2923a0c2ea2e08c69 : SUCCESS in 32m 14s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/03e1a4aadad242b1bf50ec60af580dcf : FAILURE in 12m 46s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04209cc038fb4580aa683d5dc117930e : SUCCESS in 45m 12s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c30bc1084c9a4e7ea7aaff2c681c29af : FAILURE in 12m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/fa186a4c87464eb1aad84204c6fd330a : FAILURE in 15m 21s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c1529a0dcbb46688fd341829b254b9d : FAILURE in 15m 19s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/85462c6b8a1f411c8567363c24fa3e2a : FAILURE in 14m 43s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a162a31d7a7d4c99a66b1b9da6702d1a : SUCCESS in 1h 19m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c62367988ea74d8d933de7f4c57f3a04 : FAILURE in 12m 03s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/199bb12134db43008bc566705c76a7ac : TIMED_OUT in 1h 34m 21s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/663fec2fa3f249bbbaf33c780ab97a2b : FAILURE in 12m 54s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/257dd7bccc984ba491fb907a12a45beb : SUCCESS in 25m 01s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ebfccbd8710645c3a24493b1e6bf4919 : FAILURE in 12m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff299a15a2ba4a43b9052e1264db44d3 : SUCCESS in 15m 28s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c349390cd5d24aefb7b86b37326d6cd8 : FAILURE in 13m 22s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c3d08d46b3242ab9163b2f4fb05643b : SUCCESS in 15m 39s","accounts_in_message":[],"_revision_number":1},{"id":"70e5a54901231fd4a0acc70add8b6d2e540f11b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-02 13:57:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"be04f9d7e6d51ddf805f028388bcd6493470fda0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 15:40:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fab3523d7914453c86dc67f5e6490f96 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/497b77d522944eaa86707a2954516c40 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5e8607196334cc785fbb1abcf66c4ef : SUCCESS in 8m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15a80375aafb4d788663762a88612feb : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fea8de814d14db299981faf04c30704 : SUCCESS in 9m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9f951c4a928d4f58860fde7e87b556d0 : SUCCESS in 5m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0892e0efa49e4270ab572bad1a76f0c5 : SUCCESS in 12m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/20ef7831d71e433b886736839ececa16 : SUCCESS in 31m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b2ef3f37f36a4b8593e83b4f42ca1465 : SUCCESS in 1h 09m 10s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6358a26657a347689c0414c7bd2c73b3 : SUCCESS in 48m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ee8fb88e2454446a8db3a1b87371f2a2 : SUCCESS in 1h 09m 46s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c84b175be0d743458d22fb05b56daa6a : SUCCESS in 19m 58s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f18bc0c9a79b42528f6c931964f58bc7 : SUCCESS in 16m 46s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8646498beb3549809987b55654d70279 : FAILURE in 12m 21s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e486859ad7f43f29a75d2acee836926 : SUCCESS in 1h 18m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/cc01b7e813c74fc792d925af9b8a36dd : SUCCESS in 45m 50s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54b6ec338c8e4e09aa4d6a1cca1648eb : TIMED_OUT in 1h 34m 42s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e6636e4e10964f9d9aa91ef894615c9c : FAILURE in 10m 56s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c9581a73d524a9baf3ac1493b151a00 : SUCCESS in 26m 50s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/dd84e1f5edf649a29f12d92e0d35a37c : SUCCESS in 16m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73bd9a16ba5244adb0afee3a98e85394 : SUCCESS in 12m 34s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/3be763b7ebe748089c3c072dabc0a206 : SUCCESS in 20m 14s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b28c9cf000184947b30061e6494b7d96 : SUCCESS in 15m 29s","accounts_in_message":[],"_revision_number":2},{"id":"ff26ecb64840bca0f7ca4577e210880c63f6eed9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-08 11:22:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"77ab274532f8df7969fde1472cea62d852061ba6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 13:02:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c134a0b347734b278b2c5857a8718dbe : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4135f5c6f8d4b7896d3dc021ade9a56 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d794779d387f473bb11744a63b145f7f : SUCCESS in 6m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb9b98ff1ff84ef1945e538a4a4de912 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e22a6cbe5d164d89aec585848c3abc9e : SUCCESS in 9m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bb39fe5f445148a4a95a359ab76ce9f1 : SUCCESS in 7m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9c61fe79187c46288f6812c5b8439a28 : SUCCESS in 12m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d4df3bb4bfc64e41980ce45669c577c1 : SUCCESS in 34m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2d1319f989b542a6a66b3d57da378bc6 : SUCCESS in 1h 08m 48s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dabfc2d6a1eb43cba924c0c13a71e717 : SUCCESS in 46m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e878b338df994b21a9e5c30d3c30d7cf : SUCCESS in 1h 01m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ff506a4e79d14563bf031375c37a7657 : SUCCESS in 24m 16s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5121cda34b5d481196a514cedf7f22fb : SUCCESS in 16m 12s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/51b4db1ac9524955b0eaeae7f91bb923 : SUCCESS in 1h 37m 59s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6ce3968373942438464fbfe72db9fa6 : SUCCESS in 1h 10m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f0d426cb96c6460b9b3d5fddb745fc0f : SUCCESS in 44m 43s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ce8e3d9685543ed861ac45703cc7ac1 : SUCCESS in 34m 23s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9c08f7006fb549b38eb7dbc532a27c42 : SUCCESS in 29m 39s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e3069d9cfaa4684aa66bc7095dd85b7 : SUCCESS in 25m 18s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e3a81e5ef79e4903b8cc86d7d8b13997 : SUCCESS in 13m 52s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f1eb30d204542d9b977bcd0031a329b : SUCCESS in 12m 27s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/52a99cb2db6e44a3ada754afbf4f57d9 : SUCCESS in 20m 12s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e97b325c0404b79808213df5ce2ca0a : SUCCESS in 11m 22s","accounts_in_message":[],"_revision_number":2},{"id":"c269f32e86bbb9393565a4592278d6d20703741f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-10 07:54:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"aa41c71ccf297abc2316003335b594cb241350a2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-21 11:56:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"63aaf7c3e3cc4a1d92a1700f94ea621c7c813334","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 11:57:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"20c60973e4266c446f037d732b454630f877d116","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 17:25:49.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edf2a8384bf849a58988097cc9b309ae : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77cf9e918c824cc7b6bcb2be9da02f41 : SUCCESS in 8m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6aceca906a9a495d8a2acbc9267d7b1c : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/181fb977df0b4162b313bdb6a2652306 : SUCCESS in 10m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/89c01e9ff24644468ef72c0f90daebf9 : SUCCESS in 8m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f32915323eee4b9985a39b54cda487ef : SUCCESS in 16m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d82ebe851388490a9a727b9d9a30d80b : SUCCESS in 32m 53s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/40ae3b7e5b3a482b86f7492ffd7a5fbf : SUCCESS in 1h 04m 55s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94d0abaaf9d14d85bf011df49588a614 : SUCCESS in 46m 50s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b908f9ddb7294b5a9021c232e77439c2 : FAILURE in 1h 10m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0cebc22771e1425892a1d723d6bc6fb9 : SUCCESS in 26m 53s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ea11d894e7941fea58c5dd85b0da2fe : SUCCESS in 15m 35s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/44c95596643a4769812fbbc678138378 : SUCCESS in 1h 46m 58s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf1b6c871e024540b5ca13f4d7838f27 : SUCCESS in 1h 09m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ed3ccfc744cf40e38a6f61804342d215 : SUCCESS in 52m 32s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be9e1685613c4b098de9f90581ac7880 : SUCCESS in 40m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/26716e9b7a5d40488a8c497db4641959 : SUCCESS in 28m 33s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39ed5065882048eb92b8355da7975f42 : SUCCESS in 24m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4205a2a1335c44f2b54226e015e0ecc7 : SUCCESS in 18m 31s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/998dd95703774fe492c340a64e566403 : SUCCESS in 10m 53s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/21669341ef6341789a3b74851ed28785 : SUCCESS in 17m 20s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d952f5829134b9abfa331b1aff8d612 : SUCCESS in 12m 03s","accounts_in_message":[],"_revision_number":2},{"id":"deb6705771dcf8e02d83fe1c712fe998dbbfb1e1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-21 17:33:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a3d03ccf99601076448debedf5208855aead6ad9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 19:36:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64274a432ca44f16a82141f91a944ec4 : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed4bcc268d7a4f30b35673f34004f068 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36894f0a87024bdebc30e48fe54cf2cc : SUCCESS in 9m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75c640c0f35545358e818787dd8274a0 : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/728a1926eca94a4b9c9aea61809336fa : SUCCESS in 10m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f931cf1ecdeb4f908f6330c83d4ff408 : SUCCESS in 8m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4b92f2b6d95f48c496e44da2084ada95 : SUCCESS in 14m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b3ff7e4e964943a3b799bda67a1b2a0a : SUCCESS in 32m 31s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f2b2a129bfbf46a89b20aff6d91d26fe : SUCCESS in 1h 07m 20s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff2ac746ed154ac5a4527b56afdcf28a : SUCCESS in 1h 00m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/38b7958642994764b72b8adddb6e7d6c : FAILURE in 1h 22m 46s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2e01e9e8ccf444549d74dfe6d6f9a6da : SUCCESS in 25m 02s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f1332d4de944ebb9437b8e78b85cd64 : SUCCESS in 17m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/175dd199c0214fe182f833c33ec40dc5 : SUCCESS in 1h 47m 14s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f19de252cb704c2b89ce698ca6f592a5 : SUCCESS in 1h 08m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/afc814e32bf042ba9e26ea7f434fd3e9 : SUCCESS in 48m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b174d23fac61418987c6326e55a9aba6 : SUCCESS in 36m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e8c15e7dae47471d8bc37d22a56dc2ef : SUCCESS in 29m 13s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f81bfbdc0e74d60a409e69ccfecfdcd : SUCCESS in 25m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d81f9d8242e448608c3292f8c7915f49 : SUCCESS in 16m 34s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0258ad9b5b64422bbce12f8628cf88e9 : SUCCESS in 14m 31s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e20aefe1e28c4786a4a332517fbc2046 : SUCCESS in 20m 51s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94ea9526602d4bcda0e225ae67f217e8 : SUCCESS in 15m 18s","accounts_in_message":[],"_revision_number":2},{"id":"305bfe225c444173fba1b493833ed49e0b9c76c5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-21 22:02:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a5e6dc4409b4391fa36bc049e0b76eda6c3e5fd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 23:24:31.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1ffe0cc7b844c1dadb7ea885d08c169 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31137d2064684f718d7c086f48972557 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db88fa9a60bd4ff2bb65567f4ba8a206 : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/426b7dfe939243d68dbae98f8485d081 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70b9b4c3a9f64229be1a1d5c033209e6 : SUCCESS in 9m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c538fcb22c28409b87fa7509a67a5d7b : SUCCESS in 7m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/582de07fa4d24275a578946ff106da6c : SUCCESS in 16m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/acba2f452df142609990082a352a2f79 : SUCCESS in 32m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/bb417e2204064d12975b097247c38239 : FAILURE in 14m 59s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a18696cd8fd4b16b2d8fa7529b960ce : FAILURE in 12m 05s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2d0d0d7ca8af4a80897ad461b8824147 : FAILURE in 14m 45s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2883c5a7a85a477c9474def19e9ecebe : FAILURE in 21m 42s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d57b09472c424c7ebbfd45f8d2a2af6c : FAILURE in 14m 35s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9eb12354861c43fe88b3217c17b904d7 : FAILURE in 1h 17m 49s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53f754d7668445049f2beb284326603f : SUCCESS in 1h 10m 57s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/22776907a2764b949ea3a3cf5cad8d7e : FAILURE in 13m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a6706670b224866a60de40c3ae3ad9b : FAILURE in 11m 42s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d69eacc0da464da19588d86cbb2da47e : FAILURE in 27m 20s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec8ea4e849f143a79f2d67c18ba45950 : SUCCESS in 22m 36s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/40140fd6028e44c392245b11039007b9 : SUCCESS in 17m 04s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08a3886ef85e42a494a3e05e7498fabd : SUCCESS in 13m 11s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/304ba281d77c4768bf809cd6331acb88 : SUCCESS in 19m 10s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82267eac474d4e4daf46f0cd43138e2d : SUCCESS in 14m 00s","accounts_in_message":[],"_revision_number":2},{"id":"8cbc8b38cd6b16c7ee0093ec66aae130125ec662","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-23 08:28:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dc8150c790c434abc55b20611e19f225316e559e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 10:07:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/988507c5473748e0a8109aeb3878cd69 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6a995ecc6904a6a9b03e5ed1821c8b7 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcb747445f9f44f48b12468a3ec0c6b9 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/912ef60cd9104157b516390d3b37a770 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d47409fdd37445fb74b7beebb932ac5 : SUCCESS in 7m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4dc235b650014eeba92a860acea9014c : SUCCESS in 5m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6c6750d6450f44779a410f0b95a9b33c : SUCCESS in 11m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e51c917b3e0e459f9c35a0d4b005ce5a : SUCCESS in 28m 05s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e781eff956fb4e39a91438f09dad83e9 : SUCCESS in 54m 22s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a8bf1807e7c4a94a276d90e3178b900 : SUCCESS in 47m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6bf5516037e4485787bd5bb7eccdcb5e : SUCCESS in 58m 04s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/cd1a116e577f4d6bb6ee51122586d2ed : SUCCESS in 21m 48s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f8b932c02744c4e95984c9412d60be8 : SUCCESS in 17m 35s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3b8fbc17f240416da546503372e21e22 : SUCCESS in 1h 37m 12s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/308c45300fb24d0f879425ea21ef1f7f : SUCCESS in 1h 12m 25s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1e729e9a84024658840cf5be283d2b40 : SUCCESS in 53m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3babd1071d7c46b58b4c5341ba79ebbd : SUCCESS in 36m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fa23afd7db524dca8290e0bb9021ff39 : SUCCESS in 30m 36s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf1e473f444043bbb9be58fbccae69ae : SUCCESS in 21m 40s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5a1a90ac5a264aa699d101e9f1bea614 : SUCCESS in 17m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc5a86d3b3854967a6f391a8ac747a38 : SUCCESS in 13m 15s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/81885e3a033d485fa31c151ca82d28e2 : SUCCESS in 19m 18s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8dd44f2ede534195b77653d6d0341575 : SUCCESS in 14m 30s","accounts_in_message":[],"_revision_number":2},{"id":"7c650c4071ec1dbf84239085069c06720b54a1ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 10:07:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"162b64fae4ad1a7062af9b9b3c67c1fb6b9ad57d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 11:42:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3af5778ec717497483ee546e79f5410c : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8893c746dd1426783b33aae931b7bee : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1fd61e081bf3413c8823c3e3f0993d8e : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e11bf81866846a4a699bc6cadebf52f : SUCCESS in 8m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ddeb318532f34a959bf0d11da7cac552 : SUCCESS in 5m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ca48ce5d841b48189d7c3ff3521b2ebf : SUCCESS in 11m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d4bdce308a0b40a3ab395c92b0834697 : SUCCESS in 26m 18s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4876411901d34c0f9ce8e02d44487e60 : SUCCESS in 57m 14s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1695bf9e5fd444c6bc3ca2b19e01826c : SUCCESS in 48m 46s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6860958cb0954bcd84f0e6c538dfa0c1 : SUCCESS in 1h 01m 59s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7990f1cdc4a742e99f3ef798324930db : SUCCESS in 19m 06s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18f3b243ad4249b08b732e783e55688e : SUCCESS in 11m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/39ea0e09d26949cea1309cb38a8de137 : SUCCESS in 1h 31m 43s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc09811d23374ab0acbdd09a2ab82616 : SUCCESS in 1h 04m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/90e5020520cf479cae2542a8e4a2afae : SUCCESS in 43m 10s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/967393ae74ec48f186a15ac7258bca98 : SUCCESS in 30m 45s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c467362962b8465697252051a9651405 : SUCCESS in 23m 01s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc3d94a8ee5a47fcb6229eee963c6467 : SUCCESS in 20m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5da55278b8a04aa3aa7074b1f1baf821 : SUCCESS in 13m 48s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff82b571854843f7a959b5898792ff63 : SUCCESS in 10m 22s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/64b54db19bf04d548462ca0227b67af3 : SUCCESS in 15m 07s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c252bfeff5c47eda6fd8318752a9395 : SUCCESS in 11m 39s","accounts_in_message":[],"_revision_number":2},{"id":"557bdc114dd3a2c406a05e0d391e1874c7378562","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-23 22:36:14.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"0eb860a2047440c2686d95292ed074934a6ebc81","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-23 22:36:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3eaa6dbd172ce6369143c7c0c868b2182e3df301","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 00:18:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6800ead02b6448ca80df8463ef682429 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99ebef08f3114b8f8a5022a8d7d0067e : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c05b0775d4f143bea7366181a42f1123 : SUCCESS in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f1b79cadafa43b79bda4bd8288937da : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c67960e79494cbe9b3ee74a354bf2cf : SUCCESS in 9m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/64211e48d8a346fd977a7ab6c75cfa1f : SUCCESS in 5m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9597024532ec44ca87739545bf6ca937 : SUCCESS in 13m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6043d6e9e5824757bffe6bebd8f5da0a : SUCCESS in 34m 35s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a0a7ae8e4d084f8eb1ca13c592a6550e : SUCCESS in 1h 05m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1dcd2223ca4440b4b3e1aea6eba33857 : SUCCESS in 52m 45s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6f64ff5dd8ba498bbd9a580b75a2b29d : SUCCESS in 1h 09m 40s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7d1ce7607b414623b2220adb1f3681a9 : SUCCESS in 23m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a99147c5ce24a88ab78a35b3087962b : SUCCESS in 15m 09s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f6f04879c442497daaa35ea26fb20f1e : SUCCESS in 1h 39m 04s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8380e7e8d0f49c597b63525710a041e : SUCCESS in 1h 14m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/84e25e237b8e47829c002947a8877d6c : SUCCESS in 46m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bfdd7e4f225465b88a6925d70bfb89a : SUCCESS in 34m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9c40f66a4df74615bd0d91f94b534ae7 : SUCCESS in 31m 28s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dca715335ead4d499acabd88f2c4485b : SUCCESS in 23m 11s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f09d4b92a96149388b45af2fcb3c24c7 : SUCCESS in 16m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8046f62c9e6f4f3797f1ee7e79599cbb : SUCCESS in 13m 14s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/61e1656776634a999502d961880e9ee4 : SUCCESS in 17m 17s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36b85488b6bc48cbaabd282134d44ac0 : SUCCESS in 14m 06s","accounts_in_message":[],"_revision_number":3},{"id":"53eded5f3eb05655462aa22a35fea5fc7762a3c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 00:18:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7b172c5d9b70fb71624e1064121caf30450d2ce7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 02:04:58.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baa236f804d342c8a5d28af5fc813b7b : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1379f5d930324534ad6b500148f7331f : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9eca24553424e269e9d8b4b8410f474 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/135ce3dea184491080acdb141a04be67 : SUCCESS in 8m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4b5521b8bd534767a0030fce78349b79 : SUCCESS in 6m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/40f1a41821df465bbba735c118bb7a84 : SUCCESS in 12m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b1c165806f01470fa8e13e95d088fab2 : SUCCESS in 30m 43s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3b7c2a6b58444858a5cd540d69ae1603 : SUCCESS in 1h 04m 21s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5ad0bb90ce24ae78b597b1dad70adff : SUCCESS in 54m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d363ad2b083745a3b9a0dab11ca7e364 : FAILURE in 1h 08m 26s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9b7505ee5a6b40e5bd9b3360cff1f821 : SUCCESS in 21m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b81c0923746482f87913931f36353e7 : SUCCESS in 15m 24s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d3fd5c5da09247e2b78fc6914982d3a9 : SUCCESS in 1h 43m 30s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40e14af0386e4de8a358bb0f14e3898a : SUCCESS in 1h 10m 36s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/36bbc76fe1fa4534b2aa5598b2e098ee : SUCCESS in 47m 09s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac3a898b1e224c00a02c4955d5edc736 : SUCCESS in 35m 16s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d5846e4bea07490eb1c894c778753703 : SUCCESS in 29m 57s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9653ead2b8045cf9b5a2a44f38e2c4e : SUCCESS in 24m 03s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ef2817e870e24de8a13a8d612853505e : SUCCESS in 20m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a643f20891647d990975de02a896a7e : SUCCESS in 12m 15s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ceac8d886005402e81d409b55e2c13cc : SUCCESS in 21m 27s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1e07e82c5e94961b54a5861aa8c880a : SUCCESS in 12m 15s","accounts_in_message":[],"_revision_number":3},{"id":"0c4d704010b2639301a3d3f44b62f5ea8c56129e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-24 09:46:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b3ae7d9902c37cc08ba919e8cf5cc72a0a267cb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 12:05:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29bc4ee38cf449459e7d3b1f56ab48e2 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5e3723c031849b2a28930397a91448d : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7baefd3fd2134b6da50edfff970a00f4 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8598508456024ff1a82fd0c62b0226c9 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/873788d5128745ad9b4fc3a1a1651d42 : SUCCESS in 7m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/498ca5f819c74c0982eb6ee8cff63410 : SUCCESS in 6m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/31e8dbd9537c4085a96a675110f80708 : SUCCESS in 14m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f5af87fc4cd14952b2a297df3b281d83 : SUCCESS in 32m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2edd02091e3844c9ac0523228c0a30ee : TIMED_OUT in 2h 16m 12s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8955235f001946b6aa493c961e2bf226 : SUCCESS in 55m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8fdfebeb5eae476b866d1873fcb58bd9 : SUCCESS in 1h 06m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1672e820869b4f3f96b54e26998fae1b : SUCCESS in 23m 09s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d15906fa28a54e8cb79e227cab45efbd : SUCCESS in 14m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c13c034d31f84a03a5ef705d6defef14 : SUCCESS in 1h 59m 24s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd7c621915eb4dfbbced52502247cee1 : SUCCESS in 1h 08m 09s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/63e46f87d85d409da9caece4aeca0503 : SUCCESS in 44m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eaf82f5216c2484091e9c582b9af4fa8 : SUCCESS in 40m 55s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2ad93d55e03b4178b9e229a68acff6b5 : SUCCESS in 29m 05s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c388dae9a6204190bcfce975b46c083e : SUCCESS in 21m 30s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8ef6bf9638e14b3fbcf203d611c60d22 : SUCCESS in 15m 17s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8d14f6873684c4dad587c8b46e92824 : SUCCESS in 11m 37s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/6217076cbde348bd893eb93210d2041f : SUCCESS in 19m 37s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4871d4d4686a424cae8b3659b5875190 : SUCCESS in 14m 57s","accounts_in_message":[],"_revision_number":3},{"id":"375eb6e6c9e34d7d170e8694d3d252c69a7fc102","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-24 13:02:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"70d52cae3606de05dbcace44266b32510def2501","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 14:51:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09024124cbfb48e685299a318ca7c0e0 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba3f9e7425a54adfbe01041ed7d62e2d : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3a99b59bd934525a1c9a4ae64c2b6ca : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42e3082b2302436694b1654b8291c9a8 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08ffd319ff824eee8850e9bdae9ce8d8 : SUCCESS in 9m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4fcdb038f12f4b4dafde20659e352a4c : SUCCESS in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5808e78a5d1942599842577c61acdd93 : SUCCESS in 14m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a78288462d5c4b30bc2f63acbf8476ee : SUCCESS in 31m 04s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2a44cb76b1554480a27a5784af32601b : SUCCESS in 1h 04m 09s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e327fcb601924b6b9af5084de8f0de38 : SUCCESS in 51m 08s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/11d594ed4ed849fa8b55be44a0649782 : SUCCESS in 1h 06m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f3fb474f06024d649e94fe4a68a4d40c : SUCCESS in 21m 56s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0579b6db1e1d4f8595949b99d530bfa9 : SUCCESS in 14m 16s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/641046c9aef4486aafaf7c21c582f15b : SUCCESS in 1h 43m 37s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43dc4e0b0301425bb8dcd9268354037a : SUCCESS in 1h 06m 58s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a9c129b0fe824194b1925331a23c7202 : SUCCESS in 45m 43s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5860ac0a5be48bab46f639cada090ff : SUCCESS in 48m 23s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/813bbc20fd954c71b56e7c8fe69a33a2 : SUCCESS in 28m 40s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce7f2346ecfa439ab5f4514a9d8ddebf : SUCCESS in 21m 57s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/495660919209408a9680dc0c0321de74 : SUCCESS in 19m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9e802c895c746a2ab67a58525ddf05c : SUCCESS in 10m 06s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/fa53597ce76a41b19c733530dc4fb909 : SUCCESS in 17m 45s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06165415abad47cabe2cc8b65ed10ac2 : SUCCESS in 23m 19s","accounts_in_message":[],"_revision_number":3},{"id":"3fdc4cd7bfed30621775db39290b37eb21755969","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 14:51:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c2e1a04067ab0e5ea697bf0882caf1e73eff2f10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 16:30:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/567d74baa98f4d1b9a4e216e99c453c7 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e692e017278c49a09d026207fb353c8b : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5193dbb6cda34e60a075c77a81b9123e : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a39ad60c9aa4542b4397d977bed4fc3 : SUCCESS in 11m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aadaf75d8fbf4c3aad5553e5b3fe4b0c : SUCCESS in 6m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6b31f59a995f4dc5beae8179b930badf : SUCCESS in 13m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/666513277c884494af0ce5aa63e86a20 : SUCCESS in 35m 34s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9d4082d0fc7646d8ab7d465aaad73916 : SUCCESS in 1h 06m 07s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0dc2bc6596fc4ae1866d2697baeb02e8 : SUCCESS in 50m 38s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/12231135fec94c42b79105a1cdb2db8e : SUCCESS in 1h 10m 01s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b46e585ea0a64c2392d4a548e961e8cb : SUCCESS in 25m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75b7c8850b6e40e8906e7cde5e508bb8 : SUCCESS in 16m 01s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4ac30c6a1a9148bba5a3fbcd09bf830e : SUCCESS in 1h 35m 59s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/819b7248de8441568bd9aaf113c71fdb : SUCCESS in 1h 14m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ade79908473b4745810eba2d61ac0a5a : SUCCESS in 49m 42s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67e606486ed843a1b284a2f9705bab6e : SUCCESS in 42m 19s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/78ecb12cc2d74d0ca1c6b6eaf108fc70 : SUCCESS in 26m 49s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd34ae10d49842c88fd9b52fe8126811 : SUCCESS in 24m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/76f5b4996de14852a90396b71c12bc8f : SUCCESS in 18m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89a490367ca64d38826ea16d9481fc53 : SUCCESS in 12m 59s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/8a3d5dd47bce4a2ea67c77ab061bfd88 : SUCCESS in 17m 04s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11a70cac712f4f11b40d0a682ea8256f : SUCCESS in 14m 42s","accounts_in_message":[],"_revision_number":3},{"id":"c32081be21be3e6beb439d7912eba1a721926eb6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 16:30:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1cfe5809afc0c830e124ad0e938b95c3cde4155a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 16:32:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d1f226617814a1c87fe217dd4ee91c1 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"06123b795da5da1cf38d8ba8f86f2b8f3b25f570","revisions":{"1d76f3964abf6ce5e44cd1e7c18e628cc0f884d9":{"kind":"REWORK","_number":1,"created":"2022-02-02 12:12:19.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/87/827487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/87/827487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/827487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/827487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/827487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/87/827487/1"}}},"commit":{"parents":[{"commit":"e4df5b09f6f9a9d7331b2fa81bb8bbfe94a62226","subject":"ntp: Fix service mask when service doesn\u0027t exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4df5b09f6f9a9d7331b2fa81bb8bbfe94a62226"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-02 12:11:44.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-02 12:11:59.000000000","tz":0},"subject":"ntp: Remove removal of chrony container","message":"ntp: Remove removal of chrony container\n\nThe chrony container removal was performed in the Xena cycle, so we no\nlonger require this task in the \u0027overcloud host configure\u0027 command.\n\nChange-Id: I86fcc75e844eb922f62c90c45a105519845cc1a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1d76f3964abf6ce5e44cd1e7c18e628cc0f884d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1d76f3964abf6ce5e44cd1e7c18e628cc0f884d9"}]},"branch":"refs/heads/master"},"d756b49deab4b4f90bad4eb3507e522c257d83ec":{"kind":"REWORK","_number":2,"created":"2022-02-02 13:57:41.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/87/827487/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/87/827487/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/827487/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/827487/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/827487/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/87/827487/2"}}},"commit":{"parents":[{"commit":"7aa0207cb074ec75d9461ac910987b97b8358c1b","subject":"ntp: Fix service mask when service doesn\u0027t exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7aa0207cb074ec75d9461ac910987b97b8358c1b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-02 12:11:44.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-02 13:57:32.000000000","tz":0},"subject":"ntp: Remove removal of chrony container","message":"ntp: Remove removal of chrony container\n\nThe chrony container removal was performed in the Xena cycle, so we no\nlonger require this task in the \u0027overcloud host configure\u0027 command.\n\nChange-Id: I86fcc75e844eb922f62c90c45a105519845cc1a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d756b49deab4b4f90bad4eb3507e522c257d83ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d756b49deab4b4f90bad4eb3507e522c257d83ec"}]},"branch":"refs/heads/master"},"06123b795da5da1cf38d8ba8f86f2b8f3b25f570":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-02-23 22:36:14.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/87/827487/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/87/827487/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/827487/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/827487/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/827487/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/87/827487/3"}}},"commit":{"parents":[{"commit":"01e5834ed1e301d92b47ce3452bd18651cd24e0f","subject":"Merge \"CI: enable libvirt TLS in TLS job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/01e5834ed1e301d92b47ce3452bd18651cd24e0f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-02 12:11:44.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-02-23 22:36:14.000000000","tz":0},"subject":"ntp: Remove removal of chrony container","message":"ntp: Remove removal of chrony container\n\nThe chrony container removal was performed in the Xena cycle, so we no\nlonger require this task in the \u0027overcloud host configure\u0027 command.\n\nChange-Id: I86fcc75e844eb922f62c90c45a105519845cc1a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06123b795da5da1cf38d8ba8f86f2b8f3b25f570"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06123b795da5da1cf38d8ba8f86f2b8f3b25f570"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
