)]}'
{"id":"openstack%2Fkayobe~827653","triplet_id":"openstack%2Fkayobe~master~I6d9e3acc477c9c4b3bb9db9c01a2db83b9568f59","project":"openstack/kayobe","branch":"master","topic":"ironic-inspector-routed-networks","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-05-03 17:40:27.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-18 13:42:07.000000000","reason":"Mark Goddard replied on the change"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-14 12:45:34.000000000","reason":"Radosław Piliszek replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-05-03 15:31:12.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"I6d9e3acc477c9c4b3bb9db9c01a2db83b9568f59","subject":"Multiple DHCP ranges for Ironic Inspector","status":"MERGED","created":"2022-02-03 12:31:00.000000000","updated":"2022-05-03 17:41:37.000000000","submitted":"2022-05-03 17:40:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827653-ironic-inspector-routed-networks","meta_rev_id":"72d15be57e0348c21620c4a5af0d325d813f5538","_number":827653,"virtual_id_number":827653,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-03 17:40:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2022-05-03 09:05:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2022-05-03 17:41:37.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-03 15:31:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2022-05-03 15:29:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2022-05-03 09:20:42.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-03 13:46:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-03 15:18:13.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-03-18 08:39:17.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-03-18 08:39:17.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-03-18 13:42:07.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-04-13 19:38:17.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-05-03 15:31:12.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"ab894c9eb4160ad2af27b2758b6a417f56929691","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-03 12:31:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"330bdf34dca135148bf8d2122a6d5e2efacd18a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 13:46:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/196c2b6379414581b7a02a4b65c53711 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/513cfb9f4945402ea7fd0c6c9350a38d : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5be620097dfe4268b51a57cf7d267a2b : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4a4acaab3dd424691287904c5b43169 : SUCCESS in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffda064f2f4545139c54bc2d144685fb : SUCCESS in 9m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b459bb0ddc354977869fe857dd01ab81 : SUCCESS in 7m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dd78469d2d644443a6f025cd8304a6a8 : SUCCESS in 13m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a0f86e389c1746248d6daf02c9f5358d : SUCCESS in 35m 21s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a4819ab91af94f059026dc531cdeed62 : FAILURE in 15m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/612554a96ede4eadbd261a88ab960c35 : SUCCESS in 47m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1a16a0f8cba047bd85f1454381fecef5 : FAILURE in 12m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/52da3f2148184600a6356fe6044bec00 : FAILURE in 15m 02s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/627383cf09f94575869e0433086d0feb : SUCCESS in 15m 23s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e52722af67054dfe9acd36a287933757 : FAILURE in 13m 13s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1e1008288c2480c82774181591aab67 : SUCCESS in 1h 13m 30s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8cade6c17cc44162b42f52c17b34641f : FAILURE in 14m 50s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7cb7b2cbf254074b35e9f59ff257046 : SUCCESS in 36m 25s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d9c9239ae6724d8c9d8253346919aaa3 : FAILURE in 12m 25s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d65dd57bc91b4b72ac9dade010b92260 : SUCCESS in 22m 59s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5f615d0eaf1748f5b25bb1c293e3515c : FAILURE in 12m 24s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/279db404dc0248beac94880e8c83569b : SUCCESS in 12m 29s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/553bed82ddd04eed831a459b8f344dac : FAILURE in 15m 47s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/accda5dde6bb47fbb6c12d57ef78c593 : SUCCESS in 14m 12s","accounts_in_message":[],"_revision_number":1},{"id":"78b1a287132aabba9f88bdef3c17c7140525d03a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-03 14:10:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7771904577330b25c2e7daee4b9416121b263200","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-03 15:15:40.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c4aa1fa08e020860809bc57d129568306f02c574","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-03 15:18:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"119d6cb11410bce501faa550a11cc9e43c8f839b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-03 15:30:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"31656e0722d44c2d8a1f9b89b38d0bafaeb74df0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 17:11:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4df7f3740d94152808855d9f7284221 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e8609439b474efb9c974385f88cc4f4 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb99a03b3ba0474ba66488d628bdbba0 : SUCCESS in 8m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/632e8ad2e5274394a0ee7a8f5a7233f1 : SUCCESS in 9m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3de243b3a6e4498fa851cd731e6f381e : SUCCESS in 11m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0470caf4386344a9943d172e31d32f93 : SUCCESS in 7m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a1b14042ca4a4b8e81b8a80a1e53be06 : SUCCESS in 15m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e68036969d9444cebe6170293ded91be : SUCCESS in 35m 35s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/015da3edd801460baa263c7b68d92c11 : SUCCESS in 1h 25m 30s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44fdf71f9834404699ecc74b9467f841 : SUCCESS in 48m 48s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/51312f06e6de41a8b1cc82495f50a2c1 : SUCCESS in 1h 09m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/23c0c4f4b76b4bc09ff893973fd79fc6 : SUCCESS in 24m 47s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/daf9d3b5320e44e0864cdcd4500e9feb : SUCCESS in 13m 46s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/716ded8bffc642a29749d89f3f45c448 : SUCCESS in 1h 47m 03s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33a75413c66d49ffa5920911a4b0e003 : SUCCESS in 1h 12m 04s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7ab8e6f980ee4d868a6fb15fae2350e8 : SUCCESS in 42m 58s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c7cfe80de8143ed822013a9578aec76 : SUCCESS in 36m 19s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6e4ccd00fc414084bab0d2c597286758 : SUCCESS in 28m 56s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce375ce3e95c47bb932532427b5a8d45 : SUCCESS in 21m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/bd2d10b4ee404dd3a3199e3272afe0a1 : SUCCESS in 16m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5402510e2c3746988440bb33ecbc49fa : SUCCESS in 16m 06s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/217a73cea73748b9a0d0795d12c8aa9e : SUCCESS in 21m 34s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6aaacf46e2734395adacf53047c7d130 : SUCCESS in 15m 58s","accounts_in_message":[],"_revision_number":2},{"id":"d4ee9b5b7fbb77b767b2ab70087b38b038a65d09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-04 17:47:05.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"4a9c9275e874dac39144ee2d07b7a72f45688e5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 19:24:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1621c3e1be724309abac7f4c4569fb33 : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/195e9b79f60e4679bd978d62ead59ba3 : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86e993885a1e4119a9bf2551335b96b0 : SUCCESS in 7m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44324d92342444ad814cfb7aabf901f0 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24b28aa274c84689af2b041fa02371a2 : SUCCESS in 9m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/76a0a0e4e1dd43b2a18e48574dbb47ec : SUCCESS in 7m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/43d3a4ca38f04bccbf3c74a3c9143976 : SUCCESS in 12m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dde57b1f0a2b43449de2b020da002d75 : SUCCESS in 32m 20s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1f2d42e15a4c4007b6567567dd91f8bf : SUCCESS in 1h 07m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/144262a3f9c546ad9285e404b4fa1417 : SUCCESS in 44m 25s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/daa7208555724385b39a9542862ffe7b : SUCCESS in 1h 09m 48s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d0afc08930624511bbdb04a853d5a62a : SUCCESS in 29m 51s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52f4ef4cfb59409dbab9c9e888ec0b48 : SUCCESS in 17m 23s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/689c1e6885664d4ab5e04a1887dc6b92 : SUCCESS in 1h 35m 25s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2530f8446d2942d69a3f2960f1da5173 : SUCCESS in 1h 11m 38s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c64b502f102142a5b5fa0e9258b2418d : SUCCESS in 43m 49s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/603233149e004f2d85e4b61496f6c90a : SUCCESS in 33m 28s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0463f6b9496d4c769d8c2fa2a142caca : SUCCESS in 27m 12s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c0ddebf53a8494a980e0ab207dca574 : SUCCESS in 21m 59s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b0d82c77c4c9489b98ffa573e798fa54 : SUCCESS in 16m 18s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6214644c675f460c936a215ef69e2c24 : SUCCESS in 13m 03s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/88eeb662e13c4d4d864738902ab589ad : SUCCESS in 18m 59s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83224d48d43945a182f336b68d17359d : SUCCESS in 13m 59s","accounts_in_message":[],"_revision_number":3},{"id":"f12ef9a239c8e0eed2da49b7ef082daba2a31447","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-18 10:18:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f6aac49da8e4034771f25bc81ef0f4e592c09731","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-18 12:55:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b6d373c9ff646c1f29168247e351395ae852caa2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-18 12:55:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6c60a84194bf1f463eff8f4fb74d31e1a4fd3d7c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-18 13:42:07.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e16f18ef1e44e7f034e89a11d24b27ad4e41c092","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 14:37:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/281bb4dc1ffb4335be43838eecb2cd40 : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77f3f656c7cb498cad518d904c0fc523 : SUCCESS in 8m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1823a4cdf5c4a61970939d498d35977 : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a901e061c35240b9bce1c24aea8f45e6 : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4afc0ee16d994ee9bf563d5cb7952ac6 : SUCCESS in 9m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3531760d9ad4487089eeb5eb814a8bbd : SUCCESS in 9m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/62263303bbe148b98655ca17fa6557e6 : SUCCESS in 12m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ae8d8377579b41e6832298d235f5ce04 : SUCCESS in 28m 54s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d9877d847a4448fea31dd4ff9e01de83 : FAILURE in 1h 35m 59s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e41c2ca7b8e4e95bcf933e3f81beb03 : SUCCESS in 53m 02s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e801a59aa24d4a6f8cdc8f85d8a3c6e6 : SUCCESS in 1h 10m 09s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e8448efb40b245718404ff9d4b3f7dca : SUCCESS in 23m 56s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31738a080d2048358506d5c4710c5e21 : SUCCESS in 16m 18s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6148a9197f994e118618daaf65c0cd0b : SUCCESS in 1h 39m 36s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4f654415fd04ebb9fa67fb7414393c5 : SUCCESS in 1h 12m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/eb9f5272bc2f40d884e0fab6f859e04e : SUCCESS in 49m 10s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cee5f545362e43c4b51df7574674d389 : SUCCESS in 36m 22s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f9ff0cf5121841349184930d369991aa : SUCCESS in 41m 25s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2585d26c52084fe880cdc8fbdb28ed42 : SUCCESS in 31m 29s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5d6de168e7c248879824d35256acd7ec : SUCCESS in 16m 23s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf49f8cd3ef445bbb241fd78735b4b46 : SUCCESS in 17m 23s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ff3cde5595f5481b9f4e5c5b5ecdbc0b : SUCCESS in 20m 16s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41e8ad8b597d4b9ebf60228858c38d09 : SUCCESS in 14m 17s","accounts_in_message":[],"_revision_number":4},{"id":"41c94756e153b29121d25f5d4a9a2c911605148e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-18 15:04:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3169946d331d5079cbafb4bef0437ef3fe03909b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 16:47:15.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"08ddda4b2e4a1f42f731173055b4af1c2adbf845","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-18 17:21:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c345835ea2f80e702a810043a86e731b88f5dce0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 19:35:18.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/963c1aaa11c14aa6bfd06d95d8ff15b2 : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d2bc1a5977e46f5b301b6d1be04d84f : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1a59e3bcdfb4a7da791b14151e0c6e9 : SUCCESS in 7m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37440f66fe184147bc5efddc1e2557f8 : SUCCESS in 8m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52c367e2dca9456a97f90e9e68891e68 : SUCCESS in 11m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ecc736a5098842458120750651244711 : SUCCESS in 9m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ff026066b69842fdb955bc1f403a3cf2 : SUCCESS in 16m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/820e2ec03dff478d83c9949c681e3d47 : SUCCESS in 37m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5c3c0dc4112d4ae8970dd3be458d733b : SUCCESS in 1h 12m 51s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/598f316b755f4fe2874a2ac0515477e6 : SUCCESS in 1h 00m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3eacef498a2c496eaea03722e8a7bd9f : TIMED_OUT in 2h 06m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f37bb084d5234cb7ac161a749ea1e075 : SUCCESS in 28m 06s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14f602240e4b4c98956df840696b6305 : SUCCESS in 19m 07s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2db1d61ce42b4a46ba74f1c5c1266e24 : SUCCESS in 1h 50m 51s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/536d866a7da14ca99d1d94c79e06994f : SUCCESS in 1h 29m 37s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c2b7a7e69b50490398c91ac965f11adb : SUCCESS in 1h 00m 09s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/feef13473302463084948c9eb6b05a9d : SUCCESS in 40m 35s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/771574ce8b9d41ad92adc4f1b6564d1d : SUCCESS in 42m 51s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ce8bb7da416426db68d089e535472a5 : SUCCESS in 36m 01s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/70d33822f0a244f8840fd556999911f0 : SUCCESS in 21m 47s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cebd56c6feea4621b3735ef0c3878d01 : SUCCESS in 15m 05s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/3fea21b83a994583a9d259067daffcba : SUCCESS in 17m 49s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0cb06c4ca7546db87e9cbf54a9bb3f9 : SUCCESS in 18m 06s","accounts_in_message":[],"_revision_number":4},{"id":"0020065ac54df502fc75b40371581a61aca795a7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-18 20:43:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7d6b1e3ff8d07ade15ace1d250ad4d8f13b9fc3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 22:36:08.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1991dbf3b2d54069bf0a266e4a7674a0 : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17d11c46baee456eb4941452bf06d038 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe81e0f11a2846bf87cf24fcd65a8ad7 : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/712ddc64e2af4200bd9196daa16f4354 : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82a56a65e1514efb9b0a78b05da39f68 : SUCCESS in 10m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/62476ce1bec7499f80f39fe2a638a1de : SUCCESS in 7m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8d67c2bfbb714885b1f13a88e7f8fb77 : SUCCESS in 13m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/26ed70e5f16e411b9d5a1c4ffde573a8 : SUCCESS in 27m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/63b22664dba647afa95b44f7951dc4f0 : SUCCESS in 56m 21s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/558581b95e934c16a69acdc5c72b8fee : SUCCESS in 45m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f800fee1e06e4e1a8ba9d21908076d7b : FAILURE in 1h 51m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/762f8665b3564f0cb6b0cae8142e615d : SUCCESS in 25m 15s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1527e3c38ea04437ad9cd7de506989b6 : SUCCESS in 16m 43s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/088ea7f5656f490994c7edd9a3fde878 : SUCCESS in 1h 26m 32s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de68a87d517d4bc98efd01647213d0eb : SUCCESS in 1h 05m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d94d1c0a14724713a9e2f7314af4b012 : SUCCESS in 41m 38s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00ce235ad53846e0a36d6ab4061fbfae : SUCCESS in 36m 04s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/79739c9b85e44ef6bf210561e065f601 : SUCCESS in 34m 57s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fcbb6b19f954a86af32f1c369e41dad : SUCCESS in 25m 13s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/01a6d59c2c9d4432a0040f13a3cb20d2 : SUCCESS in 16m 54s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae65400835d54608adc2266051cbc1ef : SUCCESS in 13m 30s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/1c21a40ea26146aa830050ed141fdf3f : SUCCESS in 20m 22s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a41eae4553be415fb60e76b118ee12a3 : SUCCESS in 13m 36s","accounts_in_message":[],"_revision_number":4},{"id":"e628ec6877df72b530060ec2671a1d862924971f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-18 23:20:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"55f967c5ca3fb2599a0e5e968d572767c3940393","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 01:26:55.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24835881d9f040a7a1089e8f7c49e92d : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/664ce4c92d7f45a4b82677dd30a4fc17 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56cbc24377f4431fa5931868b25ff4a3 : SUCCESS in 7m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35a41e94336f4bc5a9a6f2b11fd060f7 : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c64e88c8b5f54e56aa51eed15780497f : SUCCESS in 11m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ae10995acdc74c4fbfa917926bfa4a40 : SUCCESS in 7m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7ba67c6b74c1488491608c6f91b832ff : SUCCESS in 14m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7363a1bf7e2e4cdcbb244503dc4d74a3 : SUCCESS in 35m 03s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7a22691d3e7f4406974e9ed0539a5b32 : SUCCESS in 1h 07m 20s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e499eb584f524bb7a9ef44ef8c6c6817 : SUCCESS in 53m 46s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b4f3ce4638304ec3b08038a0cfb6466c : FAILURE in 2h 05m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b0796dad16974f5fbf2ad3592d2053ab : SUCCESS in 24m 36s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed3e25286a0e4cd49505d60c3200968b : SUCCESS in 15m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6744dfdfa2f041b388cf4a5dfe48aa91 : SUCCESS in 1h 56m 28s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae3a89199bf9402f82f504ca90b464bd : SUCCESS in 1h 31m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f134facf08d2412bb4c12f6b8eecfea2 : SUCCESS in 47m 51s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26fbf5318fcd477c8fa18b4f827f9780 : SUCCESS in 34m 37s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0788efdda177457cb23ea1fb2acbf21b : SUCCESS in 39m 59s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9350b979a93646e180846482b76ad0a7 : SUCCESS in 32m 13s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/40e2bb5535694abe98ddc8a2429b29fa : SUCCESS in 17m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fad211c970844afbb263eb870ec6e96c : SUCCESS in 14m 43s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/9376691d2cb44ea2801f0912733110c3 : SUCCESS in 19m 15s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34103bac448d464b978ebb31f9ef6cd5 : SUCCESS in 15m 51s","accounts_in_message":[],"_revision_number":4},{"id":"8a3a05c49b41a05dd4b53c8dca2bd85d288e46b8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-19 17:57:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a10702c9b3717ab6ca050e0cfbd461a3286782dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 18:48:02.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"0b9e4a371508ea4fa3b51eba26c8ee5dbb877905","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-19 20:02:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"15432da8c909a531eb6827d0fd04c30e19fa1b81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 22:09:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2442c27903941f7a429c6576d97dc70 : SUCCESS in 9m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4077c536d124576b7d99979ef8e72a8 : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d8007cbd0cf44cab4aedbe035f008de : SUCCESS in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/104829cade844fcc8c5c4f84c579610c : SUCCESS in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ade74e143db4e48a28c5fdb7ba2003c : SUCCESS in 10m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7f412481e0774a63b9350149fc752dab : SUCCESS in 7m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1c9b3adaa53445b0ace855d737ceb55d : SUCCESS in 14m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7f92a32ee3e94a4eb5d71d620efe1e34 : SUCCESS in 31m 44s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/08130536e3de43d3aab65aa758d0c80d : SUCCESS in 1h 06m 39s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e06dff1959047c7b25c8b6ea3ff2864 : SUCCESS in 1h 01m 49s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0fa4d825dab64f62b824506dc99ad055 : SUCCESS in 1h 04m 42s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5797efa4eb2f4d3190221f4926615a0d : SUCCESS in 26m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb9f4d280f3c41faa8f3ba6682d87cb0 : SUCCESS in 15m 18s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/88bd5a7df9224d5393d64fac32e61800 : SUCCESS in 2h 02m 26s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3924f454d0f4545892eccd07daa880e : SUCCESS in 1h 14m 12s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1d90bb91ab924f698352d90a32bdbf7a : SUCCESS in 34m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b235b4deb6314a7fb72c814954bb4132 : SUCCESS in 32m 42s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c8bf2250660645949bdf3d750e949c5b : SUCCESS in 29m 33s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a979d8db72bd429fbe10904f22d9b309 : SUCCESS in 29m 50s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/86d0858ee7474379878f36e51bcb592d : SUCCESS in 18m 15s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/547e8bf328754604a9769f533d89f139 : SUCCESS in 13m 34s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/38e676f3f962431b946bff96ebc4424b : SUCCESS in 19m 09s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a3c71a21e7c46999c36b8e3b7eefb95 : SUCCESS in 14m 57s","accounts_in_message":[],"_revision_number":4},{"id":"c802e708716012b2c2a38bb2ff8eadb08818b6d0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-13 19:38:17.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c400ebe0cdf7d4762025c9a3842faab908dd0902","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-14 12:42:05.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4879c1cd29c0612711721eec1d78bfa25c8464c7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-14 12:45:34.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"267e31fec37d1c7feb344912358b0eca7da0299b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-14 12:47:47.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"96c7330edb12c057500c6373c924937b820c8991","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 14:07:05.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ff50ed1d5584bd2adc0a5caff9dcb2d : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ffdae73e2af41578d3374b19eaf182e : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/998a90fbe3d04068bfdbc8ea39986177 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6caa94d4e194c5d87f9267210fcf84d : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2541daa2374e4ef59c8ce6a79fac4271 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab80d1733d2e4a1db43af69f30d51e1e : SUCCESS in 8m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d5e0b4dc13364127b31aa2fa51591960 : SUCCESS in 5m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/70773b1e730d4a5995731f91d7530b04 : SUCCESS in 12m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/55572800fd104ea6826e1bfc0057f80d : SUCCESS in 18m 17s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/71872e68c4d248218ddcd8006b2da9de : FAILURE in 22m 57s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/a6beac8bd7824fb09b59664bd21b1c89 : FAILURE in 25m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b46014ee14b4fa08309ea591a405b72 : FAILURE in 17m 36s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1d0123e3d6af48899b2b8d3993135650 : FAILURE in 25m 40s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6c9a1df3f84242b3a899a5d92591d3e3 : SUCCESS in 21m 43s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/c22514c2322848539a2b50518a1cf6de : SUCCESS in 22m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6e24a3088a44575991e227515802471 : SUCCESS in 13m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/27570f42c6064c51ac8fb0b9a7e0fb17 : FAILURE in 1h 08m 40s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ad5def4fc1c437c9ad4a8f813d84a65 : FAILURE in 52m 19s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2b1d753fdd9c4981ad03247e4b3fa687 : SUCCESS in 19m 00s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/5c845cd4fcb94cd99cb6a19fea7f5b19 : SUCCESS in 21m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b71dafeba6874af7911ccd70932f2e94 : SUCCESS in 14m 25s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/cec22298c1f143d9be1bce1e891320ee : SUCCESS in 44m 13s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/257aa86426dd426e95800796bb7c8da6 : SUCCESS in 34m 00s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30fa76ba53404843aa1355b121b0ea71 : SUCCESS in 25m 30s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6e70157c06c14e4ea5e172b7a2eacb9a : SUCCESS in 14m 37s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/bccbad131d784a14af4f86140bb94b5b : SUCCESS in 14m 26s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c510b18f40964a308014e2895ea5e2ec : SUCCESS in 10m 56s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/65b8788dd96949f7851ea7a58df05906 : SUCCESS in 20m 10s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/1a1ac03549bc48e08a26cbd571e939f0 : SUCCESS in 15m 44s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b859320030f40c9ba36b01622d7b499 : SUCCESS in 12m 11s","accounts_in_message":[],"_revision_number":6},{"id":"c06e347005001f12b6bcc4b62e6fa90a1461f570","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-14 14:49:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d8f4b4dbf5b568cc27a617973e0f4eadf2593444","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-30 20:23:13.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6ee65489bf519730c0a5d028f6eed70c307f0383","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-30 21:31:38.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37ce1f73b9e14e7b873123f9cbab9165 : RETRY_LIMIT in 1m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d5cfd931a7546c58e34f78bef38615b : RETRY_LIMIT in 1m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/545cff8162de46cbb7f0ba88f6269fce : RETRY_LIMIT in 1m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12fff05d62c24ad7a0634d4869b46670 : RETRY_LIMIT in 1m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2fdb107c27047c9bd5f001dfbc39c8d : RETRY_LIMIT in 1m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a79e2a68c86f42cbad27a006fe17126a : RETRY_LIMIT in 1m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/db5d73ed945b43d0a206313bd1d0a9a1 : RETRY_LIMIT in 1m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/70243f4cd9c3478493d1c26ade9d22ac : RETRY_LIMIT in 1m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/45bef53002d742aba376bf676351335d : RETRY_LIMIT in 1m 11s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f43391e16a5142eb8800b2292d90071e : RETRY_LIMIT in 1m 02s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/8a3793f3e13947d78cb07ac65e11d684 : RETRY_LIMIT in 1m 05s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22ba8bae1b5f449491090354cf91b7c4 : RETRY_LIMIT in 1m 03s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d66425f85b97424eb8585b2d2481506a : RETRY_LIMIT in 1m 04s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/322731e917e44e4fb65e372c57981982 : RETRY_LIMIT in 1m 04s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/31f3c12c4aaa4896866e7d64cfb13eb3 : RETRY_LIMIT in 1m 03s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd3f700692ba4cd38a77ece2b16a612a : RETRY_LIMIT in 1m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/dda06521328f4d9c8d01d7a44c5cfbf1 : RETRY_LIMIT in 1m 02s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0139462b768486c906a40486a826e32 : RETRY_LIMIT in 1m 08s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8816b87935054b7ab75ed2d6931c0fc4 : RETRY_LIMIT in 1m 29s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/db06ef8964474d969bdfc75bb8689192 : RETRY_LIMIT in 1m 03s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cfef29941534868aa93130ca7e15a71 : RETRY_LIMIT in 1m 27s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/e84d1e6b0fdf448098670ecd4fe87178 : RETRY_LIMIT in 1m 01s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0a19b6e2a6ea4ea594be9e9b9c1198a8 : RETRY_LIMIT in 1m 03s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cb779e5cb6f4f8dbfb832cade2b9ede : RETRY_LIMIT in 1m 02s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ad5e9e7ed2e34323986741b5bbaa4710 : RETRY_LIMIT in 1m 00s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/0db89b0c7a444710937e0f8531723105 : RETRY_LIMIT in 1m 09s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b3763a5438141b2b178bf6e1df3ec3d : RETRY_LIMIT in 1m 08s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/68e1b1cdb5dc45229b1f3dc765814bc2 : RETRY_LIMIT in 1m 29s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/cf68f48459c04ee6a69744e17b9fd90d : RETRY_LIMIT in 59s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb6511bef749413ba40bac5adc8bf2e1 : RETRY_LIMIT in 1m 01s","accounts_in_message":[],"_revision_number":6},{"id":"07262190f6e3fcda083c87fd86bba08a12cf7486","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-01 17:05:37.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e38db5104698e7d2fcda03c9baf8447a1d90872c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-01 19:09:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5912f46114624fefb4a0f1884ddd0b66 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14f385a0b0ef48ad94a74b044f7f6d6e : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c75998048ec645a69e06def902f76e65 : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60f7dc13e43e4b90a331681afd2af95e : SUCCESS in 6m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3947fe77098a4367915fdcd8ac55153d : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1870582519e4bb39f4792d4ec31e12d : SUCCESS in 11m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0c54a426c3b64778bbe434b51a631fe7 : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5571b9fc66aa4436828635c7bfd076ec : SUCCESS in 10m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0ef1109846ca43ba9c43ae4719341f5e : SUCCESS in 14m 04s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c3b908be80cb453b8b553e164ea48e29 : SUCCESS in 1h 16m 32s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/75b478eab25b41a581e86c02160f21e0 : SUCCESS in 1h 14m 06s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40bc9f5118924970ab1f239302ee528d : SUCCESS in 52m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7b3e89bbbc414d14a2d61ff73bc5a26a : SUCCESS in 1h 18m 50s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f2479b7820c4423f9a6c7439b90f69a6 : SUCCESS in 17m 35s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b2aaab793f534f40adb7151148b13ab2 : SUCCESS in 20m 32s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64a61713c67443bd8d4bd867c95c6d32 : SUCCESS in 14m 11s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5245385a0f9c4cf9a69b059261667b0f : SUCCESS in 1h 58m 53s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/663b842424b94ab996ef2f850b9dee14 : SUCCESS in 55m 37s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/cc8b160bd5e14bc5be777e0575856f8e : SUCCESS in 15m 27s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/f873fa56b34d4cc89abb5ed7ec24fdd5 : SUCCESS in 22m 21s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f75e29e34c01402b82c5404a87ddac60 : SUCCESS in 15m 53s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/1be11654397e4781a62b01add4ffb61b : SUCCESS in 1h 00m 21s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b8dce04e2aed4e6397f1170df52e3a1f : SUCCESS in 32m 19s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a673ac8ac0942569bf4a4de1b887036 : SUCCESS in 27m 01s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/17470606d89b45f2b0b09e81a762c979 : SUCCESS in 13m 02s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/da3e10cbc71e4b3fb72b8686c664c024 : SUCCESS in 15m 27s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e03ed0589dd4836a19ea84cf9d578c6 : SUCCESS in 11m 51s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/b74561682ee6481a9ae972df90fc2254 : SUCCESS in 14m 10s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/052c1d347ef9415fa59f9596a347a2e8 : SUCCESS in 16m 17s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a69b7d53c0404493af038106efe2d91b : SUCCESS in 11m 01s","accounts_in_message":[],"_revision_number":6},{"id":"0821141b181fc4e7569774f3cfc65e78b5f4e384","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-05-03 09:05:44.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"adec6a8b968b9fc9be4342b96c353d3ffc1fa06c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-03 09:20:42.000000000","message":"Patch Set 6: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b4a7116202bb15397cf76d9c28433e915ae41b6e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-05-03 15:29:25.000000000","message":"Patch Set 6: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"773e4c5f30acc6163fd106df131df17fcaf2cbc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 15:29:55.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e28b7765ad2cbb946f9f85443761f416eb0fc9e2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-05-03 15:31:12.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"36f150e65f565d8cd52387f8a9a5bf2439ef7c69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 17:40:27.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25b1f20413bd402b9d4d66a5953d32c9 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed856929389740d4b5498c4a8e03f26d : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef7ed9ba30924a24a185d53eff37a404 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c99167ac3c8348899fa07ede035e99de : SUCCESS in 9m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/96300a5511aa4677b9bc615ed46358ae : SUCCESS in 5m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6451979b2fc14a1ea8b00eb77091fb57 : SUCCESS in 12m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d10b480c69924f1da363a6c751b15c43 : SUCCESS in 16m 25s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b1687b08cfb1471184ce5bca8ec044c4 : SUCCESS in 1h 06m 56s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/72776f9b79b24109ac112c149220df5a : SUCCESS in 1h 01m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6324adad001a491b8026a80eafd191ff : SUCCESS in 51m 49s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/960f659d02884b95a0bdf7fa51514b9e : SUCCESS in 55m 54s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/eacd581bb44f439b844d547df646d576 : SUCCESS in 22m 21s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/8596a85f7c12417684097981880da76d : SUCCESS in 19m 40s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3aeb346bb68f4872bf459ae4fcfa5bb9 : SUCCESS in 14m 57s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ba1578a6d41446b5ac6ac9dbf9c1b2ce : SUCCESS in 2h 06m 35s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a7a17f2415b48ff8f4fd7d5f28549c7 : SUCCESS in 1h 16m 29s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/718b07c6bbcd49c9bfab53a6e31a6244 : SUCCESS in 41m 54s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/c15f53d7ee454c519f4ea6a62d3b45bd : SUCCESS in 19m 18s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bbaf715a98e4e7a80f56c2295a15e31 : SUCCESS in 15m 58s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cd0197df48e24193b96d3af64f0b1437 : SUCCESS in 35m 22s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2079a9cfb57c4a9384b116f620204437 : SUCCESS in 27m 53s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/771497cd871e4ff1ae646d61b03be0cf : SUCCESS in 14m 20s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a22c91034ac047448cab9f48aa60025b : SUCCESS in 14m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddfbaf0516764331a6e129900aeb7453 : SUCCESS in 13m 26s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/982b8430bbd6472aac739166074ef729 : SUCCESS in 15m 34s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/b9284c7ac846482ab07ddfd2de428d7a : SUCCESS in 15m 52s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b867e4a0352041788d150ce04ece2b8e : SUCCESS in 13m 33s","accounts_in_message":[],"_revision_number":6},{"id":"718c1e299f33134b90bc7629bc9c20d1b7c2e67d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 17:40:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"321f157b642f95a8fb5a3047783240f26a8ae63a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 17:41:37.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f523c54f12f14865a635f9a2dc4a0cfc : SUCCESS in 50s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"43829dadc1f782d0abbf267925df3a4b0e4acdb3","revisions":{"8cce3f23e4079bc3fb0123bdeaced016d03b4154":{"kind":"REWORK","_number":1,"created":"2022-02-03 12:31:00.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/53/827653/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/827653/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/827653/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/827653/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/827653/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/827653/1"}}},"commit":{"parents":[{"commit":"d6e836586762e41211940532b736967bad631cc3","subject":"Merge \"Support for untemplated dirs in kolla config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d6e836586762e41211940532b736967bad631cc3"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-03 11:50:58.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-03 11:54:02.000000000","tz":180},"subject":"Multiple DHCP Subnets for Ironic Inspector","message":"Multiple DHCP Subnets for Ironic Inspector\n\nSynchronize with new kolla-ansible parameters introduced in the\nIb69fc0017b3bfbc8da4dfd4301710fbf88be661a for Ironic Inspector.\n\nDepends-On: Ib69fc0017b3bfbc8da4dfd4301710fbf88be661a\nChange-Id: I6d9e3acc477c9c4b3bb9db9c01a2db83b9568f59\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8cce3f23e4079bc3fb0123bdeaced016d03b4154"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8cce3f23e4079bc3fb0123bdeaced016d03b4154"}]},"branch":"refs/heads/master"},"b8f3776248f021a3d7d29088806c28f5846f1b3b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-02-03 15:15:40.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/53/827653/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/827653/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/827653/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/827653/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/827653/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/827653/2"}}},"commit":{"parents":[{"commit":"7173a9f73126eca1cc36bd6997acf10ac75832ff","subject":"ntp: Fix service mask when service doesn\u0027t exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7173a9f73126eca1cc36bd6997acf10ac75832ff"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-03 11:50:58.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-03 15:15:40.000000000","tz":0},"subject":"Multiple DHCP Subnets for Ironic Inspector","message":"Multiple DHCP Subnets for Ironic Inspector\n\nSynchronize with new kolla-ansible parameters introduced in the\nIb69fc0017b3bfbc8da4dfd4301710fbf88be661a for Ironic Inspector.\n\nDepends-On: Ib69fc0017b3bfbc8da4dfd4301710fbf88be661a\nChange-Id: I6d9e3acc477c9c4b3bb9db9c01a2db83b9568f59\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8f3776248f021a3d7d29088806c28f5846f1b3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8f3776248f021a3d7d29088806c28f5846f1b3b"}]},"branch":"refs/heads/master","description":"Rebase"},"05cae74caf6888e9909e194e390ba364904182f2":{"kind":"NO_CHANGE","_number":3,"created":"2022-02-04 17:47:05.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/53/827653/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/827653/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/827653/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/827653/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/827653/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/827653/3"}}},"commit":{"parents":[{"commit":"7173a9f73126eca1cc36bd6997acf10ac75832ff","subject":"ntp: Fix service mask when service doesn\u0027t exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7173a9f73126eca1cc36bd6997acf10ac75832ff"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-03 11:50:58.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-04 17:46:47.000000000","tz":180},"subject":"Multiple DHCP Subnets for Ironic Inspector","message":"Multiple DHCP Subnets for Ironic Inspector\n\nSynchronize with new kolla-ansible parameters introduced in the\nIb69fc0017b3bfbc8da4dfd4301710fbf88be661a for Ironic Inspector.\n\nDepends-On: Ib69fc0017b3bfbc8da4dfd4301710fbf88be661a\nChange-Id: I6d9e3acc477c9c4b3bb9db9c01a2db83b9568f59\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05cae74caf6888e9909e194e390ba364904182f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05cae74caf6888e9909e194e390ba364904182f2"}]},"branch":"refs/heads/master"},"9cce4672e89b5fa9b76f115d7d423579ad7e5790":{"kind":"REWORK","_number":4,"created":"2022-03-18 12:55:44.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/53/827653/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/827653/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/827653/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/827653/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/827653/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/827653/4"}}},"commit":{"parents":[{"commit":"a1ed4a05bb51e89abf7841715dee8251f9fbf53a","subject":"Merge \"Use naming convention to infer VLAN tagging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a1ed4a05bb51e89abf7841715dee8251f9fbf53a"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-03 11:50:58.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-18 12:55:01.000000000","tz":180},"subject":"Multiple DHCP Subnets for Ironic Inspector","message":"Multiple DHCP Subnets for Ironic Inspector\n\nSynchronize with new kolla-ansible parameters introduced in the\nIb69fc0017b3bfbc8da4dfd4301710fbf88be661a for Ironic Inspector.\n\nDepends-On: Ib69fc0017b3bfbc8da4dfd4301710fbf88be661a\nChange-Id: I6d9e3acc477c9c4b3bb9db9c01a2db83b9568f59\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9cce4672e89b5fa9b76f115d7d423579ad7e5790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9cce4672e89b5fa9b76f115d7d423579ad7e5790"}]},"branch":"refs/heads/master"},"d179dd84a275dead9e94bf15f7ea115a10ac5389":{"kind":"REWORK","_number":5,"created":"2022-04-14 12:42:05.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/53/827653/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/827653/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/827653/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/827653/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/827653/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/827653/5"}}},"commit":{"parents":[{"commit":"a1ed4a05bb51e89abf7841715dee8251f9fbf53a","subject":"Merge \"Use naming convention to infer VLAN tagging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a1ed4a05bb51e89abf7841715dee8251f9fbf53a"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-03 11:50:58.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-14 12:40:20.000000000","tz":180},"subject":"Multiple DHCP ranges for Ironic Inspector","message":"Multiple DHCP ranges for Ironic Inspector\n\nSynchronize with new kolla-ansible parameters introduced in the\nIb69fc0017b3bfbc8da4dfd4301710fbf88be661a for Ironic Inspector.\n\nDepends-On: Ib69fc0017b3bfbc8da4dfd4301710fbf88be661a\nChange-Id: I6d9e3acc477c9c4b3bb9db9c01a2db83b9568f59\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d179dd84a275dead9e94bf15f7ea115a10ac5389"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d179dd84a275dead9e94bf15f7ea115a10ac5389"}]},"branch":"refs/heads/master"},"43829dadc1f782d0abbf267925df3a4b0e4acdb3":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-04-14 12:47:47.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/53/827653/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/827653/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/827653/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/827653/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/827653/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/827653/6"}}},"commit":{"parents":[{"commit":"341bde30b9b7e8e924fb0ce7a5918f5eb2119d79","subject":"Add Python3 zed unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/341bde30b9b7e8e924fb0ce7a5918f5eb2119d79"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-03 11:50:58.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-14 12:47:32.000000000","tz":180},"subject":"Multiple DHCP ranges for Ironic Inspector","message":"Multiple DHCP ranges for Ironic Inspector\n\nSynchronize with new kolla-ansible parameters introduced in the\nIb69fc0017b3bfbc8da4dfd4301710fbf88be661a for Ironic Inspector.\n\nDepends-On: Ib69fc0017b3bfbc8da4dfd4301710fbf88be661a\nChange-Id: I6d9e3acc477c9c4b3bb9db9c01a2db83b9568f59\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/43829dadc1f782d0abbf267925df3a4b0e4acdb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/43829dadc1f782d0abbf267925df3a4b0e4acdb3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
