)]}'
{"id":"openstack%2Fkayobe~828367","triplet_id":"openstack%2Fkayobe~master~Icb45116708bd31581c50a758180a531144eeac8f","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-14 09:55:05.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Icb45116708bd31581c50a758180a531144eeac8f","subject":"DNM: use correct libvirt roles","status":"ABANDONED","created":"2022-02-08 15:39:59.000000000","updated":"2022-02-14 09:55:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b10b8512c2596bbd3b79cc54391b850c969ae5e7","_number":828367,"virtual_id_number":828367,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-02-10 12:13:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-08 17:50:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6e0de0e1a7d010f04c744cc0cf1d838816815d55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-08 15:39:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7292d3db1abfe62a32dbe4c8980a1a582f697bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 17:50:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/eb23274bc76e4c3e9c3dd9dac3aead28 : RETRY_LIMIT in 6m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc2a46d2bec842feab39d4cce199b07d : RETRY_LIMIT in 4m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4655b91ee70b461abfc08e90d2f1f0dd : RETRY_LIMIT in 6m 35s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9e711a50abd54ac8a479a71a8bd6be96 : SUCCESS in 1h 54m 48s","accounts_in_message":[],"_revision_number":1},{"id":"fc6334f1eff8e03b8aaa49808849c4523138f5a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-09 08:55:48.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"cc52fa1b5364857beb91ad8b7ea2f6a6a92f50ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 11:10:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/590b23ab40f342dc89b6cc39c4166a61 : SUCCESS in 1h 00m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23f8804004a64be6a3995d053098adc5 : SUCCESS in 49m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f8225736f5df43ee9a8031b271897956 : TIMED_OUT in 2h 05m 48s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/eac4dfcf008b4b14ae92998785ac7868 : SUCCESS in 1h 34m 07s","accounts_in_message":[],"_revision_number":2},{"id":"9d3ebf9f7f95afeb87796bf3daf0c1f192d2cc57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-09 12:15:17.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b2e8c52d2ac0e1b0636b4f701fd3a8273ed4a4bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-09 13:33:22.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"119f06bd298a439e9bca018e0e3817b6843aa4ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 15:57:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6f4d370504e2425493cb4a6c9f1a91de : SUCCESS in 1h 11m 03s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66aadce6026e4972a74f79e1638dc020 : FAILURE in 40m 29s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4d16c3898152491e8667bab3b3c30495 : FAILURE in 2h 02m 33s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3d447219e4a2487d8881828c7b696db9 : SUCCESS in 2h 09m 59s","accounts_in_message":[],"_revision_number":4},{"id":"2b6d209ca1bae91ee3d74ccc392d6c13e402a11f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-10 10:40:35.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"70a93fb86f667970f318182c1fab673706847047","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 12:13:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/14e5184ddd4d434d84b1f6cc2654ff2c : SUCCESS in 1h 06m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c80a94ce6dc44d2abba15ac13403b70 : FAILURE in 48m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8341deeeed71459186befaf4f1e09e1d : SUCCESS in 1h 09m 17s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e195569ff4c44fa88e9646aab4468d11 : SUCCESS in 1h 30m 21s","accounts_in_message":[],"_revision_number":5},{"id":"b10b8512c2596bbd3b79cc54391b850c969ae5e7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-14 09:55:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4cd9ab97b09e34b57bd25855a129239fda22776e","revisions":{"21f755a0a92a0fc590e92211e13cea1c02d5ab4e":{"kind":"REWORK","_number":1,"created":"2022-02-08 15:39:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/67/828367/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/828367/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/828367/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/828367/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/828367/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/828367/1"}}},"commit":{"parents":[{"commit":"b57b94bbad4333020cc6906d124d6f01f75e1e2c","subject":"CI: test ironic UEFI boot mode in overcloud TLS jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b57b94bbad4333020cc6906d124d6f01f75e1e2c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-08 15:39:52.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-08 15:39:52.000000000","tz":0},"subject":"DNM: use correct libvirt roles","message":"DNM: use correct libvirt roles\n\nChange-Id: Icb45116708bd31581c50a758180a531144eeac8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/21f755a0a92a0fc590e92211e13cea1c02d5ab4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/21f755a0a92a0fc590e92211e13cea1c02d5ab4e"}]},"branch":"refs/heads/master"},"fe21cc72af4fc0d403427d68bf6532928e634239":{"kind":"REWORK","_number":2,"created":"2022-02-09 08:55:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/67/828367/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/828367/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/828367/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/828367/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/828367/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/828367/2"}}},"commit":{"parents":[{"commit":"b57b94bbad4333020cc6906d124d6f01f75e1e2c","subject":"CI: test ironic UEFI boot mode in overcloud TLS jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b57b94bbad4333020cc6906d124d6f01f75e1e2c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-08 15:39:52.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-09 08:55:48.000000000","tz":0},"subject":"DNM: use correct libvirt roles","message":"DNM: use correct libvirt roles\n\nChange-Id: Icb45116708bd31581c50a758180a531144eeac8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fe21cc72af4fc0d403427d68bf6532928e634239"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fe21cc72af4fc0d403427d68bf6532928e634239"}]},"branch":"refs/heads/master"},"a034c9c8b750dbf891aad44f5a49fe376a723cd4":{"kind":"REWORK","_number":3,"created":"2022-02-09 12:15:17.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/67/828367/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/828367/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/828367/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/828367/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/828367/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/828367/3"}}},"commit":{"parents":[{"commit":"b57b94bbad4333020cc6906d124d6f01f75e1e2c","subject":"CI: test ironic UEFI boot mode in overcloud TLS jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b57b94bbad4333020cc6906d124d6f01f75e1e2c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-08 15:39:52.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-09 12:15:17.000000000","tz":0},"subject":"DNM: use correct libvirt roles","message":"DNM: use correct libvirt roles\n\nChange-Id: Icb45116708bd31581c50a758180a531144eeac8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a034c9c8b750dbf891aad44f5a49fe376a723cd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a034c9c8b750dbf891aad44f5a49fe376a723cd4"}]},"branch":"refs/heads/master"},"cc5e66618b9ee2ddc7ea3e559d98ada841c0f040":{"kind":"REWORK","_number":4,"created":"2022-02-09 13:33:22.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/67/828367/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/828367/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/828367/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/828367/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/828367/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/828367/4"}}},"commit":{"parents":[{"commit":"b57b94bbad4333020cc6906d124d6f01f75e1e2c","subject":"CI: test ironic UEFI boot mode in overcloud TLS jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b57b94bbad4333020cc6906d124d6f01f75e1e2c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-08 15:39:52.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-09 13:33:22.000000000","tz":0},"subject":"DNM: use correct libvirt roles","message":"DNM: use correct libvirt roles\n\nChange-Id: Icb45116708bd31581c50a758180a531144eeac8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc5e66618b9ee2ddc7ea3e559d98ada841c0f040"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc5e66618b9ee2ddc7ea3e559d98ada841c0f040"}]},"branch":"refs/heads/master"},"4cd9ab97b09e34b57bd25855a129239fda22776e":{"kind":"REWORK","_number":5,"created":"2022-02-10 10:40:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/67/828367/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/828367/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/828367/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/828367/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/828367/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/828367/5"}}},"commit":{"parents":[{"commit":"b57b94bbad4333020cc6906d124d6f01f75e1e2c","subject":"CI: test ironic UEFI boot mode in overcloud TLS jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b57b94bbad4333020cc6906d124d6f01f75e1e2c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-08 15:39:52.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-10 10:40:35.000000000","tz":0},"subject":"DNM: use correct libvirt roles","message":"DNM: use correct libvirt roles\n\nChange-Id: Icb45116708bd31581c50a758180a531144eeac8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4cd9ab97b09e34b57bd25855a129239fda22776e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4cd9ab97b09e34b57bd25855a129239fda22776e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
