)]}'
{"id":"openstack%2Fkayobe~828658","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I42026fafb1be0071f5ec94e81881c4a3bdd34af8","project":"openstack/kayobe","branch":"stable/xena","topic":"story/2009810-stable/xena","hashtags":[],"change_id":"I42026fafb1be0071f5ec94e81881c4a3bdd34af8","subject":"Set requirements branch for IPA build","status":"MERGED","created":"2022-02-10 10:32:21.000000000","updated":"2022-02-11 11:25:03.000000000","submitted":"2022-02-11 11:23:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828658-story/2009810-stable/xena","cherry_pick_of_change":826735,"cherry_pick_of_patch_set":4,"meta_rev_id":"cd8df047557a5e15d9fa1d57c60d340887f5fa05","_number":828658,"virtual_id_number":828658,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-11 11:23:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-02-11 09:03:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-11 09:28:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-11 09:28:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-10 10:32:21.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2022-02-10 12:19:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-11 09:28:51.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"2503edfc5abcc4925577c36c529c972e2e1e960b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-10 10:32:21.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"869904ca5313d884c6b01a405e59f12851aa76ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 12:19:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b62dd5d749404b9ba12ce86ed3db3c1f : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6664a30e96a842d08759e2e9f84f02f7 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c86f44a9eb4844728f3b42aaf6bda4fe : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3032ce77c46d4db2b4fdac5fad8f5ce8 : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb78061847d0496f9f09e04c0b9359ce : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b190932d03554ed683fd5bf4f68ee461 : SUCCESS in 9m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8502a11dc754bdbb05c1d1c1ed8c801 : SUCCESS in 3m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fb61c92a0a78405d9495355e6d921819 : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8dc2202aa042471cb2adf0da3361485f : SUCCESS in 11m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ab6e37cb1bf044b5b00ee4d3b31c5746 : SUCCESS in 28m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ebb3e8c0f5e64465b49ea52bf00b575e : SUCCESS in 1h 03m 51s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48e6a8eeb86c494cbea2f0060cc40354 : SUCCESS in 57m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c94b3a44d32a42598e8d8a55b4f0c1d8 : SUCCESS in 1h 06m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d953e1e3d41241d7867c31ad61d97de3 : SUCCESS in 21m 45s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68726b3294ed4c08874b61e27c23a235 : SUCCESS in 17m 31s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9d524215a6e344bab44b67df221efc1c : SUCCESS in 1h 44m 37s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f09c144bf1e483c872aad5344e989ec : SUCCESS in 1h 18m 14s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c0dcae6c3f14420389725e5123f7b264 : SUCCESS in 35m 32s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4aa54aec4884bd9a767f18308e5a587 : SUCCESS in 29m 59s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/db45463309ae45ac9e37509a7a52ca8b : SUCCESS in 25m 50s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48aa7ccbe6514a6aabb7c2eebd5f7ce2 : SUCCESS in 26m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/13c1f4058f9e4dcc80082c8014da2062 : SUCCESS in 16m 56s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c4d3ed4c5244a2eae6c90f86a284925 : SUCCESS in 12m 16s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/4776b5161ddc4bc4aa58c25d7779c39f : SUCCESS in 17m 38s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3eef40cfb81c4af7a9d956b458f2ab7f : SUCCESS in 15m 33s","accounts_in_message":[],"_revision_number":1},{"id":"4e9abcd4980349507b28959e342492f8495a160a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-11 09:03:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0c05421d7b2d0e00389fb295f9390fa8fffdc198","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-11 09:28:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5f0424775fd3635a4f55d39d83b01784ba1a3620","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 09:29:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6cea92876f74f1c31d052c505c2aef9b5e6afff8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 11:23:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e6bbe52836c4ddb8d80c3ea3e8c4f2b : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c954f5c1bbb4e2a866b0580acd647cf : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2575be613cf84ddb986762cc8ba7c400 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed6c71431c02475e9c40d613aaee8435 : SUCCESS in 9m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d730e918db9249948700fab4ceeec806 : SUCCESS in 4m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7681728a5f1743ca9ee24236d2532d12 : SUCCESS in 6m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4800321653094af9bf07220b53dd323d : SUCCESS in 12m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dbf9f12607a64f0ba3ffd20167418748 : SUCCESS in 34m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6f72f23b47f64e06b8ce7ba7ce9091fb : SUCCESS in 1h 09m 44s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/802759ce72c542a9a955b3b56e4f4721 : SUCCESS in 49m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/330de946b31441949d4bc23a0b4d5aa7 : SUCCESS in 1h 10m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0c29c68a07524eb9995c6421c5c1359f : SUCCESS in 24m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8498d381ed2f4e10bdf7f54b570384c5 : SUCCESS in 15m 16s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/be6094ac3f344c76a11eca75798bfe2b : SUCCESS in 1h 52m 45s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a7f6d7c88bd447e856efa08eebc2050 : SUCCESS in 1h 26m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/55128778613e4d5690bb1abfba115134 : SUCCESS in 35m 59s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f96d45a00d44605b8d34cb0a9c0dffe : SUCCESS in 30m 51s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/dc699afd78894d11b84217e2537a98c5 : SUCCESS in 29m 27s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/928c6c4f848143cba4c8a18acc5e58c7 : SUCCESS in 24m 13s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2caa70ea3123460087d03b600debd214 : SUCCESS in 19m 15s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e610e093e0ee4e58ab81d1547c41d0a3 : SUCCESS in 12m 04s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/80a02cdd143a45e0bd40a3a4f4904a26 : SUCCESS in 19m 20s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52478927fa824107b32efdd3b167a593 : SUCCESS in 13m 29s","accounts_in_message":[],"_revision_number":1},{"id":"0e350bea8286053faba7bb8db63144c4e69dd619","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 11:23:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cd8df047557a5e15d9fa1d57c60d340887f5fa05","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 11:25:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bae790ab04d4eb4906e20e4dfe7907a : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c642233a9b142cf92fa87c866d5d5b5 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"deb969e597608838730d47cfa8693d431661446c","revisions":{"deb969e597608838730d47cfa8693d431661446c":{"kind":"REWORK","_number":1,"created":"2022-02-10 10:32:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/58/828658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/58/828658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/828658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/828658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/828658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/58/828658/1"}}},"commit":{"parents":[{"commit":"816456972eb48adc03135b455928fb3b3ea58fdd","subject":"ntp: Fix service mask when service doesn\u0027t exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/816456972eb48adc03135b455928fb3b3ea58fdd"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-01-27 16:51:12.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-10 10:32:21.000000000","tz":0},"subject":"Set requirements branch for IPA build","message":"Set requirements branch for IPA build\n\nCurrently we use the HEAD reference for OpenStack requirements. This can\ncreate images that are incompatible with your version of OpenStack.\n\nSee:\nhttps://review.opendev.org/c/openstack/ironic-python-agent-builder/+/688911\n\nChange-Id: I42026fafb1be0071f5ec94e81881c4a3bdd34af8\nStory: 2009810\nTask: 44371\n(cherry picked from commit 156ae2bb8ba9df35696018b8321e6d1ab9060ef0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/deb969e597608838730d47cfa8693d431661446c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/deb969e597608838730d47cfa8693d431661446c"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
