)]}'
{"id":"openstack%2Fkayobe~828660","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I42026fafb1be0071f5ec94e81881c4a3bdd34af8","project":"openstack/kayobe","branch":"stable/victoria","topic":"story/2009810-stable/victoria","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-03-02 11:38:14.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-02 10:16:04.000000000","reason":"Mark Goddard replied on the change"}},"hashtags":[],"change_id":"I42026fafb1be0071f5ec94e81881c4a3bdd34af8","subject":"Set requirements branch for IPA build","status":"MERGED","created":"2022-02-10 11:37:01.000000000","updated":"2022-03-02 11:40:07.000000000","submitted":"2022-03-02 11:38:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828660-story/2009810-stable/victoria","cherry_pick_of_change":826735,"cherry_pick_of_patch_set":4,"meta_rev_id":"79eb52018dc33b60d19d12a54bc54ead655c9b58","_number":828660,"virtual_id_number":828660,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-02 11:38:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2022-03-02 11:40:07.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2022-03-02 10:16:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-02 11:40:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2022-03-02 10:16:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-10 11:37:01.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2022-02-10 14:17:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-28 21:51:57.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2022-03-01 08:52:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"78a6094d448ad9c7e3de451c1a577c7a23cbc727","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-10 11:37:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* doc/source/configuration/reference/ironic-python-agent.rst","accounts_in_message":[],"_revision_number":1},{"id":"b09f96ba1095eda42c165923754945ca324ffdcc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-10 11:52:39.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"bac7dbd3b9944c42b93aa49d88a4f3b3e0002c57","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-10 11:53:14.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e996cd41665467640061e2aa7e2b1a015bfd6988","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 14:17:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9b38fc85ae3452eb768903871f5e330 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47a3c0171afc44f98dc4e43912b315bc : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab70015ef8074424aacb1c590ba401ae : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d205314a38049259c5e79792adad468 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1542be40588e45268170c6f873d3fb51 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fce02adfd095477b9a45d8561326d6cf : SUCCESS in 5m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9e70c640aad8465790bcccaa16cc7850 : SUCCESS in 5m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d81992f369944f14a4a002bc4d619e0b : SUCCESS in 9m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1b42b37eacec44e6b274d66fca894f68 : FAILURE in 6m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c882e615f54348778f54811c03525b2f : SUCCESS in 57m 57s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/79e4e3ed04da4553945569d7acdb799a : SUCCESS in 59m 32s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/48372e9976c94498b141e87843132fcb : FAILURE in 8m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/606fdb9e78814207ba035a4c0bf284a9 : SUCCESS in 2h 18m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bd1e9b0081c7444dbf7a497dbe3c9f4c : SUCCESS in 26m 16s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b8d6d41be7014437b30ff10b55ec1e0f : FAILURE in 18m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d7dcf4f08249489c8e8bd804825ced55 : SUCCESS in 13m 33s","accounts_in_message":[],"_revision_number":3},{"id":"b091e9505085f66ee3237e02a4f7e57f512b568d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-11 09:04:25.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"88feb3399a6f6146a281231c48742acda0fcac8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 13:34:23.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b2a42fff9ea46379fe09382cffd5f18 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/153bacb91b914bd7a42814497a87cd3d : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d642739655594c8aa2e7f0bb7a98e905 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c91d0a6d04514950b1bfcbd30b8d4092 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7d89d91b757479a9c698dbf47f64560 : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eccbb4207ed14e07946e1d5cb008db1a : SUCCESS in 4m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8119f93d541f439faeaa7271f3398dd6 : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1417f844a0c24cb1b90872b6d2ab10fa : SUCCESS in 11m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ad97365121524cf38422c887bc89a2cc : FAILURE in 6m 08s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/46c527cdd8884aa28aec28cb22751339 : SUCCESS in 1h 33m 44s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5c24b31586e84b848cd4a5727d0dd329 : SUCCESS in 1h 02m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f7a391b5c0224fd1be39e1ce3e699165 : FAILURE in 9m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d0a12be3f3a54a0b816b458abe2a07c0 : SUCCESS in 1h 53m 58s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e1bb608395af4078bcdbcd69ac4f83e0 : SUCCESS in 30m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6b2c350355a4464eb63c71aeaf1bde84 : FAILURE in 18m 52s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/088d8b8c18af484d873c6ecf19203b19 : SUCCESS in 13m 59s","accounts_in_message":[],"_revision_number":3},{"id":"e973400759db1d7f86f7d0cc2ed46f9c6a666380","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-28 19:21:49.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"792648f5a198ffa2755bbbfda25f3814176b1f05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 21:19:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56315aab46564016b05439ffc9e57152 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d8b771cc8a0400b9cd6a93833bff313 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/782f0120ec5747bd991badd166fd3056 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5390f45511ca4aa78fc33ebfdf25cadd : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f50bf46dcc504a88af60b6ced79a9b2f : SUCCESS in 9m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f439a7a895845ce934816c108d0066a : SUCCESS in 5m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f2dca225f5ea4efdb998f5890466f725 : SUCCESS in 7m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9854a81a3de2464abbbe7c0ea328ba2c : SUCCESS in 10m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/131f4a41328a4b3a99be10af4ea2911d : SUCCESS in 15m 57s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8ae918821c2d42c0bceeb0b617c19296 : SUCCESS in 58m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6969fd56a7274654983b89ba37f259f4 : SUCCESS in 54m 01s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/93c62c5f769e4507a9a698a3a7ccafac : SUCCESS in 19m 02s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/159a76496a51417d93bf61629a9b0282 : SUCCESS in 1h 54m 36s (non-voting)\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/85369ecd495a4b7eb1a20e25eba321fb : SUCCESS in 27m 44s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/efb9764b53864596ad06784f28c781c9 : FAILURE in 19m 57s (non-voting)\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2dd43071fcad4fa093a1f926ae7f7b07 : SUCCESS in 13m 56s","accounts_in_message":[],"_revision_number":4},{"id":"e8bc6ee0bb250b1154dfe1d227f2c98405da90e6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-28 21:51:57.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"177a30ebcc0753ceed52a64bd241dba8d8f91556","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-01 08:52:35.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"85c14ce9793cbdec95cfe5e6ab54e67f757d8d72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-01 09:02:08.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"22f4d6f01a3622f800f04c277309ccff76e15017","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 13:17:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e8aa2c60743459fb7a861e42bf532c4 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f34e83b0b2641999f4d836d8166999a : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e00b400a09a048c2acb2e7220f4027a4 : SUCCESS in 7m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d51d184e2d534aaf880d64c5d133f0d8 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e05490988c314c4183228ffe47750072 : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70466d92198d430aa6fcf5779a4a8fb2 : SUCCESS in 5m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1c2ad3a53da54d8f811622949a08f6c5 : SUCCESS in 6m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bffb5fb5b29e440bb85e3ddb3d3a5366 : SUCCESS in 10m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4155b0b9bb5448a093b0805a1c5398a6 : SUCCESS in 20m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fe0e12d7ab8948738c05fb0e13a21569 : SUCCESS in 1h 02m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/454c523073e34eddab16fadb905415ea : SUCCESS in 1h 04m 12s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a7de5d82be3d4ab79ae5bc7a40aa1873 : SUCCESS in 17m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6e74a88cfa03458eb9310dbd14997ba3 : RETRY_LIMIT in 1h 26m 00s (non-voting)\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3eb9cee0bf2e434195a7de479f290d49 : SUCCESS in 30m 16s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b50cc66c422b40cab0dceae3c3b6c61a : FAILURE in 22m 53s (non-voting)\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/bd7086654d894e53aba5181a76a2e4f8 : SUCCESS in 15m 11s","accounts_in_message":[],"_revision_number":5},{"id":"22586e8c60f9379412a8da12ad7236d343183e25","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-02 10:16:04.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"51dff1474e6898edfcfd49818de75ddcb66b3bba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 10:16:44.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b4fc9767c092d8a11d909a2bf0a406e31bd710de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 11:38:14.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/916f343d3a1542aeb22b62c19eb15eb4 : SUCCESS in 11m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fc9a5518c1e48f6a9b546d3f60d8dff : SUCCESS in 8m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f4ede2f7a5e42a9b80b931973efa753 : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87f0a6780a2f4d7789f0eee591ef9e8f : SUCCESS in 7m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcc1f1a4a2514dfdbeebc949f4eb5c72 : SUCCESS in 6m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8a08a5c256794d4eb4815944cac72dbc : SUCCESS in 6m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/625949f117a54562ac841303e1d071cc : SUCCESS in 10m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8b96c6fc70be450cafd72eef90f2c537 : SUCCESS in 20m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/de73c24e714a4e0a8b4e67f65ce68cf4 : SUCCESS in 1h 03m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/883d64118caa4add804f9bbe56ed2e6a : SUCCESS in 1h 15m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e13076756b704f1f8cfe013184a6ad2d : SUCCESS in 17m 53s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a39dc94a25074015b930947a65b06955 : SUCCESS in 33m 16s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/da2ad17c96f94894aadd24523116d2cf : SUCCESS in 14m 09s","accounts_in_message":[],"_revision_number":5},{"id":"af73db78fad018dfdacd48a2a055ada6bb38c424","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 11:38:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"15e2f5084ed591b8771369f6633d5e9cc8a5a710","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 11:40:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bd31525e3414fc09f16aeb1528c6ea0 : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f750b80651aa43e283f681e1854ea1c2 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c4b68fc7394516eb06d7d0537d25ee3b88234fa7","revisions":{"55c7a3c36e03c4997e59c04d244068763b3d2f4c":{"kind":"REWORK","_number":1,"created":"2022-02-10 11:37:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/60/828660/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/828660/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/828660/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/828660/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/828660/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/828660/1"}}},"commit":{"parents":[{"commit":"adce4a0417ff97ea0abbf361521a020bb2d4f9a2","subject":"Merge \"Generate kolla config when running database commands\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adce4a0417ff97ea0abbf361521a020bb2d4f9a2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-01-27 16:51:12.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-10 11:37:01.000000000","tz":0},"subject":"Set requirements branch for IPA build","message":"Set requirements branch for IPA build\n\nCurrently we use the HEAD reference for OpenStack requirements. This can\ncreate images that are incompatible with your version of OpenStack.\n\nSee:\nhttps://review.opendev.org/c/openstack/ironic-python-agent-builder/+/688911\n\nChange-Id: I42026fafb1be0071f5ec94e81881c4a3bdd34af8\nStory: 2009810\nTask: 44371\n(cherry picked from commit 156ae2bb8ba9df35696018b8321e6d1ab9060ef0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55c7a3c36e03c4997e59c04d244068763b3d2f4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55c7a3c36e03c4997e59c04d244068763b3d2f4c"}]},"branch":"refs/heads/stable/victoria"},"c8a2e19b1c40815ac9377f0df07e80b759026bf4":{"kind":"REWORK","_number":2,"created":"2022-02-10 11:52:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/60/828660/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/828660/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/828660/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/828660/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/828660/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/828660/2"}}},"commit":{"parents":[{"commit":"adce4a0417ff97ea0abbf361521a020bb2d4f9a2","subject":"Merge \"Generate kolla config when running database commands\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adce4a0417ff97ea0abbf361521a020bb2d4f9a2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-01-27 16:51:12.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-10 11:52:39.000000000","tz":0},"subject":"Set requirements branch for IPA build","message":"Set requirements branch for IPA build\n\nCurrently we use the HEAD reference for OpenStack requirements. This can\ncreate images that are incompatible with your version of OpenStack.\n\nSee:\nhttps://review.opendev.org/c/openstack/ironic-python-agent-builder/+/688911\n\nChange-Id: I42026fafb1be0071f5ec94e81881c4a3bdd34af8\nStory: 2009810\nTask: 44371\n(cherry picked from commit 156ae2bb8ba9df35696018b8321e6d1ab9060ef0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c8a2e19b1c40815ac9377f0df07e80b759026bf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c8a2e19b1c40815ac9377f0df07e80b759026bf4"}]},"branch":"refs/heads/stable/victoria"},"0cebcde448027815ef059dfa8a4a594847ced0c3":{"kind":"REWORK","_number":3,"created":"2022-02-10 11:53:14.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/60/828660/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/828660/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/828660/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/828660/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/828660/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/828660/3"}}},"commit":{"parents":[{"commit":"adce4a0417ff97ea0abbf361521a020bb2d4f9a2","subject":"Merge \"Generate kolla config when running database commands\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adce4a0417ff97ea0abbf361521a020bb2d4f9a2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-01-27 16:51:12.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-10 11:53:14.000000000","tz":0},"subject":"Set requirements branch for IPA build","message":"Set requirements branch for IPA build\n\nCurrently we use the HEAD reference for OpenStack requirements. This can\ncreate images that are incompatible with your version of OpenStack.\n\nSee:\nhttps://review.opendev.org/c/openstack/ironic-python-agent-builder/+/688911\n\nChange-Id: I42026fafb1be0071f5ec94e81881c4a3bdd34af8\nStory: 2009810\nTask: 44371\n(cherry picked from commit 156ae2bb8ba9df35696018b8321e6d1ab9060ef0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cebcde448027815ef059dfa8a4a594847ced0c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cebcde448027815ef059dfa8a4a594847ced0c3"}]},"branch":"refs/heads/stable/victoria"},"1c8e72b3072fb392293abc2b00928d8dc935343e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-02-28 19:21:49.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/60/828660/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/828660/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/828660/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/828660/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/828660/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/828660/4"}}},"commit":{"parents":[{"commit":"a3c170c08634ecfb6d544d0d42840a3d387f5eff","subject":"Revert \"CI: temporary disable voting of some jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a3c170c08634ecfb6d544d0d42840a3d387f5eff"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-01-27 16:51:12.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-28 19:21:49.000000000","tz":0},"subject":"Set requirements branch for IPA build","message":"Set requirements branch for IPA build\n\nCurrently we use the HEAD reference for OpenStack requirements. This can\ncreate images that are incompatible with your version of OpenStack.\n\nSee:\nhttps://review.opendev.org/c/openstack/ironic-python-agent-builder/+/688911\n\nChange-Id: I42026fafb1be0071f5ec94e81881c4a3bdd34af8\nStory: 2009810\nTask: 44371\n(cherry picked from commit 156ae2bb8ba9df35696018b8321e6d1ab9060ef0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1c8e72b3072fb392293abc2b00928d8dc935343e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1c8e72b3072fb392293abc2b00928d8dc935343e"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"},"c4b68fc7394516eb06d7d0537d25ee3b88234fa7":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-03-01 09:02:08.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/60/828660/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/828660/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/828660/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/828660/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/828660/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/828660/5"}}},"commit":{"parents":[{"commit":"1264ac29cb2afc4a2325174cec75d4b7b6da7e36","subject":"Revert \"CI: temporary disable voting of some jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1264ac29cb2afc4a2325174cec75d4b7b6da7e36"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-01-27 16:51:12.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-01 09:02:08.000000000","tz":0},"subject":"Set requirements branch for IPA build","message":"Set requirements branch for IPA build\n\nCurrently we use the HEAD reference for OpenStack requirements. This can\ncreate images that are incompatible with your version of OpenStack.\n\nSee:\nhttps://review.opendev.org/c/openstack/ironic-python-agent-builder/+/688911\n\nChange-Id: I42026fafb1be0071f5ec94e81881c4a3bdd34af8\nStory: 2009810\nTask: 44371\n(cherry picked from commit 156ae2bb8ba9df35696018b8321e6d1ab9060ef0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4b68fc7394516eb06d7d0537d25ee3b88234fa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4b68fc7394516eb06d7d0537d25ee3b88234fa7"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
