)]}'
{"id":"openstack%2Fkayobe~829174","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I96827fc32c1594ca9a0535e259929c49d3f0e704","project":"openstack/kayobe","branch":"stable/wallaby","hashtags":[],"change_id":"I96827fc32c1594ca9a0535e259929c49d3f0e704","subject":"CI: Enable bare metal testing for Ubuntu","status":"MERGED","created":"2022-02-15 09:09:17.000000000","updated":"2022-02-15 15:09:45.000000000","submitted":"2022-02-15 15:07:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829174","cherry_pick_of_change":782273,"cherry_pick_of_patch_set":16,"meta_rev_id":"6eb507a7840f586ff32c802b11754278b84819d8","_number":829174,"virtual_id_number":829174,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-15 15:07:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-02-15 11:26:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-15 12:42:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-15 12:42:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-15 10:50:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-15 12:42:58.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"54160db3e4d3c38cd2dad2301ecf7808420f279a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-15 09:09:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"41adff0912f1239191bc92010bc0419757cb5121","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-15 09:09:29.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b2bcaa55dbb531599d7cdc3b053cd0b1954bbc58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 10:50:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a6b6e902ef345828001143983095ba7 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9536f8135a1c43deb9961f7ea10357e6 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28bb7bf903824ba48a4534b5f2e3ea3b : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35c103b37c114d9491008eaad1e7d33e : SUCCESS in 7m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ac66ec3a82043a38ab710b8d3d351f4 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c5c2075fbb642369405a61b4873798a : SUCCESS in 8m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c7880c9fbde1451d9a02fc41eb987a97 : SUCCESS in 8m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/48baa54e7e5b40eca467eaa427fa1103 : SUCCESS in 10m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7f602c75fb7b42f0a5576d7c665a8039 : SUCCESS in 18m 32s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/48f32c6e3aa7471baceba77dc461c58c : SUCCESS in 1h 30m 10s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68493ba3c1a84110b247c62daf8d0fbd : SUCCESS in 47m 38s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/29edaa24d10040d2837c96a224a25615 : SUCCESS in 1h 33m 00s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c03b01de29aa42328eb7f57a75e3b677 : SUCCESS in 20m 08s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dcf0ed1076df4dee9a41464f2e47f6ee : SUCCESS in 16m 20s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b3fd4bd54ead41f2835c80470f9d694f : SUCCESS in 1h 32m 01s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/20dcea21850847ba9504c08ae38d0e45 : SUCCESS in 40m 51s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae2f2dbc3f544de28192108e15bc3bfe : SUCCESS in 29m 05s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cf83a9f1d9854d2e8c202d73ed1dd903 : SUCCESS in 37m 25s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/18c39f42cb564cf8920e4c03a0f7c4c4 : SUCCESS in 13m 35s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78add855bbde471499419aa66e96c631 : SUCCESS in 11m 34s","accounts_in_message":[],"_revision_number":2},{"id":"67c648252f63dab1d32d45f92ee65d7b263dba71","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-15 11:26:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"22de351f95d262c8b49aed9252f6c1a418324000","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-15 12:42:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3b85c76dca6bd97713333e0b6297cabe3bc5fbe9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 12:43:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"180ae16af7069c393703240f6cd2b26f1c7142f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 15:07:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e30a506ab534c1e878c70d8d44fa03b : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05595bdd2424487abfa3cc08576040f3 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c63846fecc29454ab7b16a161cee1640 : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/924e5cf099ea40009a9d625d389eb903 : SUCCESS in 8m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c58d8211edf0441994d6ec759cc4f30c : SUCCESS in 6m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9f4d11b6434b43deb64ef9c06f607858 : SUCCESS in 11m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9f3fa05203c0422994a13b5605f0bde9 : SUCCESS in 20m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4b7ce21c7f3f42f4bd2546e912e2fa64 : SUCCESS in 1h 22m 44s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4eab806fd794f959c61a64ec08f153a : SUCCESS in 51m 34s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7ad3dd3702884c8c96381c3dda49def9 : SUCCESS in 1h 35m 38s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b81f23c2a9b94cc1a37e49d5ec712adc : SUCCESS in 21m 47s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f2846c8d14f4205ad03fa855f3e6e53 : SUCCESS in 15m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8d8210a5bb094d61bf4070387768b66e : SUCCESS in 1h 41m 01s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c4bfe320a1424f299ad070062d3df2d7 : SUCCESS in 39m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60ec9a20912e49ae994b8812991cc232 : SUCCESS in 30m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2176b016834f41cfbe39afd6894e1c13 : SUCCESS in 27m 34s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b715f8c6a3a748709badd40ed04affe0 : SUCCESS in 15m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48c5a7151ca847ad98ca52778c038b51 : SUCCESS in 11m 39s","accounts_in_message":[],"_revision_number":2},{"id":"d74d7fff28b2e58189609ddf6cbcd1b258710a87","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 15:07:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6eb507a7840f586ff32c802b11754278b84819d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 15:09:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33bcc7f2b68f478da54da987747d7f49 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8a859c2363520f40b4f7a93eccf4977c8ebdc15d","revisions":{"cc9a0472788979d6a25314bf4877064204d80af6":{"kind":"REWORK","_number":1,"created":"2022-02-15 09:09:17.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/829174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/829174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/829174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/829174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/829174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/829174/1"}}},"commit":{"parents":[{"commit":"52973754493db472a0b5f058ae76c8af35c2e14c","subject":"Set requirements branch for IPA build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52973754493db472a0b5f058ae76c8af35c2e14c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:10:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-15 09:09:17.000000000","tz":0},"subject":"CI: Enable bare metal testing for Ubuntu","message":"CI: Enable bare metal testing for Ubuntu\n\nPreviously we were seeing issues with ipmitool and virtualbmc on Ubuntu:\n\n  Error setting Chassis Boot Parameter 5\\nError setting Chassis Boot Parameter 0\n\nThe dependent change fixes these issues, and this change enables bare\nmetal testing in Ubuntu CI.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/829006\n\nChange-Id: I96827fc32c1594ca9a0535e259929c49d3f0e704\n(cherry picked from commit 883027afb0656a46369d3f4965a0c65b4a3e8dd8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc9a0472788979d6a25314bf4877064204d80af6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc9a0472788979d6a25314bf4877064204d80af6"}]},"branch":"refs/heads/stable/wallaby"},"8a859c2363520f40b4f7a93eccf4977c8ebdc15d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-02-15 09:09:29.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/74/829174/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/829174/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/829174/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/829174/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/829174/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/829174/2"}}},"commit":{"parents":[{"commit":"52973754493db472a0b5f058ae76c8af35c2e14c","subject":"Set requirements branch for IPA build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52973754493db472a0b5f058ae76c8af35c2e14c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:10:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-15 09:09:29.000000000","tz":0},"subject":"CI: Enable bare metal testing for Ubuntu","message":"CI: Enable bare metal testing for Ubuntu\n\nPreviously we were seeing issues with ipmitool and virtualbmc on Ubuntu:\n\n  Error setting Chassis Boot Parameter 5\\nError setting Chassis Boot Parameter 0\n\nThe dependent change fixes these issues, and this change enables bare\nmetal testing in Ubuntu CI.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/829173\n\nChange-Id: I96827fc32c1594ca9a0535e259929c49d3f0e704\n(cherry picked from commit 883027afb0656a46369d3f4965a0c65b4a3e8dd8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8a859c2363520f40b4f7a93eccf4977c8ebdc15d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8a859c2363520f40b4f7a93eccf4977c8ebdc15d"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
