)]}'
{"id":"openstack%2Fkayobe~829580","triplet_id":"openstack%2Fkayobe~master~I6f603c9304d836012244bb92ec88695c443d7e9b","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-04-06 16:09:21.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I6f603c9304d836012244bb92ec88695c443d7e9b","subject":"kolla-target-venv: remove permissions workaround","status":"ABANDONED","created":"2022-02-16 17:08:33.000000000","updated":"2022-05-30 11:13:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"96b5b09e9890905226e9431782777f1f723d4f9f","_number":829580,"virtual_id_number":829580,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-02-22 10:41:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-16 18:44:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"40f9bc52f3b7ec568420a3c39bcfc8cd978c5aa2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-16 17:08:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d32f0671f654d9e833e19f793b674b4a16849f4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 18:44:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1133b4fff954d5bb7a5f5d4b56087e7 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5b388c4886f498188dd333076562ecf : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/258d017d30354377bf71a1ae811bf9a2 : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f7ff3d5c39d4570bccacb4e4976ccc5 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/640bdd93cca949768db1e34d3066520d : SUCCESS in 9m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d4301a5fda874f8ca970074676f4464b : SUCCESS in 7m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/978c081d6df64c2197440a7bff5b086d : SUCCESS in 12m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5ed60a2a9d8a49bba9f1b778a5a797f6 : SUCCESS in 37m 22s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/07d99a87604e428388d81367512fa625 : FAILURE in 12m 00s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6d585ed7c06427eaa0f26ac0ad1b53a : FAILURE in 9m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fa675a03e35f42b887e222bdde7394f5 : FAILURE in 13m 18s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e881e8e99682428eb7326680feebd2ee : FAILURE in 16m 01s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43dee99674c243fbb9920ff5515cddf1 : FAILURE in 11m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6b80acba238940ddbf653253118d9f18 : SUCCESS in 1h 33m 31s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a35bab2181a436ab2e5ac965b6bb100 : SUCCESS in 1h 07m 01s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ce8dd3950abb44d9860ec31e909bb99c : FAILURE in 12m 20s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e47513633c448ea82edb700cc019e17 : FAILURE in 9m 55s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e18c007c13b04f4c9ae3e85e6e9d68d0 : SUCCESS in 27m 54s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfc76700d0a943de9e91ffc89791866f : SUCCESS in 20m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3c19425f3eca46ee9dfa03cf0759494a : SUCCESS in 16m 02s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f82d191c5bf34577ae1ab45eda4c9998 : SUCCESS in 12m 58s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/143853cc824643afad7f54ddf3f10775 : SUCCESS in 17m 09s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/caa523031d294e5d8f932ced974def86 : SUCCESS in 15m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a756eddaab8b7529b6ebbed1e7f3e0ad757ac46c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-22 10:09:49.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"bde4487420cacde81f514cecfff63b4a15613c32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 10:41:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39f1ec9ba93b401baf48ed36d9eed350 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f736a447feab4f8c823c8abd65970b91 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7707a826ca64be69d16acd2baad5781 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/510bad90aad647d38599db975102695c : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fa435f3fc5b407090f89a4ce57c80f7 : SUCCESS in 8m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/25db4400d0774c1aad97cb2b7e8473ed : SUCCESS in 5m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6292bdc6eed2445a90cd7f8efab1b6cf : SUCCESS in 12m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/927b7fff34154a8fa796bd98b246fba7 : SUCCESS in 29m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/64592e608bca490cad18602e6a15a989 : SUCCESS in 18m 10s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4cbd1774857d4429b0159ec6f93448e9 : SUCCESS in 10m 27s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/6bf418cb39e348d2a5507e6e7262dfa0 : SUCCESS in 15m 39s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11346de11328497aa39ea88d317cb74c : POST_FAILURE in 11m 26s","accounts_in_message":[],"_revision_number":2},{"id":"afb6f4b4a33619570cb2c25c03edeff32ba0c7e5","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-06 16:09:21.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"96b5b09e9890905226e9431782777f1f723d4f9f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-05-30 11:13:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d2e650f53932d4c3e6ef7997d6e1ec3e44230de4","revisions":{"ea05633ab4ac00d43cdc4efe18537f8026595ceb":{"kind":"REWORK","_number":1,"created":"2022-02-16 17:08:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/80/829580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/829580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/829580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/829580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/829580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/829580/1"}}},"commit":{"parents":[{"commit":"37a9e87041d0bfb2ae43a0e4ae597b3047454e96","subject":"Merge \"Support elements repositories for overcloud host images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/37a9e87041d0bfb2ae43a0e4ae597b3047454e96"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-16 17:05:16.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-16 17:08:27.000000000","tz":0},"subject":"kolla-target-venv: remove permissions workaround","message":"kolla-target-venv: remove permissions workaround\n\nPreviously we used become when creating the virtualenv for\nkolla-ansible, then recursively changed its ownership. This is a little\ninefficient. This change removes the workaround, using become_user to\ncreate the virtualenv as the kolla user.\n\nTrivialFix\n\nChange-Id: I6f603c9304d836012244bb92ec88695c443d7e9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea05633ab4ac00d43cdc4efe18537f8026595ceb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea05633ab4ac00d43cdc4efe18537f8026595ceb"}]},"branch":"refs/heads/master"},"d2e650f53932d4c3e6ef7997d6e1ec3e44230de4":{"kind":"REWORK","_number":2,"created":"2022-02-22 10:09:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/80/829580/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/829580/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/829580/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/829580/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/829580/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/829580/2"}}},"commit":{"parents":[{"commit":"37a9e87041d0bfb2ae43a0e4ae597b3047454e96","subject":"Merge \"Support elements repositories for overcloud host images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/37a9e87041d0bfb2ae43a0e4ae597b3047454e96"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-16 17:05:16.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-22 10:09:49.000000000","tz":0},"subject":"kolla-target-venv: remove permissions workaround","message":"kolla-target-venv: remove permissions workaround\n\nPreviously we used become when creating the virtualenv for\nkolla-ansible, then recursively changed its ownership. This is a little\ninefficient. This change removes the workaround, using become_user to\ncreate the virtualenv as the kolla user.\n\nTrivialFix\n\nChange-Id: I6f603c9304d836012244bb92ec88695c443d7e9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2e650f53932d4c3e6ef7997d6e1ec3e44230de4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2e650f53932d4c3e6ef7997d6e1ec3e44230de4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
