)]}'
{"id":"openstack%2Fkayobe~830061","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I697e83fbc6b1bd044b0e7e42e7c3c6c7468ccfb2","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-02-21 11:49:50.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-21 09:53:55.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-02-21 08:46:21.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"I697e83fbc6b1bd044b0e7e42e7c3c6c7468ccfb2","subject":"Bump up manage-lvm role version to v0.2.6","status":"MERGED","created":"2022-02-21 07:53:47.000000000","updated":"2022-02-21 11:51:26.000000000","submitted":"2022-02-21 11:49:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830061","cherry_pick_of_change":825671,"cherry_pick_of_patch_set":4,"meta_rev_id":"8b477cb9538270fc8d299e03ceaf3fbcb3d534b3","_number":830061,"virtual_id_number":830061,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-21 11:49:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-02-21 09:53:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-21 08:46:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-02-21 09:53:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-21 07:53:47.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-02-21 07:57:14.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-02-21 07:57:14.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-02-21 09:51:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf28e5ee482610816c6761a4c4661ef0f5f95dca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-21 07:53:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"848d571c4dc2d2b3ddfe76c69b60d4a52e0d8483","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-21 07:57:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9059d95c24922784d3e19c97d0ea087049a6bc6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-21 08:46:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"406e25b210d93a5d2415fd4f9cf66c76003d412d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 09:51:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fb796e0f876451a8999a99e941e7793 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f14314bdcbbc4386a875fca222dcb3c2 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1511fc5b13d441208930621c11b6bd9c : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec6201bd06c44d4e9c2d98d60babfaff : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a3415311b19458f8ed4661735ef9ac8 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/776804abd79b40d0b3c9974d5c0febc0 : SUCCESS in 9m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/859a096963ce43569f778f605b1be522 : SUCCESS in 7m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/85c47fccd1ea4981807c8b1520642213 : SUCCESS in 14m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2239eff65088475c89139a57f22934af : SUCCESS in 29m 27s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e2a36c03f6af41c6a6eb01afd2290aa7 : SUCCESS in 1h 03m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0963410889047c0af023dee644e5140 : SUCCESS in 52m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9e2f2e968e864dfcad7db781d7320042 : SUCCESS in 1h 09m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/75ca25623943480dbae33c0ca630bd57 : SUCCESS in 24m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc880318402f41dc991e957dd3e7402a : SUCCESS in 16m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/27885e60b77844799348f66c3168d4dc : SUCCESS in 1h 55m 21s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0de0e1001cbc4e688f37a5b15151a1df : SUCCESS in 57m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e36a7ba72abe4f9fbf990b6e5d2fa160 : SUCCESS in 38m 35s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59496a22a46e4a9888b76e8377c80791 : SUCCESS in 26m 52s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/57da0ab1e045450499410fe0215f0d3f : SUCCESS in 31m 56s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b4104ec68194fefa0e6a10109726d07 : SUCCESS in 25m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4a34191a956b40499b5837a12a737c23 : SUCCESS in 20m 44s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/caaa5980b2394f29a296944a7de44852 : SUCCESS in 14m 04s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/fbb6090c66594ebdbd546635dfc99a03 : SUCCESS in 21m 38s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ec7731fe7bc4db78735436f550e2ce1 : SUCCESS in 15m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a713ed01f4a46c94b9d7de9d282f92252c6baae4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-21 09:53:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"75b1954ea76a86e86e666d4e31040be9e6c8fa32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 09:54:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dbf0159c71f6f6b0f5546f5fd695031f2fe673a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 11:49:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0728c0391d7a44ebbccb1029f915e40c : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fb705fdf95e49539aaa0e05f13afb46 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a8ba444022d48159c342cf5aa4d32bf : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d822ec988c74c8d95c54b158fc44e09 : SUCCESS in 9m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ccd6904462114acab74f5509a8186957 : SUCCESS in 6m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8b7d0c858ab545a1aa1916dc4a653beb : SUCCESS in 13m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f4446f2f07c54e5ab682df930917daf9 : SUCCESS in 32m 17s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b052a4a76624488f8b95eb7ce41e468a : SUCCESS in 1h 07m 19s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f20760d9aaf4033bb2672921fbc8851 : SUCCESS in 54m 06s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b3b25097a2414dd19d5f35cb9ee3dc10 : SUCCESS in 1h 09m 25s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3b284084f4064ecdbd479f1cd83f071b : SUCCESS in 24m 14s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3aa066f200ce433db183f45ffd221d07 : SUCCESS in 12m 51s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b0f73b997e724fa694aa610514616762 : SUCCESS in 1h 51m 28s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7455e9dd2ec847c79edfecb75b36a3d9 : SUCCESS in 1h 14m 10s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/76c2b3b12c71493cad307cbcb5c9283e : SUCCESS in 36m 33s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32510834f49d4952ad68adc219294c0f : SUCCESS in 31m 25s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8e14afea764a4690b4d4d2f8258f75d0 : SUCCESS in 27m 06s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b410c384b3ee44e199e2d26e338f375e : SUCCESS in 27m 05s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1f4f286479bf47858ccebc0b70fd4da8 : SUCCESS in 18m 07s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c43fa867bf81455d99192c466320ceed : SUCCESS in 14m 19s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e8ac2e25f677456eb730a43157202436 : SUCCESS in 18m 24s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6ffcc5c019a4263a681e8944a81d520 : SUCCESS in 16m 30s","accounts_in_message":[],"_revision_number":1},{"id":"0fdb11352f736c45abc77185fb10a2eacc5c3e05","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 11:49:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8b477cb9538270fc8d299e03ceaf3fbcb3d534b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 11:51:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5664ee334c24d3892e1b39ce6983315 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"451d1c3ac6fb31beee4b68e9c3b95575ae808603","revisions":{"451d1c3ac6fb31beee4b68e9c3b95575ae808603":{"kind":"REWORK","_number":1,"created":"2022-02-21 07:53:47.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/61/830061/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/61/830061/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/830061/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/830061/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/830061/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/61/830061/1"}}},"commit":{"parents":[{"commit":"11bb76b637f960c9cb07df2b71b7efcf177cfa7e","subject":"Merge \"Set correct gateway for the bifrost provision network\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/11bb76b637f960c9cb07df2b71b7efcf177cfa7e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-01-20 20:18:41.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-21 07:53:47.000000000","tz":0},"subject":"Bump up manage-lvm role version to v0.2.6","message":"Bump up manage-lvm role version to v0.2.6\n\nThis fixes a bug that caused the role to fail on NVMe only hosts.\n\nChange-Id: I697e83fbc6b1bd044b0e7e42e7c3c6c7468ccfb2\n(cherry picked from commit 0ce066ca3652688cb3486f29390f9451ed9437e7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/451d1c3ac6fb31beee4b68e9c3b95575ae808603"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/451d1c3ac6fb31beee4b68e9c3b95575ae808603"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
