)]}'
{"id":"openstack%2Fkayobe~830067","triplet_id":"openstack%2Fkayobe~stable%2Fxena~Id1b4e9775c834b8b97e086241ee8b247977225a2","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-02-21 14:05:13.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}},"hashtags":[],"change_id":"Id1b4e9775c834b8b97e086241ee8b247977225a2","subject":"CI: test ironic UEFI boot mode in overcloud TLS jobs","status":"MERGED","created":"2022-02-21 14:00:45.000000000","updated":"2023-02-21 16:20:48.000000000","submitted":"2023-02-21 16:18:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830067","cherry_pick_of_change":827667,"cherry_pick_of_patch_set":4,"meta_rev_id":"c2f0f548f75cd6b3a7dc5a749e2206e992e733fd","_number":830067,"virtual_id_number":830067,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-21 16:18:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-02-13 07:51:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-02-21 14:05:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-02-21 14:05:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-21 15:57:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-16 11:37:36.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2023-02-13 07:51:22.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"ffe5ae14e337b0ed8489ee0cfd395419d34fce51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-21 14:00:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5e7d41ac8a0e551cccafd1d65bdb7755e113049f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 15:57:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57e15e845f3d42929f9018f11f0c3745 : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce0b00b232d44e44a36d59ec70ddb615 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56b08c200d134fe1bca301c4b0953195 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44edd97a170a4c969bb7dcb2e73c8b00 : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a93220aabf0741ad85713176d5118049 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/273da101b7ad40b9b68f2177d486e0b7 : SUCCESS in 10m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0abad51bb7de4a0c913bd72eeaf763be : SUCCESS in 7m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/52c6a4dd32814ac39cca55bb8acf5796 : SUCCESS in 13m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/57fab55321d84424b6acf4230397ddcb : SUCCESS in 31m 45s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c66a3c2d85f64a70860b2f416b4eea1d : SUCCESS in 1h 12m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e2943e35ed341999e0b59d3f237538b : SUCCESS in 53m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d83604e861e5460594b4b3be7849d6bf : SUCCESS in 1h 18m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/69b68f2294594026b036f768c6661598 : SUCCESS in 23m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7afab81b3daf4d18b41ca7d3aa99e13e : SUCCESS in 17m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/da9fe983c34145f5be1113039072fea0 : SUCCESS in 1h 48m 58s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39ae9e3b3f9643cca51ce7db42bd4db7 : SUCCESS in 1h 32m 09s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ac53fb39ac464129b0642cd07cdb3db8 : SUCCESS in 36m 40s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/814a798d2eba413d8ca99dd5a0045260 : SUCCESS in 30m 53s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cb22020f589648a49fd79f31b2750083 : SUCCESS in 27m 30s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de2e95947b334221a835cdba89973efe : SUCCESS in 22m 45s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/80e12355d5a94dafb96bd5359a205318 : SUCCESS in 18m 05s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdc8648a44ba431a967fd5027359e574 : SUCCESS in 13m 36s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/b7cd83b9267f4e34b49ae734e14c086a : SUCCESS in 19m 26s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/983afbbabb39479a8f4ff4a161b61888 : SUCCESS in 14m 35s","accounts_in_message":[],"_revision_number":1},{"id":"fe5f9dc1c869d841ac295ee97976b25b0c188d7b","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-21 17:36:57.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"fce7358f415855afad76d80f3ee48fdc4d1db443","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-16 11:37:36.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"f39114a7d34ca36907ee8fdc3833076f6fd9e8dd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-13 07:51:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"735dcadd4e9dce914ab910d2da22edaacdfd87f0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-21 14:05:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"18475800d893c1db885390dc23640e89ae609687","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 14:05:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"96fd3175d30819eb8213fc350b15f40c81382d71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 16:18:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d498d2f4e954186a0549e13fec98fe3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/295872967f204308baf45333fd1077fd : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54d7f244078845dfb2d2860a2220fa78 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47315e3aa68c454fa32c5185f6d2b301 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d3900bae1b440b4b634fd4a1022ccf3 : SUCCESS in 8m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e1d5a5677e704fa5bc87fe8cf89031c6 : SUCCESS in 5m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e2b734d5c2944bafb37afe7508f0eecc : SUCCESS in 13m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8b27c6d6b2ab4b47af2b95dbbb38a2fb : SUCCESS in 20m 53s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8a63fefaf8c04d7c8abcfd5e68697807 : SUCCESS in 1h 14m 47s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/8cac46db22bb48d58bdafae1229f325b : SUCCESS in 1h 09m 37s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0dcf111ae3dc4d0681fbb68ada7d2b88 : SUCCESS in 56m 40s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ed56ddb0c89645d9a82525a9292aac7c : SUCCESS in 1h 20m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7a6dd2e0ab3e40c0b6b4c67afb5bd276 : SUCCESS in 25m 39s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/10edaef140a94b01a5d1605dd3fe03e1 : SUCCESS in 24m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b7ce68a6c6142c3b44ab26b291e2d0e : SUCCESS in 17m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/096461e73bfa4df08d161359bb385f8b : SUCCESS in 1h 58m 04s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69a6d8b658a24c90992ba6620813f44f : SUCCESS in 1h 15m 05s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/707dd13b9fa9466da65cc848cd0a6f27 : SUCCESS in 21m 14s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/c0e739b3694a495f8d1601de98ff42d4 : SUCCESS in 23m 15s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f16ef8d42e6442d858212829bc9a273 : SUCCESS in 12m 33s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e5f9e4538969499ba851bdd2a348fb1e : SUCCESS in 38m 33s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8acb43f4b6b44c4aa4e8ee8357792bad : SUCCESS in 20m 57s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1c07bb9cb87a440b9fadce93a18b6f2f : SUCCESS in 42m 40s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c30b1e0bc7b54fb9b82eff9aea890011 : SUCCESS in 16m 36s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b7b9fa5f28146aa8456b0c07aca7915 : SUCCESS in 13m 20s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/9d5fc5267ea645639b6600d79a138543 : SUCCESS in 19m 37s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/b2e0771fef60457fbe9c4f639fae2ca7 : SUCCESS in 20m 40s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4e169012f234ee3b4ded3b5f94f3af1 : SUCCESS in 13m 51s","accounts_in_message":[],"_revision_number":1},{"id":"0fd047429bc27f177b0e452b8da865765d10694f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 16:18:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c2f0f548f75cd6b3a7dc5a749e2206e992e733fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 16:20:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d647ebe057d84c658ee29088f0bc2f1d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc43fbbb65734ffebc8b2ee30e771011 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f7a4e6ac38564c9c97b71e333e3aa530f8443e35","revisions":{"f7a4e6ac38564c9c97b71e333e3aa530f8443e35":{"kind":"REWORK","_number":1,"created":"2022-02-21 14:00:45.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/67/830067/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/830067/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/830067/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/830067/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/830067/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/830067/1"}}},"commit":{"parents":[{"commit":"451d1c3ac6fb31beee4b68e9c3b95575ae808603","subject":"Bump up manage-lvm role version to v0.2.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/451d1c3ac6fb31beee4b68e9c3b95575ae808603"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-03 13:57:46.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-21 14:00:45.000000000","tz":0},"subject":"CI: test ironic UEFI boot mode in overcloud TLS jobs","message":"CI: test ironic UEFI boot mode in overcloud TLS jobs\n\nSet the Ironic boot mode to legacy BIOS explicitly in Tenks config in\nanticipation of an upcoming change to the default boot mode.\n\nOverride the boot mode to UEFI in the overcloud TLS job to improve\ncoverage. This requires enabling iPXE booting.\n\nDepends-On: https://review.opendev.org/c/openstack/tenks/+/827479/\n\nChange-Id: Id1b4e9775c834b8b97e086241ee8b247977225a2\n(cherry picked from commit b57b94bbad4333020cc6906d124d6f01f75e1e2c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f7a4e6ac38564c9c97b71e333e3aa530f8443e35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f7a4e6ac38564c9c97b71e333e3aa530f8443e35"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
