)]}'
{"id":"openstack%2Fkayobe~830068","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~Id1b4e9775c834b8b97e086241ee8b247977225a2","project":"openstack/kayobe","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-02-22 14:40:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id1b4e9775c834b8b97e086241ee8b247977225a2","subject":"CI: test ironic UEFI boot mode in overcloud TLS jobs","status":"MERGED","created":"2022-02-21 14:00:56.000000000","updated":"2023-02-22 14:43:14.000000000","submitted":"2023-02-22 14:40:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830068","cherry_pick_of_change":827667,"cherry_pick_of_patch_set":4,"meta_rev_id":"ba4d74af1322fc4a38654e6d0edca87f914ca6fb","_number":830068,"virtual_id_number":830068,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-22 14:40:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-02-13 07:51:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-02-21 14:05:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-02-21 14:05:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-21 15:40:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-16 11:37:45.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2023-02-13 07:51:16.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"8e414ec3b7160ffceb61b868d07f58a06835b952","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-21 14:00:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cd100a0acf6a7fc323be475f2b9938d44b9a9dcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 15:40:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45dffd9447154e6fa1e939d821e2c79c : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42db810033304befad84d99d26c3fd97 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aab8ea227db241e6b25b6398ee1712e4 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/372e8981eabb40b6b52ed8efa11285bf : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6cd244e5b87440081942b982d314279 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13ff1fe32f31402dbfbde68f061999e5 : SUCCESS in 8m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0a5688abcb9949d89126a24122fd0dde : SUCCESS in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/901f3a52666a4022800fbdb595dac8fb : SUCCESS in 11m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/12e62e9c490b42d9ba836c62787781f9 : SUCCESS in 17m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/30912337e44f41af8f2f097d1201f1a7 : SUCCESS in 1h 09m 03s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af2ecdcb12a84fa18f5581e6951e2960 : SUCCESS in 52m 36s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9d48097009e94f018506368cd860d824 : SUCCESS in 1h 09m 11s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/93a93874a7684ee5b321a1a57726cb21 : SUCCESS in 20m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9434f503643e4eb7bad9342f6bb7efd8 : SUCCESS in 15m 23s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e84465553c60450fb1e683040070b226 : SUCCESS in 1h 34m 55s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1477d14346264384a1b90dccf548ec12 : SUCCESS in 36m 38s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/466506156fce407b9cc873e77c589468 : SUCCESS in 33m 39s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/00f4e24777884b15b0ac6a1dec01d942 : SUCCESS in 26m 04s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/52d82ab7647d4152bdad052f7ed83f85 : SUCCESS in 16m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f254c1ad31e45b983f0634b0e17ad9b : SUCCESS in 12m 11s","accounts_in_message":[],"_revision_number":1},{"id":"809415cf2ddcdb8b34f372b167f3a4a6e5477bd9","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-21 17:36:52.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"c781d28d323f8d8577fea425a8a1d7803f8c44a3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-01-16 11:37:45.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"8cecc50b9d1d5ce6f931b4678304bbcd25ac4a22","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-13 07:51:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2b1be4590b25c4b028fd6a5c34861e2f102dfeec","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-21 14:05:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"922d4574236a1ce2909f397fece1eb6a378dacfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 14:06:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e5106da0a36bd55aac6ea0b2c1c0c2e3825685a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 16:18:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/63f35d476bf443bcaadab8ebcb4199a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e595b173124a4c2f9f50d41423ccc530 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91b6943a1d364b4e8535e61c81c3c8fc : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c78deb14ee1c452ba5d2deaf488fddb3 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbfed14c04bf40eea304d9b59de48fd2 : SUCCESS in 7m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f82497a118f84e9f9d18ac102a3893b9 : SUCCESS in 4m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cb15aab6a23c423f845be05c378818f2 : SUCCESS in 14m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8be0c769b36f42eba6c60c01516f6e1f : SUCCESS in 19m 32s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/50aaace2b902401da6d71588d202e5d0 : SUCCESS in 1h 09m 55s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f916e7de328e41e094e20be19ef05e85 : SUCCESS in 52m 29s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d7d168a505d043df9d8d8d74078b1072 : SUCCESS in 1h 16m 02s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/922e2557ca7343789b66ce781c55d29a : SUCCESS in 20m 00s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/769f7f0e0a62453c84c30b7af4eaa707 : SUCCESS in 14m 50s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b239d7522b96419a9f91f03b54462bc8 : RETRY_LIMIT in 3m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/dfd82e4b8977467bbdb957377fd1e402 : SUCCESS in 20m 36s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/760bb5a3781441b0aaf873986e117fa7 : SUCCESS in 19m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b08fece1adf04fcc8c3dc73c8371a6c2 : RETRY_LIMIT in 4m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5ff6c7ce0ed04345943b21491d7c60ac : SUCCESS in 14m 58s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8f76d19fb484da7b258d4440ce652af : SUCCESS in 9m 37s","accounts_in_message":[],"_revision_number":1},{"id":"c6b15ea6dc3f3889303015a9119a93b55de4d67a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-02-22 11:49:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f0f47213faf48dbe1a4b82cf45090c72de837bbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 13:10:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e7081e653954cc08d614fca5ba96c9f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2b2bbe127e34a179b5a21e391f9136c : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4843f09669154fc288c90bfb04b36784 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/153e41a0634a4e04864bcd87ab5d71e7 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ba76469ce7d4d2483f88c81d818ccea : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2769ad58318047c7933e2cabcb7d3afd : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d49d4be8e1884cdc9fb44d39ad3a6500 : SUCCESS in 8m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d4da246c3d134fb0b5d690a48690616c : SUCCESS in 4m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/77343771520f43f49dea278b76109767 : SUCCESS in 13m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bbf726e5ad7d43d28dbf066588a5a9c8 : SUCCESS in 18m 59s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1747e9faa78b4cef8f067c4b8b445a62 : SUCCESS in 1h 10m 51s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef64fd6d9ae0416da1056b4f5df6096e : SUCCESS in 54m 58s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/12d6ddaaecb34128b0af23123cd05a4d : SUCCESS in 1h 13m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/236c019db246463fbd8279e528baf584 : SUCCESS in 23m 06s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b484b1c85224bd289978821824f482b : SUCCESS in 14m 19s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3c60eb189c954ac5ae7d685c7f801c3c : SUCCESS in 20m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b57c449bef8a4f61a9c5d124639d5421 : SUCCESS in 14m 51s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/e83f5a1830e5482b8f8823d0e491f00d : SUCCESS in 49m 40s (non-voting)\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/416d885cf0494eaeb8ad51bb1e4b4da7 : SUCCESS in 17m 13s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca49d963611c4fa7b03d63543a49a1ff : SUCCESS in 11m 29s","accounts_in_message":[],"_revision_number":1},{"id":"0807672a41a4460f372178b15dcdfee8c0bd0a84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 13:10:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a4ee17106a938a627a0658842ec046a92be208d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 14:40:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07f51a11633f4d49a9baca550a4f7c0a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13df6bbdf16b4614a787b18afd8138c6 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/251093ee1b3b4faeb0ae4fc92661f6c4 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ef061a92a2849d6963425744ecf40ec : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e94adde040742b0b2343a49d37fb1b8 : SUCCESS in 10m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c63e90a19233436ab1d334ac6bd67ac8 : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c57b5a60117c4e88aacf4c0667e69cd7 : SUCCESS in 10m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1275ec76747e4c2eb78575593822ce80 : SUCCESS in 18m 36s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9482bc931fcd49b8938ed1c1dcb2cdbb : SUCCESS in 1h 24m 39s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06096151e6ae4c479868d53468658036 : SUCCESS in 51m 32s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b1458a7e5cc34bc7912f035caf64d93c : SUCCESS in 1h 17m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c18b8e28295f4b64a4a4c6c38b929df9 : SUCCESS in 23m 23s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bae0f098fc474f7286d88daf9e702559 : SUCCESS in 14m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f8b2cb613a6b4846b364715e884549c4 : SUCCESS in 20m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/456b8aa470cf41ff8cb92adbd9a6bebf : SUCCESS in 15m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/854f29ec985341c3a5568f80604e84eb : SUCCESS in 16m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2a2ac7a54da4b46875e02630a234611 : SUCCESS in 11m 10s","accounts_in_message":[],"_revision_number":1},{"id":"2a720b4fce55fc7c3e1e244ce6534ebf68f5a77c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 14:40:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ba4d74af1322fc4a38654e6d0edca87f914ca6fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 14:43:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9aab9b42a8f14af19975158ed40155c4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19a46d37d7b445afa7b24a014f28a06d : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dc4c4a6fc8df8b729d30edd511954379c6b3155f","revisions":{"dc4c4a6fc8df8b729d30edd511954379c6b3155f":{"kind":"REWORK","_number":1,"created":"2022-02-21 14:00:56.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/68/830068/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/68/830068/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/830068/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/830068/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/830068/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/68/830068/1"}}},"commit":{"parents":[{"commit":"47628a8dbee79a120baa84301fa157fc56345831","subject":"Merge \"Set correct gateway for the bifrost provision network\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47628a8dbee79a120baa84301fa157fc56345831"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-03 13:57:46.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-21 14:00:56.000000000","tz":0},"subject":"CI: test ironic UEFI boot mode in overcloud TLS jobs","message":"CI: test ironic UEFI boot mode in overcloud TLS jobs\n\nSet the Ironic boot mode to legacy BIOS explicitly in Tenks config in\nanticipation of an upcoming change to the default boot mode.\n\nOverride the boot mode to UEFI in the overcloud TLS job to improve\ncoverage. This requires enabling iPXE booting.\n\nDepends-On: https://review.opendev.org/c/openstack/tenks/+/827479/\n\nChange-Id: Id1b4e9775c834b8b97e086241ee8b247977225a2\n(cherry picked from commit b57b94bbad4333020cc6906d124d6f01f75e1e2c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc4c4a6fc8df8b729d30edd511954379c6b3155f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc4c4a6fc8df8b729d30edd511954379c6b3155f"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
