)]}'
{"id":"openstack%2Fkayobe~830069","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~Id1b4e9775c834b8b97e086241ee8b247977225a2","project":"openstack/kayobe","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-01-09 13:56:22.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id1b4e9775c834b8b97e086241ee8b247977225a2","subject":"CI: test ironic UEFI boot mode in overcloud TLS jobs","status":"ABANDONED","created":"2022-02-21 14:01:08.000000000","updated":"2023-01-09 13:56:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"cherry_pick_of_change":827667,"cherry_pick_of_patch_set":4,"meta_rev_id":"061c6c7b3dd2365db252446a9949027def1fb914","_number":830069,"virtual_id_number":830069,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-03-01 10:58:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2022-02-21 16:30:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d3987ea422c70983e8eed8a8e6338a7891b6b5d5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-21 14:01:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* playbooks/kayobe-overcloud-base/pre.yml","accounts_in_message":[],"_revision_number":1},{"id":"6deda3466d14b913823193d686ae41a1a369d9d6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-21 14:01:48.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"aa63a78b73c39773870a09070fd063e653112424","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 16:30:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/267dfa26d2d84b5ea84b90e4dfd76292 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/674118b78be646d7881d5e520271853a : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b3ad3b5023f4c3f8c03ff9964d0f7d7 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a424fc0a9d094bab8df7a89a9801eedd : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dcf4c4ef8a5497e9ac113d72f0b0f9e : SUCCESS in 7m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b601377ae5454f62840d8b0a7e737ba0 : SUCCESS in 5m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2b9835674fcd43cbb11ade79b452ba01 : SUCCESS in 9m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3902148f100b4c0cb0b9ba521f5b3fcb : FAILURE in 8m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b8953a8cf6fe49d7a436f3a7f7064d0b : SUCCESS in 1h 01m 11s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/65f3ceb3a52941998bf96a835d99f1dc : SUCCESS in 1h 05m 11s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5f1dd8eb70c8434091e2d12ddac6cee9 : FAILURE in 10m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/149d77878e814184999274c4e95d24d9 : SUCCESS in 2h 24m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/62b2ea19729a4a539fdff3051f46082d : SUCCESS in 32m 06s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/47a150be63cd4106aed84f0aaad88f94 : FAILURE in 20m 25s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/597b359dba424e958b0a874082297e4e : SUCCESS in 13m 21s","accounts_in_message":[],"_revision_number":2},{"id":"53f4d09b9ef88327695195eef4ae9fdb2345cf64","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-28 19:21:56.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a2d8e02378e0010d4e434a23eef14c7613b10810","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 21:26:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ceb9ef4a91b34d408d7fd1d4c16ff367 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9040636dada549f9b33b3054a47cfacc : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a2f4e7237834bb0be8efb9018e5c7d3 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6ad08b08d914297865fd3774de2380e : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbd6f10da33640eead784edbc0ca56ad : SUCCESS in 8m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e82bd1bbfbc0480fbf361fda12893c86 : SUCCESS in 6m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b114eb49ecaa499d918c0a3c4cb6dff5 : SUCCESS in 9m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f9ad65340d7d4df39ceb69dba5369541 : SUCCESS in 16m 06s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a86471192a25443cb967953755ef3afd : SUCCESS in 41m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d42b0b6385654908b31cb08f2373ea26 : SUCCESS in 1h 46m 39s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ffd070a79b284910bceeba3e775c78d7 : SUCCESS in 16m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e53cb09402ed48018c7e81c1b55c78f9 : SUCCESS in 2h 02m 07s (non-voting)\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/170c091fa9ff4227973a24369e2c2b3a : SUCCESS in 30m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0cd7f53feffb4daa914e0694e9fc6c3d : FAILURE in 28m 24s (non-voting)\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2e8c757bd6af47ada1b557a6027f148d : SUCCESS in 10m 33s","accounts_in_message":[],"_revision_number":3},{"id":"30020e9810f7c80a2d9e381ae950837eee201904","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-01 09:02:32.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"9372941c3e322e56bf77fe9bf04260426b0635d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 10:58:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51a5ae0cf89647d49becc57f30f723ed : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89d297b93f72427b98fa88edde9638ee : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/456fbe2889d247ccb804b3ba5c63e9b7 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7ae0df8849b495f8023a52ad94d477f : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ba9c0be56994017a4451d2d35013931 : SUCCESS in 8m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/91cafdcdbe3e45e986c9e43fb01447f5 : SUCCESS in 6m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dd84da0f536842558d65e650473d5514 : SUCCESS in 10m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e41b7ef612084224acab686d44a0fc4b : SUCCESS in 17m 20s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/00dbcb5321a845b1a8fda65c4a41ee2d : SUCCESS in 50m 00s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/456a4acafd174a22841f74a5cced1cf4 : SUCCESS in 1h 05m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c3f0dbb9d66045918870fc06b0a9d607 : SUCCESS in 18m 43s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/dd3b4ea6e329429a92988095080a6527 : SUCCESS in 1h 39m 37s (non-voting)\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/67f7ea543ad44cce93add748babe24cb : SUCCESS in 27m 18s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/31a34300043e4f74a858433327cdf532 : FAILURE in 19m 09s (non-voting)\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9f7d46f61f29400e9909921988347be4 : SUCCESS in 11m 15s","accounts_in_message":[],"_revision_number":4},{"id":"061c6c7b3dd2365db252446a9949027def1fb914","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-01-09 13:56:22.000000000","message":"Abandoned\n\nEOL","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4e513776361fc9d125cf0ddcceb59711e7f5a077","revisions":{"3bffe803e82f57c473c675062c27c0e46a448eb3":{"kind":"REWORK","_number":1,"created":"2022-02-21 14:01:08.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/69/830069/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/830069/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/830069/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/830069/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/830069/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/830069/1"}}},"commit":{"parents":[{"commit":"adce4a0417ff97ea0abbf361521a020bb2d4f9a2","subject":"Merge \"Generate kolla config when running database commands\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adce4a0417ff97ea0abbf361521a020bb2d4f9a2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-03 13:57:46.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-21 14:01:08.000000000","tz":0},"subject":"CI: test ironic UEFI boot mode in overcloud TLS jobs","message":"CI: test ironic UEFI boot mode in overcloud TLS jobs\n\nSet the Ironic boot mode to legacy BIOS explicitly in Tenks config in\nanticipation of an upcoming change to the default boot mode.\n\nOverride the boot mode to UEFI in the overcloud TLS job to improve\ncoverage. This requires enabling iPXE booting.\n\nDepends-On: https://review.opendev.org/c/openstack/tenks/+/827479/\n\nChange-Id: Id1b4e9775c834b8b97e086241ee8b247977225a2\n(cherry picked from commit b57b94bbad4333020cc6906d124d6f01f75e1e2c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3bffe803e82f57c473c675062c27c0e46a448eb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3bffe803e82f57c473c675062c27c0e46a448eb3"}]},"branch":"refs/heads/stable/victoria"},"f3a791e01e1b8f05a1dd4a46b15c28d720eabb22":{"kind":"REWORK","_number":2,"created":"2022-02-21 14:01:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/69/830069/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/830069/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/830069/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/830069/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/830069/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/830069/2"}}},"commit":{"parents":[{"commit":"adce4a0417ff97ea0abbf361521a020bb2d4f9a2","subject":"Merge \"Generate kolla config when running database commands\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adce4a0417ff97ea0abbf361521a020bb2d4f9a2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-03 13:57:46.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-21 14:01:47.000000000","tz":0},"subject":"CI: test ironic UEFI boot mode in overcloud TLS jobs","message":"CI: test ironic UEFI boot mode in overcloud TLS jobs\n\nSet the Ironic boot mode to legacy BIOS explicitly in Tenks config in\nanticipation of an upcoming change to the default boot mode.\n\nOverride the boot mode to UEFI in the overcloud TLS job to improve\ncoverage. This requires enabling iPXE booting.\n\nDepends-On: https://review.opendev.org/c/openstack/tenks/+/827479/\n\nChange-Id: Id1b4e9775c834b8b97e086241ee8b247977225a2\n(cherry picked from commit b57b94bbad4333020cc6906d124d6f01f75e1e2c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f3a791e01e1b8f05a1dd4a46b15c28d720eabb22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f3a791e01e1b8f05a1dd4a46b15c28d720eabb22"}]},"branch":"refs/heads/stable/victoria"},"1c60fbf0ad95929c300294a950bd43e70fcc688c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-02-28 19:21:56.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/69/830069/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/830069/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/830069/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/830069/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/830069/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/830069/3"}}},"commit":{"parents":[{"commit":"a3c170c08634ecfb6d544d0d42840a3d387f5eff","subject":"Revert \"CI: temporary disable voting of some jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a3c170c08634ecfb6d544d0d42840a3d387f5eff"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-03 13:57:46.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-28 19:21:56.000000000","tz":0},"subject":"CI: test ironic UEFI boot mode in overcloud TLS jobs","message":"CI: test ironic UEFI boot mode in overcloud TLS jobs\n\nSet the Ironic boot mode to legacy BIOS explicitly in Tenks config in\nanticipation of an upcoming change to the default boot mode.\n\nOverride the boot mode to UEFI in the overcloud TLS job to improve\ncoverage. This requires enabling iPXE booting.\n\nDepends-On: https://review.opendev.org/c/openstack/tenks/+/827479/\n\nChange-Id: Id1b4e9775c834b8b97e086241ee8b247977225a2\n(cherry picked from commit b57b94bbad4333020cc6906d124d6f01f75e1e2c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1c60fbf0ad95929c300294a950bd43e70fcc688c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1c60fbf0ad95929c300294a950bd43e70fcc688c"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"},"4e513776361fc9d125cf0ddcceb59711e7f5a077":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-03-01 09:02:32.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/69/830069/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/830069/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/830069/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/830069/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/830069/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/830069/4"}}},"commit":{"parents":[{"commit":"1264ac29cb2afc4a2325174cec75d4b7b6da7e36","subject":"Revert \"CI: temporary disable voting of some jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1264ac29cb2afc4a2325174cec75d4b7b6da7e36"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-03 13:57:46.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-01 09:02:32.000000000","tz":0},"subject":"CI: test ironic UEFI boot mode in overcloud TLS jobs","message":"CI: test ironic UEFI boot mode in overcloud TLS jobs\n\nSet the Ironic boot mode to legacy BIOS explicitly in Tenks config in\nanticipation of an upcoming change to the default boot mode.\n\nOverride the boot mode to UEFI in the overcloud TLS job to improve\ncoverage. This requires enabling iPXE booting.\n\nDepends-On: https://review.opendev.org/c/openstack/tenks/+/827479/\n\nChange-Id: Id1b4e9775c834b8b97e086241ee8b247977225a2\n(cherry picked from commit b57b94bbad4333020cc6906d124d6f01f75e1e2c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e513776361fc9d125cf0ddcceb59711e7f5a077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e513776361fc9d125cf0ddcceb59711e7f5a077"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
