)]}'
{"id":"openstack%2Fkayobe~830176","triplet_id":"openstack%2Fkayobe~master~Ib46834fd226c3f59146bbaada53c8d079481d315","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-03 13:16:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib46834fd226c3f59146bbaada53c8d079481d315","subject":"CI: stop setting libvirt_host_require_vt in overrides","status":"MERGED","created":"2022-02-21 14:14:38.000000000","updated":"2022-03-03 13:19:08.000000000","submitted":"2022-03-03 13:16:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830176","meta_rev_id":"b5ac98992b9bf3aa3cbbea1afda761bfa6d47837","_number":830176,"virtual_id_number":830176,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-03 13:16:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-03-03 10:56:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-21 14:36:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-03-03 10:56:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-21 14:36:15.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-02-21 16:13:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-03 10:56:59.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"e200d655c23dab532ad3fd30a2f105afcfaa679c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-21 14:14:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68ba89d11c16529a7369f049eeb526a3836cb136","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-21 14:36:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"91e7f15ce4979af8c01b791c1052ab2f4e4b49d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 16:13:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e698bbbe0744303b0bd3a7ddfcc263f : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e51cd81dcae34bffbb456f44d3e01126 : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/307124a5ccad43ff85f59b6c26859277 : SUCCESS in 7m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8dac0508418f40539378d7c46114bbfe : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ef8b59ed19942b7afae7db25c9a9ce4 : SUCCESS in 10m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/57ed17e1f1404488b533c72b64fd7529 : SUCCESS in 7m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/be1eb6c1ccda466ba6a29fc10f310422 : SUCCESS in 15m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/77e84956e4474ca4930685d9a98dd5de : SUCCESS in 35m 08s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/16584eac3a874176a80efc0fef3bee45 : SUCCESS in 1h 09m 33s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1ea42ae416e42369db7568eff62ee60 : SUCCESS in 58m 17s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9f9fd7a6a1be4a28b7a738d9a4b1aa5a : FAILURE in 1h 13m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/178c46d3279d4c098b27e82c45df8ecb : SUCCESS in 26m 13s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc91f55813ac4a4c9ff79260ebbfe12d : SUCCESS in 16m 03s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/60e59253b13e437eb2c7358df01973cd : SUCCESS in 1h 52m 34s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d5805cb7f65406cb49294a8a04d4afe : SUCCESS in 1h 32m 30s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9aaac2201e194e5fae199736d468f214 : SUCCESS in 51m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/135756fe7aed472fb1a412361a558eb9 : SUCCESS in 37m 26s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/929ff6ff0a094a8b86be75b7cefcabc4 : SUCCESS in 29m 14s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff6896d5035c4edfa3eba58ede55015b : SUCCESS in 22m 05s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/fb3c53a909304a559673940d2c56a833 : SUCCESS in 18m 26s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4a5a0b36a7d46b9bbd08fb49de49760 : SUCCESS in 13m 07s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/cb7663594e104d02901696398c1a05d6 : SUCCESS in 22m 08s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a156c3c2768541e89a80d72a037ed08e : SUCCESS in 20m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f1246fd3aece3356b1b8cc9a8f45596217911c62","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-21 16:55:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"099f634ff34d47d4c4a9729f6e3a50805b9055fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 18:45:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3885df447dba4d31aea4887553097a69 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2b7f8eb55774301ab87989a8f3cc969 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e04f38cfd884170a2b103d32c21ebb7 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d63625b348b4e6b9e201a6bd7b5f98f : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd4ce2a6f0034e35847d4cc782876bb9 : SUCCESS in 10m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9b8ee3b454294d9e8699f5ac014ab94b : SUCCESS in 8m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b49aed80a9424db09950afee1276d9b8 : SUCCESS in 13m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ee260a972bbc473ba2938b816ce81cc3 : SUCCESS in 35m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d2be688ba78149cd8070001ad8a71418 : SUCCESS in 59m 07s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a19ce536eaaf4bf3b166df47d8c8e262 : SUCCESS in 59m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/14ffc984447a479d87cec6b1523dd1fc : FAILURE in 1h 32m 30s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/93d105fc1c1a44febb925d6cb72953bc : SUCCESS in 24m 18s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbb6d2ec888245b6ba60b27993171e40 : SUCCESS in 14m 53s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f00b2c42e788477ca49472d900614721 : SUCCESS in 1h 41m 54s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/054410567ef34723bedbc2242ee04166 : SUCCESS in 1h 02m 24s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0ce5c746fe9f4445ba1b0f9e28b1c29e : SUCCESS in 51m 57s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3695067538104d039ad786a0f8017ef1 : SUCCESS in 38m 41s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e0ae2b6d15264aeeaf7f50418afd40a8 : SUCCESS in 25m 46s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42a31a718d614ee8ae22873da8f86052 : SUCCESS in 22m 52s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0b89d89deae14bef86a90f95e54b562b : SUCCESS in 19m 46s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c7c969ba7bb42a68c54856a8f5a65d4 : SUCCESS in 12m 07s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/6ba8d4898bda415e9fe7d971aaebfa25 : SUCCESS in 19m 56s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06e2ce4038f0437db94b721922c903eb : SUCCESS in 14m 43s","accounts_in_message":[],"_revision_number":1},{"id":"95b0c6bf2fe3dbf3d963ff82fbf4bbf3dedda312","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-21 22:02:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8eb8d32066a788c0bb8d1aef9a6aa13f97389743","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 23:59:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44039d1926964855b2ad490ec4a19a89 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a456f7d5278454281303c43ec33bfbc : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0879af0daca841b2be3e3c38bfa63a95 : SUCCESS in 7m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd9363f44bb045db99e57a9abfafaa20 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c94d0b878bb548678feca0a9d5e1614f : SUCCESS in 9m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/af01fab2583149ac8dd27d942c10603a : SUCCESS in 7m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a6207584967e42829794822e3f870886 : SUCCESS in 14m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0ff423b14d754c9195cb5f48b3bd82aa : SUCCESS in 37m 21s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3f1dfac140b94b7695069eb1e0d6c343 : FAILURE in 14m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebbcaa05cc7d4bafbbd63476c9195754 : FAILURE in 12m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ed006a3d3071402f84ae820b1efa2ccb : FAILURE in 15m 02s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/474889e2c8334a21a8fc0a97776a0a34 : FAILURE in 18m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af42d53e767c4804836ec84068627f69 : FAILURE in 13m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c299ad2f6cbd4f60bb002c3cbfe2d30e : SUCCESS in 1h 54m 17s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b037cd474f84f588bf910646d47661c : SUCCESS in 1h 19m 24s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c88558123d084218a84dc08088072f67 : FAILURE in 15m 18s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a39f80b0f6a45458733beb67109d5e1 : FAILURE in 12m 48s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/490e6367b54447b4a7a065df98b1cb8d : SUCCESS in 33m 25s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f96d0d8027f34fa48cfd3a9e0ff708f7 : SUCCESS in 28m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f65e92b6ff6e4fe1a9c072caf1377a2f : SUCCESS in 18m 13s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd037d4293f64c9fbb06309702c69a79 : SUCCESS in 15m 21s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/7fc03b955d2e464db86637c93b6b3188 : SUCCESS in 19m 49s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3eff4c2d9cd4cc98a49aaa3df71f1ed : SUCCESS in 20m 34s","accounts_in_message":[],"_revision_number":1},{"id":"742e8772cca6c6a31f06ed745ab941ed85f426fe","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-23 08:30:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b4dc17fd61b89898de7a3562f32ab8a282cba7a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 10:09:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9d6a3d2dbc746f596ecc4a4599334d9 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b92291da72eb41948f11f6a1098c4243 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/723fa9dbf52c43bf9628c14f6c394ecf : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efbd9879319c45289be28a2938d3034e : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c22c57234821498986526b7024065942 : SUCCESS in 9m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e787e2e283fc49a4834bc8ffd079468d : SUCCESS in 6m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/90f8b1c638fb4a6c978fc67780c45abb : SUCCESS in 13m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d9450b9456524b24914007d1e81da8c0 : SUCCESS in 34m 36s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/276e361480c5469a8c5ea6bede7536ba : SUCCESS in 1h 09m 29s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b96dde1d985f45d5b3eea8a528872175 : SUCCESS in 56m 19s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d5da44cee5bb47d2ab67e5f3d3a35640 : SUCCESS in 1h 08m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0b458dd311f34975a5bedd12688b7915 : FAILURE in 17m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65a9d09877674375990883958189a53e : SUCCESS in 17m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5d0b6d307adb4572aebe056d2560d40c : SUCCESS in 1h 32m 59s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7063e3bc98104cbaaad72c15f9511957 : SUCCESS in 1h 07m 20s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b2fad7dc8f864c57b3c52dea4249c6c0 : SUCCESS in 47m 53s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7350a45203c44591b9faefda3170e07a : SUCCESS in 37m 25s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7443519473c34fe9a2fc4cd4453639f2 : SUCCESS in 33m 48s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3d0b39710ac468aaec751ec5f4236e7 : SUCCESS in 23m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4edf924e88b24e709f39a1139bc1adfb : SUCCESS in 16m 35s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cbaaa330b5043e99cbca685d0e0a189 : SUCCESS in 12m 24s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/4fc6b99c066a48f4bb37a19885e8e302 : SUCCESS in 20m 14s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d808d6f87cc54e058cb4f894172a431e : SUCCESS in 13m 55s","accounts_in_message":[],"_revision_number":1},{"id":"c8a563fdb068216cbb5b6c285cd45dc10decf1fd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-23 10:24:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f4bcda81ec719ceefaffdc2bb261a8187b3702df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 12:15:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/616d99ce95c84daca3c72b61f258cb4a : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72f716c0aa3c4b37a432c99ae0a2afb7 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7900601b0a54bb89c3b5496db09ceec : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97125ea4a4cb4445b5402ef9ed9ade3c : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de912392033645ccbb8dee657296c6d3 : SUCCESS in 8m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/29aa5b5238f34797828c045caaa2354f : SUCCESS in 5m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/87767737bf9f45bead77c9e70c3decd4 : SUCCESS in 11m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d5a48a0b67714927b7ffec8b8b893166 : SUCCESS in 27m 42s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9ec32d80d9e148518b812e87e5298ea5 : SUCCESS in 54m 43s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1911044f7a904a6abdf45d1f7a55ece3 : SUCCESS in 56m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1ad94084bd9647c4b2a039eb951ed930 : SUCCESS in 59m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6eb464bc493d4adab87fbc353af460e2 : SUCCESS in 24m 36s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d5d1ebd726848048e5b09b676390918 : SUCCESS in 12m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/af5df87073d54cacb9a6b1a76462582e : SUCCESS in 1h 47m 02s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1220f51db87f4471bdaa76446c70709d : SUCCESS in 1h 00m 14s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ded96f736bd0482387b67945c3529c83 : SUCCESS in 50m 48s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78f0ecaa1e064b24be4ee1636eb2d765 : SUCCESS in 35m 44s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7e0076b98b214b8990c1650ed1fcd5d7 : SUCCESS in 26m 04s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2440cca77a50476b92dad218b3a947ab : SUCCESS in 20m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f6614978336f4ecc8136a0b2e05a41fe : SUCCESS in 14m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94276c3724014a65800c5a0b12982efa : SUCCESS in 10m 48s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/1cd7b018a4034fc98ad3d61e0467f0c0 : SUCCESS in 16m 26s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b83905093004bc18f80c534ad17a204 : SUCCESS in 36m 15s","accounts_in_message":[],"_revision_number":1},{"id":"1fa74ef8bb7b803d863fc11d6fad96321abe66f6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-03 10:56:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0c821268d176faf1ffef00d7bffa62ddf1419a5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 11:06:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"97f68ee2e60bff11759ae5fcd0e2009a8cc5b3d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 13:16:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c62cb5e0aff24f75a81671f54f0db724 : SUCCESS in 9m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c18fd2ecd2040c6b9c2c1e6502ba083 : SUCCESS in 9m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58c2eb53b86d4a39a05a97bc40355084 : SUCCESS in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e47b6546978045f3b39a4f40e8a5a6e3 : SUCCESS in 11m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/31e33aec7ccc455da7ca0a4c63409d12 : SUCCESS in 8m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5a368be3ff4d4141b24647aab5846d70 : SUCCESS in 16m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b87858e129054829a9123f2d37e36ff4 : SUCCESS in 32m 02s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/81129d9af6114cdab8fbdc157f7e49c8 : SUCCESS in 1h 30m 29s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f810eabfe093457086cf174a00903bef : SUCCESS in 56m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e1311315abbc45068b28ce2bafa77e64 : SUCCESS in 1h 13m 40s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/848f85653c6042b6929143bc85a742e7 : SUCCESS in 23m 34s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f2f2eb67f5a49deb985f3f1276b7fbb : SUCCESS in 16m 02s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f5ff20beb3aa46738b3022a032868bf2 : SUCCESS in 1h 48m 26s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49c60660626b49e4a980087ad403b33e : SUCCESS in 1h 25m 10s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3aa92ad30aa74838967d857ca1e522a0 : SUCCESS in 48m 08s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21caba1d258b43799d4825d53858f99b : SUCCESS in 38m 52s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3f4129c906b64ac7a422c4d00d311540 : SUCCESS in 33m 07s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9dd56104e7b42748b069b1241838664 : SUCCESS in 27m 04s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a4d4721b3954443393d1381f49789ce6 : SUCCESS in 17m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f96d85c4e8214f2497d778eeac85e3d0 : SUCCESS in 15m 22s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/8545d0c18b4e40d598bfcbe1bad1dd85 : SUCCESS in 22m 34s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1638d24afd84e66ac080d7aea30c4c9 : SUCCESS in 15m 02s","accounts_in_message":[],"_revision_number":1},{"id":"055a3714c3927cb274855fe5f41fa25cb79bcaf8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 13:16:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b5ac98992b9bf3aa3cbbea1afda761bfa6d47837","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 13:19:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4af7791426ea400ca22cb8886c8420f6 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5cf03e9870e9211ff65da94f23e120991111546b","revisions":{"5cf03e9870e9211ff65da94f23e120991111546b":{"kind":"REWORK","_number":1,"created":"2022-02-21 14:14:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/76/830176/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/76/830176/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/830176/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/830176/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/830176/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/76/830176/1"}}},"commit":{"parents":[{"commit":"e027e5fc10da32d4c379f48674960719214d8473","subject":"Merge \"CI: test ironic UEFI boot mode in overcloud TLS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e027e5fc10da32d4c379f48674960719214d8473"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-21 14:13:44.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-21 14:14:30.000000000","tz":0},"subject":"CI: stop setting libvirt_host_require_vt in overrides","message":"CI: stop setting libvirt_host_require_vt in overrides\n\nIt is now set globally in kayobe-config-dev.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/829225\nChange-Id: Ib46834fd226c3f59146bbaada53c8d079481d315\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5cf03e9870e9211ff65da94f23e120991111546b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5cf03e9870e9211ff65da94f23e120991111546b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
