)]}'
{"id":"openstack%2Fkayobe~830183","triplet_id":"openstack%2Fkayobe~master~I863e18841924e88c8943c1df0c6753fd90c90ef3","project":"openstack/kayobe","branch":"master","topic":"bond-ad-select","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-03-03 13:16:22.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-23 09:25:00.000000000","reason":"Mark Goddard replied on the change"}},"hashtags":[],"change_id":"I863e18841924e88c8943c1df0c6753fd90c90ef3","subject":"Add the bonding 802.3ad aggregation selection option","status":"MERGED","created":"2022-02-21 14:45:25.000000000","updated":"2022-03-09 14:35:19.000000000","submitted":"2022-03-03 13:16:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830183-bond-ad-select","meta_rev_id":"38742c8195ae4fa65c01ae7dd5c4a2cc7d978065","_number":830183,"virtual_id_number":830183,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-03 13:16:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-03-03 10:56:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-02-23 11:43:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-03-03 10:56:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-21 17:14:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 09:59:23.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-02-23 11:43:10.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-03-03 10:56:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"e1e92d59dead549c29e1c8c353ff87ffe6c50c57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-21 14:45:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"111b398a5bb754b427a6916380701fa5c3e6f8af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-21 16:00:48.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"4b0fb788d40e5ff020846e1320f83bfd9566db0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 17:14:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa7c75bf2a9645ffa3331cad7c5021df : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4af12bc454b042a7a9bf79c415e1d244 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/047b9f27fbce4f1dadde928dc0b2cb56 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/473ea7eae9b842a2890abb9e33f54fc3 : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f06ebb89629e42ed92b2d45bda17de81 : SUCCESS in 9m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/181c928156bf4d25a0247e71dc80a3b3 : SUCCESS in 5m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7a4e9efd840f4fe48b65cf49efc46b9e : FAILURE in 8m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d6bef433eb6d477f80000af31e248a5c : FAILURE in 8m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/be378d23ad6646f18cd22130e015f26f : SUCCESS in 37m 20s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c8abfc3d0cfd4c2f9f7f844cc385ec35 : FAILURE in 11m 36s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7727ae928e744033bfc733a0b622ac8a : FAILURE in 8m 44s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/21512e2431e04f74ae862fcb2ab1dd5b : FAILURE in 10m 34s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/29b0f05c56d84da5ae29114317eb26e0 : FAILURE in 10m 29s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47626219b23841c88a1065141c7d4705 : FAILURE in 10m 19s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d8d69273e0544c169f514aaffd114bbc : FAILURE in 1h 08m 22s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c07a944692d4a04ac18c2e4ec2cef75 : FAILURE in 56m 45s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5508c9e92e604945817b608e244dc77a : FAILURE in 11m 49s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/069e0590c1b44a9a9394f52de2064977 : SUCCESS in 41m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ecf7ad615c5c4e67863147572ad1ac93 : FAILURE in 23m 45s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7191fb6230e749df9c668c3050324899 : FAILURE in 20m 45s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/fdba7da1a01f4ddeafc7e3f1bcd3dd6a : FAILURE in 10m 52s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7062af74abe94ebaa646721e0271ca04 : SUCCESS in 16m 38s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/8f979fbc7e5f410e87650099b0c5d295 : FAILURE in 10m 15s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9417abe14a714bd786fc35e9dbe7e5da : SUCCESS in 15m 21s","accounts_in_message":[],"_revision_number":2},{"id":"694e1a2118a313a5c93ead15743cf3b8328ae27b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-22 09:59:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e644e50e48172ef0224a20826625dbd8c2920bc3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-22 14:30:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6335855d49fb05cb6a05a3ac45907cce47d1d88b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-22 14:43:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b7095d84fa3256e330070fb4a5deaadc43d7c845","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 14:59:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/515285be68824996af3193b2fa9ce726 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17a8071733fd45f2af2006100afdb902 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71b63999358f4784a1dec4842922d455 : SUCCESS in 9m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1d55876fdfd4699b82e362690f5cbb7 : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eae4e1e1a9c645a89744f35101f9ea14 : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb0c28818e164b20b592f0fa3cd27703 : SUCCESS in 4m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bb9949d5002840699ca5b0171c266c14 : FAILURE in 9m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1fe36a2000fb494c963156a821b6361e : FAILURE in 7m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f6cf05b05d0e4a68a106def928094e2f : FAILURE in 9m 47s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e41c29f3c2fa45fbac4b1fad411215ff : FAILURE in 9m 42s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/838b3bbedcd84954a027f8338496dc52 : FAILURE in 7m 22s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/5eedd6b40ff84c0d8c925c0189fbee11 : FAILURE in 12m 11s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3094b97af2634c7cacc8d123e83b529a : FAILURE in 9m 32s","accounts_in_message":[],"_revision_number":3},{"id":"27d1f212c68a3d393731e3339aef1dbad0c6e588","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-22 15:45:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"92cff2adab2eb675c6e45fd9d56a15438c8a4895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 16:24:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77808eb0f76548fe88eb0b0c367b9528 : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7f9d68b4e35491d868222a841361c49 : SUCCESS in 7m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48e506d4c8ba4de5bb1af4cf378da431 : SUCCESS in 7m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96b9d2d4f0014d9eab5e82098cdbbd86 : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18877e8e6f3d49a8aff35630c88e6250 : SUCCESS in 11m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ec651ab3eb5482e98f149d56806be04 : SUCCESS in 4m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c16a1a6fd0fe43918e160f64cf87a529 : SUCCESS in 6m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a5d497b781f94ba2836d5b87688ec605 : SUCCESS in 15m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8d570d54452148239e55f40820554f97 : SUCCESS in 31m 02s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/64ed96029ed44657a563abf79100b86f : SUCCESS in 19m 31s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a73a7f82d70f4e7483dc528a3df415e5 : SUCCESS in 15m 23s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e3ea893e37e14abf9f82477dcb619d9f : SUCCESS in 22m 21s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7020cebcb5534ca49c65b4aa263ea081 : SUCCESS in 15m 58s","accounts_in_message":[],"_revision_number":3},{"id":"45ef937873da62c404e31bf6ce654ad0d4f40a10","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-23 09:25:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3b8f259f27b48256e00a6894d3377732d14bff90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 11:05:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9da52c9fcaf246c1b8cc5bb8994f7aa2 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f98e245c6b634486bd05a922a52b8611 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a12114b7ded426a9e1253cd77565edd : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/139e1eed6bf6477988e686792d925302 : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6645d05b323242bebbe89c52240abcd6 : SUCCESS in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e80c307c91e64b66bdd5aedcd71b2fac : SUCCESS in 4m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/edf85c22eae14ff684cb6642e32279ce : SUCCESS in 6m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7b64dbc0f97a42eb93a8cb80444d1f44 : SUCCESS in 13m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ae0fcfa2d57c4054b58617cfbfa5e23e : SUCCESS in 35m 26s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2ec75b2a9b2448d387d1286b49242450 : SUCCESS in 1h 03m 48s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40fc6f77da3645f38a27dd934e5f0aa4 : SUCCESS in 52m 42s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8af258d139a6443994f81347215fa3d3 : SUCCESS in 1h 13m 38s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/31dbadba02454617a01ea237ebe31e76 : SUCCESS in 19m 51s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7290b54ba8db44c7a6643a1624eade18 : SUCCESS in 12m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/81223b79601c4a3cbab4c89a7f00c32b : SUCCESS in 1h 35m 27s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1db479f53b6b41f79fb978461c53e3dd : SUCCESS in 1h 09m 35s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/94072886886e46278786ef8ba1259ebf : SUCCESS in 51m 20s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c499a68c1c148f0b58ac2ce2283066d : SUCCESS in 36m 45s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0e0386f5cb1648cc9b6cd0b302227adb : SUCCESS in 55m 51s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e241757c45a47c2b2aef62a09230cad : SUCCESS in 24m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9e4ca2e11866488d9a92443af4784e0b : SUCCESS in 17m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/238ebb92f728457e997b801026747489 : SUCCESS in 13m 59s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/51d1316fcb2946ac9457ad95a2af722a : SUCCESS in 21m 01s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29dbfed7f2dc4e788b5e338b8e813239 : SUCCESS in 13m 55s","accounts_in_message":[],"_revision_number":3},{"id":"d3dc43ab9b85cb44cfaf52f1a23ff3565d75ec95","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-23 11:43:10.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0050ae9518ffef5ef94ecf1f0d12b0738a5b77a1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-03 10:56:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9722f9420b5564de811c8462a0c6241b49d9c203","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 11:06:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"350ccbdf08d5c257eafca0e1d266541cbb088e9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 13:16:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a489cacb7df74d5eb8437c7fbdbc3bfe : SUCCESS in 9m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40635bd2029249558c96148c2c91597d : SUCCESS in 9m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a545d30c8574a468a292710cc4dd067 : SUCCESS in 9m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a37241ad74574dae91ca1c9de05fa982 : SUCCESS in 12m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab39943534664d1ea69b9f37f37bd145 : SUCCESS in 7m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9033e38936c64adebea0e5a960bada5b : SUCCESS in 9m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1775bc8555de466a94aefec50ef9163e : SUCCESS in 15m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7bd2ace85d03442e9017da3d79f568b2 : SUCCESS in 32m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9011fba3642e4345a484238830d1dc70 : SUCCESS in 1h 08m 11s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf1536466e214f7d94d4f7c30325c605 : SUCCESS in 59m 14s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d6bc7a00783d40cab59a60bcc476c542 : SUCCESS in 1h 09m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0c4760c0aed54c64bbaec2ab32138d17 : SUCCESS in 23m 53s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f10a0a20815a4e6b8e835a8a4026c106 : SUCCESS in 16m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b74ca1f46dd6453092eda72f5669e5a9 : SUCCESS in 1h 49m 22s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50cd80aadae246f7a5a34d32c3bb2fab : SUCCESS in 1h 17m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/92da3148fb384672ade80b49dbfcbb14 : SUCCESS in 47m 03s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecc1320f47b04580b9ac0875916a9568 : SUCCESS in 39m 02s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/19f6664ed9de43b08695d8ff530d562c : SUCCESS in 30m 53s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f4a38d037454c2b93b139ee603ae837 : SUCCESS in 27m 02s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/49df5f77bc4b4c29ab4cf8174d791da4 : SUCCESS in 17m 30s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01fdc555ae204f0f9c5ed1b6ae499cf9 : SUCCESS in 13m 57s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e1c53eb5727e4c1282a526685da3f85d : SUCCESS in 19m 24s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/673568b17ebb499db5846c2091a364c8 : SUCCESS in 14m 55s","accounts_in_message":[],"_revision_number":3},{"id":"28affd2259728c9cd7977663dc03329721282637","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 13:16:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"38742c8195ae4fa65c01ae7dd5c4a2cc7d978065","tag":"autogenerated:gerrit:revert","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-09 14:35:19.000000000","message":"Created a revert of this change as I78e63704cbdb3f63f042a755d369875bca722c3d","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8e36626e3ed41793c79d95951be6000d9b1e480e","revisions":{"c23c6f73fe128a32d206f0a1867ac3d8f77e3df7":{"kind":"REWORK","_number":1,"created":"2022-02-21 14:45:25.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/83/830183/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/830183/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/830183/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/830183/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/830183/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/830183/1"}}},"commit":{"parents":[{"commit":"abb2edd8a0de089f219b845ddef8a2c24d0fafb3","subject":"Merge \"Remove chrony cleanup from overcloud host configure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abb2edd8a0de089f219b845ddef8a2c24d0fafb3"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-02 13:09:28.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-21 14:44:39.000000000","tz":180},"subject":"Add the bonding 802.3ad aggregation selection option","message":"Add the bonding 802.3ad aggregation selection option\n\nChange-Id: I863e18841924e88c8943c1df0c6753fd90c90ef3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c23c6f73fe128a32d206f0a1867ac3d8f77e3df7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c23c6f73fe128a32d206f0a1867ac3d8f77e3df7"}]},"branch":"refs/heads/master"},"5aceb97721c18329e13fb26414f7cffe3b49624d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-02-21 16:00:48.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/83/830183/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/830183/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/830183/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/830183/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/830183/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/830183/2"}}},"commit":{"parents":[{"commit":"e027e5fc10da32d4c379f48674960719214d8473","subject":"Merge \"CI: test ironic UEFI boot mode in overcloud TLS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e027e5fc10da32d4c379f48674960719214d8473"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-02 13:09:28.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-21 16:00:32.000000000","tz":180},"subject":"Add the bonding 802.3ad aggregation selection option","message":"Add the bonding 802.3ad aggregation selection option\n\nChange-Id: I863e18841924e88c8943c1df0c6753fd90c90ef3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5aceb97721c18329e13fb26414f7cffe3b49624d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5aceb97721c18329e13fb26414f7cffe3b49624d"}]},"branch":"refs/heads/master"},"8e36626e3ed41793c79d95951be6000d9b1e480e":{"kind":"REWORK","_number":3,"created":"2022-02-22 14:43:21.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/83/830183/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/830183/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/830183/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/830183/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/830183/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/830183/3"}}},"commit":{"parents":[{"commit":"3efc551e7ef97a351350ac54791a17aa938fe9aa","subject":"Install Ansible collections for kolla-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3efc551e7ef97a351350ac54791a17aa938fe9aa"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-12-02 13:09:28.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-22 14:42:39.000000000","tz":180},"subject":"Add the bonding 802.3ad aggregation selection option","message":"Add the bonding 802.3ad aggregation selection option\n\nChange-Id: I863e18841924e88c8943c1df0c6753fd90c90ef3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e36626e3ed41793c79d95951be6000d9b1e480e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e36626e3ed41793c79d95951be6000d9b1e480e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
