)]}'
{"id":"openstack%2Fkayobe~830190","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I1a161a4f241cccac3f39ab7a3e11bccac02f9a07","project":"openstack/kayobe","branch":"stable/xena","topic":"chrony-rtcsync-stable/xena","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-02-22 11:37:06.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-22 09:47:39.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-02-22 10:01:51.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"I1a161a4f241cccac3f39ab7a3e11bccac02f9a07","subject":"Enable rtcsync in chrony by default","status":"MERGED","created":"2022-02-21 14:13:46.000000000","updated":"2022-02-22 11:39:17.000000000","submitted":"2022-02-22 11:37:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830190-chrony-rtcsync-stable/xena","cherry_pick_of_change":820657,"cherry_pick_of_patch_set":6,"meta_rev_id":"ac020b854d7166754d7ddde79c890b89be30e874","_number":830190,"virtual_id_number":830190,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-22 11:37:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-02-22 09:47:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-22 10:01:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-22 10:01:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-21 14:13:46.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"CC"},{"updated":"2022-02-21 14:15:22.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-02-21 14:15:22.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-02-21 15:47:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"34e22e8f0a6b4b14c45b581155deed63fe718317","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-21 14:13:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"67588ae1def45d5159751ad1f6fececb3f558763","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 15:47:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcc7e0dfc01448c09527ada19003c344 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ffdf5d8d34c4ca3bd5f7d2b791ac0e7 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ee517a2168f47d781010ff85ea70801 : SUCCESS in 8m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90ea6f60d09744dfbbb87fb6c4f3fcc6 : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0329aa1994142dbaf5f3da107ec8ce2 : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4319a28d06de48108d19c8536998603b : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c30fe20dc94404ab87c7331af3ab897 : SUCCESS in 4m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a406cd6d2b914a068123c6f8eddf8336 : SUCCESS in 7m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/415232755c5f4bf2a13e11500c5bf2c4 : SUCCESS in 14m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/440ff048bf044a87bab581f89e9cb20d : SUCCESS in 33m 08s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b8da73c3d0ae4d4eb70d485764ef5348 : SUCCESS in 55m 35s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb6bfd267bc5459b855f2fe153d1bb2a : SUCCESS in 45m 10s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c2aaf9955dd8403d98e24429486f9e39 : SUCCESS in 58m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5017f764fb1342c183c3baa97838a72d : SUCCESS in 21m 03s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11c5fc7f27cf4b48a9c7196c5b082b27 : SUCCESS in 14m 28s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/af498d03f0a7456fa84e60c7f3ae6a14 : SUCCESS in 1h 30m 56s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d40cf677604e4e4d9f23103be092fdb4 : SUCCESS in 1h 02m 35s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/852ab68415bc408c8d6184d1bf070ccc : SUCCESS in 38m 43s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/735a5d3856eb41bab34524493981783b : SUCCESS in 32m 58s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/edb597fce7d94f278057a7be3340914c : SUCCESS in 32m 23s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36abc0a469c14a1cbeed405b07945b5f : SUCCESS in 24m 07s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5ed39c833cc74cc1a09c7260d8fc7dc4 : SUCCESS in 19m 42s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96038e58d2e44dfdb326c6d099ceda81 : SUCCESS in 14m 41s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/f1af656b78284ec4ba5f548d33732235 : SUCCESS in 19m 56s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a12b841fd9c94b50af5551e726dc638e : SUCCESS in 14m 32s","accounts_in_message":[],"_revision_number":1},{"id":"3d559ceca3d9863291ded5605c716fdeec33e62d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-22 09:47:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1686c4998822085272e2413cfe8525093afb3a4d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-22 10:01:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8245d6e27f32834928d486eb486c56880af35084","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 10:03:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b349da763bf81d2a048e3eaebf7158195e403e41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 11:37:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbae66c058be46d089d316187495cbff : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/716df612e8b0486a935755db310c3370 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eed5078751f8465aac05873705da65b4 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39662e0e758b4579b3d1a618c6ad63cf : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ca19d434c93437db9a39795788db9fc : SUCCESS in 5m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eb3afe1efc9541b581c7fedf2b038a5d : SUCCESS in 5m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/223fedfc3b2f42368f67844a8ec276ce : SUCCESS in 11m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3f190a7128b74e26ab3a984c3a19ea44 : SUCCESS in 32m 47s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c78d5f93c2cc46c988828a7c3bc82374 : SUCCESS in 59m 33s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a85d971639ae4b23a2ff6b6fcef46eb8 : SUCCESS in 54m 04s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b4b327a98e174147965e36544ed7d642 : SUCCESS in 1h 17m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/adbedc8d2f82405992c9a445f9c7fde3 : SUCCESS in 23m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a281d33667a5468badeae2c9d371a5dd : SUCCESS in 15m 43s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/55b82bf708ec487caa85f03a7a3da311 : SUCCESS in 1h 32m 57s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d14133979614b0a9d2aca23e142a19e : SUCCESS in 1h 19m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0cd7ccff5b9a4ee6be3efb8fed021068 : SUCCESS in 37m 59s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6e352321a1b43b88592ebbde186c3ec : SUCCESS in 29m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0da701140dad4985ac1f1763abc7ec7c : SUCCESS in 39m 01s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/837754ceeeb14872a5ec07f8ebcfac80 : SUCCESS in 25m 01s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b7115bdb23ac460c96a12a967b9eeacc : SUCCESS in 17m 12s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eef78d27ed764fd995dd163a3b0d9cbd : SUCCESS in 11m 55s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/12a986b152ad42218e75b999ab5b6d11 : SUCCESS in 17m 19s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abb85dedfd554d35a3fc378a94189f55 : SUCCESS in 10m 36s","accounts_in_message":[],"_revision_number":1},{"id":"558e9f0dec82aaa6fa76e9c5bb874ec956527269","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 11:37:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ac020b854d7166754d7ddde79c890b89be30e874","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 11:39:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02c3103523da4ce0884b2cfd2076deda : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36602022f30642cd87527ef9c5659c3b : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f4a81e48306ce1ae0a611caa8c728b4ccccc4941","revisions":{"f4a81e48306ce1ae0a611caa8c728b4ccccc4941":{"kind":"REWORK","_number":1,"created":"2022-02-21 14:13:46.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/90/830190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/830190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/830190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/830190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/830190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/830190/1"}}},"commit":{"parents":[{"commit":"451d1c3ac6fb31beee4b68e9c3b95575ae808603","subject":"Bump up manage-lvm role version to v0.2.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/451d1c3ac6fb31beee4b68e9c3b95575ae808603"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-12-07 07:26:19.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-21 14:13:46.000000000","tz":0},"subject":"Enable rtcsync in chrony by default","message":"Enable rtcsync in chrony by default\n\nStory: 2007872\nTask: 44139\n\nChange-Id: I1a161a4f241cccac3f39ab7a3e11bccac02f9a07\nCo-Authored-By: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 84284946d75214b5654e0e7bc18eed01319ff8c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4a81e48306ce1ae0a611caa8c728b4ccccc4941"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4a81e48306ce1ae0a611caa8c728b4ccccc4941"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
