)]}'
{"id":"openstack%2Fkayobe~830191","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I1a161a4f241cccac3f39ab7a3e11bccac02f9a07","project":"openstack/kayobe","branch":"stable/wallaby","topic":"chrony-rtcsync","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-02-22 11:50:27.000000000","reason":"Maksim Malchuk replied on the change"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-22 09:47:26.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-02-22 10:01:59.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"I1a161a4f241cccac3f39ab7a3e11bccac02f9a07","subject":"Enable rtcsync in chrony by default","status":"MERGED","created":"2022-02-21 14:23:42.000000000","updated":"2022-02-22 15:23:07.000000000","submitted":"2022-02-22 15:19:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830191-chrony-rtcsync","cherry_pick_of_change":820657,"cherry_pick_of_patch_set":6,"meta_rev_id":"f01fb0d584476f5521cd37fe81d8e0e323a8c0ed","_number":830191,"virtual_id_number":830191,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-22 15:19:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-02-22 09:47:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-22 10:01:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-22 10:01:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-21 14:23:42.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"CC"},{"updated":"2022-02-21 14:31:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-02-21 14:31:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-02-21 16:24:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7d1f33c9d02ec6f9cd832ff8486013253da5c6b1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-21 14:23:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* requirements.yml","accounts_in_message":[],"_revision_number":1},{"id":"539ecaab6c7dfa9339f1256083f6e3f23cf8b9e9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-21 14:29:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1d89867a05d0541728b83eb7cafcfd526f0ac101","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-21 14:31:07.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"0a802456c48f3accf6610f67bef721610993f4cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 16:24:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5eb52182b18848bf91dd6485ba5edf6d : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b9340f8d7fd452ea81d5796c956f388 : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9c00bf2921c42c8afd9c744b3438844 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75ce49f0c1494befae48f23dd088a6c1 : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d6ed66fdd194eaab4085995371d02d7 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2ebfbacab6d424e81faff517f3dbd41 : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39c2fc1f4d05474ca55bab4481ea8f0c : SUCCESS in 5m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c6900a12b9e54d5c9b839daee73cafb5 : SUCCESS in 7m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/489b2c1cead84588a843fb72db913bbe : SUCCESS in 12m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e491cca907c14565a5336fb9fbc5a2d1 : SUCCESS in 21m 51s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/be777360395745ebbb478669cce61395 : SUCCESS in 1h 06m 41s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e84f59586a89404fa1ff9880c34c045b : SUCCESS in 56m 35s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c7a91c9d14504880a6e04bceec0ccb45 : SUCCESS in 1h 08m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a0d03e6aac954635be8404118c58b6a1 : SUCCESS in 19m 09s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/927384003af94359b10ef9ab2a4634ab : SUCCESS in 15m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/712702f769614e64989500fb449b8384 : SUCCESS in 1h 47m 37s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4c1ebcaa73ed41c08ca9a70f6a635e2f : SUCCESS in 32m 44s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d540e241a20452b8969ab7330e2e651 : SUCCESS in 31m 04s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9c8b36fe7c9c4b4a8cce31814c2c86a9 : SUCCESS in 24m 02s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d32cfb004915492d82f110bc58593a6a : SUCCESS in 14m 03s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a642d7572f64beda8e1994b1da2ea10 : SUCCESS in 11m 00s","accounts_in_message":[],"_revision_number":2},{"id":"f6d590453782d0344e959738a840f2b89cbb124b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-22 09:47:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a626f306819a507862bbe6fb19f59752831890bf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-22 10:01:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"90e30c0e352480cffac56834a6f6312797a35903","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 10:03:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"33a4abe177175573db4f77607e6dcadb516a1f71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 11:37:49.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45c748567bb94a1e8745ad4b8dd930e5 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f04a5f08c3d4e6498cf6416b28150ad : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6556c5a12df47c3be3a4832486085f1 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c210d7b5c2d740fb9c73b0fe010fd984 : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db2e8f71723f48379feae7e6d625e18f : SUCCESS in 4m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c62173d9c0924156bc0ed44db3fea3dd : SUCCESS in 6m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/45604e2930344d0d8c241ea3dd0696ea : SUCCESS in 11m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/55f93745e2f0484f8ee9c2a6e6b3c361 : SUCCESS in 17m 51s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3097e78e95774a7894d1d845ea1d4850 : SUCCESS in 1h 09m 25s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcde20d8122f40248c842cf341cc4cce : FAILURE in 41m 13s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1acfb10995114d6185beae569c08c112 : SUCCESS in 1h 03m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ef7f6e336bd24453ad908a39c05bd470 : SUCCESS in 20m 56s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbe82ee70d364ba9b74ca5102cb0553b : SUCCESS in 14m 20s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/909a91a068c34b7daf262b7dbf436492 : SUCCESS in 1h 32m 49s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ec7e9ad2ed5d45f9aadf0909e74c3dfa : SUCCESS in 39m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6073c068ad514bc4a337ea7d1adbbe6e : SUCCESS in 24m 37s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/252b35d25f1f41fbb676c144b10b9c82 : SUCCESS in 21m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/895b448f92cf49ed99396c1620081b20 : SUCCESS in 14m 34s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/034404bfbfc646b79e2fa33b62bed6b9 : SUCCESS in 12m 06s","accounts_in_message":[],"_revision_number":2},{"id":"f2953b96ccc544122f18970e64d56f7f88b03a42","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-22 11:50:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e54b2ec047867b1601102e10b6866eb0224272e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 13:44:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59b11bb368d64b669b3dbc9b7253430f : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd343783ccca4681b09b3b1ad46a7cb5 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c94c28fcd55425a84139aabdcc12144 : SUCCESS in 7m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f27272677cb8422ebd47006371d70cda : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1bcc15985f44228822553faa3516cb8 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64cdf2bd5f2f4e7f9e4a219e90efa686 : SUCCESS in 8m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eaa1fa31a9844db7bc8fbdcc11a5db81 : SUCCESS in 4m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d1a10124586845c4a6bf29766f053f02 : SUCCESS in 7m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/702c6ea3d9eb458f82ca1d76d9a6a73d : SUCCESS in 13m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c5bd36b680064996ac5cd12fc307cb56 : SUCCESS in 18m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c063a5c113404a32847cb6a8128a2bed : SUCCESS in 1h 11m 57s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86d0e680f32c408c903971a75ed0300a : SUCCESS in 1h 01m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0da6842022cd475fae6d7d7a5aa06830 : SUCCESS in 1h 05m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/14704fd4170b4d3b90988cb77259706c : SUCCESS in 20m 25s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f33a069d2f346b4a658fc28fc768a3e : SUCCESS in 14m 24s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7795a64edfab42c1a0b010387024ee12 : SUCCESS in 1h 50m 12s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0a8d4885d11041c49c49f4a21d20c424 : SUCCESS in 39m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c0a4a581d18462ab2191f70eb33dced : SUCCESS in 31m 12s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c36cd674af0e45c2ab86bbe6b1684d26 : SUCCESS in 31m 01s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ab2bc44992dc42c9928656fc3376f6d5 : SUCCESS in 17m 47s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03f395217ac540e9b7f5f2d183d94b7e : SUCCESS in 11m 21s","accounts_in_message":[],"_revision_number":2},{"id":"df9a75092e0e52a0221eda9382b7f0f7de188d8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 13:44:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1fd73a05bd2ea573afa6e533e15cad7cf1e69272","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 15:19:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/420b2a0a966a4ae59fc3f011f6aeb9ea : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d65db7e7cd4d4bac9df22cbfedd697e9 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48b7da09231f4902ad831afe9e1ae6c7 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e01c00a7e6a499ea06d8bf28c2cc0b6 : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3da5cfd733e2413b9b3c09c056388b81 : SUCCESS in 3m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9ec5ecf085e34ebba3f0a5313cbb752d : SUCCESS in 5m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/625f534f7b264b59817ee680d2c90325 : SUCCESS in 10m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/93516bada726432295dfd52050325f0f : SUCCESS in 16m 44s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/71717e7013bb4e1ab65cb042f5babf9e : SUCCESS in 1h 00m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2afa26afadb438681f609cff8a4a8ef : SUCCESS in 1h 06m 11s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6065f4cfb51840a9bd2a1e779a6d3808 : SUCCESS in 1h 02m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b52cf0e575b24fe28b78a5df2196d66d : SUCCESS in 18m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6976187abfa64cb2b1ea36b0f281a9b4 : SUCCESS in 12m 22s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/51aeb4f5224141d2b0d92b64c8796d52 : SUCCESS in 1h 28m 37s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2f234633f01944a29c15f1640d6c9955 : SUCCESS in 36m 23s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b378d25f925149c29adc140a023242c5 : SUCCESS in 35m 14s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/650c237df40b4ca18c4a191c8cb739aa : SUCCESS in 32m 47s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/508e21ce2872473e9e0a06ce9bb80eff : SUCCESS in 15m 15s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f791fdd18b7d4d9bb7100cb38f589efe : SUCCESS in 12m 19s","accounts_in_message":[],"_revision_number":2},{"id":"bc6cac44a3a6927c8d1758ddd355262671e3fd56","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 15:19:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f01fb0d584476f5521cd37fe81d8e0e323a8c0ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 15:23:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8483a97ddeab4e1081482a10e28f6673 : SUCCESS in 2m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f73d22de7fd45c9b5dde6a997928e70 : SUCCESS in 1m 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ae617ae760ab308e8f56e58555bb412553ba05cf","revisions":{"236e158ae6dcf1f36c85290a628a9ed843fb8550":{"kind":"REWORK","_number":1,"created":"2022-02-21 14:23:42.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/91/830191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/830191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/830191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/830191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/830191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/830191/1"}}},"commit":{"parents":[{"commit":"47628a8dbee79a120baa84301fa157fc56345831","subject":"Merge \"Set correct gateway for the bifrost provision network\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47628a8dbee79a120baa84301fa157fc56345831"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-12-07 07:26:19.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-21 14:23:42.000000000","tz":0},"subject":"Enable rtcsync in chrony by default","message":"Enable rtcsync in chrony by default\n\nStory: 2007872\nTask: 44139\n\nChange-Id: I1a161a4f241cccac3f39ab7a3e11bccac02f9a07\nCo-Authored-By: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 84284946d75214b5654e0e7bc18eed01319ff8c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/236e158ae6dcf1f36c85290a628a9ed843fb8550"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/236e158ae6dcf1f36c85290a628a9ed843fb8550"}]},"branch":"refs/heads/stable/wallaby"},"ae617ae760ab308e8f56e58555bb412553ba05cf":{"kind":"REWORK","_number":2,"created":"2022-02-21 14:29:49.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/91/830191/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/830191/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/830191/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/830191/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/830191/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/830191/2"}}},"commit":{"parents":[{"commit":"47628a8dbee79a120baa84301fa157fc56345831","subject":"Merge \"Set correct gateway for the bifrost provision network\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47628a8dbee79a120baa84301fa157fc56345831"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-12-07 07:26:19.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-21 14:29:18.000000000","tz":180},"subject":"Enable rtcsync in chrony by default","message":"Enable rtcsync in chrony by default\n\nStory: 2007872\nTask: 44139\n\nChange-Id: I1a161a4f241cccac3f39ab7a3e11bccac02f9a07\nCo-Authored-By: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 84284946d75214b5654e0e7bc18eed01319ff8c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae617ae760ab308e8f56e58555bb412553ba05cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae617ae760ab308e8f56e58555bb412553ba05cf"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
