)]}'
{"id":"openstack%2Fkayobe~830222","triplet_id":"openstack%2Fkayobe~master~I8af3bd198b06f3fd92d09851c045f018cc7aa70f","project":"openstack/kayobe","branch":"master","topic":"centos-remove-stale-interfaces","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-03-03 15:49:01.000000000","reason":"Maksim Malchuk replied on the change"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-10 10:15:31.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8af3bd198b06f3fd92d09851c045f018cc7aa70f","subject":"Remove stale interfaces","status":"ABANDONED","created":"2022-02-21 15:54:39.000000000","updated":"2022-03-10 10:15:31.000000000","total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1afba1d75e631715ddd9e9da3424dc0332545bc5","_number":830222,"virtual_id_number":830222,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-02-24 13:43:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2022-03-03 15:49:01.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-21 17:09:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 09:51:23.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"066b85bf5d9ebcd39d62c367ea777559ca092d96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-21 15:54:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"396704f98ade781a7252e7ef80cb8c5213e44c5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 17:09:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/056a3ecdde5a4fb2b98c095e6f94050c : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2437a94642134ab2a6edaae4d44dd2a4 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d4bbeb5fe9349ba8ac55edf40b178f0 : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09d1e2d6795b45e081a94a42e034bcbb : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b57dee38b8c48baafbbff05ac62a343 : SUCCESS in 9m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7078f86504846beadfc6fd27d2bf559 : SUCCESS in 3m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6cf328ff30bf48efb88835eab38163ec : FAILURE in 8m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2e079cbfd9d4482a8055314f679f271c : FAILURE in 8m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1b2ee39dc81443628065797d6d7f4699 : FAILURE in 8m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5b1a18fee4fd4147b3e6506ed456d390 : FAILURE in 10m 53s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5c53160018c4c4b8cb69855800ea703 : SUCCESS in 55m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c3e0f57a5f1446198bea09035ce5e6af : FAILURE in 12m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c9dc7e159bff42efacedb09e897621cc : FAILURE in 11m 12s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3afedb1a5ccb426490fa5d992c6842d2 : FAILURE in 11m 22s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/332d2cf2814a42cda77001af56215f5a : FAILURE in 1h 07m 46s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a128bba7f2e142caa3e7de0c957cb18e : FAILURE in 47m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e9b215217ea148b994c9e00e7da3197a : FAILURE in 11m 28s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f53b00c099b94d2684c6d339f6dc358f : FAILURE in 9m 28s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c2023771cc34473289f5ddb4142bb37e : FAILURE in 26m 24s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f855dc9ef804b45a812aef04903a5f5 : FAILURE in 21m 27s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c3c0dca8cce247c4b1c1f1245e1618c3 : FAILURE in 12m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11074877dfa244a8bc5a5bac801172a7 : FAILURE in 8m 41s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e3dfec12affa458180b3c97820652e04 : FAILURE in 10m 46s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0d878c656e24440a9f73a137762551c : FAILURE in 13m 27s","accounts_in_message":[],"_revision_number":1},{"id":"1fd20de333445edfddd2d66c039fada5f41a51f1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-22 09:51:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"917017fa896470ddb208241583f01a8821aec4bb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-22 14:52:57.000000000","message":"Patch Set 1: Workflow-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d633e351c86cbdbd6a5d50425347027840adff45","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-22 15:03:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67cf08e8ec1110b7db66de0e6636f9a63e25ec03","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-22 15:18:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"faa80d4454e1144d4ae4dff083b3fdb6b9c423ef","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-22 15:20:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4228cda5c142e9c080316c8617d4c96c682f35d1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-22 15:43:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1a3aa48f6f5f68ab2935216cd0b93e82589cb4b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-24 11:35:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aa2e71ee7cc6777844db2be1d68e3cc915815744","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-24 11:38:30.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"45e8dc1cb354ba49ef033ae7489af356738d7c8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-24 11:55:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9ac6aac56df57f2414bd964d40318b7027ffdaad","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-24 11:56:19.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"ec234424d8bab31107487b0cec7ccbf76718e944","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-24 12:22:24.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"9dfa116c869f3862080c6656ab78f960714907a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 13:43:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/834ebddbc77241a0abdf154b2b956039 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8e96072420e4bc99c071f427ae3b3fb : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af9fd06ef96346d7aea19c855ace8c38 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f863b1a131f47dfbceb6f7ed958b3c6 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67359593db7e4de9a0508aaef0555164 : SUCCESS in 9m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8ea19336245c453487262a920727976f : SUCCESS in 6m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/07f0974af04345f1babf3d6d1793bf42 : SUCCESS in 14m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1fdeba9a548e451bbefeced8c40f3729 : SUCCESS in 31m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ac7524da881d4d42ad7ab14e2c04fe27 : SUCCESS in 1h 11m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93c953af48b74175b14f22ab399e080b : SUCCESS in 1h 01m 02s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9369c21b0a1b473bae1d27978edab136 : SUCCESS in 1h 16m 17s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e260dbe6c60f446caa0b9be728290b2c : SUCCESS in 24m 31s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce794283ea184cc9ac2bbbd053faa2e9 : SUCCESS in 18m 12s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/60d49d3d81cb48749a575d34ddb78c76 : SUCCESS in 1h 43m 00s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45cf8e2165624a5c89f489a17abfb449 : SUCCESS in 1h 14m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/81730d224d10429bb5d7a4659b8f7738 : SUCCESS in 46m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cf28de96323458fad362689eb2c8801 : SUCCESS in 35m 21s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/af17677f0c74477085644a3f3b529088 : SUCCESS in 33m 51s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed18d844db3e437c89148de23767b769 : SUCCESS in 24m 02s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/366263955811463ca58e52927b7e5a03 : SUCCESS in 18m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33b2cb2b16f74a21a9f651a170d89cc3 : SUCCESS in 13m 41s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/d9996f805c264db1a526b248884c9a3b : SUCCESS in 19m 48s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/514e3b59353c415ab2fba2d63a0efcba : SUCCESS in 14m 36s","accounts_in_message":[],"_revision_number":3},{"id":"84f838ad9b78babbbbe5ec4491fc7202e01a697b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-03 15:25:15.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1ba3fd0ba2b9bda79c82406261b378edd7faa8cf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-03 15:49:01.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1afba1d75e631715ddd9e9da3424dc0332545bc5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-10 10:15:31.000000000","message":"Abandoned\n\ndue to https://github.com/michaelrigart/ansible-role-interfaces/pull/122","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"96744a812ecd091d016e4dadd84946f074a07e77","revisions":{"d315fee9ab28f579b059ffaa65e0bd4a0a103cb9":{"kind":"REWORK","_number":1,"created":"2022-02-21 15:54:39.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/22/830222/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/830222/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/830222/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/830222/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/830222/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/830222/1"}}},"commit":{"parents":[{"commit":"e027e5fc10da32d4c379f48674960719214d8473","subject":"Merge \"CI: test ironic UEFI boot mode in overcloud TLS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e027e5fc10da32d4c379f48674960719214d8473"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-11-08 21:28:58.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-21 15:54:15.000000000","tz":180},"subject":"CentOS: remove stale interfaces","message":"CentOS: remove stale interfaces\n\nSince [1] is merged we can use the \u0027MichaelRigart.interfaces\u0027 role to\ncleanup stale interfaces via \u0027interfaces_workaround_centos_remove\u0027\nvariable.\n\n1. https://github.com/michaelrigart/ansible-role-interfaces/pull/118\n\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nChange-Id: I8af3bd198b06f3fd92d09851c045f018cc7aa70f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d315fee9ab28f579b059ffaa65e0bd4a0a103cb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d315fee9ab28f579b059ffaa65e0bd4a0a103cb9"}]},"branch":"refs/heads/master"},"5c2b8d8d7b0c04aa7e1f14fd0c80de9963932db1":{"kind":"REWORK","_number":2,"created":"2022-02-24 11:35:19.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/22/830222/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/830222/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/830222/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/830222/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/830222/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/830222/2"}}},"commit":{"parents":[{"commit":"01e5834ed1e301d92b47ce3452bd18651cd24e0f","subject":"Merge \"CI: enable libvirt TLS in TLS job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/01e5834ed1e301d92b47ce3452bd18651cd24e0f"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-11-08 21:28:58.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-24 11:34:52.000000000","tz":180},"subject":"Remove stale interfaces","message":"Remove stale interfaces\n\nSince [1] is merged we can use the \u0027MichaelRigart.interfaces\u0027 role to\ncleanup stale interfaces via \u0027interfaces_workaround_centos_remove\u0027\nvariable on CentOS systems.\n\nAlso this change adds the documentation with case when this should be\nused, and describes manual workaround for both CentOS and Ubuntu.\n\n1. https://github.com/michaelrigart/ansible-role-interfaces/pull/118\n\nChange-Id: I8af3bd198b06f3fd92d09851c045f018cc7aa70f\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c2b8d8d7b0c04aa7e1f14fd0c80de9963932db1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c2b8d8d7b0c04aa7e1f14fd0c80de9963932db1"}]},"branch":"refs/heads/master"},"96744a812ecd091d016e4dadd84946f074a07e77":{"kind":"REWORK","_number":3,"created":"2022-02-24 11:55:42.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/22/830222/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/830222/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/830222/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/830222/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/830222/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/830222/3"}}},"commit":{"parents":[{"commit":"01e5834ed1e301d92b47ce3452bd18651cd24e0f","subject":"Merge \"CI: enable libvirt TLS in TLS job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/01e5834ed1e301d92b47ce3452bd18651cd24e0f"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-11-08 21:28:58.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-24 11:55:33.000000000","tz":180},"subject":"Remove stale interfaces","message":"Remove stale interfaces\n\nSince [1] is merged we can use the \u0027MichaelRigart.interfaces\u0027 role to\ncleanup stale interfaces via \u0027interfaces_workaround_centos_remove\u0027\nvariable on CentOS systems.\n\nAlso this change adds the documentation with case when this should be\nused, and describes manual workaround for both CentOS and Ubuntu.\n\n1. https://github.com/michaelrigart/ansible-role-interfaces/pull/118\n\nChange-Id: I8af3bd198b06f3fd92d09851c045f018cc7aa70f\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/96744a812ecd091d016e4dadd84946f074a07e77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/96744a812ecd091d016e4dadd84946f074a07e77"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
