)]}'
{"id":"openstack%2Fkayobe~830269","triplet_id":"openstack%2Fkayobe~master~Id35894be893183b0b84e6dacb81e1b72e4a828f9","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-21 21:59:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id35894be893183b0b84e6dacb81e1b72e4a828f9","subject":"CI: Revert from UEFI to BIOS mode in TLS jobs","status":"MERGED","created":"2022-02-21 17:34:19.000000000","updated":"2022-03-18 14:18:36.000000000","submitted":"2022-02-21 21:59:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830269","meta_rev_id":"5838c4d80a88bfe17a00dbce825189e49a0bf9c6","_number":830269,"virtual_id_number":830269,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-21 21:59:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2022-02-21 17:56:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-21 21:48:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2022-02-21 20:48:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-21 17:56:53.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2022-02-21 18:57:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-21 21:48:44.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"406fcd5d4d361b5ffa92294268c374595dd1e9e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-21 17:34:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5b731afcd5b711224248d88fad8928ce2460592","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2022-02-21 17:56:53.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4fb742bd0983c2159c8d3de25bcde3317ec88962","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 18:57:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b546783b67284fe39f1d1885893247d5 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13f52a89db5244cc883a3f3514fe4025 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82fb29ccd65f4b65a9f5211e2014963d : SUCCESS in 7m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b259d8d5ab1c40a281480bf3e792c148 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6896d0ff5224da4906a5ca242944eed : SUCCESS in 10m 02s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5a19edc5909e474f8f0b815d53457828 : SUCCESS in 1h 08m 50s","accounts_in_message":[],"_revision_number":1},{"id":"a328926a96743547c86c979c30087796f4989914","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-21 20:48:10.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0439a0d38c2d0d7244a9178549a9abc52386a237","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 20:49:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"29a85f44cf427a23a69c195fc54941b652fb71fb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-21 21:48:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e54520d68fa5397f300835628bc209d9117dd2d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 21:59:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce5478d515064e52a648629e1d88a744 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d42a285cc8154d69b82c980ab5c74fa5 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc246f51f1764f458831e4d0ce485eae : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0b5ca6c9a7243afaf63c2a22eda3346 : SUCCESS in 8m 36s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/adef81b9fa3b4a96997d4bf6a1b20625 : SUCCESS in 1h 07m 08s","accounts_in_message":[],"_revision_number":1},{"id":"2fca0f6a4a90725f28aaba12f98a2eb35c22d623","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 21:59:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"434f819190c21fd3441ef330c14c52a741ed8da5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 22:03:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c598dd18e1884d0eb39d87a9d31046fb : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1},{"id":"5838c4d80a88bfe17a00dbce825189e49a0bf9c6","tag":"autogenerated:gerrit:revert","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-18 14:18:36.000000000","message":"Created a revert of this change as I4987480b4ed1bd4bd95a0ae72861f471cc229e20","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2e8e715538b4f7dcfda7d657497818828a66dae","revisions":{"b2e8e715538b4f7dcfda7d657497818828a66dae":{"kind":"REWORK","_number":1,"created":"2022-02-21 17:34:19.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/69/830269/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/830269/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/830269/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/830269/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/830269/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/830269/1"}}},"commit":{"parents":[{"commit":"e027e5fc10da32d4c379f48674960719214d8473","subject":"Merge \"CI: test ironic UEFI boot mode in overcloud TLS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e027e5fc10da32d4c379f48674960719214d8473"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-21 17:33:02.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-21 17:34:08.000000000","tz":0},"subject":"CI: Revert from UEFI to BIOS mode in TLS jobs","message":"CI: Revert from UEFI to BIOS mode in TLS jobs\n\nWe keep hitting the OOM killer in the TLS job, and the switch to UEFI\nmode in bare metal testing seems like the most likely culprit.\n\nThis change keeps in place most of the changes of\nId1b4e9775c834b8b97e086241ee8b247977225a2, but removes the override of\nironic_boot_mode in the TLS job.\n\nChange-Id: Id35894be893183b0b84e6dacb81e1b72e4a828f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2e8e715538b4f7dcfda7d657497818828a66dae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2e8e715538b4f7dcfda7d657497818828a66dae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
