)]}'
{"id":"openstack%2Fkayobe~830350","triplet_id":"openstack%2Fkayobe~master~I361c68904eb66e9728b4a902a0998e4bc06d2dcd","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-02-22 10:46:40.000000000","reason":"Michal Nasiadka replied on the change"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-22 21:02:42.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-02-22 10:47:09.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"I361c68904eb66e9728b4a902a0998e4bc06d2dcd","subject":"Revert \"CI: drop jobs requiring kolla-ansible\"","status":"MERGED","created":"2022-02-22 10:12:03.000000000","updated":"2022-02-23 00:35:46.000000000","submitted":"2022-02-23 00:31:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"revert_of":830326,"submission_id":"830350","meta_rev_id":"b05eca7039bda58158845517212bfafdce4ad1c9","_number":830350,"virtual_id_number":830350,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-23 00:31:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-02-22 10:46:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-22 10:47:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-02-22 12:08:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 10:12:03.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-02-22 10:12:03.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-02-22 10:12:03.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"258cf41d49027f5d049730ab6c87fc38e0b26ce8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-22 10:12:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2cfceb00a593bc791c1c88fdf4b193d23bece09d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-22 10:13:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"def02a4e1ad1f8aa324bad47c50535a78f22a7f5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-02-22 10:46:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"06588d4d085160bd70ffe429b5fc31b8ed8087c3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-22 10:47:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"64874e27909d7f185c335a9f563a846ba134fedc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 11:58:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16ed3d33915a4bf5b004d5c70ad0e1d2 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed9ce43bf5a94cb9a24394869d439e8d : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e029a825eb6340278e8f16f909691efa : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8bba2e6532924f59b9cc4b301d525cdd : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e25523d5e02444082d57af198482c09 : SUCCESS in 9m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d76af919424e4f679408602c84b1f6da : SUCCESS in 1h 04m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/342e0ed25a464601a6bc330f61274acc : SUCCESS in 54m 57s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7ff84666315742829dee8e2007bd8ccd : SUCCESS in 1h 21m 27s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/16e480d7dbd543ba980eba7ba2787c4e : SUCCESS in 22m 33s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/842f3e866fc348339fbb5bc04e5a2450 : SUCCESS in 16m 57s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e4a839d077ae406d82ebd7c6264518d8 : SUCCESS in 1h 43m 54s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/601ad586779b409babfca2d3c593a205 : SUCCESS in 1h 16m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/de63d81e27b24240a2be65b73b6d8459 : SUCCESS in 46m 00s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8521a3d4ae2449b3a98100880137ab0c : SUCCESS in 38m 21s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ccbce2852b0049f7a5c08e12da982409 : SUCCESS in 29m 36s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dda3d4585744248b8cc09b238f5ee4c : SUCCESS in 24m 31s","accounts_in_message":[],"_revision_number":2},{"id":"a66f7bbd9dbbb2605c47a53fad7f58ea94a88f94","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-22 12:08:55.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"208479fbdc602bbdb0a177deedeb597d36dafd83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 12:33:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"37de2263f1743d33dae550309ca4fb2b7434bada","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 14:25:15.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14479dc1a5514880a92c1b4d4e472532 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf997736e5fa470693037db43c1c0fc5 : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d91a68661ac7495ab686c8f84214acc7 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f5ec6e93ac74fdd9767863cf2bdb156 : SUCCESS in 8m 11s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4b6bf9cb76fa466383e4d2f6eb2eb1db : SUCCESS in 1h 06m 01s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47f6b7f43f8f48fd99ef0c8e17030cd0 : SUCCESS in 1h 02m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3875abb36e564525908aeb7f3d37bb4b : SUCCESS in 1h 00m 45s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4ca394a72fe2440b962e11b865d096f3 : SUCCESS in 28m 49s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8835fa38b5894731ac2eee209e610cee : SUCCESS in 12m 21s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/eab3c6a4320042428b71842932a3d509 : SUCCESS in 1h 47m 39s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73efc5d1280740559c3f3745f42e509c : SUCCESS in 1h 00m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7a56b9c5dda54405a768bd76f4eedb69 : FAILURE in 39m 57s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffc5cdfe3c3a4af089954d56a1208cab : SUCCESS in 35m 54s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/12e55d422ac246029d8c124fafeaf027 : SUCCESS in 25m 18s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c95a28afef4445c888c33c815a529006 : SUCCESS in 22m 06s","accounts_in_message":[],"_revision_number":2},{"id":"44affebfdc16655615d27185ca769b5968a3122f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-22 14:33:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4bc49cf816da293938d86b2d7c3b9dfa49b3e956","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-22 14:33:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1150fd19d210c0f8c1ee91ffabf55a2e8586dea6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 16:50:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30f75894c61a455fbc37fe5f79c899f5 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc93e44fad0947be819baf5e5a1f6891 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b76c99b7b814187a072d6691f7a98b9 : SUCCESS in 7m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e1caf2f4eb24a3696b776c8ddf4b17c : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52efc3f27de0434f9a3a18bb60cd0837 : SUCCESS in 8m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f6dfd4f46844418a82fad482081c29d4 : SUCCESS in 1h 06m 53s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a539de0010cf469e8c6bbd00b38ab8e6 : SUCCESS in 56m 05s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/acf01c13b62c41e4a031190a4bffc460 : SUCCESS in 1h 13m 35s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1a99ccb44c83465280f2889a791af8ed : SUCCESS in 24m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c622e3153564a4aa5e06b008035266a : SUCCESS in 16m 02s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0c8b224dfc714ce49426b096e701653a : SUCCESS in 2h 11m 14s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03f14977c3e14532b8e92932d9ae57ee : FAILURE in 37m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d7c4a15f5afb440989d01fa132577066 : SUCCESS in 45m 51s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06e88b5c5a694fa69d94dd0de4b6ae4e : SUCCESS in 45m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/075fada24c774ef48c2ecbad0da3f861 : SUCCESS in 26m 15s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e615c8f0e4594f72bec1bd9aa93f653a : SUCCESS in 23m 06s","accounts_in_message":[],"_revision_number":2},{"id":"4c261713933c22f38a9a674de2829f1e21206261","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-22 16:53:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"176e25213e89e84ffe04271d0597a03aa926ebf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 18:48:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f5020ff52c84e8fb42cba070876e191 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e72b38cd94d44f8bd9795067b65e53c : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0537ba0d680e411f84fed59753e888ef : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f52d5021d424eebb086b799ba278a17 : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/877fed6bf2eb47cf86b3f55a177e8b4b : SUCCESS in 9m 53s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/df22edee5c524ef7bbb78922ac1461df : SUCCESS in 1h 18m 41s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a6ef736be5740cc850d096091cafe61 : FAILURE in 38m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2a1765b35c8f48678e780995709ade2d : SUCCESS in 1h 12m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7192d0ff5b9f473c912a7aa1a37a4dee : SUCCESS in 24m 04s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9f947654313463b9119b9c1eda3cb18 : SUCCESS in 15m 18s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9ab5f1da552f4edead05c53ec0cabbb6 : SUCCESS in 1h 46m 09s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16530ec512af410ea830de9822e69bfc : FAILURE in 50m 53s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/453136d6499145e38c684bb9e9c0e417 : SUCCESS in 52m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/168f9012ce044af7a8f8a0e051ad8366 : SUCCESS in 37m 33s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fff687d40c3049eeb90ba2b073b7141c : SUCCESS in 34m 20s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91f24a863cb8432693f745aed21b8660 : SUCCESS in 23m 19s","accounts_in_message":[],"_revision_number":2},{"id":"73d3719157c2d85675979d1279e4fa1d323fafcc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-22 19:14:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ef8aa7b599b0df6458e568a17a95324547c54999","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 20:48:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4681b97937794e0e95d70bde96e5a744 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da41f3a7d7b94faf92d35b43a67fdee5 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dec9f0ba68a74fb9bed9189f95db514e : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccf87cbe8ab346249bbbe4c043e29bb2 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/734f551377944688b05f0918f05bec04 : SUCCESS in 8m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6e22b429061c45c5838f86a1f74c6ccd : SUCCESS in 56m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffb2644c9b084738a43dc2265ef51d66 : FAILURE in 48m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7a352214d53641cda16b6f8ad2e6450f : SUCCESS in 55m 34s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/135ba7b18a4e4a84bb17ec727166d636 : SUCCESS in 18m 09s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b943e3e66d794d998ccf4f50b458f975 : SUCCESS in 13m 12s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/59cd7a3d0df84702a88a014abcc59c0a : SUCCESS in 1h 31m 21s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a5b5e6bcc7e4ff9b7063cde4d4fc92e : SUCCESS in 55m 53s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e1840372ed4a413a9966495e600b4e58 : SUCCESS in 46m 28s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7f73886764740ffba9e4f6347846991 : SUCCESS in 31m 21s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c1dd4ece40854a58830557fa55160f62 : SUCCESS in 23m 08s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f097a7fd9f0d418f9900c3f853caa62f : SUCCESS in 20m 04s","accounts_in_message":[],"_revision_number":2},{"id":"d0a956e69479a19878be4bc1c98fb600e9421476","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-22 21:02:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e75d07767a3de628bad8415edee30dd6ade7aa87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 23:06:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62f640bbe9a847c5a920bf0d5b0f8435 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdac47c6d8d3413eac47b2d60a5f55f8 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/285690f5906c49cf983c776af5485ffc : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12f48ed332fc43f3b9b086b28d5bb8e0 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfb3947500c1411590a18ec210b53546 : SUCCESS in 8m 32s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8cc365fbcfdd4f3eab1247b927c68d97 : SUCCESS in 1h 04m 25s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab6ed7d86b11425a9672de8eac148faf : SUCCESS in 41m 03s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a4a3cdeef2d74e2295583522e7100ad4 : SUCCESS in 56m 34s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0fbfcbe437b14a2e91cc915bf708209f : SUCCESS in 22m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98897b33e72c43939d76e510823c2593 : SUCCESS in 11m 02s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d9acefcf53024f71b190a56c306da2e5 : SUCCESS in 1h 56m 43s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fbe16be075045908a032c929a7d0d2a : SUCCESS in 1h 14m 55s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d44874273d714672a92611d5adfb6ac7 : SUCCESS in 44m 34s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f700fc173aa435aa74df1ac11d7785d : SUCCESS in 43m 41s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e9e2abd6f3e945b992ae4f893344fb41 : SUCCESS in 28m 32s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb7e390a71bf449cbd88c574ccf23865 : SUCCESS in 22m 10s","accounts_in_message":[],"_revision_number":2},{"id":"350c1d349186503b94d86f3904450a00ac93d71f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 23:06:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"17bd12de51da70b3b82abcb021d47750af955500","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 00:31:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c31f157665fc443abc5c318dc86757ed : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76b762739e4a4029ba5d0ab091829cfe : SUCCESS in 6m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad8142a9d71540afb76798f750a4ff04 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/327afbac0ba1417aa48be477d3840516 : SUCCESS in 7m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ddc1eac5100c425ba952109f103b2017 : SUCCESS in 56m 36s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39dce797b37341af96dd07906b215c12 : SUCCESS in 45m 53s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c548836bdbcc4a4e8bd92c51d3fa964e : SUCCESS in 51m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/16e36ce496f44a79b50085334a3614ec : SUCCESS in 18m 29s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1cc3d2ded344558a23eb2e0b71bb285 : SUCCESS in 11m 18s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/41038d104b0f4fde87b11f7feaeded74 : SUCCESS in 1h 14m 45s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5e0245cadea4894b928e7578606bb3f : SUCCESS in 58m 33s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/68395fd6c2d0475fb2c69b373508456c : SUCCESS in 43m 20s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b54b585bc7e24b0eb0b718d098dc7429 : SUCCESS in 32m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1c142db9deb5444ab0efc0a71c4b5820 : SUCCESS in 21m 00s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/394bd2aed2d64e0e8022c2881b16cc7b : SUCCESS in 19m 30s","accounts_in_message":[],"_revision_number":2},{"id":"665e4e7ffbdf121967de66840126b63df09cfb18","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 00:31:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b05eca7039bda58158845517212bfafdce4ad1c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 00:35:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e63e5ea018b7443eb81798871a78e4b2 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f477473bd700cc4453ba17d487f17dfc2ad6ba48","revisions":{"6e8cb2b3c4a674ecbe26ee423545b13d99f3d33f":{"kind":"REWORK","_number":1,"created":"2022-02-22 10:12:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/50/830350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/50/830350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/830350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/830350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/830350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/50/830350/1"}}},"commit":{"parents":[{"commit":"f00a65ead9f3e7431c7c54c2eb7132c2ff32a089","subject":"CI: drop jobs requiring kolla-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f00a65ead9f3e7431c7c54c2eb7132c2ff32a089"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-22 10:12:03.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-22 10:12:03.000000000","tz":0},"subject":"Revert \"CI: drop jobs requiring kolla-ansible\"","message":"Revert \"CI: drop jobs requiring kolla-ansible\"\n\nThis reverts commit f00a65ead9f3e7431c7c54c2eb7132c2ff32a089.\n\nReason for revert: fix for kolla-ansible installation with\nopenstack.kolla has merged.\n\nChange-Id: I361c68904eb66e9728b4a902a0998e4bc06d2dcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e8cb2b3c4a674ecbe26ee423545b13d99f3d33f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e8cb2b3c4a674ecbe26ee423545b13d99f3d33f"}]},"branch":"refs/heads/master"},"f477473bd700cc4453ba17d487f17dfc2ad6ba48":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-02-22 10:13:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/50/830350/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/50/830350/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/830350/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/830350/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/830350/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/50/830350/2"}}},"commit":{"parents":[{"commit":"3efc551e7ef97a351350ac54791a17aa938fe9aa","subject":"Install Ansible collections for kolla-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3efc551e7ef97a351350ac54791a17aa938fe9aa"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-22 10:12:03.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-22 10:13:04.000000000","tz":0},"subject":"Revert \"CI: drop jobs requiring kolla-ansible\"","message":"Revert \"CI: drop jobs requiring kolla-ansible\"\n\nThis reverts commit f00a65ead9f3e7431c7c54c2eb7132c2ff32a089.\n\nReason for revert: fix for kolla-ansible installation with\nopenstack.kolla has merged.\n\nChange-Id: I361c68904eb66e9728b4a902a0998e4bc06d2dcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f477473bd700cc4453ba17d487f17dfc2ad6ba48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f477473bd700cc4453ba17d487f17dfc2ad6ba48"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
