)]}'
{"id":"openstack%2Fkayobe~830578","triplet_id":"openstack%2Fkayobe~stable%2Fxena~Ia9ade04b1aa9ad102e637d2ddf638b28b2d11305","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-24 09:25:30.000000000","reason":"Mark Goddard replied on the change"}},"hashtags":[],"change_id":"Ia9ade04b1aa9ad102e637d2ddf638b28b2d11305","subject":"Sync enable flag defaults with kolla ansible","status":"MERGED","created":"2022-02-23 13:58:49.000000000","updated":"2022-02-24 11:29:20.000000000","submitted":"2022-02-24 11:27:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830578","cherry_pick_of_change":829114,"cherry_pick_of_patch_set":1,"meta_rev_id":"579f85ed1640c2718757215d46ef9ad59f8c0b17","_number":830578,"virtual_id_number":830578,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-24 11:27:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-02-24 09:25:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-23 16:25:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-02-24 09:25:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-23 16:06:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-23 16:25:18.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"c29c539fd786f199978c2dea0f162b844fea09da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-23 13:58:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2e9c961650e1c084f5f13ac8df8ad1e7f06a04a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 16:06:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2a9637bf7774f8eaf11f8b4dd92f276 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2da420edb5ba42798e710b13de49e808 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd83e73e30494999a0724b15b359246e : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de2a6bd9aecc43d9877e8ce293e9bdeb : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92b8d6d2591d414591886a21fea494fe : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59708f3c5fae4d9bb0be49df2b60aa29 : SUCCESS in 9m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c696e2c0ab7b418aa8acf12e1a5031b5 : SUCCESS in 5m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4185acce89de446ebd9371e0937f0af6 : SUCCESS in 12m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a2ce7c98bf70421eaba4a61e36b6a25e : SUCCESS in 31m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c9b36360d45f4ff988948c21cbafab5e : SUCCESS in 1h 21m 40s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83df7765c1a346f9bfb7c36e91d289a0 : SUCCESS in 47m 03s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c3aab72478ae403a8c38c5bf708d100d : SUCCESS in 1h 07m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9213fb1e0d9d4206bc95fe418d874893 : SUCCESS in 24m 24s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f0e53fed3784367b2046471332b8dee : SUCCESS in 16m 33s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e6c25316c8e249208ae684ba7261118a : SUCCESS in 2h 01m 49s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e505d735a3b34004a5a873c126657f08 : SUCCESS in 1h 26m 49s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9ed1bdefe0854e6aafdf7b8089da0f6b : SUCCESS in 40m 56s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/989f42c6764c42a5826437801d090226 : SUCCESS in 29m 41s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/22752c3d872342dd809887628466c084 : SUCCESS in 29m 10s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6c2a7c0135b4b3a9aded47fb96c2e97 : SUCCESS in 24m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2b0959f436294389b3651f21c96aa3ea : SUCCESS in 11m 56s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9748e114810f4207a756e42139886c64 : SUCCESS in 14m 12s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/261314947e32433381c8a3c888684db2 : SUCCESS in 18m 22s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16378654681d4d05a663794965651df5 : SUCCESS in 11m 48s","accounts_in_message":[],"_revision_number":1},{"id":"27bb4ee1bb57c7c87f6d30e7ae5b6de8e5176c78","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-23 16:25:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3a117d115a2b8628d8acec5780069965a98a1f97","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-24 09:25:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"04642e71e8a384e2804e756f9e3c830c1719e39e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 09:26:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2c9f57e126bbd5b847cd19f49bc3716e6e74639a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 11:27:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db3446ccf55c49f39b0e1b3b9fdf1a31 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/263e17f789084477aace2fa3e729aa70 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/338e5213283c4b719095e90a7eae0000 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b01046e5499c484b80fa80c1ccd0c621 : SUCCESS in 11m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/30e72051690f4f18bf98f10d0e1f0427 : SUCCESS in 5m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4d09adfb19b549aaafa9ef2522741cc9 : SUCCESS in 12m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a87c877d21d14852a9a6db3fdd85a08c : SUCCESS in 29m 54s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ae938fad0b544fec8cec980e992a9443 : SUCCESS in 1h 11m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/292304560d02467cbb2d359ddf67fdd9 : SUCCESS in 54m 57s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c93e668834c943538338edd9d468cf6b : SUCCESS in 1h 13m 44s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/24a84cf4a25f49e1900f4de106cb2f25 : SUCCESS in 23m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72ff0ba26cbf4898847c96e8c748e294 : SUCCESS in 14m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0523a6ff122946deb721aebaa2aaa6fd : SUCCESS in 1h 56m 24s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2659260f6927415fa51a92b59534ce65 : SUCCESS in 1h 08m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/eefca57482d04d2dbed1838d26661eae : SUCCESS in 37m 03s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8aa0054867b14ae799cf3ffda1ea2236 : SUCCESS in 27m 27s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/825a5f0f1a974a53a228b50f61cd034a : SUCCESS in 32m 41s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d630d281cad34beca3cf6c43b76e8388 : SUCCESS in 21m 41s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5ad6ebcb7f144a24ba06aa6b8a841b3c : SUCCESS in 15m 51s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59e8ef505ec04283ad437fe780bdc8fe : SUCCESS in 10m 24s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/359856bf9f304d3aa0395c311bbf1263 : SUCCESS in 19m 55s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2225d26cbfac441fbd7d6243a2f11585 : SUCCESS in 14m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b6ace640573485e48106f9c7f7edfc1cecabb901","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 11:27:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"579f85ed1640c2718757215d46ef9ad59f8c0b17","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 11:29:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ae7822adb3145878a85acbf208198bb : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c69a808a8ed5ea1d2501ec12ff40eed72ac28ce8","revisions":{"c69a808a8ed5ea1d2501ec12ff40eed72ac28ce8":{"kind":"REWORK","_number":1,"created":"2022-02-23 13:58:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/78/830578/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/830578/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/830578/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/830578/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/830578/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/830578/1"}}},"commit":{"parents":[{"commit":"f4a81e48306ce1ae0a611caa8c728b4ccccc4941","subject":"Enable rtcsync in chrony by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4a81e48306ce1ae0a611caa8c728b4ccccc4941"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-14 17:15:13.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-23 13:58:49.000000000","tz":0},"subject":"Sync enable flag defaults with kolla ansible","message":"Sync enable flag defaults with kolla ansible\n\n* Cinder ZFSSA backend was dropped in Xena\n* iscsid is no longer required for Ironic since the Ironic iscsi deploy\n  interface was removed in Xena\n\nChange-Id: Ia9ade04b1aa9ad102e637d2ddf638b28b2d11305\n(cherry picked from commit 52d4458b072e89a9f563946ca866f734ecd6bfab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c69a808a8ed5ea1d2501ec12ff40eed72ac28ce8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c69a808a8ed5ea1d2501ec12ff40eed72ac28ce8"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
