)]}'
{"id":"openstack%2Fkayobe~830873","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~Ifcc3781d771e150b5771f2bf20cd8bcbc5faa443","project":"openstack/kayobe","branch":"stable/victoria","topic":"centos-repos","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-02-28 19:14:53.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-28 14:00:32.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-02-28 13:24:51.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"Ifcc3781d771e150b5771f2bf20cd8bcbc5faa443","subject":"CentOS Stream: update DNF repo file names","status":"MERGED","created":"2022-02-24 17:06:05.000000000","updated":"2022-02-28 19:17:12.000000000","submitted":"2022-02-28 19:14:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830873-centos-repos","meta_rev_id":"d7e266fed53fb5d9d7774b5b49414b45c89d12f8","_number":830873,"virtual_id_number":830873,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-28 19:14:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-02-28 14:00:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-28 13:24:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-02-28 14:00:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-24 17:06:05.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-02-24 19:20:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-24 19:25:02.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-02-24 19:25:02.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"fbf52216a09bd4a9710c72921609ed1f4ba98124","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-24 17:06:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f80fbd45ba84b7ea1988a2cb129e94904c275751","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-24 17:10:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4767cd8c85ac67f0b952b59a8f7c48baa398921e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 19:20:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/514a972a96be437e899fec2fb2dbdc50 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e455f583e234654a7c4008be2c2bc25 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03365c7c660c4045b100a6a7cd08b082 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d6976d65b8d48ed905e358438b7f77e : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d93ab314164945b28f83076517c70647 : SUCCESS in 8m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bbe48cb5bfd24672b871358d029ecebd : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/86f3326c27b0421cb83c38c0834e2a09 : SUCCESS in 9m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d70b3e632ed4415b96ce44abacf9eda7 : FAILURE in 6m 13s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4e3ac5e9ec81409fb5ed8bf3fbd73b0c : SUCCESS in 1h 15m 19s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6b1af99b41fa4f23a2fe2c83c25b6633 : SUCCESS in 1h 08m 42s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8833f55d0b8b420fad6403b5a736f320 : SUCCESS in 16m 38s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cb69e9c91ba8416eb964958b85f30a9c : FAILURE in 2h 01m 05s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/990d4ee21eb44cd2bcb558c1be3a9a18 : SUCCESS in 29m 25s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/38641a3f66bb45a8b8b06af11d14fd52 : FAILURE in 19m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c1daaf6bb1c44c24801eb1df165ab995 : SUCCESS in 12m 58s","accounts_in_message":[],"_revision_number":2},{"id":"21df07be76bea263d1d744ef18bd870c508665cf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-24 19:25:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4cd52e17d2afdd4ddb08ecbceb0a82d478a3345a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-24 22:29:22.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ecbb5c59d77cedff8ae0080889551ba240a771ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 00:41:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30b321f2f7eb4f85a8edc4ff1a9aa926 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82449782f82248feadabeb7aa0995593 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3aedf37133fc4a7a8676c05eac4c67e0 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4b37f1ce92a413cb653cf1d82c13bca : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f0eb9a6eba94983a1547e47715ac736 : SUCCESS in 6m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b2c1dfbcc5134be291a3cd1bbea3ec5e : SUCCESS in 4m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9420993d491c40498ec863dabb34acfb : SUCCESS in 8m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b056dda72f7e4a07adbf6a9cb09cc7f7 : FAILURE in 6m 53s (non-voting)\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c3a7fc3f15e841518050dbd85e9a259c : SUCCESS in 56m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1322e75bdebe4c8ab9bb4b6ad171abd0 : SUCCESS in 1h 03m 16s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6117ac75dfec43c39cb77c9103864ce3 : SUCCESS in 20m 42s (non-voting)\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c707ffb6dda9457c9ebd536a2f58eb82 : SUCCESS in 2h 08m 13s (non-voting)\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ffc9c8f57ae64eaca6ba9790b73400ae : SUCCESS in 33m 19s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cafea4d8ad094f19a598cb66d7f1265b : FAILURE in 19m 44s (non-voting)\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4156a7b7174d4523a40efe5a4afd4981 : SUCCESS in 12m 42s","accounts_in_message":[],"_revision_number":3},{"id":"7edf214dba62506292c3a3e0092e8efccb2b8cce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-28 10:33:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8884b83a12e98f8df9801f9c3ba3b47d97da1343","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-28 10:37:20.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"db7b61fc0e08e921fe63a99e84155d7e45e052dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 12:56:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0e2d346f7b3440cbc5d6c58aacb7d72 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9df3926017ad4200b6fd44a0dc46068f : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bee6884752547e2b0cada8c1c8af7df : SUCCESS in 8m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1aff74111ad84af385116d987c732298 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9252b9dd958040188ac9033f336a89ec : SUCCESS in 9m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/636abd3acf534a60a65ae7f410ebab4f : SUCCESS in 6m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/00f86e72f63d40fb8e2831dea2ca333f : SUCCESS in 11m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/30188f47a17b4833a07e9bb175c2630f : FAILURE in 7m 47s (non-voting)\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0800e93546fc4e7486744216edbcaeaf : SUCCESS in 1h 02m 10s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/172ee26c3cc14bfba479d89a1bcbf012 : SUCCESS in 1h 04m 22s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d0d8a3e8d2df4034b029eb004725dc3f : SUCCESS in 18m 29s (non-voting)\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0f283c48406c40578fcbf0c29537f3a8 : SUCCESS in 2h 17m 01s (non-voting)\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2fd0e46ab5d24e1281d5dcacc4d96a6c : SUCCESS in 32m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0c7ab7800b424b3dbcc5425bf4038c6e : FAILURE in 20m 39s (non-voting)\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/28859fb286404ffb8d1f5cbcbb3861af : SUCCESS in 13m 23s","accounts_in_message":[],"_revision_number":5},{"id":"e8c66450b38e0ad2486f789a62c33d9c0244e2b5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-28 13:24:51.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"b3b1903a8fcfd87f865f83aa8cd924aead49ab69","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-28 14:00:32.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"50b6ca7946510d1d5a26ceb1e788e13a8e76150e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 18:06:14.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"96a4222e2a7aae658eead625ffcc63aef2c46da1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 19:14:52.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d04ae083c224d6e807b6267dd93e09a : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5364a7c640684f1599f7333a1c5447e4 : SUCCESS in 7m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f91cd888f304d72993709785f0fcebc : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32358bd227b94aceac8b7febdbd376e2 : SUCCESS in 8m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ca1075b821674faf9f58e578735eb2a6 : SUCCESS in 6m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8f26383117f845af9e935f4dc15e40f1 : SUCCESS in 9m 40s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3a12e1e4696e4486b3d2485d59f74ea2 : SUCCESS in 1h 07m 34s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/aac2da74615b4f268dac8238deaf6b22 : SUCCESS in 1h 01m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a3752acf07564a3dba1058347cb9df62 : SUCCESS in 30m 34s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d7434416138a40b5b3cf09c7fce2fccf : SUCCESS in 16m 04s","accounts_in_message":[],"_revision_number":5},{"id":"5a39a407e4d09b8d86702eff7a424609664b0d45","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 19:14:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"d7e266fed53fb5d9d7774b5b49414b45c89d12f8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 19:17:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/967fb6af7f0c4ab99606f33bdbee4e8b : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"aa10df3cc6e30d5f600636ab54334249bd8a4623","revisions":{"24efabef10548f335e8a1db472977c7da2d55090":{"kind":"REWORK","_number":1,"created":"2022-02-24 17:06:05.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/73/830873/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/830873/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/830873/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/830873/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/830873/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/830873/1"}}},"commit":{"parents":[{"commit":"adce4a0417ff97ea0abbf361521a020bb2d4f9a2","subject":"Merge \"Generate kolla config when running database commands\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adce4a0417ff97ea0abbf361521a020bb2d4f9a2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-06 14:02:21.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-24 17:05:19.000000000","tz":180},"subject":"CentOS Stream: update DNF repo file names","message":"CentOS Stream: update DNF repo file names\n\nCentOS repositories were renamed following the separation of CentOS\nLinux and Stream. This change updates their names when using a local\nmirror. It also removes the old files.\n\nChange-Id: Ifcc3781d771e150b5771f2bf20cd8bcbc5faa443\nStory: 2008795\nTask: 42435\n(cherry picked from commit ece41b7a677341d5023eb3f5e4db7f5326cfc86c)\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24efabef10548f335e8a1db472977c7da2d55090"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24efabef10548f335e8a1db472977c7da2d55090"}]},"branch":"refs/heads/stable/victoria"},"700d7030827577393929e083fe044eec5c90d9d5":{"kind":"REWORK","_number":2,"created":"2022-02-24 17:10:19.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/73/830873/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/830873/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/830873/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/830873/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/830873/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/830873/2"}}},"commit":{"parents":[{"commit":"adce4a0417ff97ea0abbf361521a020bb2d4f9a2","subject":"Merge \"Generate kolla config when running database commands\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adce4a0417ff97ea0abbf361521a020bb2d4f9a2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-06 14:02:21.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-24 17:10:08.000000000","tz":180},"subject":"CentOS Stream: update DNF repo file names","message":"CentOS Stream: update DNF repo file names\n\nCentOS repositories were renamed following the separation of CentOS\nLinux and Stream. This change updates their names when using a local\nmirror. It also removes the old files.\n\nChange-Id: Ifcc3781d771e150b5771f2bf20cd8bcbc5faa443\nStory: 2008795\nTask: 42435\n(cherry picked from commit ece41b7a677341d5023eb3f5e4db7f5326cfc86c)\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/700d7030827577393929e083fe044eec5c90d9d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/700d7030827577393929e083fe044eec5c90d9d5"}]},"branch":"refs/heads/stable/victoria"},"892db2e66486a89caba6f52bbc9d01dc9d40e90f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-02-24 22:29:22.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/73/830873/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/830873/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/830873/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/830873/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/830873/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/830873/3"}}},"commit":{"parents":[{"commit":"e2c797626f4d431a14aa7dc4fae10d3d4a42653a","subject":"CI: temporary disable voting of some jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2c797626f4d431a14aa7dc4fae10d3d4a42653a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-06 14:02:21.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-24 22:29:22.000000000","tz":0},"subject":"CentOS Stream: update DNF repo file names","message":"CentOS Stream: update DNF repo file names\n\nCentOS repositories were renamed following the separation of CentOS\nLinux and Stream. This change updates their names when using a local\nmirror. It also removes the old files.\n\nChange-Id: Ifcc3781d771e150b5771f2bf20cd8bcbc5faa443\nStory: 2008795\nTask: 42435\n(cherry picked from commit ece41b7a677341d5023eb3f5e4db7f5326cfc86c)\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/892db2e66486a89caba6f52bbc9d01dc9d40e90f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/892db2e66486a89caba6f52bbc9d01dc9d40e90f"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"},"efcf7bd2b0de7eedffd5dc5ca74df32719ab1f96":{"kind":"REWORK","_number":4,"created":"2022-02-28 10:33:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/830873/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/830873/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/830873/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/830873/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/830873/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/830873/4"}}},"commit":{"parents":[{"commit":"e2c797626f4d431a14aa7dc4fae10d3d4a42653a","subject":"CI: temporary disable voting of some jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2c797626f4d431a14aa7dc4fae10d3d4a42653a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-06 14:02:21.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-28 10:32:49.000000000","tz":0},"subject":"CentOS Stream: update DNF repo file names","message":"CentOS Stream: update DNF repo file names\n\nCentOS repositories were renamed following the separation of CentOS\nLinux and Stream. This change updates their names when using a local\nmirror. It also removes the old files.\n\nThis backport has been adapted to support both CentOS Linux and Stream\nhosts.\n\nChange-Id: Ifcc3781d771e150b5771f2bf20cd8bcbc5faa443\nStory: 2008795\nTask: 42435\n(cherry picked from commit ece41b7a677341d5023eb3f5e4db7f5326cfc86c)\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/efcf7bd2b0de7eedffd5dc5ca74df32719ab1f96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/efcf7bd2b0de7eedffd5dc5ca74df32719ab1f96"}]},"branch":"refs/heads/stable/victoria"},"aa10df3cc6e30d5f600636ab54334249bd8a4623":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-02-28 10:37:20.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/830873/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/830873/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/830873/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/830873/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/830873/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/830873/5"}}},"commit":{"parents":[{"commit":"abff9e6c1b4ad2b38fe40ce3d68eb8c3c674d692","subject":"CI: temporary disable voting of some jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abff9e6c1b4ad2b38fe40ce3d68eb8c3c674d692"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-06 14:02:21.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-28 10:37:14.000000000","tz":0},"subject":"CentOS Stream: update DNF repo file names","message":"CentOS Stream: update DNF repo file names\n\nCentOS repositories were renamed following the separation of CentOS\nLinux and Stream. This change updates their names when using a local\nmirror. It also removes the old files.\n\nThis backport has been adapted to support both CentOS Linux and Stream\nhosts.\n\nChange-Id: Ifcc3781d771e150b5771f2bf20cd8bcbc5faa443\nStory: 2008795\nTask: 42435\n(cherry picked from commit ece41b7a677341d5023eb3f5e4db7f5326cfc86c)\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aa10df3cc6e30d5f600636ab54334249bd8a4623"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aa10df3cc6e30d5f600636ab54334249bd8a4623"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
