)]}'
{"id":"openstack%2Fkayobe~831270","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~Ia24ce02b5609b97e89a45a3355cdb1f50ea9ef7a","project":"openstack/kayobe","branch":"stable/victoria","topic":"fix-ci-voting-jobs","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-03-01 14:33:13.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-01 10:12:20.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-03-01 11:10:13.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"Ia24ce02b5609b97e89a45a3355cdb1f50ea9ef7a","subject":"Revert \"CI: temporary disable voting of some jobs\"","status":"MERGED","created":"2022-02-28 19:19:34.000000000","updated":"2022-03-01 14:39:40.000000000","submitted":"2022-03-01 14:33:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831270-fix-ci-voting-jobs","meta_rev_id":"af13e4df9e453b1d7e57ec599715fc8147b6ea9a","_number":831270,"virtual_id_number":831270,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-01 14:33:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-03-01 10:12:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-01 11:10:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-01 11:10:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-28 20:56:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-01 08:54:07.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-03-01 09:54:29.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-03-01 11:10:13.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"9a1e20cfafc5dbb40c984f411d0a5129030acac3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-28 19:19:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"931b08df376af3e47a79ad26184559ab9bc80615","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 20:56:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e343a88b65f545a5be55b9669dd26b74 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8566647432ae409297a315f0cc6223b2 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f812ac6f0da84aca9beb430435ec2ac7 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0a7b5beef2c42448f7b0078869b3d56 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e56f17d2faf4720be33a6674741263a : SUCCESS in 7m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/370d718379da4bcf997acee77b8915f3 : SUCCESS in 6m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7804c2f8c3424448aa098766065a2985 : SUCCESS in 10m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5089dc822b414f0e9cff775318b5b233 : SUCCESS in 19m 53s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5f1f360c3e57491095b1589062cec89f : SUCCESS in 58m 14s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4aa00fdff7ed47248b8a8023df1633d1 : SUCCESS in 1h 01m 45s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/481f04cca7eb444b9e548466e1a5697c : SUCCESS in 18m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8addd580b57346f5a5f72db3890ce9ca : FAILURE in 1h 32m 31s (non-voting)\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/251a64fc36c14d69a12647766444daec : SUCCESS in 29m 06s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a5ed2029e19c422dafc3345b9b16a84f : FAILURE in 21m 02s (non-voting)\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/068c255131014d818aef7b2b94ca10e9 : SUCCESS in 16m 06s","accounts_in_message":[],"_revision_number":1},{"id":"2766a63ada82de2d10e9bb2ab2a7b7bd7a570902","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-01 08:54:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fab4546d536c9f9a516aa43f1be85b3bd5221feb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-01 08:59:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"aa222e374f49a332ab002595c13dcf46a5ef4438","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-01 09:01:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8ab4064abaeebae9b9103c1b4267cba0358fefd2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-01 10:12:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"39b3ef19d105c9fec836f4ca807c8f4bff09eceb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-01 11:10:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3d65159479a4bc096735fb57522895dc98d33839","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 12:53:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5338cc11cfe04b47a7131d3f85a7099f : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2def20371d6d4a74b17c70cf57bb3e12 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7faf0f1986be40a499032953d3424e1a : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9d63d0b4dd6492aac0813fb15c81cd5 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25b974018ce7442e8624ce91d24ca0ff : SUCCESS in 8m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/729b5eede49240b1a08631db1605c37e : SUCCESS in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6a72f584021c444d8e884ce84467b5e2 : SUCCESS in 9m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/563151773f2b48cda2dcfb32ffaa2b1b : SUCCESS in 20m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/76c3c830fe0c4b6981e932b3e32aef10 : SUCCESS in 1h 10m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8495c1c875de48fbb5d785e21816048c : SUCCESS in 1h 01m 12s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e3e833db9f0d4556b47124b7918b78d6 : SUCCESS in 19m 50s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e8ae2bf116ad41c5bb57bba5217d4951 : POST_FAILURE in 2h 21m 39s (non-voting)\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ccb477cfc8fd4a07a95612d0fc8b3612 : SUCCESS in 29m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8ce38b1c653246f79e080272a2ac7dbc : FAILURE in 20m 16s (non-voting)\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a0d88a94fe364ce992d9ef32a7984289 : SUCCESS in 14m 23s","accounts_in_message":[],"_revision_number":2},{"id":"0e215d046405c4eb9a6fdc303f75cad0e72348bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 12:53:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5c6bef08d600332cbea98c44c6e0b0dabbfa1776","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 14:33:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fafdb1c672dc4a59bfd8c20e11462bbf : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c22aa37f52904295acc5b6ce9952d74e : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dabbd535388642d2859e2bcf10390934 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e90b4ff003d4c1d8e562dbb3a20e604 : SUCCESS in 8m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4d3dc12c18b64165a9caf1fd4215a3ce : SUCCESS in 6m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5ae178f3073049f9b5e781fbf9938136 : SUCCESS in 10m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8ab89dceaedf48a8916d742341325735 : SUCCESS in 23m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/10d4b4f590a44b6bb0ad4270f8bcc24a : SUCCESS in 1h 35m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a3af2c4892324bf99ce0c1b241727a18 : SUCCESS in 1h 34m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d4d802f8074741a49931f4bd0c75912e : SUCCESS in 17m 19s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d789ef884cf2441ba89c4da3c1ee0bc2 : SUCCESS in 32m 42s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5c89ac49774b4491ba0bce59b4a59389 : SUCCESS in 14m 19s","accounts_in_message":[],"_revision_number":2},{"id":"94ec188abfa572dcdaadcf84cae0d4c660ac22e5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 14:33:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"af13e4df9e453b1d7e57ec599715fc8147b6ea9a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 14:39:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79b8b431bb7e450986d06b888a816b2d : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1264ac29cb2afc4a2325174cec75d4b7b6da7e36","revisions":{"a3c170c08634ecfb6d544d0d42840a3d387f5eff":{"kind":"REWORK","_number":1,"created":"2022-02-28 19:19:34.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/70/831270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/70/831270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/831270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/831270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/831270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/70/831270/1"}}},"commit":{"parents":[{"commit":"71ed929cb089ca4caf079f326cfb094cf932bc4b","subject":"Merge \"CentOS Stream: update DNF repo file names\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/71ed929cb089ca4caf079f326cfb094cf932bc4b"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-28 19:16:04.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-28 19:18:47.000000000","tz":180},"subject":"Revert \"CI: temporary disable voting of some jobs\"","message":"Revert \"CI: temporary disable voting of some jobs\"\n\nThis partially reverts commit abff9e6c1b4ad2b38fe40ce3d68eb8c3c674d692.\n\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nChange-Id: Ia24ce02b5609b97e89a45a3355cdb1f50ea9ef7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a3c170c08634ecfb6d544d0d42840a3d387f5eff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a3c170c08634ecfb6d544d0d42840a3d387f5eff"}]},"branch":"refs/heads/stable/victoria"},"1264ac29cb2afc4a2325174cec75d4b7b6da7e36":{"kind":"REWORK","_number":2,"created":"2022-03-01 08:59:54.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/70/831270/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/70/831270/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/831270/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/831270/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/831270/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/70/831270/2"}}},"commit":{"parents":[{"commit":"71ed929cb089ca4caf079f326cfb094cf932bc4b","subject":"Merge \"CentOS Stream: update DNF repo file names\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/71ed929cb089ca4caf079f326cfb094cf932bc4b"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-28 19:16:04.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-01 08:59:36.000000000","tz":180},"subject":"Revert \"CI: temporary disable voting of some jobs\"","message":"Revert \"CI: temporary disable voting of some jobs\"\n\nThis partially reverts commit abff9e6c1b4ad2b38fe40ce3d68eb8c3c674d692.\n\nChange-Id: Ia24ce02b5609b97e89a45a3355cdb1f50ea9ef7a\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1264ac29cb2afc4a2325174cec75d4b7b6da7e36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1264ac29cb2afc4a2325174cec75d4b7b6da7e36"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
