)]}'
{"id":"openstack%2Fkayobe~831407","triplet_id":"openstack%2Fkayobe~master~I8f5f93340642d055cce7ef306d942e75b10c86a9","project":"openstack/kayobe","branch":"master","topic":"qemu-img","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-03-03 18:00:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8f5f93340642d055cce7ef306d942e75b10c86a9","subject":"CI: remove qemu-utils installation","status":"MERGED","created":"2022-03-01 16:43:25.000000000","updated":"2022-03-03 18:02:21.000000000","submitted":"2022-03-03 18:00:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831407-qemu-img","meta_rev_id":"ec00f5a1d9bc4eb76f7bfff4dfcd66dc9022454d","_number":831407,"virtual_id_number":831407,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-03 18:00:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-03-03 10:28:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-03-03 09:14:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-03-03 09:14:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-01 18:35:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-01 18:47:57.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-03-03 09:14:30.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-03-03 10:28:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"c899108f59d5a902d1f16181a98b8b6ef118fe79","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-01 16:43:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3cbc3173ffeefce35bd47d08ea1adb3c015a5300","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 18:35:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e18f0e34a1d24cd99a1b5b3f6aca74d3 : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b331d56459449cf8a76de1807ca6b0b : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fcc2577520844289c8d570967ce61c8 : SUCCESS in 10m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6d406fd15564410bb68bf1a6baecc61 : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18a4cab5e1e9437ebd5bf8909bcd639c : SUCCESS in 11m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/97c59bd00c3e46f5b85cb9a357f4c47e : SUCCESS in 7m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f804933626ff4bfe898123057fd0dcd9 : SUCCESS in 14m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/976cacd8ba844cb5a8aac9f832d23873 : SUCCESS in 33m 45s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c0e40eb17eb5498a92a8dbe6a8e48433 : SUCCESS in 1h 08m 29s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57f9f14eae6440eba4fcd5e9407699cf : SUCCESS in 56m 07s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7b1a4f648fe5478aae3e41752f8d8702 : SUCCESS in 1h 11m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/edfe4ae3617f4bce9e14328cdf8fed86 : SUCCESS in 30m 16s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0a1dc4f171e4f038a3ca069ab5e9016 : SUCCESS in 17m 34s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8328296f164b4ae099a64fad33ae92a8 : SUCCESS in 1h 46m 30s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b459759081b540a3a9fd4c1d84fdf6d9 : SUCCESS in 1h 16m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/cd4bbc7633f24c9e87300fee70d3292f : SUCCESS in 50m 22s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9708ba95d9a6424d81453e0d5fe19ee0 : SUCCESS in 39m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b1cbbbba6e3a4e76a1568e5049c4487d : SUCCESS in 34m 08s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32baf262b17945129d86b19b35998880 : SUCCESS in 22m 57s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/079c7fe781914e5b96299a2b40705b43 : SUCCESS in 18m 30s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bce68d327084fae93c98ca2a88dac3f : SUCCESS in 15m 43s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/421ab064b3214e9e9b6096d1cb1c713f : SUCCESS in 19m 33s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c79a17f7ddc849b7a61235f5478e0a52 : SUCCESS in 16m 44s","accounts_in_message":[],"_revision_number":1},{"id":"836f70a618701e38024517f930d3b605b897306e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-01 18:47:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"16a4b942cd1d5069de8f652f138358ba67942dd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-03 08:36:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d829adcb18693b16d31689c74648a09fe5b86e7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-03 09:01:32.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"9b5c4ae88d3517c450ce5939d0ea88201c3f1fcd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-03 09:14:30.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5419d62c77bc9fa02bd1f64d11292f19a2ee70c9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-03 10:28:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"526f2083b3d09f2fcbc69aa5d113a59edcb33357","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 11:00:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcf02b770a1c4515b248831f067b3a5b : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eacb96959814e32a8dbd819d2a14c12 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4add51d61d664cea98d2af84f9a1b2e7 : SUCCESS in 9m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4edd8e4684224a669dc5e741cc133f9e : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cbec186b7b140c6b936a0f17d4d1066 : SUCCESS in 10m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9c12a235f10a4d61b40edf73bb24f1c0 : SUCCESS in 8m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aa085d864d194c41af068a724ee72573 : SUCCESS in 14m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5bef860bfe5f485dbbac14789071d401 : SUCCESS in 35m 13s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c3b60f36c32648dba554c14e97d74735 : SUCCESS in 1h 07m 10s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1294b641a45468d9bb51c1a2de63d4f : SUCCESS in 55m 51s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f98c04f1310e459eab349bab0db1f0dc : SUCCESS in 1h 13m 22s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c5a6fcde8a1440468d0c0cebef463af0 : SUCCESS in 23m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/208aa781bb374045ab2517721eca83f9 : SUCCESS in 17m 57s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ac4b65a7f7ad45a99384444b9e573cce : SUCCESS in 1h 53m 20s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5dfbfef41f24586b1e9b7213321e6d4 : SUCCESS in 1h 16m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1f8183836e0c470b9af430d4741b1aae : SUCCESS in 1h 01m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cfed540dfd7743a5ad9a64913b046d14 : SUCCESS in 40m 59s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/22361b2bd26042f59d70d7fecaf29dbb : SUCCESS in 29m 52s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb96bac5b7fc4eb1ab7d73f438011be9 : SUCCESS in 28m 05s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/09849f2ab1714aaa94eaaa40cab5d31a : SUCCESS in 17m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3198e0231474004a1fd238df9d2ebdb : SUCCESS in 13m 42s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/8f057b0a1f244fe7b136120b0f1b50da : SUCCESS in 18m 48s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39291f6eb9b44887af14d19303383929 : SUCCESS in 14m 47s","accounts_in_message":[],"_revision_number":3},{"id":"e3f24e4ac42d233580536a289852edac622f4ac1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 11:07:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2b47ab097a54ab9861884070268eff80818bc15b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 13:16:27.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93d95dff3864471cb34574457cf6d85e : SUCCESS in 9m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c8fff981ed841958a7d1d02a6391906 : SUCCESS in 9m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5744a60017444af6be7a6374233cf586 : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10972825ade34f1985824464fbf277bd : SUCCESS in 10m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4c86df65a489426a917b7d868218cbfa : SUCCESS in 8m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e7d883a58ec64169b7bdeaf157005eda : SUCCESS in 15m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1fba1c1400174571b037677358097701 : SUCCESS in 31m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/54b8dc0c38e14c1a98cce3b5df2353a8 : SUCCESS in 1h 11m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c97f1f592dad4d9abb180556409a8cac : SUCCESS in 58m 02s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c16fe3f93bca4f0f92414fa0a964a6a3 : FAILURE in 1h 18m 59s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b76f0f4d34394f3ab61e4e9cbf5bb846 : SUCCESS in 25m 24s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/988c62ebd13e4b899e5250d60c22cda6 : SUCCESS in 21m 09s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a684599309024c7f99f490d96da39bf9 : SUCCESS in 1h 54m 51s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f779f3ee1e44c828bdcb50ab0863438 : SUCCESS in 1h 21m 14s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f0b1bb2f742e47d5b9a1c2acc4644549 : SUCCESS in 1h 01m 56s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a05f2343fe348f9a9588baea79f1264 : SUCCESS in 39m 49s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/198b53d6aa374e378931a0da2c2bb5ea : SUCCESS in 30m 05s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5ff3b1a17234915aed703c5594a3c26 : SUCCESS in 29m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6eff87c0d4fe477583de315743432f98 : SUCCESS in 18m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c8784ecf54a4a98b66e199651bd193b : SUCCESS in 13m 37s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/97a7df7f3def417999cfea152747a2c7 : SUCCESS in 20m 33s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fc110ddaa9c4d278993efd39326a2d5 : SUCCESS in 15m 08s","accounts_in_message":[],"_revision_number":3},{"id":"6236a1d95748f2580a77b2ef311e458f7c233b2d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-03 13:34:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0816740bbc54dc58c3d155177a36c442727b78d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 15:51:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/773572ee8ed9461aac6c22712cfafc94 : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6737f5a8554487d92fa60058d7d0f42 : SUCCESS in 8m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26e519e1b9854981bda531836f66af1d : SUCCESS in 7m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b611b0a9a0c849ceae2c6175097a2a04 : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/866e392923914ea481987b4870d9e016 : SUCCESS in 11m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8a3dc94e7e404be3a2251a92562417f2 : SUCCESS in 8m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/647a37b5db3f410d9861e1ada017f72e : SUCCESS in 14m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/541216b1c902409789b90c824a63ccf2 : SUCCESS in 33m 47s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0d71851f08c34513b670aa6bc3f579b5 : SUCCESS in 1h 08m 03s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8bb2531de9f47268425d511015a1978 : SUCCESS in 53m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/186019b0de6d4d6a91f5ea0c5cd41f88 : SUCCESS in 1h 18m 01s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2c838374ad5c44fc85cd5558121dbe57 : SUCCESS in 25m 33s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e39bae848e5482f88719f22470d6f3f : SUCCESS in 16m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ee3a59f1ebd442efa007e5002f02db36 : SUCCESS in 2h 05m 13s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad1c79a8efbd41bd9456c611ced9d5b1 : SUCCESS in 1h 19m 02s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/da6a43d5b63a484f8322c36d8f9bbd0b : SUCCESS in 1h 00m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8ecbb4da2ba47caa02e3f13dfc6c9b5 : SUCCESS in 41m 05s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0c37c6b71a3d4f26817b2eefc0bdaebd : SUCCESS in 33m 55s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c73e42e4970144548359e0e5e6d8b2c0 : SUCCESS in 27m 18s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7387095826df4463aed9195b9726d2dc : SUCCESS in 17m 18s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0a3fcd41d94457d846294452387760f : SUCCESS in 13m 31s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e042bc67e5eb44368091298c44032387 : SUCCESS in 18m 56s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/601283441f4d4dd5bfe7c334e70de93b : SUCCESS in 14m 00s","accounts_in_message":[],"_revision_number":3},{"id":"6c1528e23c579c94cebc34f87c646a905c633317","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 15:59:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5a05e9e38df6a0bb3d59ad119cc3292e9ef16438","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 18:00:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1465a85c0f94991afba2a984a511613 : SUCCESS in 10m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f98f95833e4049daae8bc904b75fa4a0 : SUCCESS in 8m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/619e17ae1bf54121874d5bbdf11d61de : SUCCESS in 10m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ac1164fb88a48d4b2521d56855661a0 : SUCCESS in 12m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a3740fbcd55c48b9b0cdb9d622de5200 : SUCCESS in 10m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e9acf897a73d4d9697f2f4cdcf837ee9 : SUCCESS in 14m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3ce91a8bca204fcda7dcb81f3039f69d : SUCCESS in 35m 36s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/40a416ca262f41f185b97a145c850fb5 : SUCCESS in 1h 19m 02s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/233b89f712ef423a8402d42400846013 : SUCCESS in 56m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/09958959d7074095b5d7d98c7e1d4bef : SUCCESS in 1h 18m 12s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8305687c3a2c48b49ee534f37415417c : SUCCESS in 25m 06s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78950657f49a45bf94d40f09787eaa60 : SUCCESS in 17m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/845265c599b74396abd3c6dd36ff77a7 : SUCCESS in 1h 59m 48s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d0c750fbebb45b6b2eb8dac304311a6 : SUCCESS in 1h 26m 14s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f721e5ce351b4825b54113e6fd4014e1 : SUCCESS in 51m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddbc3155885d465ea0823922a327d1a6 : SUCCESS in 38m 05s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c20993ae186c4f909a24b8d45a9999a8 : SUCCESS in 29m 57s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a10127a7f38a433285b9546ad1a5831f : SUCCESS in 27m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/eb01adccb6004da29a7b438e7cc02993 : SUCCESS in 18m 57s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6da13d81fe124ca58f2580269cf692ff : SUCCESS in 14m 12s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/27a9d342b7f247e683a1d9c4bfe77817 : SUCCESS in 20m 42s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dec6e5918a03496ca499a3ac0f47f653 : SUCCESS in 17m 09s","accounts_in_message":[],"_revision_number":3},{"id":"1543369b8516dbe06e9e86c4dc6e305b7c32e9aa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 18:00:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ec00f5a1d9bc4eb76f7bfff4dfcd66dc9022454d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 18:02:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23adfa20a1334471a062055519671e43 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e453f3c57053bf52b4b70d0a1c1776d360170157","revisions":{"acf31fe5ffa5b2eb922f53ee3bc08b95008ebeab":{"kind":"REWORK","_number":1,"created":"2022-03-01 16:43:25.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/07/831407/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/831407/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/831407/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/831407/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/831407/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/831407/1"}}},"commit":{"parents":[{"commit":"082f9c5477c1c9ae2ab86484618691d757ba6de0","subject":"Merge \"CI: test fact caching\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/082f9c5477c1c9ae2ab86484618691d757ba6de0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-01 16:41:42.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-01 16:41:42.000000000","tz":60},"subject":"[WIP] Remove qemu-utils installation","message":"[WIP] Remove qemu-utils installation\n\nKolla now installs qemu-utils in nova_libvirt images [1].\n\nOnly available in master for now, we will need it in stable branches for\nthe upgrade jobs.\n\n[1] https://review.opendev.org/c/openstack/kolla/+/830401\n\nChange-Id: I8f5f93340642d055cce7ef306d942e75b10c86a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/acf31fe5ffa5b2eb922f53ee3bc08b95008ebeab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/acf31fe5ffa5b2eb922f53ee3bc08b95008ebeab"}]},"branch":"refs/heads/master"},"365c9f87f9e6978886b7d25f6a2b43f284bbfef3":{"kind":"REWORK","_number":2,"created":"2022-03-03 08:36:38.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/07/831407/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/831407/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/831407/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/831407/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/831407/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/831407/2"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-03 08:34:14.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-03 08:34:14.000000000","tz":60},"subject":"Remove qemu-utils installation","message":"Remove qemu-utils installation\n\nnova-libvirt images now include qemu-utils on master [1] and xena [2].\n\n[1] https://review.opendev.org/c/openstack/kolla/+/830401\n[2] https://review.opendev.org/c/openstack/kolla/+/831411\n\nChange-Id: I8f5f93340642d055cce7ef306d942e75b10c86a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/365c9f87f9e6978886b7d25f6a2b43f284bbfef3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/365c9f87f9e6978886b7d25f6a2b43f284bbfef3"}]},"branch":"refs/heads/master"},"e453f3c57053bf52b4b70d0a1c1776d360170157":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-03-03 09:01:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/07/831407/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/831407/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/831407/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/831407/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/831407/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/831407/3"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-03 09:01:21.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-03 09:01:21.000000000","tz":60},"subject":"CI: remove qemu-utils installation","message":"CI: remove qemu-utils installation\n\nnova-libvirt images now include qemu-utils on master [1] and xena [2].\n\n[1] https://review.opendev.org/c/openstack/kolla/+/830401\n[2] https://review.opendev.org/c/openstack/kolla/+/831411\n\nChange-Id: I8f5f93340642d055cce7ef306d942e75b10c86a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e453f3c57053bf52b4b70d0a1c1776d360170157"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e453f3c57053bf52b4b70d0a1c1776d360170157"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
