)]}'
{"id":"openstack%2Fkayobe~831532","triplet_id":"openstack%2Fkayobe~master~I875f8976df75dee68ba00842fe624c29cc1b123c","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-03 13:16:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I875f8976df75dee68ba00842fe624c29cc1b123c","subject":"CI: stop using zuul as kayobe_ansible_user in TLS jobs","status":"MERGED","created":"2022-03-02 13:40:25.000000000","updated":"2022-03-03 13:17:48.000000000","submitted":"2022-03-03 13:16:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831532","meta_rev_id":"2eb0f25198256c547731ddadb13d8e7b0ef57857","_number":831532,"virtual_id_number":831532,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-03 13:16:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-03-03 10:55:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-03 08:39:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-03-03 10:55:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-02 15:34:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-03 08:39:13.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-03-03 10:55:18.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"de2bac025a095015624ae4da8796688ae9d69f7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-02 13:40:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d1877fb19a6f0125f0382ec4bb1495c6be5d596","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 15:34:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68258497ba13455d9bcda7bf1cefbb38 : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a147451a4b684663a265f0aad0062dab : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b50f95c12b3a44d0b1b7cb6b981fdf26 : SUCCESS in 7m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ef2bd8ca5784917a92b7973839a12a4 : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5610a86cbbb24b36978defb384ecb11f : SUCCESS in 10m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bb735ed51f454c4986fb9754aa7d4b4c : SUCCESS in 8m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a3cb79b5c315474fb3c12589ebf303e0 : SUCCESS in 14m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1f93e4f78f6849cc93b05eab5ccc7f48 : SUCCESS in 34m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f6aae345b82e440097ea65b33f0eaaaf : SUCCESS in 1h 08m 06s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9a385d2c9554fbeb88b3c63502b827d : SUCCESS in 56m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5cfcf8fed3d243ea8fa48b50e95e1607 : SUCCESS in 1h 08m 04s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6aa9ae1baa0a4573a7e6e9bba3ac33a9 : SUCCESS in 22m 26s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03c3aeb3b37f487597dda065c788cb9f : SUCCESS in 15m 23s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d53b00db80ee45f1968a43eb70c04614 : SUCCESS in 1h 51m 00s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19325fc98243471488b40abb5d71c30d : SUCCESS in 1h 25m 02s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a1f59c970b2f4d44a4105a5039ea851f : SUCCESS in 46m 12s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/efc9ae17ad2e4e50b45574d96acc145a : SUCCESS in 41m 55s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/87d8ed6c995646d786da2dcf84dcb556 : SUCCESS in 33m 52s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb30893254904339ad03f91b039ef5ec : SUCCESS in 23m 39s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/efb708d2b7b5474098767fea701fc195 : SUCCESS in 18m 04s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4c92919d91d4988b666c72331d42636 : SUCCESS in 13m 11s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e457128d83d44078a5c3c88580cc0558 : SUCCESS in 19m 45s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03c5eda733e34321917b51844f182eff : SUCCESS in 15m 33s","accounts_in_message":[],"_revision_number":1},{"id":"0b71f04cfb2d55bbb6e195b145eec507c7e88e43","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-03 08:39:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b8ad123bfc7f9c48e41d2953006336e655ca5f3e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-03 10:55:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"36ef28acd60938fc96bc5b33d081cc94ca363b02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 11:06:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42ff04a592929fd08fb90498652b938d1fa674e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 13:16:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50eb87bcd17746ec9ee16c20c536a504 : SUCCESS in 9m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c9fc9d49ae5490a97ca2e8c20e4d72e : SUCCESS in 9m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e49fce7c40b4b00b78e0fe835847852 : SUCCESS in 9m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e8ed212ffd14de98ea12947646d8e7a : SUCCESS in 10m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/616458b808ce4277a39548f038bcab18 : SUCCESS in 9m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/650aedd15a8b42148be6897ffab4d112 : SUCCESS in 16m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f30a6ba41ad2432bb44c5aeda513376f : SUCCESS in 33m 03s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/bb75c2c1a55c46018c6a656580bb8a12 : SUCCESS in 1h 07m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ef3dfffaccb463988a4e6e32d6ac4ec : SUCCESS in 57m 40s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/869c859f2e5b4efc92f8ab7a8f466f3b : SUCCESS in 1h 17m 00s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/44f1792a81a94348a2ee1471b7352268 : SUCCESS in 24m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/831355d66e9d4d6e9db0f17a52f3f976 : SUCCESS in 17m 25s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b95f1a979e8546a796456411f0d8ad3e : SUCCESS in 2h 03m 40s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d435fd3c9c84e9b8be004977b41a271 : SUCCESS in 1h 27m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e9cf1f1e240b49278fd71cf8055dae81 : SUCCESS in 58m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8edac3b1b37741eeb02ab39af58fe1c1 : SUCCESS in 47m 54s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/949a1a66fab24be5836ffbe276d7211e : SUCCESS in 29m 42s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fd773177c034775bee86a11c775c5e4 : SUCCESS in 31m 18s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3a5589c58f99425a8fd62a42fcf3ef95 : SUCCESS in 18m 14s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8e5c04227a24a18be22e022a320433b : SUCCESS in 15m 45s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/a8be7792a94547709f1144950afd55cb : SUCCESS in 20m 51s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de53ec48c038475aa8c6cdf2214985be : SUCCESS in 15m 29s","accounts_in_message":[],"_revision_number":1},{"id":"50fbd7fd8d5789d727b278f5e9b5772cf6befd9f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 13:16:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2eb0f25198256c547731ddadb13d8e7b0ef57857","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 13:17:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8cd392d927a45be83c7f4a59aa8e6e5 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a0665cd9c63735b794f200514d90ff1b9e6dd239","revisions":{"a0665cd9c63735b794f200514d90ff1b9e6dd239":{"kind":"REWORK","_number":1,"created":"2022-03-02 13:40:25.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/831532/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/831532/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/831532/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/831532/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/831532/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/831532/1"}}},"commit":{"parents":[{"commit":"082f9c5477c1c9ae2ab86484618691d757ba6de0","subject":"Merge \"CI: test fact caching\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/082f9c5477c1c9ae2ab86484618691d757ba6de0"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-02 13:25:21.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-02 13:34:12.000000000","tz":0},"subject":"CI: stop using zuul as kayobe_ansible_user in TLS jobs","message":"CI: stop using zuul as kayobe_ansible_user in TLS jobs\n\nPreviously we were using the zuul user in the TLS jobs. This was due to\na permissions issue when accessing the CA certificate in kayobe-config\nin the zuul user\u0027s home directory.\n\nThis change reverts to the default of using the stack user for the TLS\njobs. In order to make this work, the generated CA cert chain is added\nto the trust store.\n\nChange-Id: I875f8976df75dee68ba00842fe624c29cc1b123c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a0665cd9c63735b794f200514d90ff1b9e6dd239"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a0665cd9c63735b794f200514d90ff1b9e6dd239"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
