)]}'
{"id":"openstack%2Fkayobe~831652","triplet_id":"openstack%2Fkayobe~master~If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2022-03-20 22:20:24.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-03-20 22:20:24.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-18 17:14:36.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-03-18 13:13:21.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad","subject":"Add support for Rocky Linux 8","status":"MERGED","created":"2022-03-03 10:27:30.000000000","updated":"2022-03-20 22:25:48.000000000","submitted":"2022-03-20 22:20:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":74,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831652","meta_rev_id":"5706febfc9089e0aacfc177207dd112f44a898ed","_number":831652,"virtual_id_number":831652,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-20 22:20:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-03-18 17:14:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-20 20:33:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-20 20:33:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-03 11:05:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-07 14:23:53.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-03-17 09:46:15.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2022-03-17 12:14:13.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2022-03-18 10:48:49.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"b67cf002b4f8aab77195112bedfe773242b25167","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-03 10:27:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdaff29c6c29cf5d4ceb5f7d9a6128fdb9ba8322","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-03 10:39:24.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7f6adb3fd913f20297f1482725db210d6b07056d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-03 10:39:51.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"a050ee78e0b1f4af6298c91797b7baba89a1a375","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 11:05:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a00b6560d1141509ad4956bb42ba774 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/727573e096f1482195af5711f715e9cc : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1db8aaf142324f70b0434fa8defe4672 : SUCCESS in 8m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63f12de66c37477a915a3e9ad25fc912 : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b81a790cc104423dad554e44b31991ed : SUCCESS in 11m 16s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/9139cb6bb5ff4ca39cf1cc779030c241 : FAILURE in 16m 28s","accounts_in_message":[],"_revision_number":3},{"id":"58428c68737a0232f37749ba1664f8ba58953e56","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-03 11:23:02.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"613fca8f029e5412218af86a6f8aed159a6ef5ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 11:48:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad3d5bd824864f958fdaa4707e41ab7b : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4a63a8be9cc4f0fbc96893970ef54f6 : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96f9c49de50d4a4ea0ae3454d5a52bb2 : SUCCESS in 9m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f015c5e628b4f56a61c1f4b934c8cd2 : SUCCESS in 8m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58f08bf0cb0340d784f8186d95ca3355 : SUCCESS in 11m 22s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/ed815b3930c6496d97bb5ed79d82fb63 : FAILURE in 18m 01s","accounts_in_message":[],"_revision_number":4},{"id":"779778ebb4c8dffbd4e9303c8883fb7fb13832f6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-03 13:50:47.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"c2ef4ed5350dff5d8dd8ce5d855b323820e9d811","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 14:14:35.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c45f5c54ddc04f1abfe47121ddf36437 : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66aecac2156445f996d8584109704a6b : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56cfff354206417fb7ace7be14549f12 : SUCCESS in 7m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be4c97ed59d2434f8e071498ec81c0da : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daa54e19f8b4487281e3d7085d77e60c : SUCCESS in 10m 00s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/2c612d2645e54b2db841e5516f3e40cf : FAILURE in 19m 03s","accounts_in_message":[],"_revision_number":5},{"id":"f439fe3412054e3e451833f72763c2b7409f9851","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-03 14:34:45.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"f8514d57d67daad7574e612aa148e4a4fdafaba1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-03 14:52:49.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"4205caab89fa24cf369b3112478330a68aa812ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 15:13:50.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd070e7f332f4828bd8434b8b8623d6f : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/050efded6405469399aa773350c610fc : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acb3eaf5d2034dc88730edd140695474 : SUCCESS in 8m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27a63ef3c56c432eabb5e81aa3b22e89 : SUCCESS in 8m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5bc6285a0094998920b7eeb6aba0409 : SUCCESS in 11m 05s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/4117aa8df54e484197933716b1728503 : FAILURE in 15m 22s","accounts_in_message":[],"_revision_number":7},{"id":"3c9a15b668c98a3d970720bdb53aa6943500e443","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-03 15:17:38.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c6aa936de368501a2cba587cdbe0483998b02d63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 16:35:39.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6abd1bb092f47feaeba2e286fe5495b : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5da6d57ca8a44cb8b7778f9fc26e01c0 : SUCCESS in 8m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7f5d46a05dd4c7c8818c244f3e32259 : SUCCESS in 8m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6740fd294ec74b969f3fd83c6203009f : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b86760c6664471e98aa016eb75068cb : SUCCESS in 11m 03s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/961a89dbe6ad42e18e4542151c51daae : SUCCESS in 1h 10m 44s","accounts_in_message":[],"_revision_number":7},{"id":"5d173063e6c1c6525756e5eead97aea0beb7f1d7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-04 07:28:01.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"24acdf2519078a9a59c9b947f29daeec1f4d4971","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 08:48:13.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d68fc3e0e374a1aa02209080a74affb : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/769333b3237d4a0a9676baddf9ba936f : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f160c6aa55ee4fad9f5853bf2703a074 : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6de127e992b0437fb033190ea335566a : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37dc02e1ec574a15bf0dd39bbffa2f56 : SUCCESS in 10m 01s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/319bb46f53ed44b2b82dd832ae1e9ae7 : SUCCESS in 1h 12m 56s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/656158e9b01c4c30b5706ae82d8d8e71 : FAILURE in 12m 23s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/054e88aafd824d399f222d39d74963b2 : SUCCESS in 45m 50s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/59cd023e868b449faae3f28821427207 : FAILURE in 14m 16s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/fb6622d995d4483281707d059bf722a4 : FAILURE in 16m 16s","accounts_in_message":[],"_revision_number":8},{"id":"2888e28a2f32aaf6765f529bdfd3ccb41b8c539a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-04 09:31:32.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"c7ac4e4bf3d5d6447d9d9dab372b56963cfc250a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-04 09:33:48.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"df0652e4776b1c23681ec6995a6813ee850c2cf2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-04 09:59:29.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"77177c396ec01bbf5c703435e311cbdd82801ccf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-04 10:00:51.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"98edf879d8f92e8db0174190f33e64a4f1bc9d05","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-04 11:22:07.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"6c91db917f908a7e334c02fa831cd4dd4ea4e752","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-04 11:37:49.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"7ed7d6e493391a0a001e1e444804a29cb14b977c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-04 12:02:49.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"175c89f2a3afe04415f5fe0c3ce838e528999492","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-04 12:36:07.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"03f83dd124c6a6af3898bcf30acb0ac667c2de68","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-04 12:50:11.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"42913fd3a4c8472d985718265bb9c82ff796174a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-04 13:14:50.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"f35a6a028a687ffc85cb486ec320c2de968f8794","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-04 13:42:32.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"a2a8fdab5d79258ea837d9aa9e116edcb32bc31b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-04 14:09:54.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"64cef81cb05e8c6be74c4d067d1886fbc221d98f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-04 14:40:13.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"384bbdd201d6a8839299c61562594504fd2dedaa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-04 15:12:22.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"e44f259666b13d5676d7c39444fec82ba04e3088","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-04 15:34:24.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":22},{"id":"4ebdbacb174dd7b1345b70356d3976505cc6118f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 17:05:08.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a485330286443a899d080fe785708af : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/094a64ed9b8b4bd887094a897d4236b0 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d510cb1e1eee45429edb7466dbaf8cf2 : SUCCESS in 9m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c44d62a75a024c0ea6027dfe3a66061a : SUCCESS in 8m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aac93101c444efd96dcb40b6ea4bec3 : SUCCESS in 10m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26688827cae148dcbe5a620c5574e367 : SUCCESS in 6m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/508327a0bf2046b288b17b8c6f4962e5 : SUCCESS in 7m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e0a7f582838c4e3b8d9c38e1ee5ac6f8 : SUCCESS in 13m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/53899b5a7310462481441932e5b93d21 : SUCCESS in 33m 26s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a76548c1788f474c9c56226ae8e64b6c : FAILURE in 50m 19s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/ec296916647047f5a4ca1f1624bbe59d : SUCCESS in 1h 12m 25s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c761480c103a47cfac63001de5deb21d : SUCCESS in 44m 57s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a48e55cd2e164a0aa11325ea96a51fde : FAILURE in 1h 14m 08s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b8c77c6a031643a1b5e26d10fb14468e : FAILURE in 24m 36s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/07604f17cc14466cbb22514c4f9b574b : SUCCESS in 22m 36s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ac360543ab44687ba62b38aa8cf8a56 : SUCCESS in 17m 09s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4db49ac8430140489218c9f0bde009d6 : SUCCESS in 1h 50m 27s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ffae68f86004b72b093c639dc133f05 : SUCCESS in 1h 24m 24s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/973d9b8d919a4390bce26ad6c48ca642 : SUCCESS in 53m 05s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/f573eb6aaba34c5ebbd66b2dba3376e6 : SUCCESS in 47m 31s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/139bcd8838ae437586c115dec37afcab : SUCCESS in 41m 36s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cec2231c5ad14604bb1d26f4f7c1c926 : SUCCESS in 31m 42s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91ab803d911e4ffd99b51b87617059fd : SUCCESS in 36m 53s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/703130aa18164e6b9d187c0112113a9e : SUCCESS in 20m 01s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/0486070014ee47fa8c76cc095ba2c153 : SUCCESS in 16m 31s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63d785ef425e451f8095f37c6d75c38b : SUCCESS in 15m 32s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/63efc38856eb4942bc48c54173846770 : SUCCESS in 22m 26s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/85cb7042eab2413eb3287e0fbe72701e : SUCCESS in 19m 44s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f53890abcf1455ca329cfecdc217f40 : SUCCESS in 16m 46s","accounts_in_message":[],"_revision_number":22},{"id":"79c5fe5f7b53190ac8e9a43ea0a4193a1a597177","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-07 11:40:23.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"b9bc7bd0b253ec665926b4b9f9610f5de87d2778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 13:13:29.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3d6417b93de43e0bd91d29e1ffe3852 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5bb6f956837406e8d82a3b29ff7c27d : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/515b040fec0a4945b00b135c36a7b4f0 : SUCCESS in 8m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d7508486ec84e47b7281ea8d527d416 : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beb0452cbf0f419b868ececac0efc07b : SUCCESS in 10m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96117ec211eb484c92b730e5f1071779 : SUCCESS in 4m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/70c2fac2737446b797047188e3ef6907 : SUCCESS in 7m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/67c00bb3ced74a4b8356be71a5159e67 : SUCCESS in 15m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d9817eb65e544f57867a787a1df2d790 : SUCCESS in 35m 49s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8c9ba91ba466490e8dbe0c48b034ca84 : SUCCESS in 1h 05m 32s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/47463332077240f4a1664d99a9745177 : SUCCESS in 1h 07m 23s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11fc0e3cb1ca4990b1bec9d613148956 : SUCCESS in 56m 49s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7bab4a00bc314c47b5782ec9a9ee11c7 : FAILURE in 59m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/364d4c7f77cd4857917801e16c468ea0 : FAILURE in 22m 20s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/4cc2f26fa70f4f0c9c681e9ccb36972d : SUCCESS in 21m 33s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3095221f91154955aa779c7f42a813fe : SUCCESS in 17m 16s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ff3c67ffc39e42b8bf1f2fb1d70f17c8 : SUCCESS in 1h 29m 42s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/600ea1fd58d34c6dac5e6f9a4f8e790f : SUCCESS in 1h 15m 02s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e54cc5345ecb499cb9223c7622759cf9 : SUCCESS in 48m 08s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/3783d0aceda84b148765530e3d1fd760 : SUCCESS in 50m 10s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa902e58a7664a3bb6728a0379556967 : SUCCESS in 34m 00s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a37e991fd0734590a5cab6509c525aa2 : SUCCESS in 39m 16s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1952004a175942689e67282d19ff808b : SUCCESS in 28m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/215fe25cd5344b3e848ed850e8b490df : SUCCESS in 19m 40s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/0fa233341c384a54b6a97f4a384c964f : SUCCESS in 15m 18s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6275c59650c4471fb1af0faf33808af2 : SUCCESS in 15m 00s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/0e7e52152b2640da973827ec76951f03 : SUCCESS in 16m 12s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/88f23822daba411fa58ed558151d6e65 : SUCCESS in 17m 17s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32c0872d498c42c99dc79c16216b3950 : SUCCESS in 14m 19s","accounts_in_message":[],"_revision_number":22},{"id":"f53f5305ccedaa5b596b1d13bbe5848b6aa6000c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-07 13:49:13.000000000","message":"Patch Set 23: Published edit on patch set 22.","accounts_in_message":[],"_revision_number":23},{"id":"ab35ff000792f351f742f8f8d4ebd9687a2bf82c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-07 14:17:00.000000000","message":"Patch Set 24: Published edit on patch set 23.","accounts_in_message":[],"_revision_number":24},{"id":"6d2d11336dd67733edf1713471900e844726e01d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-07 14:20:09.000000000","message":"Patch Set 25: Published edit on patch set 24.","accounts_in_message":[],"_revision_number":25},{"id":"8d0885f04194a1a2ff8fc9088202fb38dee44546","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-07 14:23:53.000000000","message":"Patch Set 23: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":23},{"id":"e640a2f2d4eed7b0283bf0605c445dbdc3e815bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-07 14:56:26.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"90d138b5a99e11c11b7b9dcb281701a1dad5871a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-07 14:58:38.000000000","message":"Patch Set 26:\n\n(6 comments)","accounts_in_message":[],"_revision_number":26},{"id":"1912e23f8de58e57db9dd91ae84898b8b37531ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 16:40:33.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b547b110cce4ef393d07bc381853ef4 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04e356956ae048be8d405730b6535a87 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/356602a4440a4495a15d7e9db21865a6 : SUCCESS in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/608f1d52e8ab46fda070b9e52832940a : SUCCESS in 8m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4181ebcb94874bc99737ec5bf30fab7a : SUCCESS in 12m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41f2b269960847708ea79bd3d8e5189b : SUCCESS in 5m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/19aaa967fd71494f975941c89d4a5955 : SUCCESS in 8m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7cd8af8f95a54b6b81e9d0734f12f3b0 : SUCCESS in 15m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/79ae6304f62a4785914a33956aeb1981 : SUCCESS in 33m 55s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/707e515b180a43759aaf78157fb48931 : SUCCESS in 1h 10m 15s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/1ec1c93f2bec49b2a7d32115c663f554 : SUCCESS in 1h 07m 07s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9dfbe8e4eac4a22a7e26953cbebd745 : SUCCESS in 1h 04m 29s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a9a1229e65754b75a76c3386adcca89d : SUCCESS in 1h 13m 11s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/33f9d684f93c4f0eaa79a0d1fb323b79 : FAILURE in 13m 17s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/749141768f2c478aad60dc91a0e230c9 : FAILURE in 11m 44s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c78342c373694122a8c440fb91671a7a : FAILURE in 15m 28s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9a05003c65e4426ca149411991deb0c4 : SUCCESS in 1h 42m 34s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98e7a666b82d4f1fa1b5d442dc77aa4e : SUCCESS in 1h 31m 00s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/cd05ec1ade9f4e05bd13c85479c76475 : SUCCESS in 50m 16s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/42d6f45836ad4b278755c960a7ba5c40 : FAILURE in 43m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7411a3554c04174857d1e3ab37c21f6 : SUCCESS in 37m 42s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6cc246ffaf8c4591b6b62acf1ebb7060 : SUCCESS in 26m 29s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/772e6d0e84494750a0b00358c74602e0 : SUCCESS in 29m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8e09b4fe5f654bac872d1f72f37464ff : SUCCESS in 18m 23s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/abc35a792c7b487dabf91f1365f7115e : SUCCESS in 19m 32s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/521cdd308b4d4a898d354124ea83fbe4 : SUCCESS in 13m 40s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/6d08ef1884bf490babdbea65a7b22a91 : SUCCESS in 18m 54s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/b8c78d0f1b964cc397255248a85d77c0 : SUCCESS in 18m 38s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d084876acf904e7b84592b3484e172de : SUCCESS in 14m 24s","accounts_in_message":[],"_revision_number":26},{"id":"a5ebd53a276bf41b6b9c5a50213d9efbc0fdd094","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-08 08:04:54.000000000","message":"Patch Set 27: Published edit on patch set 26.","accounts_in_message":[],"_revision_number":27},{"id":"5714ab467baac940cbdae8157ff0dea44591f4ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-08 08:11:29.000000000","message":"Patch Set 28: Published edit on patch set 27.","accounts_in_message":[],"_revision_number":28},{"id":"148f714c3519c23bec3337e184d3f612abc8a8e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 10:31:49.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61bd4e4bbf6b4c9a98443746a15043ec : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/518880a53a14438082cd34a9a8f3f8e6 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70ebff7420e14d2bad50863db8ae5e02 : SUCCESS in 7m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5c92d0aa5d54f309408a27f9610cc23 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a8ce6158e6a486e86446ae7a4ceb471 : SUCCESS in 10m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8dfd2111ebe40b1bb87880196c4330b : SUCCESS in 4m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/39dcfa7df2de4dd0a71c3815fb149d3b : SUCCESS in 8m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/65962610b2e9495090c67ae3e638d022 : SUCCESS in 12m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ae30f93c03f942fe9f85bb6c6b76c373 : SUCCESS in 33m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8e33d09d22e94a22b5f2b0b0cce76543 : SUCCESS in 49m 50s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/a5b3c887bf1241b4a3291885cf00e003 : SUCCESS in 1h 12m 43s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12566a860b484aa199bfd046c5846ead : SUCCESS in 58m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/73285f13bc754752a283fc59b5ed32ce : FAILURE in 1h 18m 18s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8db4a887bee943bfa3a40bfd7225e2a9 : FAILURE in 23m 39s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/dcd8874be8554d689906802f261dd930 : FAILURE in 32m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04882cfac6a6483aa5b6a2a4c3b7c5fe : FAILURE in 16m 21s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f8cb6123f0cc47f382feeeac499199d5 : SUCCESS in 2h 14m 35s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ef01855d76448409c8db2c20a2cf6f5 : SUCCESS in 1h 10m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/19dd533e06c7442690487e9c14388b4d : SUCCESS in 54m 23s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/4d7afcf0b5ba456e85e07718bdc3e484 : FAILURE in 34m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2553bf8bf8d54b37a58153fd3117ca1c : SUCCESS in 35m 28s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/325cea20c6c14b978d38c1fc5171901f : SUCCESS in 29m 51s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb5e2ae560c74200b2fe7418fdc44883 : SUCCESS in 26m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ce8876fdcb5e4e0b834a6aea60c7afce : SUCCESS in 16m 41s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/90ff67feab2f4781bf316cfd22e40848 : SUCCESS in 15m 06s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4796cca0c7f742bcaff203ab33766f5d : SUCCESS in 13m 10s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/af55c709a989404d9edc33004adc881b : SUCCESS in 19m 37s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/ed52fab2d561417f8db0ca593a570997 : SUCCESS in 17m 39s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12e3c4d5f5594d92801f3d91cf01ad62 : SUCCESS in 14m 22s","accounts_in_message":[],"_revision_number":28},{"id":"f8c4c66e82e84fe39ab2b64014ee51e63dc65b58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-08 13:13:43.000000000","message":"Patch Set 29: Published edit on patch set 28.","accounts_in_message":[],"_revision_number":29},{"id":"fa3ffb0c80faaca9349564c4610ca1fd57ec2a2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 13:49:02.000000000","message":"Patch Set 29: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":29},{"id":"f74fde42e213b5164c29067ee0f14e30bbad1d4a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-08 13:50:25.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"9f985aa03633729e8e7c8f909aac0a4c29d09c71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-09 08:38:05.000000000","message":"Patch Set 30: Published edit on patch set 29.","accounts_in_message":[],"_revision_number":30},{"id":"00bc4fb69c752e79ecaff913bd6ca8bd2fd50718","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-09 08:44:30.000000000","message":"Patch Set 31: Published edit on patch set 30.","accounts_in_message":[],"_revision_number":31},{"id":"5cdba18c84d4795504617c818075560c800ac34a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 10:45:47.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4abe02e5318e489fad5c896096087beb : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c7041ecc9a54d5c96922f7a06002cf1 : SUCCESS in 9m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d67f5845ccf34cf383ca4386fa7a35fb : SUCCESS in 7m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c6b0b7899e349b5b545d3bec66af96b : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5d84202a46b454fbdb0425148dbbaca : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1047aabac9124b5584c9d7902f91f720 : SUCCESS in 4m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/079a5b8f084048c689a2d81cf15f773d : SUCCESS in 7m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c96b2df3decf4f46a5d3da78245afe70 : SUCCESS in 14m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0614d2dfa11c43908ccfae7e6a84433d : SUCCESS in 33m 55s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a5573235e77f426abb5ab15e6ef8c653 : SUCCESS in 1h 06m 48s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/641f1e6963a04d3fa21515712d774def : SUCCESS in 1h 24m 03s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b51cce57daa248e09688e40c9ac96428 : SUCCESS in 51m 23s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/765db0cf468844c9b585bffd64c05752 : SUCCESS in 1h 14m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/292a633fe5f6436fb407d0177aac4bf8 : POST_FAILURE in 21m 41s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/7c769a1b3e244dc6ad89fb00cfe9f99a : SUCCESS in 20m 21s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/510958869bbd425e9ef5af8d7885a63a : SUCCESS in 17m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ed01da508d9a465b8356642e798172ee : SUCCESS in 1h 54m 26s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d33f90e2789d41a89846d0c86be93a33 : SUCCESS in 1h 20m 43s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/10e92a44d5d5468480a8c7349f7bbae0 : SUCCESS in 45m 26s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/6a7c12d9e6bd4b0eafba015a322d861b : FAILURE in 35m 35s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/489e1f12dac74683a36cf9923d110199 : FAILURE in 1h 01m 54s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1aee9b70f5b74fcbbe0b643d53a1f3c2 : SUCCESS in 30m 10s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fbec8ca5a82453da70dc4113b9aebad : SUCCESS in 29m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/87ec6602de444058a4ddcd8f68835267 : SUCCESS in 16m 51s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d8c2af3e90f349e29061d5ad0423f957 : SUCCESS in 18m 24s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b400fb0c76b4426dbac58edbf220eff6 : SUCCESS in 14m 21s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/012b538afdf34e2abfc214a712b5aa03 : SUCCESS in 22m 50s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c1a2cd7c3fea46bda856bb0708ebf7d9 : SUCCESS in 19m 44s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86fa55cb45f94f09945d519332d18957 : SUCCESS in 16m 14s","accounts_in_message":[],"_revision_number":31},{"id":"e5e72f678e837473f69c25d4b3055ff776c2a09f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-09 10:54:50.000000000","message":"Patch Set 32: Published edit on patch set 31.","accounts_in_message":[],"_revision_number":32},{"id":"5d02b83f4f6e4add359534cff4c2e734647a826f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-09 11:27:44.000000000","message":"Patch Set 33: Published edit on patch set 32.","accounts_in_message":[],"_revision_number":33},{"id":"347a45cdf5c6ffabf783acfdcb7d010951bfabf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 13:24:09.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15c910bdc0564f7d951c3f6136681dbb : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26e3561ea12847cfa8a51efe711c13a5 : SUCCESS in 8m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1680507a77745d299f7ac2fa4c65370 : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26fd7906b33f4d798bea65f5dd649691 : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8761d6f1434344e69008346b07ecbb5a : SUCCESS in 10m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a5204ca0bb24057a8522512a140556b : SUCCESS in 5m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d259f930a68640a48332a483eb2c2cce : SUCCESS in 8m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3f1e2495034949958cdca930edbb561a : SUCCESS in 16m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8ce4f97bcb7b4b519e0c84ae90cb20a6 : SUCCESS in 34m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f96783f8d287435bb30d5cd500585b54 : SUCCESS in 1h 21m 14s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/b8387b726fcd43b19c37df0f6344db90 : SUCCESS in 1h 13m 07s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb3844df4cc746e89bf192be5b28a567 : SUCCESS in 1h 00m 03s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/747947636af64454813c01acc2f8a88d : SUCCESS in 1h 15m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/44c95abd59554528a61a99717faef2c4 : SUCCESS in 25m 05s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/c09171ad14304eceab5037f18973f218 : SUCCESS in 22m 53s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94523354fd3648a9a6d4bc6fb03c412d : SUCCESS in 17m 18s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/98cef8d7702a4a7ea951efc478f9cb42 : SUCCESS in 1h 54m 28s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a8f924c02f848009ccce7cc7d15b6ed : SUCCESS in 1h 33m 05s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/96e78f9fcbde47efac245c18d7ae692e : SUCCESS in 54m 02s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/40efc946afee411ca8041e49a21fb5c2 : FAILURE in 37m 09s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3e4790646dd459ba2fd11f3f5822253 : SUCCESS in 52m 29s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3e2a10775f4444b08fea84721bdc2233 : SUCCESS in 30m 10s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/510488692e8a46af9f23fe3b4cbd7861 : SUCCESS in 29m 26s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/df5abfba0c0043e3b57c254530873b8d : SUCCESS in 17m 45s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/9785fb0793dc4704b6e8359037c1f19d : SUCCESS in 20m 16s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13de1e0a7d9a4b238e27403ba41b53ca : SUCCESS in 14m 25s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/21193d0b87c34312b22c3477a8d34954 : SUCCESS in 20m 38s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/7ee51e6a4cbd40c6af3129609a30d53d : SUCCESS in 18m 59s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/deb02bd1e43e4ff29dc6684b08ccf663 : SUCCESS in 16m 07s","accounts_in_message":[],"_revision_number":33},{"id":"3397c7c5e4ce9850ef3a7ff46d3c8975e81aeffa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-15 08:15:06.000000000","message":"Patch Set 34: Published edit on patch set 33.","accounts_in_message":[],"_revision_number":34},{"id":"4a777890dbef016f579de6831a4675464a98e202","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-15 09:00:08.000000000","message":"Patch Set 35: Published edit on patch set 34.","accounts_in_message":[],"_revision_number":35},{"id":"6fd6f05c35bb6f9b8f02fd1c4a712588efada12e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-15 09:01:11.000000000","message":"Patch Set 36: Published edit on patch set 35.","accounts_in_message":[],"_revision_number":36},{"id":"86f6b4747074ce7c0bb319a55db44e34be352bd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-15 09:49:41.000000000","message":"Patch Set 37: Published edit on patch set 36.","accounts_in_message":[],"_revision_number":37},{"id":"bacff660efa5f6a1da2ad85b648e89261eb77e31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-15 10:32:58.000000000","message":"Patch Set 38: Published edit on patch set 37.","accounts_in_message":[],"_revision_number":38},{"id":"2cccec7524ea3f653d54147d926a0a4a5b193e34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 12:28:54.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92001d9cb61b4980810a0e4984691ba8 : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/660b02babfc640e4ba664d3a26c37254 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e15b6c982db4b1986ebc953857c6de7 : SUCCESS in 7m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c271524d618642919414ad83c587dbab : SUCCESS in 7m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/343386ceebbf46bab3a4bdcf17b6840d : SUCCESS in 10m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fd086d97843443985de1a7767e99dfd : SUCCESS in 5m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f7220d3b116e42089482976f3b0a70f1 : SUCCESS in 7m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7ef178e4a2274b15923c7aca0230f59f : SUCCESS in 14m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f8beaeefda1c4df285f075e121e9c75a : FAILURE in 8m 22s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/45bd68607c4446a5a3b6108555dd3a4a : SUCCESS in 1h 15m 51s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/a87584ebd9b746c2b10333e41d904bcc : SUCCESS in 1h 11m 22s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/795366c112cd4752bb97e59c21c8a610 : SUCCESS in 52m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/015835a1f4314c64bf457e28b68c1b29 : SUCCESS in 21m 51s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/fe63792a11604f52a0aff44936bfb988 : SUCCESS in 22m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6e20fbd22144b429b11997753fe6a68 : SUCCESS in 15m 27s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/77975dff308140898947c90a5ec56493 : SUCCESS in 1h 54m 28s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d368718d8e049638f110a8a61cba33f : SUCCESS in 1h 12m 20s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/fed8bd694d8f47388238c3eb582db9e2 : SUCCESS in 46m 32s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/8f4aca0f900348c48399c8690fc92691 : FAILURE in 36m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ebd54de219d46f29d93be463a08b971 : SUCCESS in 41m 58s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c1d56c8ccd134d8db3dcd3019ee6290f : SUCCESS in 29m 25s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/defd512f50444ddc968072544df329fd : SUCCESS in 26m 10s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/af74990ce8724afcb11ea241bb243366 : SUCCESS in 19m 11s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2ac92d77747e41d58ddce6fe0a414a97 : SUCCESS in 15m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82cc9987d1fc4119b98e133d67b8f2fa : SUCCESS in 13m 11s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/476161ec201c4ccd99ccf7f618ed2dfd : SUCCESS in 20m 04s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c2267d1557b04ecbb439f4a7139ba025 : SUCCESS in 17m 31s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23387be60bbd473c84077959162f75ab : SUCCESS in 15m 23s","accounts_in_message":[],"_revision_number":38},{"id":"c35223136d3391dfdcbf5902d719bdaf8bc11864","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-15 13:31:17.000000000","message":"Patch Set 39: Published edit on patch set 38.","accounts_in_message":[],"_revision_number":39},{"id":"cf9291576080b8f540a2695eeaf88aecce2ece11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 15:32:15.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/063fc51ca5224a12a640e4b0f20fa5ef : SUCCESS in 8m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbed74954c2b4f7eb7b784519cfea065 : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29479e927b244a8b883f5bcc4cd8494b : SUCCESS in 8m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e521e66b025e4c66b3f820669a3b931d : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4f6e08a4a914120958c5798b573499f : SUCCESS in 11m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab8e1db62a6741af9ffe4a728cb8cea2 : SUCCESS in 5m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/be020a53a4824177b01fc42af57a5329 : SUCCESS in 9m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0e2ce7602b0f4ebca5ce78bd88c84ad2 : SUCCESS in 16m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4dc5bf7e3752450198f9bc4d2032dba3 : FAILURE in 9m 14s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2c7142681aa9415f840ad94e822c6140 : SUCCESS in 1h 04m 08s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/b438674a0c484e3fb9d9db43cd3977d4 : SUCCESS in 1h 03m 37s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e133c99b1e0c4760bf14e73d5c9d8fd8 : SUCCESS in 51m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5b0d5a933c9841c4bc490f400e846968 : SUCCESS in 22m 07s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/7f9acaacb5804e4790be7665b7e2c0cd : SUCCESS in 21m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5da3e1589d3e4dd9a3f68edb533fb860 : SUCCESS in 16m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/11ce731a8dec48a0b7600dd34d9a0c90 : SUCCESS in 1h 48m 31s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fd60c0527af42f48c8f9469c69f7888 : SUCCESS in 1h 15m 08s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3ac8dd8f19684e3eb928da6000c87b01 : SUCCESS in 49m 43s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/4357b4500fa843a9a8bb7454d37947f2 : FAILURE in 41m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09bfbb1cd20f48859c202a3be399bb43 : SUCCESS in 1h 01m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/42792c0e71e249239ab192c8676128c6 : SUCCESS in 32m 04s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90cd94fd66124804988710c2ca0651a1 : SUCCESS in 31m 04s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ec6680388a444dab8f4d0186ac835e5e : SUCCESS in 17m 15s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/7f8181c9920447c2a2c942dac0a3a16f : SUCCESS in 15m 27s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd772db614014be785e20c93f1e349e8 : SUCCESS in 15m 46s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c279542ecfae46ef8577ee8a06b6e34b : SUCCESS in 20m 29s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/250fd9a3fe1643bd9b9ed3b51c69b94a : SUCCESS in 19m 31s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82e0317bccc249a991fca82db0357d09 : SUCCESS in 15m 49s","accounts_in_message":[],"_revision_number":39},{"id":"483ed613ca0ecf2a7e3c7c86aaba5d98508f0b0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-17 08:20:15.000000000","message":"Patch Set 40: Published edit on patch set 39.","accounts_in_message":[],"_revision_number":40},{"id":"3b6c995958d56249636ec34cf2248ff400bb4def","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-03-17 09:46:15.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"f95240a72a378cdb8a55459b21c63c1e38f6821f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-17 10:10:07.000000000","message":"Patch Set 41: Published edit on patch set 40.","accounts_in_message":[],"_revision_number":41},{"id":"b05e894b6c602ac25a66451db777cd666e1e7f7e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-17 10:16:27.000000000","message":"Patch Set 41: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":41},{"id":"99aab95bf2539058c274c560ae1ab1fe3a512711","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-17 10:27:33.000000000","message":"Patch Set 41:\n\n(5 comments)","accounts_in_message":[],"_revision_number":41},{"id":"6d246418a00c88955d706889f52fe2f758837c28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-17 11:04:50.000000000","message":"Uploaded patch set 42.","accounts_in_message":[],"_revision_number":42},{"id":"591f6d3740d64682fc871121334fa0193b68bcff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-17 11:11:27.000000000","message":"Uploaded patch set 43.","accounts_in_message":[],"_revision_number":43},{"id":"8211fac6d2f5f9315a1958e8f8d736a332021309","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-17 12:11:14.000000000","message":"Patch Set 43: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":43},{"id":"4f11d217c73169a511a7fa331c7de43d6bfce874","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 13:08:22.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e64d1ec2ccf4c99ac84c585f0c35db0 : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae3781daf17c456bbaf65eca65bd635f : FAILURE in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e7ff5a3d52243908ed133933ba28d97 : SUCCESS in 7m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/474e2d1cb6ae4971b52da8bd21454111 : SUCCESS in 9m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7638c63ce06c438a8e3542651aee228e : FAILURE in 10m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9518914566f42b0844a35f23649ec9b : SUCCESS in 6m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5efc185f2de3477cb213a2bf28f3426d : SUCCESS in 8m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c671383aae39471c82747a1f904026e2 : SUCCESS in 13m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8946e324874f4fe9bc72115621572d1b : SUCCESS in 35m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1071bcbabd3a47aeaeb41f000ce5647a : SUCCESS in 1h 11m 42s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/fd14a7c8696946cfa5203b06345b9190 : SUCCESS in 1h 07m 23s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ba8db54340541c89ef8428ce04b8e74 : SUCCESS in 59m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b15f75b3cbc34e97b86a8fd9f27b8a95 : SUCCESS in 1h 43m 54s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bd1ca5de6b8a490abd8a3b04efbf6af0 : SUCCESS in 24m 06s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/77b675b9eba74576b07d8a30f639f5dd : SUCCESS in 20m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd075dca78d54a4983c4fd674b9f11e4 : SUCCESS in 19m 58s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7e16129e089e49feac7fa957168124a0 : SUCCESS in 1h 52m 43s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba9a1251879d4631a09a1477b38d3afd : SUCCESS in 1h 15m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/192a2ddadd2344459471d223da38e8ff : SUCCESS in 49m 08s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/12170bed9d234851998cc507cbb9515d : SUCCESS in 40m 16s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e4d1ecdba9541c2973b295caecd0cfb : SUCCESS in 41m 51s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/979a4906836748fd9c7bba09e6e9eef1 : SUCCESS in 41m 19s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1380f93868734824a67c1d0444d8b3d3 : SUCCESS in 33m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/fdff4ad554dc40a095fea206fbb1265f : SUCCESS in 17m 58s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/1abd15e515504bddbe7d571d801cf5e5 : FAILURE in 16m 07s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f5ef109173a497abb2b5fe59495f273 : SUCCESS in 14m 55s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/065d86b1cd3c47f98f4fa681e2ab72fb : SUCCESS in 21m 07s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/cabb385e8b4641a199f624077cad1497 : SUCCESS in 17m 31s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8cf6b26c91f46e58e204f47d86b3851 : SUCCESS in 16m 24s","accounts_in_message":[],"_revision_number":43},{"id":"6869db0765ab17df2995ba06eb5e12e75c2099b2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-17 13:17:02.000000000","message":"Patch Set 43:\n\n(1 comment)","accounts_in_message":[],"_revision_number":43},{"id":"e73c4abccd76083c1e9d611348c70f04ca3e4a29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-17 16:10:55.000000000","message":"Uploaded patch set 44.","accounts_in_message":[],"_revision_number":44},{"id":"95a69ad49c19a948e10146428f67206f0f99032f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 17:57:42.000000000","message":"Patch Set 44: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71c2aeb25fc4453781db85e28817105c : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20016900fc9243efbd24a9db28b5b9d0 : FAILURE in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3090d018ecd4dbca1a599bec9307dd4 : SUCCESS in 7m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67d4b5710fed465cbebddeb5f42e6c24 : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bee91254bbc4f558dbab98b9abcb4b3 : SUCCESS in 10m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5df5b1e154634e3ab923cb3b72f2d25c : SUCCESS in 5m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/10b4eb747a6c4a40aa4dc871213ff7b0 : SUCCESS in 7m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d2a21bdca8564a16bf7c7c4b92467b8a : SUCCESS in 16m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1c1d5fae05684576bcc96ba6c441b74f : SUCCESS in 31m 43s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8f71284f5bb14742ae409df14b013d9a : SUCCESS in 1h 04m 27s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/2fce4aa4d3a145c68c23422149019814 : SUCCESS in 1h 04m 05s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6c9a04b1cc54cbd9290a42107420f60 : SUCCESS in 53m 21s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d46254b01f8442018f0148700e5b39bf : SUCCESS in 1h 16m 33s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b774054d581a4d35b2e304e421eca5ba : SUCCESS in 22m 00s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/e3166583ea2949e1a24e13842b27cd44 : SUCCESS in 20m 18s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68c6f13c4ac34eac850e0e716d976fba : SUCCESS in 15m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d6f4cd24be2f4050b5e5ad2b52b60fee : SUCCESS in 1h 41m 50s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59870f06c78c41e4a7f19ef14184b511 : SUCCESS in 1h 11m 54s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/05be42f7e32c4b8d8b2da7aed164d770 : SUCCESS in 47m 04s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/5e1da9ea52bc4acf833adbcbd18d7502 : SUCCESS in 39m 18s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ae61526f5694a828caa5176c1fb7997 : SUCCESS in 36m 01s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a51704a5259a4b65bd02dfaf31884c3c : SUCCESS in 38m 39s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b824a4eb3fc944df88642e78924cbede : SUCCESS in 31m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/127a28dbbc7f45369f7dfd051f156d94 : SUCCESS in 16m 33s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/94c6790c89df44a4b47d67cda3c8db62 : SUCCESS in 15m 36s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83a3e1275db248b0bafe08b204248e5c : SUCCESS in 13m 08s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/f23f6a135c2748a7aa891412d32e947d : SUCCESS in 19m 07s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/76c4b60af8664451ae0003dfffd3c78c : SUCCESS in 16m 47s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a2223640d9f456cbc435d533dc12c1d : SUCCESS in 15m 54s","accounts_in_message":[],"_revision_number":44},{"id":"85bee168daa887dfc353d5a4a59310773ca231a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-17 18:02:27.000000000","message":"Uploaded patch set 45.","accounts_in_message":[],"_revision_number":45},{"id":"6d1cfa2eaff1c549afd96b2e894f7866d08f89cf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-17 19:25:22.000000000","message":"Patch Set 45:\n\n(4 comments)","accounts_in_message":[],"_revision_number":45},{"id":"60ad702bc16f2815562f90bfaa26156a9560d264","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 19:56:50.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56d9e6da6c7e49248460793c10c1ca78 : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e935af1101db417992ec89e912f63f5a : SUCCESS in 8m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a465bb064a7d44539f009eb3b6efa7f5 : SUCCESS in 8m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab43d047d76b4baf8ba1f930813c43be : SUCCESS in 8m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f99a0d4118d34c5da67399c0908ba664 : SUCCESS in 10m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe279fd47cbb4c1ca3c715be55023e14 : SUCCESS in 5m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/223e32d807064bbeb7a51f4da6b0af27 : SUCCESS in 7m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/accceba83cff40b1859742abc75b2850 : SUCCESS in 16m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dd426f498124439a8de4c49c15bdbca4 : SUCCESS in 31m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/dbbf4e9c8cd44c34a7d98ca287ac66f7 : SUCCESS in 1h 10m 30s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/ea44683a4e124a379de8f12ede50d20b : SUCCESS in 1h 07m 36s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e38eb0cf74df42009c48dec1f99352a3 : SUCCESS in 57m 23s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9d5135c69fb2488788a7091873e2523d : SUCCESS in 1h 08m 06s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ea1b8c8383da428e81118e1b80e6230b : SUCCESS in 24m 50s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/9f8a8a319d5342ad8f2e15e84e9ccc8a : SUCCESS in 24m 04s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c80359b7aaf43739bf544579e446a1f : SUCCESS in 18m 31s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3bff938ee7c14e82a9186daa682b0ab1 : SUCCESS in 1h 51m 53s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b93664450c8842e487e91d09ab025763 : SUCCESS in 1h 23m 57s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/be00268bc2b54194b7e83ba25f664a7b : SUCCESS in 49m 44s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/a0e7ff50d90541e091f4d31e8a700d8f : SUCCESS in 43m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94010200244a49c4a20fe1ccea1feb76 : SUCCESS in 38m 46s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2c4fc39d9bdb4ba59baa642b4a62e965 : SUCCESS in 42m 09s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3d96b6e5eb54e22a58910395926975f : SUCCESS in 36m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d5db2431fb684fc591d3a9d97a51d395 : SUCCESS in 19m 14s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/758446f26fc04035addfed15065c70b4 : SUCCESS in 16m 22s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24262943e1e74955b3dbdbc1eccd5ad7 : SUCCESS in 13m 31s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/d5daa71fd8204f9eb264ef2e15fe9c04 : SUCCESS in 20m 10s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2df00314fbce4a06be4ee9379c415dd9 : SUCCESS in 18m 10s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/041086ea7a224d0e9391b5a8d3356db1 : SUCCESS in 15m 30s","accounts_in_message":[],"_revision_number":45},{"id":"695445e08065637707891dd0b7dbdb7c2589e927","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-18 08:54:27.000000000","message":"Patch Set 46: Published edit on patch set 45.","accounts_in_message":[],"_revision_number":46},{"id":"650ca9447472fb25e974722e69ffbef5e5ab0ad5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-18 10:25:07.000000000","message":"Patch Set 46:\n\n(5 comments)","accounts_in_message":[],"_revision_number":46},{"id":"292d62ddaf98d5b1bdd776bc6eeaed0b03e7ca3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-18 10:26:45.000000000","message":"Patch Set 47: Published edit on patch set 46.","accounts_in_message":[],"_revision_number":47},{"id":"6eeb473f5c517161867856e1004e4cce28c1d1f8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-18 10:27:01.000000000","message":"Patch Set 47:\n\n(1 comment)","accounts_in_message":[],"_revision_number":47},{"id":"1d979107f25867406850c7964131a62cb3368f35","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-18 10:44:01.000000000","message":"Patch Set 47: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":47},{"id":"0f4c69ae2f9fef603b33f4ca7174958b2c0a87c4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-18 10:48:49.000000000","message":"Patch Set 47: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":47},{"id":"3945688fddc4b5b4c1080c2036f547a295ce8f43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-18 11:47:02.000000000","message":"Patch Set 48: Published edit on patch set 47.","accounts_in_message":[],"_revision_number":48},{"id":"410b42de696707965c6cd7b7552bb866a29b5732","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-18 11:47:47.000000000","message":"Patch Set 49: Published edit on patch set 48.","accounts_in_message":[],"_revision_number":49},{"id":"84a857f82eb3fc3e7415bb2262bb7cbac114f8cc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-18 11:48:02.000000000","message":"Patch Set 49:\n\n(10 comments)","accounts_in_message":[],"_revision_number":49},{"id":"92d6c49fffde991841551785a7db86273d7052a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-18 11:51:02.000000000","message":"Patch Set 50: Published edit on patch set 49.","accounts_in_message":[],"_revision_number":50},{"id":"d767380e78bc27e4cc9064c6c3830b0c650b5475","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-18 11:51:18.000000000","message":"Patch Set 50:\n\n(1 comment)","accounts_in_message":[],"_revision_number":50},{"id":"97e45433f132bf18f847a9199621cbfa4b47fc98","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-18 13:13:21.000000000","message":"Patch Set 50:\n\n(2 comments)","accounts_in_message":[],"_revision_number":50},{"id":"c9c40692b30da61aaa0d2faba1de125c1188a92e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-18 13:14:58.000000000","message":"Uploaded patch set 51.","accounts_in_message":[],"_revision_number":51},{"id":"7042e0f6b61ff14da98a0600ec8c3f12b3213134","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-18 13:14:59.000000000","message":"Patch Set 50:\n\n(3 comments)","accounts_in_message":[],"_revision_number":50},{"id":"3e50ad56a7357eb96f9ec4fb9437b18dba62284a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-18 13:15:36.000000000","message":"Patch Set 51: Code-Review+2","accounts_in_message":[],"_revision_number":51},{"id":"9234ae612c4e26e056b8b376363f54fb772ac19e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-18 13:52:58.000000000","message":"Patch Set 52: Published edit on patch set 51.","accounts_in_message":[],"_revision_number":52},{"id":"c3da2459cf63b5f3f8900fb643733e2eb8289ec9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-18 13:56:38.000000000","message":"Patch Set 52: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":52},{"id":"2239ff0144bca8068dc80fc9ef6d59b79d9360e4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-18 14:57:38.000000000","message":"Patch Set 52:\n\n(1 comment)","accounts_in_message":[],"_revision_number":52},{"id":"202eb4e0577477fe5d7f3a926621ce7ad15453ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-18 15:04:21.000000000","message":"Patch Set 53: Published edit on patch set 52.","accounts_in_message":[],"_revision_number":53},{"id":"459767db948f40cda8b7fad0ced86ac2b303a395","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 17:08:32.000000000","message":"Patch Set 53: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1514d58da33481e87d2740ea0a200c2 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21d457dca8724cde898f66cd5f293af1 : SUCCESS in 8m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4798b080e13d421faf2d489230f4eb5e : SUCCESS in 8m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91bbc42cb4c6403793337779bc2b327a : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f13194472584f07b7dca3211c36dfdb : SUCCESS in 12m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48a5a83f69034a3c988383115a4744cc : SUCCESS in 5m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2c9d0f06e4344207af7e035d4cc9f6df : SUCCESS in 9m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f1b16c8ca8844088b626133308739bb2 : SUCCESS in 15m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/02a3dfc08f304911bc59a30acb697a6e : SUCCESS in 34m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7084ca1ffeb84b028065443eb9854c26 : SUCCESS in 1h 03m 00s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/765fef52eb1444d98ce8fd57a80bf3a2 : SUCCESS in 1h 06m 08s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea768af3f83f4d228b30d4ca53780e73 : SUCCESS in 59m 17s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b73d7557b2c140938fa98637e2f9ebbe : FAILURE in 2h 01m 47s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/621a99b6c15541feb645319061c77e7e : SUCCESS in 26m 32s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/ba51fbb3253a492e882ded276dab5c6b : SUCCESS in 21m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5efe96f2a6fa4084b937bf3d8957b24f : SUCCESS in 19m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d8378a3530254672846f0dbdb463848d : SUCCESS in 1h 42m 17s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be135d9f1d8f463ab96011149b0b5755 : SUCCESS in 1h 22m 06s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d8b670743a6f476783e08432d5e55518 : SUCCESS in 54m 20s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/3ee41b864c9e4feababe2bcffb2deec2 : SUCCESS in 42m 18s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b993503d51c64f99be7751bf5704aadb : SUCCESS in 36m 16s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3a7e63836db44746816f78567630ac7e : SUCCESS in 39m 42s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c6e4d1685ed4472ba19c2a675fc78f0 : SUCCESS in 36m 05s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/29cdc8b0296a4077b2c1458c828b74c0 : SUCCESS in 18m 31s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/916608add1c946589c7c37fb86f62bec : SUCCESS in 14m 46s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f0a5b6f70714cca8082b9b99857342e : SUCCESS in 14m 43s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/af2a32ae0c014e819a1b327afb2a552f : SUCCESS in 20m 43s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/8b22bf639d13490f843ac17b22ac9bc4 : SUCCESS in 18m 08s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c2b710e679b4792ba817e42e13f36be : SUCCESS in 16m 46s","accounts_in_message":[],"_revision_number":53},{"id":"cee0af82d830d02eb416b40823e6731001d3b83a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-18 17:14:36.000000000","message":"Patch Set 53: Code-Review+2","accounts_in_message":[],"_revision_number":53},{"id":"0c4d8cfe1639e67d0a7a67ba6c16bee20ea33fa5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-18 17:25:11.000000000","message":"Patch Set 53:\n\n(1 comment)","accounts_in_message":[],"_revision_number":53},{"id":"31e7826ca3d7c07bc988313cda951a5b2c51f1e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 19:37:35.000000000","message":"Patch Set 53:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d9942bf6c524573b6b11e2e0bc1e16f : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e0057e70ffc45f7bf83f803d9f97a56 : SUCCESS in 9m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c902b822dcdb49aaafacdd50b7bf647c : SUCCESS in 9m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f91026d1cc974a7a8328660fcc370822 : SUCCESS in 9m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23cabe8cd09a4e6e9c92b76ae7ee06f7 : SUCCESS in 12m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99df83c78dc040bb97238b157f1c7bd1 : SUCCESS in 7m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/515d606acf304b0e982c7f798c64b8ff : SUCCESS in 9m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/875835c800b547d1b939de445938c079 : SUCCESS in 16m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e1535164e4574bc8966924c746e731a2 : SUCCESS in 36m 05s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/631fcc130dcf461a9af9b8164f98fb8f : SUCCESS in 1h 25m 50s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/304a58d28f554be29c7b163de75b4766 : SUCCESS in 1h 08m 05s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9db47494b49b498595bb6d18287f15e8 : SUCCESS in 56m 00s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/98bd39099f484b6b96dc4ab7b3819992 : FAILURE in 2h 04m 28s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1be3905759c6473f9f6e3ddf3cce49c4 : SUCCESS in 26m 37s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/7b7b1dda86274ca1930cd8fbc05d8c89 : SUCCESS in 23m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e892833ddd914eb39af72ec308a4b04b : SUCCESS in 17m 51s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c8bbe4af8f7f42fdb77f4b9c2e7970e7 : SUCCESS in 1h 53m 28s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24608e6037d7417a9b2a83c647c77b9e : SUCCESS in 1h 20m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/eb0f33de6b4049d6af4c9ed0e8ce6d87 : SUCCESS in 1h 03m 55s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/a3ebb3c2871440fd93e68f1dbc040ed0 : SUCCESS in 47m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecb333eb281d457d99d31462b819930a : SUCCESS in 46m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/21e3a13cbea746b98daadb7d9afeab83 : SUCCESS in 45m 36s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51cd863aa7334c25a78cf4da4484ca51 : SUCCESS in 34m 49s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d8b459d1b34d432591c58edc47dcf76f : SUCCESS in 20m 51s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f0617949ab5649a1abe9ebf021127edd : SUCCESS in 17m 18s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85b6b778b3bf4f0aa4c739990e66aba0 : SUCCESS in 16m 29s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/20997e8ae41445d6bb21077d814dea58 : SUCCESS in 22m 04s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/84d4ce5559e941b4ae30e570f5190f13 : SUCCESS in 20m 26s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e83d6f7f51334b5bbd9c64f229d86a4b : SUCCESS in 17m 55s","accounts_in_message":[],"_revision_number":53},{"id":"a1cde35be8fbacbf3980713a17e658ccda7ff6c5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-19 21:36:08.000000000","message":"Patch Set 53:\n\n(1 comment)","accounts_in_message":[],"_revision_number":53},{"id":"64a75136f7593192391e44da9dc97d192184f242","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 22:59:18.000000000","message":"Patch Set 53: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a954ef4ec72c46ada9adfcd3719e3271 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a15156720944c259c75578c93643097 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f092833bb3944961a54f9956f0df6d9d : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b0b70a44eb3432ab8e0b8c0d212c7e2 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75cbeadd5d534ff582edc4af9f6e2f1c : SUCCESS in 9m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a140494c737640b4b1d8e6c98df77108 : SUCCESS in 5m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/60b735a631874695b9a9b1d1639dce50 : SUCCESS in 6m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/04aabb197b544bef9e10cfad17a967b0 : SUCCESS in 12m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/499ecb10d98145ee8e574ef000a41c98 : SUCCESS in 26m 48s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/51e8bca444874de29f9b608f45d8b750 : SUCCESS in 52m 19s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/7220c04f46b64238ae9f4fd18e560d63 : SUCCESS in 1h 06m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3f068a566624c7e9ad3356054d485cd : SUCCESS in 43m 12s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5df72324a44a4a81a6338e6909dd12fe : SUCCESS in 52m 26s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f2fc10e1d8bd42899e6f49f4f18d780a : SUCCESS in 20m 32s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/15d9aafbd0ce48fe83f9d4a0a32736f3 : SUCCESS in 22m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f52c3be7922247d3922c9a809cad8817 : SUCCESS in 14m 59s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a454bf51a0ec4472bb872902955d9511 : SUCCESS in 1h 21m 32s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e50d7cfedb8e4cb8820744ee34b7e4b7 : SUCCESS in 57m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/71540f265f254eb6898635d3226c69ef : SUCCESS in 39m 24s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/831d38d9ba104df3b7df335cfbfdd570 : SUCCESS in 43m 06s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28185264ce5944dab5f480fd67fa7746 : SUCCESS in 30m 29s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bf2600e3385e4fd6b6871a4ab536dedb : SUCCESS in 34m 36s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c9085c9ddf348b9a827e03dbef01474 : SUCCESS in 26m 49s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3db97d49e8494fc9992f1b387f572821 : SUCCESS in 15m 44s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5897029b66c44ddabb79eca0f971fd99 : SUCCESS in 16m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/155556c17a064e1f88692d391a7aebe2 : SUCCESS in 12m 10s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/65dbfb82e0714e37be85bfccceb0da50 : SUCCESS in 17m 27s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/517723928722489bbb681c13964b09fc : SUCCESS in 17m 32s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8c9eb5f7ebe4e70a5c93292a14847c2 : SUCCESS in 13m 17s","accounts_in_message":[],"_revision_number":53},{"id":"b2e1272bacd8197064f312e8037f25e4cf8b2157","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-20 20:33:37.000000000","message":"Patch Set 53: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":53},{"id":"bcb069f9a165bd8bcd4fea59a572ef35e878c341","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-20 20:34:14.000000000","message":"Patch Set 53: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":53},{"id":"8aad1ea41789f5ce28a3cbfb5821137a8ce3b9ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-20 22:20:24.000000000","message":"Patch Set 53: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df2d828b89f54363b2b808186596108e : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e0f09c0f5124b57bf7e346a2a579e0c : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ede5b9d31e8493a96254607f62069e7 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae758340d9fe4b77994aec60336cc82a : SUCCESS in 10m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f859106f7f0542da9cc9ba808a124723 : SUCCESS in 5m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/75b901cfa9804cf4b9b7b7fa9c79ceaa : SUCCESS in 8m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a5ffdc54a90e45dea1c32a051040de93 : SUCCESS in 11m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aa70a2f587704d32b1b9bea82272d40a : SUCCESS in 25m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c312b01b198642a5a8a8c423212e669d : SUCCESS in 51m 36s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/a90fd169db494a2ca957c3109474d944 : SUCCESS in 1h 05m 13s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9949faa4ffe45bf8aaaa6d7d072b285 : SUCCESS in 42m 46s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/462b0d4276524e0fa2b80cd4b29d202a : SUCCESS in 1h 01m 46s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/21b540a0d7ac4d84a7a1ff72276d757a : SUCCESS in 19m 55s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b3ac8b75a6ee463da47387399b1c9f8d : SUCCESS in 21m 36s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12bfb5f4a126417b94b42f7b576d116d : SUCCESS in 14m 28s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/eb9c73cce31048af876a8363288ec7eb : SUCCESS in 1h 44m 03s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d0c3de834004a2f8e9eaed1c27bfb9e : SUCCESS in 57m 35s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/24c84955ad884b8393b42a98b9f9dd97 : SUCCESS in 50m 28s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/11eb5b8db1c14957a402692226f75363 : SUCCESS in 40m 04s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce7e96a46b7e4d17adc5499856eb5c32 : SUCCESS in 31m 36s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5bbf81c294554217be75de2f6412175c : SUCCESS in 32m 17s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b565b7a542f47628458074d52940d53 : SUCCESS in 32m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d57c63ac473643e190672ebc8cad6c85 : SUCCESS in 14m 43s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/7cf322fb281446eaa82877a7fd0b94d1 : SUCCESS in 16m 52s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3371ba68415427ab415d1d40e9cb6c6 : SUCCESS in 12m 56s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/b2340b45a74e474a9dcda5a8111135fc : SUCCESS in 17m 28s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/b70e13e39ba049e4bd7062f1f4df8914 : SUCCESS in 20m 35s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd7a0a665fa44159a83c4766b6f766f6 : SUCCESS in 12m 27s","accounts_in_message":[],"_revision_number":53},{"id":"207b8b5de103dfe839dd0a8cd8a4098a400c29bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-20 22:20:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":53},{"id":"5706febfc9089e0aacfc177207dd112f44a898ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-20 22:25:48.000000000","message":"Patch Set 53:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/771db3a82f4c4d269a4c8e5b8523be76 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/704414b7e19949329fd1a5c50b4e7bc7 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":53}],"current_revision_number":53,"current_revision":"8e55ea08a474d3138fed160129acb8322f9eeae9","revisions":{"b7945123fcb56ba417f48f578bb38291229c93e5":{"kind":"REWORK","_number":1,"created":"2022-03-03 10:27:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/1"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:26.000000000","tz":60},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b7945123fcb56ba417f48f578bb38291229c93e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b7945123fcb56ba417f48f578bb38291229c93e5"}]},"branch":"refs/heads/master"},"e17079d50f43423be5b3a2615e5c1bc2e404c9e3":{"kind":"REWORK","_number":2,"created":"2022-03-03 10:39:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/2"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:39:24.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e17079d50f43423be5b3a2615e5c1bc2e404c9e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e17079d50f43423be5b3a2615e5c1bc2e404c9e3"}]},"branch":"refs/heads/master"},"966fe6a513bc74d091e2798f1e9fb8e28ff6116a":{"kind":"REWORK","_number":3,"created":"2022-03-03 10:39:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/3"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:39:51.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/966fe6a513bc74d091e2798f1e9fb8e28ff6116a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/966fe6a513bc74d091e2798f1e9fb8e28ff6116a"}]},"branch":"refs/heads/master"},"fcb4ce8fe83281ba6e288050c58876d4544debf0":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-03-03 11:23:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/4"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 11:23:02.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-collection-kolla/+/831642\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fcb4ce8fe83281ba6e288050c58876d4544debf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fcb4ce8fe83281ba6e288050c58876d4544debf0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c70b0e8e38877e231febaff9ec888dd417839af4":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-03-03 13:50:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/5"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 13:50:46.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/815104\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c70b0e8e38877e231febaff9ec888dd417839af4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c70b0e8e38877e231febaff9ec888dd417839af4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c9a4cd06e1a62e296dadadf4d70d4c8b5e9d0a7a":{"kind":"REWORK","_number":6,"created":"2022-03-03 14:34:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/6"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 14:34:45.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/815104\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9a4cd06e1a62e296dadadf4d70d4c8b5e9d0a7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9a4cd06e1a62e296dadadf4d70d4c8b5e9d0a7a"}]},"branch":"refs/heads/master"},"b23d773045abfd326c42ae1fe4f610c8274e71f7":{"kind":"REWORK","_number":7,"created":"2022-03-03 14:52:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/7"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 14:52:48.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/815104\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/831788\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b23d773045abfd326c42ae1fe4f610c8274e71f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b23d773045abfd326c42ae1fe4f610c8274e71f7"}]},"branch":"refs/heads/master"},"f1dda61ec52c63f2443a141f2e54aa6cc0f19940":{"kind":"REWORK","_number":8,"created":"2022-03-04 07:28:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/8"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-04 07:28:01.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/815104\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/831788\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f1dda61ec52c63f2443a141f2e54aa6cc0f19940"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f1dda61ec52c63f2443a141f2e54aa6cc0f19940"}]},"branch":"refs/heads/master"},"7d7db1319b8b1ca58cd7d9902733c4ab5edbdadd":{"kind":"REWORK","_number":9,"created":"2022-03-04 09:31:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/9"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-04 09:31:32.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/815104\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/831788\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7d7db1319b8b1ca58cd7d9902733c4ab5edbdadd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7d7db1319b8b1ca58cd7d9902733c4ab5edbdadd"}]},"branch":"refs/heads/master"},"987d58eab693c7910d409d10f18c3dd86481cdd2":{"kind":"REWORK","_number":10,"created":"2022-03-04 09:33:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/10"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-04 09:33:48.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/815104\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/831788\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/987d58eab693c7910d409d10f18c3dd86481cdd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/987d58eab693c7910d409d10f18c3dd86481cdd2"}]},"branch":"refs/heads/master"},"471c9af40ee7b5cd3a9e51296bdc9e5250e0f7cd":{"kind":"REWORK","_number":11,"created":"2022-03-04 09:59:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/11"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-04 09:59:29.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/815104\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/831788\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/471c9af40ee7b5cd3a9e51296bdc9e5250e0f7cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/471c9af40ee7b5cd3a9e51296bdc9e5250e0f7cd"}]},"branch":"refs/heads/master"},"9515e2d000695d4201292eeb55e9318bd7d9e1a0":{"kind":"REWORK","_number":12,"created":"2022-03-04 10:00:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/12"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-04 10:00:51.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/815104\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/831788\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9515e2d000695d4201292eeb55e9318bd7d9e1a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9515e2d000695d4201292eeb55e9318bd7d9e1a0"}]},"branch":"refs/heads/master"},"eba609f49baf3ba7286bb5d090b6e02a939a45d2":{"kind":"REWORK","_number":13,"created":"2022-03-04 11:22:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/13"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-04 11:22:05.000000000","tz":60},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/815104\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/831788\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eba609f49baf3ba7286bb5d090b6e02a939a45d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eba609f49baf3ba7286bb5d090b6e02a939a45d2"}]},"branch":"refs/heads/master"},"d47b4cbc9640133f319ad35a2e9307a1de09a727":{"kind":"REWORK","_number":14,"created":"2022-03-04 11:37:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/14"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-04 11:37:39.000000000","tz":60},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/815104\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/831788\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d47b4cbc9640133f319ad35a2e9307a1de09a727"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d47b4cbc9640133f319ad35a2e9307a1de09a727"}]},"branch":"refs/heads/master"},"75170dcb7c653cd8ddee73187b1f71b5d8b60e65":{"kind":"REWORK","_number":15,"created":"2022-03-04 12:02:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/15"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-04 12:02:43.000000000","tz":60},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/815104\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/831788\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/75170dcb7c653cd8ddee73187b1f71b5d8b60e65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/75170dcb7c653cd8ddee73187b1f71b5d8b60e65"}]},"branch":"refs/heads/master"},"b911d66bae30d6795477f3a1999bf7fa391122c1":{"kind":"REWORK","_number":16,"created":"2022-03-04 12:36:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/16"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-04 12:35:44.000000000","tz":60},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/815104\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/831788\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b911d66bae30d6795477f3a1999bf7fa391122c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b911d66bae30d6795477f3a1999bf7fa391122c1"}]},"branch":"refs/heads/master"},"250260639b7b9fe466c9ffe6b26ff19611553da4":{"kind":"REWORK","_number":17,"created":"2022-03-04 12:50:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/17"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-04 12:50:05.000000000","tz":60},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/815104\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/831788\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/250260639b7b9fe466c9ffe6b26ff19611553da4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/250260639b7b9fe466c9ffe6b26ff19611553da4"}]},"branch":"refs/heads/master"},"01076d2d223fa92029f21b448567d4a01549d20b":{"kind":"REWORK","_number":18,"created":"2022-03-04 13:14:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/18"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-04 13:14:40.000000000","tz":60},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/815104\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/831788\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/01076d2d223fa92029f21b448567d4a01549d20b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/01076d2d223fa92029f21b448567d4a01549d20b"}]},"branch":"refs/heads/master"},"3f257c8b788a895698e60aa225a7925c9c66146e":{"kind":"REWORK","_number":19,"created":"2022-03-04 13:42:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/19"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-04 13:42:24.000000000","tz":60},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/815104\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/831788\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f257c8b788a895698e60aa225a7925c9c66146e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f257c8b788a895698e60aa225a7925c9c66146e"}]},"branch":"refs/heads/master"},"07d25a47fa9f257c6a9ca3f6fb4bc1c92871014f":{"kind":"REWORK","_number":20,"created":"2022-03-04 14:09:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/20"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-04 14:09:48.000000000","tz":60},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/815104\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/831788\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/07d25a47fa9f257c6a9ca3f6fb4bc1c92871014f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/07d25a47fa9f257c6a9ca3f6fb4bc1c92871014f"}]},"branch":"refs/heads/master"},"343b4b93c3129ecc7d6ddb20233522544077f461":{"kind":"REWORK","_number":21,"created":"2022-03-04 14:40:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/21"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-04 14:40:07.000000000","tz":60},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/815104\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/831788\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/343b4b93c3129ecc7d6ddb20233522544077f461"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/343b4b93c3129ecc7d6ddb20233522544077f461"}]},"branch":"refs/heads/master"},"4bf73907a39ce1429701b3c50fc2b125450c7788":{"kind":"REWORK","_number":22,"created":"2022-03-04 15:12:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/22"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-04 15:12:11.000000000","tz":60},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/815104\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/831788\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4bf73907a39ce1429701b3c50fc2b125450c7788"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4bf73907a39ce1429701b3c50fc2b125450c7788"}]},"branch":"refs/heads/master"},"c465f0cd47c63b7b89461a451ebc1e8f9ff75c12":{"kind":"REWORK","_number":23,"created":"2022-03-07 13:49:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/23"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-07 13:49:13.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/815104\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/831788\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c465f0cd47c63b7b89461a451ebc1e8f9ff75c12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c465f0cd47c63b7b89461a451ebc1e8f9ff75c12"}]},"branch":"refs/heads/master"},"22031b5355dac79e9199b37894e81abd82fe658c":{"kind":"REWORK","_number":24,"created":"2022-03-07 14:17:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/24"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-07 14:17:00.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/815104\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/831788\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22031b5355dac79e9199b37894e81abd82fe658c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22031b5355dac79e9199b37894e81abd82fe658c"}]},"branch":"refs/heads/master"},"09e121f6792058804fe56dba2711324a53b4d717":{"kind":"REWORK","_number":25,"created":"2022-03-07 14:20:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/25"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-07 14:20:09.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/815104\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/831788\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/09e121f6792058804fe56dba2711324a53b4d717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/09e121f6792058804fe56dba2711324a53b4d717"}]},"branch":"refs/heads/master"},"0ca5186dc890e9860abc5559544378e14ebc656a":{"kind":"REWORK","_number":26,"created":"2022-03-07 14:56:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/26"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-07 14:56:20.000000000","tz":60},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ca5186dc890e9860abc5559544378e14ebc656a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ca5186dc890e9860abc5559544378e14ebc656a"}]},"branch":"refs/heads/master"},"69c832a5f704a111c81ecaeff3c1c003504fbb53":{"kind":"REWORK","_number":27,"created":"2022-03-08 08:04:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/27"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-08 08:04:27.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/69c832a5f704a111c81ecaeff3c1c003504fbb53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/69c832a5f704a111c81ecaeff3c1c003504fbb53"}]},"branch":"refs/heads/master"},"ddbf6201cfe86cc741864dfc96b3b3fc95867c69":{"kind":"REWORK","_number":28,"created":"2022-03-08 08:11:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/28"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-08 08:11:28.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ddbf6201cfe86cc741864dfc96b3b3fc95867c69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ddbf6201cfe86cc741864dfc96b3b3fc95867c69"}]},"branch":"refs/heads/master"},"9f1c1f4377a72acd7be26c870ab98233a8bf9b90":{"kind":"REWORK","_number":29,"created":"2022-03-08 13:13:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/29"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-08 13:13:42.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/832630\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f1c1f4377a72acd7be26c870ab98233a8bf9b90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f1c1f4377a72acd7be26c870ab98233a8bf9b90"}]},"branch":"refs/heads/master"},"e8dc6e752b8f7ff7da91fc6d0ce4dc4562fa52ea":{"kind":"REWORK","_number":30,"created":"2022-03-09 08:38:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/30"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-09 08:38:05.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/832630\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8dc6e752b8f7ff7da91fc6d0ce4dc4562fa52ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8dc6e752b8f7ff7da91fc6d0ce4dc4562fa52ea"}]},"branch":"refs/heads/master"},"0d3ec4ee66d5189639cf95cad874e475bd787a3b":{"kind":"REWORK","_number":31,"created":"2022-03-09 08:44:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/31"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-09 08:44:30.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/832630\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0d3ec4ee66d5189639cf95cad874e475bd787a3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0d3ec4ee66d5189639cf95cad874e475bd787a3b"}]},"branch":"refs/heads/master"},"56af17213f6d426d482ac42608c5220bd16c0a3a":{"kind":"REWORK","_number":32,"created":"2022-03-09 10:54:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/32"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-09 10:54:50.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/832630\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56af17213f6d426d482ac42608c5220bd16c0a3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56af17213f6d426d482ac42608c5220bd16c0a3a"}]},"branch":"refs/heads/master"},"d9f33911fa119944fa11992b2ae89dc9ad4e8881":{"kind":"REWORK","_number":33,"created":"2022-03-09 11:27:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/33"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-09 11:27:38.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/832630\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d9f33911fa119944fa11992b2ae89dc9ad4e8881"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d9f33911fa119944fa11992b2ae89dc9ad4e8881"}]},"branch":"refs/heads/master"},"da171a7cd20aafc2acc8ea69a467f4b718c35056":{"kind":"REWORK","_number":34,"created":"2022-03-15 08:15:06.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/34"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-15 08:15:05.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/832630\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da171a7cd20aafc2acc8ea69a467f4b718c35056"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da171a7cd20aafc2acc8ea69a467f4b718c35056"}]},"branch":"refs/heads/master"},"ab8899258cc91967f13b527823645d4051d9b6cb":{"kind":"REWORK","_number":35,"created":"2022-03-15 09:00:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/35"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-15 09:00:08.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/832630\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab8899258cc91967f13b527823645d4051d9b6cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab8899258cc91967f13b527823645d4051d9b6cb"}]},"branch":"refs/heads/master"},"2435bc5f16f069b58d8b1bc3d116fe967b266094":{"kind":"REWORK","_number":36,"created":"2022-03-15 09:01:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/36"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-15 09:01:11.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/832630\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2435bc5f16f069b58d8b1bc3d116fe967b266094"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2435bc5f16f069b58d8b1bc3d116fe967b266094"}]},"branch":"refs/heads/master"},"14b23dd644fece35e9b06def04589f75f3fbc663":{"kind":"REWORK","_number":37,"created":"2022-03-15 09:49:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/37"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-15 09:49:41.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/832630\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/14b23dd644fece35e9b06def04589f75f3fbc663"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/14b23dd644fece35e9b06def04589f75f3fbc663"}]},"branch":"refs/heads/master"},"b96e60f1fcf3ed9a876bcdec1cdaf5ea4cd74715":{"kind":"REWORK","_number":38,"created":"2022-03-15 10:32:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/38"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-15 10:32:57.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/832630\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b96e60f1fcf3ed9a876bcdec1cdaf5ea4cd74715"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b96e60f1fcf3ed9a876bcdec1cdaf5ea4cd74715"}]},"branch":"refs/heads/master"},"989a676d62b7825c77e906002655f9e97c526488":{"kind":"REWORK","_number":39,"created":"2022-03-15 13:31:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/39"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-15 13:31:17.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/832630\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/989a676d62b7825c77e906002655f9e97c526488"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/989a676d62b7825c77e906002655f9e97c526488"}]},"branch":"refs/heads/master"},"9f6dad43eb48728647ea3797e53157dbb1b9cdd7":{"kind":"REWORK","_number":40,"created":"2022-03-17 08:20:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/40"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-17 08:20:15.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/832630\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f6dad43eb48728647ea3797e53157dbb1b9cdd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f6dad43eb48728647ea3797e53157dbb1b9cdd7"}]},"branch":"refs/heads/master"},"bc382af834c3254559101d84a5cce95fbe16a262":{"kind":"REWORK","_number":41,"created":"2022-03-17 10:10:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/41"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-17 10:08:56.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/832630\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bc382af834c3254559101d84a5cce95fbe16a262"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bc382af834c3254559101d84a5cce95fbe16a262"}]},"branch":"refs/heads/master"},"dbb3e75185672913d24c6e68387ff61f444aec3e":{"kind":"REWORK","_number":42,"created":"2022-03-17 11:04:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/42"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-17 11:04:48.000000000","tz":60},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dbb3e75185672913d24c6e68387ff61f444aec3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dbb3e75185672913d24c6e68387ff61f444aec3e"}]},"branch":"refs/heads/master"},"9217f09c9d3cf510226319cccb7365c39bd1ab18":{"kind":"REWORK","_number":43,"created":"2022-03-17 11:11:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/43"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-17 11:11:21.000000000","tz":60},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9217f09c9d3cf510226319cccb7365c39bd1ab18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9217f09c9d3cf510226319cccb7365c39bd1ab18"}]},"branch":"refs/heads/master"},"47f33a945b5f520eaa8ffb6631b3c05d83356112":{"kind":"REWORK","_number":44,"created":"2022-03-17 16:10:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/44"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-17 16:10:48.000000000","tz":60},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47f33a945b5f520eaa8ffb6631b3c05d83356112"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47f33a945b5f520eaa8ffb6631b3c05d83356112"}]},"branch":"refs/heads/master"},"8ebe1eefbe5091cf1b8272f800e2075b3f29a7e0":{"kind":"REWORK","_number":45,"created":"2022-03-17 18:02:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/45"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-17 18:02:21.000000000","tz":60},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8ebe1eefbe5091cf1b8272f800e2075b3f29a7e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8ebe1eefbe5091cf1b8272f800e2075b3f29a7e0"}]},"branch":"refs/heads/master"},"adef61131739bac3477d74d55e1c46a1b7d4f8d1":{"kind":"REWORK","_number":46,"created":"2022-03-18 08:54:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/46"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-18 08:54:27.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adef61131739bac3477d74d55e1c46a1b7d4f8d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adef61131739bac3477d74d55e1c46a1b7d4f8d1"}]},"branch":"refs/heads/master"},"f40e2a0ebcc2a2a83fed8e4fea66017410ba7369":{"kind":"REWORK","_number":47,"created":"2022-03-18 10:26:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/47"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-18 10:26:42.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f40e2a0ebcc2a2a83fed8e4fea66017410ba7369"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f40e2a0ebcc2a2a83fed8e4fea66017410ba7369"}]},"branch":"refs/heads/master"},"218afecc9dbbb64e3bbed251dd6cca0a631f63a8":{"kind":"REWORK","_number":48,"created":"2022-03-18 11:47:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/48"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-18 11:45:49.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/218afecc9dbbb64e3bbed251dd6cca0a631f63a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/218afecc9dbbb64e3bbed251dd6cca0a631f63a8"}]},"branch":"refs/heads/master"},"abc5710f2cd3a5aaa9aba893ddb19bce1a27ec51":{"kind":"REWORK","_number":49,"created":"2022-03-18 11:47:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/49"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-18 11:47:45.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abc5710f2cd3a5aaa9aba893ddb19bce1a27ec51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abc5710f2cd3a5aaa9aba893ddb19bce1a27ec51"}]},"branch":"refs/heads/master"},"e3fcc3d68cdb1c8535c445a3a146ef0cbc3b5dae":{"kind":"REWORK","_number":50,"created":"2022-03-18 11:51:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/50"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-18 11:51:02.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e3fcc3d68cdb1c8535c445a3a146ef0cbc3b5dae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e3fcc3d68cdb1c8535c445a3a146ef0cbc3b5dae"}]},"branch":"refs/heads/master"},"3c781a4a3139481ba0ae1097382c1ef702d5de72":{"kind":"REWORK","_number":51,"created":"2022-03-18 13:14:58.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/831652/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/51"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-18 13:14:48.000000000","tz":60},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c781a4a3139481ba0ae1097382c1ef702d5de72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c781a4a3139481ba0ae1097382c1ef702d5de72"}]},"branch":"refs/heads/master"},"5ef058b70f6aa805d93849e8ddc3ac240a271b40":{"kind":"REWORK","_number":52,"created":"2022-03-18 13:52:58.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/52/831652/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/52"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-18 13:52:58.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ef058b70f6aa805d93849e8ddc3ac240a271b40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ef058b70f6aa805d93849e8ddc3ac240a271b40"}]},"branch":"refs/heads/master"},"8e55ea08a474d3138fed160129acb8322f9eeae9":{"kind":"REWORK","_number":53,"created":"2022-03-18 15:04:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/831652/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/831652/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/831652/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/831652/53"}}},"commit":{"parents":[{"commit":"b2b013c1cfccacaf103aecb2b3b685a739a26e39","subject":"Merge \"Bump up Ansible supported versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2b013c1cfccacaf103aecb2b3b685a739a26e39"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-18 15:04:21.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e55ea08a474d3138fed160129acb8322f9eeae9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e55ea08a474d3138fed160129acb8322f9eeae9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
