)]}'
{"id":"openstack%2Fkayobe~832225","triplet_id":"openstack%2Fkayobe~master~I615707976454a91c8f6aecc5eda1852def7197d4","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-03-16 01:34:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I615707976454a91c8f6aecc5eda1852def7197d4","subject":"CI: Move to pytest-testinfra","status":"MERGED","created":"2022-03-07 13:32:50.000000000","updated":"2022-03-16 01:34:36.000000000","submitted":"2022-03-16 01:34:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832225","meta_rev_id":"03b44f3069be3e4cdb942a88b1cdba337acddf65","_number":832225,"virtual_id_number":832225,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-16 01:34:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-03-07 13:34:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-09 20:35:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-09 20:35:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-07 13:34:55.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-03-07 15:18:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-09 20:35:56.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"2a7e11371828eb5bd5dce34287bcf1dee2921659","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-07 13:32:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"180f66895fcaf0a6903e336049d7ab7cace315df","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-07 13:34:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3b47a645f3cac51a95033af94b361aaf096398a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 15:18:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca429b51175a41559894b04936242a11 : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa5ebf0b0410404889b9e240d3f7af54 : SUCCESS in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aac05e9ae4ba4af9b20c07d2a3e37c87 : SUCCESS in 7m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ef50f290b9e4b61864b85c206bdbcac : SUCCESS in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0f7b391b2e44c1b84d1c4afd69da7ab : SUCCESS in 10m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a6076608a84a4e4c8e254d83f5367389 : SUCCESS in 8m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fcbdfeef32b24c68bb2465789cacad3f : SUCCESS in 16m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/86e61372869b4c12b33302a1eee5dc0a : SUCCESS in 37m 35s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/13d488f407d24ef29ddf5dfd4428e233 : SUCCESS in 1h 10m 03s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35b806d591cb4048be079d14f14eab77 : SUCCESS in 59m 45s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9849356aa89d4c35b6e0cd5e65d5a1ad : FAILURE in 1h 13m 50s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/69f156e0a1054a9e98180086b4ee2c00 : SUCCESS in 26m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ed03401b75d4661af1e075b4dac3270 : SUCCESS in 16m 39s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b75176f7337a4582ae53a9766b9a05d8 : SUCCESS in 1h 43m 48s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3bdded8715f4b4887744ab2b85bc4af : SUCCESS in 1h 17m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/99868614e8de494493aeceea69df0b76 : SUCCESS in 50m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c539dc4748d8426c90ba564628feb50b : SUCCESS in 55m 59s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d7317c66fe384adabff3a8ceac9c472d : SUCCESS in 25m 02s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ed87f24fb44431d8970a096fe8aaed4 : SUCCESS in 20m 53s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f0eb543a2ee24cd4a3eb71fc978835ec : SUCCESS in 17m 06s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a56c854acc6a446dad8eec67029c1917 : SUCCESS in 16m 52s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/fd90b4ddcbd04a24a5f95ebd4c5b4cee : SUCCESS in 15m 11s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3100fd426e124447a70994d83f00da21 : SUCCESS in 12m 42s","accounts_in_message":[],"_revision_number":1},{"id":"696f77f58a1bd2b19d632d6f003ddd4857f84c7e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-07 16:03:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e43fa0ef9d967f5f58c1435fa8811a57142da19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 17:59:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3c235c668134e11a0ebeb9a0d6e2e37 : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/589e925f8b044f0c8006bd85f78d6938 : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39de11d29b4040baa65227386b7a6bce : SUCCESS in 7m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9de3a1832d734aa085d1021d52921d23 : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/593ff11874b643f5944781cb812e2870 : SUCCESS in 11m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1e18a8e9e87f411daeac890c0bfe610c : SUCCESS in 8m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8963d63f3f4a4cc1815852296f261966 : SUCCESS in 14m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dced17ec58954ecd9d5de0044ac0c8ec : SUCCESS in 32m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/cb4983fc2a32476bb6389a23a1eba5d7 : SUCCESS in 1h 11m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf03bb6ee3884454b024cbf119111eea : SUCCESS in 52m 48s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2a3c2c5132a3491db7abe57f69471e17 : SUCCESS in 1h 08m 26s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/32220b01af364ea49b33b32b18a40df6 : SUCCESS in 24m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ebabdcd311f4ff794906e2b3dda89f6 : SUCCESS in 15m 25s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/605ea13bd8384d178be5b541ce15e6b4 : SUCCESS in 1h 49m 24s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99f0946e9bd74469b5c8aa71186b1a20 : SUCCESS in 1h 12m 24s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9da52d18ddf847f39018b1e54b483990 : SUCCESS in 50m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9a9058b140348ff91738e9c9e282bcf : SUCCESS in 36m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5eb06ef92567456a82cff06171904e3e : SUCCESS in 30m 46s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf0c3d59d96d45b3aca2cccb345f397c : SUCCESS in 29m 04s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a2a49ea441f040b88d733145787599f9 : SUCCESS in 17m 08s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd4bffc22dee4302849d381312ad59ae : SUCCESS in 13m 35s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/a565578869544ce48e2faa787b3c7e95 : SUCCESS in 18m 45s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e08ade54e5ad4f7785d91b801df6183b : SUCCESS in 14m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d63bb910a84d5992cba453a85aa0404d01558109","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-09 20:35:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3332631e3ee5e8116b762158d1c7a2b77e36c96c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 20:36:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1dbe371acf570d61a7e860a2ba7df757b4baecd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 22:53:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e2bb36133a742df8aa7a03f6a3f62f5 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d3cc311e9284e62b417c05fadb70716 : SUCCESS in 8m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73bcfa055698402ab1f6540601f450f0 : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b1fdcb5d094402da9c4f4f45adc935f : SUCCESS in 10m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/38a374873fb24864b0b29eb0097ce8da : SUCCESS in 7m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7e776fe61b33443797f9a2c6b7581213 : SUCCESS in 12m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/853e42e08680445880b37c46e000a72e : SUCCESS in 32m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/92c60fc511f64a9d8b21d2e236b43cae : SUCCESS in 1h 06m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5579b14f958f414b8d93a018dcc3ddaf : SUCCESS in 52m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5cee409c9faf4c2d9e3a8ed66a9cd965 : SUCCESS in 1h 10m 05s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0546be7be2e741228142b54caa4d590e : SUCCESS in 24m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4db7349ec5e1443fb53e8374f9255e0e : SUCCESS in 17m 38s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/975c6d16f2ff4f2eb17fe646310d9f2a : SUCCESS in 1h 48m 39s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c78c13ba2e5541cb81eccb00554616ec : SUCCESS in 2h 09m 48s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/59219c46439b4deeac5ab99ef4a7dbd4 : FAILURE in 19m 59s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad25f69fe6e14beeb398ec8268050b3e : SUCCESS in 37m 25s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/31f01678822c44b5a0ab695b29e60a14 : SUCCESS in 28m 27s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a64250cc254545c0996059f0e299da78 : SUCCESS in 26m 17s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8da6df93a7f7475ba32ced2e68ad64f9 : SUCCESS in 22m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22c975f8ce644301a4f6b20a0271c399 : SUCCESS in 13m 13s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/fb4ec40da5aa44df8d24f244b3bc9107 : SUCCESS in 22m 02s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d2a94721ebb4a84bfa37c07f4907e98 : SUCCESS in 14m 20s","accounts_in_message":[],"_revision_number":1},{"id":"0c9f46765b60dae4af33c8c1c65f52e4311c0bd0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-10 07:45:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d5e0571273f79177fc4385efb4300f4415e4a9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 09:09:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4566fe3ca2b04970a73ab5c734729dc2 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93918cb97f2a483ca68c6490e449f36e : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50a131e5b27b4b3b9f4375fa3ee717f0 : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/114643342498451fa6dced232972bfeb : SUCCESS in 8m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6be4278423c4d7aa51c114523d4cd5f : SUCCESS in 11m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9161122916844f9099bb4f09b1976621 : SUCCESS in 8m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fa1b4d16d1c8437691ad6dbb30f7e556 : SUCCESS in 14m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/852ce2a48a604ae8932840882dc9d81b : SUCCESS in 29m 31s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/31db0c173a624fc5982a0adb097262e2 : SUCCESS in 58m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7da02a467fe740d3a0c1874a1c585242 : SUCCESS in 56m 53s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2660c01741e448f3a9ef880ec0c8215b : FAILURE in 1h 02m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8d96b650122b4d539d506e2becf4e117 : SUCCESS in 21m 25s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b50385a9d94404da89971a01ab645b9 : SUCCESS in 17m 34s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/51d0364514b746519c0b3580e4da5d77 : FAILURE in 1h 03m 44s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58be43178a1249edaa4fe4bb472abf2f : SUCCESS in 1h 22m 46s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/506cd9893087462383d0262a96e70773 : SUCCESS in 45m 06s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40f1329dc7114c3791fbef792cb4e79a : FAILURE in 38m 23s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e6ddac1ed023471397a201ab4ebf7dc6 : SUCCESS in 28m 37s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25f117760fb0450e9086fa09a512016e : SUCCESS in 29m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1a1b1c95fc54483899f4e3ba5be9753f : SUCCESS in 15m 38s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0132b7a9155e4e03b6cf427c0e81c078 : SUCCESS in 15m 14s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/0c0a7da5723949628bd8f0bb964de7c0 : SUCCESS in 16m 10s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f34025973d1a42aaa060507411a27f22 : SUCCESS in 15m 12s","accounts_in_message":[],"_revision_number":1},{"id":"4c5bdc6be2a958945b3f85ec11de33eb0fa01fa9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-10 09:44:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81f2124c4c665df592856d31ca0512ce32c805ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 11:46:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73583c7e09d1418f869960e2d46c4a80 : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8692491e2094891a312acdbb146c87d : SUCCESS in 9m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec95a57096364a098a4b63ca779370f2 : SUCCESS in 8m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d256907a47943dba529bae72c78a2f1 : SUCCESS in 9m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43dc1253302c4834bec9b7644a14a9fc : SUCCESS in 11m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2dce319c8ea34d1289ed4b4b6656bc7e : SUCCESS in 9m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ae5ea9b55e52404daf25b8af7b6ba27a : SUCCESS in 15m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c29f57138d374f698be54b7983653888 : SUCCESS in 32m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f3db27b9fbf7448fbe7c554c8157f177 : SUCCESS in 1h 06m 30s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4bc4602fb894352aa617adc433445cc : SUCCESS in 1h 01m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/43991d902bfb425499950bc997b48e6f : FAILURE in 1h 16m 17s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/479eb8a7d5df4e4abd5ad0df8375aedb : SUCCESS in 28m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ff042d6b36f44489953e1ac5e2f8218 : SUCCESS in 18m 33s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/138d1d4acc6044ca9115f4748ac09c90 : SUCCESS in 1h 46m 58s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01c38d72c9e8423993024ce9a9889a03 : SUCCESS in 1h 13m 25s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/24f924f287784c98b7854a90d187e4ce : SUCCESS in 51m 31s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f748202921e4c479c3c9304f7c25057 : FAILURE in 42m 30s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e7bde892fa734d22a8289bee86f89d44 : SUCCESS in 31m 44s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/386d478f062c45da9b401d42ff696676 : SUCCESS in 26m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/26df4d10e6b34766b147ee90608bbe3c : SUCCESS in 18m 13s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aafd959382db4a148dea1ec222f6ab9b : SUCCESS in 14m 52s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/82f894fea86c45a7bb7edf67871ba499 : SUCCESS in 20m 22s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bde19370897b40fba59cdfc27936b956 : SUCCESS in 16m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e12f7dfbee2f56ade134894bbccd3848fdbd2d25","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-11 16:53:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ada6eaa6184a070d1813f5d64ea9c2901fe5548b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 18:49:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3acbcc798c4d49f493bdd11956517bea : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be80ff589c9947439e90bc7a06d6ad04 : SUCCESS in 9m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcc8e6bb33ba431fa97c4e1690d245bf : SUCCESS in 8m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e16945c8bd9148ac87d807e89609abec : SUCCESS in 9m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/033465c6da994ce69b869a15dfb4e419 : SUCCESS in 12m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6d61c53153fb4bf299a2c4f671d08cd7 : SUCCESS in 11m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6036830b8fc548759ed7cf3dded541a6 : FAILURE in 20m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5f114c6a25e94f7ca627fce73ff8fb88 : SUCCESS in 39m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3ce58af98985418ea9b6fb37c1b8640b : SUCCESS in 1h 08m 59s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32084a998cb2412d98034728127b0df2 : SUCCESS in 56m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5c71977da40043239e6361bae0201054 : SUCCESS in 24m 24s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ceab7d83f7174ed48a37cd9701cce621 : SUCCESS in 16m 42s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4fd0f45c5a9e4a4da2856dac83b42b04 : SUCCESS in 1h 48m 44s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ac2974e95084b759c985076a877b86c : SUCCESS in 1h 15m 37s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5ee83091c63d4072a2cbc63c46f58737 : SUCCESS in 59m 23s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b8ee7b2dc3c442f8beb170373020eeb : FAILURE in 47m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d60690b7137947e4800d96ee68b0dd1f : SUCCESS in 29m 40s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/530f2fcb22f94ced93a1ab3febbd03f1 : SUCCESS in 36m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b0b818fb7dc24058af9f5760e4284eb5 : SUCCESS in 18m 45s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f958fb6696f4ee78938379f4e3e90bd : SUCCESS in 13m 25s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c094d14517ac4f0bb2320cc3da6c52eb : SUCCESS in 21m 33s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59063d7d353d4d6b94220e2e6ee07b17 : SUCCESS in 15m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b74528fbb12cb592abeb45403ad0fd5a3dac0ef3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-11 21:18:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d79fde6ba62b4300c1f995416b3ed79939fd980e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 23:18:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9df4442ffe5443198c7a967b56deefd1 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c95c88f611a4651affb9b16732d9ca8 : SUCCESS in 8m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bd3fa3facb7487188815f91fa33b46a : SUCCESS in 10m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/897b698e956f4703a5b33401009cccd8 : SUCCESS in 8m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18a9e25698b0449bbf735501d286fc6c : SUCCESS in 10m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/258308a5dc4643428487635d6e951c48 : SUCCESS in 7m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/27271b60a2fb4890a8b28092e88848e7 : SUCCESS in 14m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e0a7d56435e040d3978762e76a607046 : SUCCESS in 34m 26s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/79e86c74a9d446deb74bc53b4964fefe : SUCCESS in 1h 19m 17s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/939d748f10554741a511d3fa5e2f0b1a : SUCCESS in 1h 00m 37s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8edcef210c824f1294cd49aeb21262c6 : SUCCESS in 27m 44s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/044cdcd71b5f4814a6de7961a1e49543 : SUCCESS in 16m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6edf634b8b744b978c2faf6570124ad7 : SUCCESS in 1h 51m 51s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d40b88c11384f0e981975266a13e707 : SUCCESS in 1h 29m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ed4b1f38ce9845eb9173b913220c729f : SUCCESS in 52m 10s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65dfe1db7b8149bba48fdaf94d5f3220 : SUCCESS in 37m 57s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bfac8ecca6ad4c968c095e28dcb22462 : SUCCESS in 31m 46s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8dc790b94ade43648c81390ce2df7c78 : SUCCESS in 27m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b39af31acf25494cad9611729b0b8268 : SUCCESS in 20m 06s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f491ecd83e324f73aa82257dac1ab8cd : SUCCESS in 13m 48s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c4c415a2254b40eabe085737e8594e13 : SUCCESS in 20m 38s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab8eabbd72d14439bb186ebe440c147e : SUCCESS in 14m 29s","accounts_in_message":[],"_revision_number":1},{"id":"79ec92dfd96d7f6f9b132c73f306d886b40ca876","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 23:18:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab3de91069b2d385ff23eb4377fa12b604973007","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-12 00:25:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bbc04090e824da2840abfe39c3bbc04 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebe717e2ce014fa6ad0a185e59c1b9b5 : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b6cf027a7a14a9ca3ac70451b75d0b7 : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1ae046f53e343b0b5ffa3253367a493 : SUCCESS in 10m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d74a8c4012ed46989668582c493ac317 : SUCCESS in 6m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e2afacb046c04f378014f413b0263c28 : SUCCESS in 13m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d9809e34262d48658bd993056682a5a9 : SUCCESS in 28m 49s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/00e1a0bf455240b38cc311b8dae31a3c : SUCCESS in 57m 05s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ad2d1fabc23434a939c8ab3207ac752 : SUCCESS in 47m 08s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/89d0fff871454d9983b2e7d9b0261661 : SUCCESS in 19m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5df43de10eef41cd8a95cb797d3563bd : SUCCESS in 12m 03s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2ded3a2c35aa445ab699932eb17e1a0f : FAILURE in 43m 06s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f6fd61aebc64d18ab234c12aad9f430 : SUCCESS in 1h 01m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d2dd771cbb81494cb2e119e1388a16f0 : SUCCESS in 45m 40s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd26681c644c419cb57d4e3787d7683f : SUCCESS in 35m 44s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/67220d3a30184c61b93c22ec1e9872d2 : SUCCESS in 26m 23s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/caabbd9dc669463caae2ba7da6b113ed : SUCCESS in 23m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2f62352e8a1f4609a65a917148e4163c : SUCCESS in 16m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90517d615ddb478cb1b69fbea712a2bc : SUCCESS in 11m 37s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/9003783517dd4ae2b4a5130ad794ee01 : SUCCESS in 17m 32s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c40efbefc0184dbbb555433f4eed5b3f : SUCCESS in 15m 57s","accounts_in_message":[],"_revision_number":1},{"id":"b072cd95e2a57564e90c0074420faeedcea252bc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-14 09:12:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"87785f24a37770889eef737ff9dee3c77b54f519","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-14 09:12:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9055497b47c63e1e9843dd78ece428beb275db1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 11:19:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae639c98e48b4431b54847d84707cb7b : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/058b03f50a2a4f03809cb0bc0bf14b4e : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55e94bd1b9554f439fa8b6c45db4d8a2 : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0d34eff58594704bcad376df6896b41 : SUCCESS in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09120da29f4843ea9b51f352f782b437 : SUCCESS in 9m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a8c5d78f040e4b89a88b51813ea3b359 : SUCCESS in 6m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/479c6507064e478994685e18b7014d40 : SUCCESS in 12m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0bf7ce2a43a94b91a2cfbcabc65ebd60 : SUCCESS in 27m 55s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5f84e818014449cfa2fd349dbfcbc4b5 : SUCCESS in 56m 01s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6681e135c57a447994d2bce2effe61fe : SUCCESS in 46m 35s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2d98138bf9e64869b8ec156b81b0bd99 : SUCCESS in 21m 25s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90ede9befea147788eaaf37fdedbe54b : SUCCESS in 13m 10s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7a64b20d45c74a5993b99b3318f1e172 : SUCCESS in 2h 06m 48s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee0f3230229d48588c59c1d35fff93fd : SUCCESS in 1h 15m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/46f93b6c55b7496e85d0ebba8bf77a9b : SUCCESS in 54m 22s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/172edd3ba251426b9e32fcb8617dde67 : FAILURE in 37m 26s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2a07dd62281046528b19c0c38f6af95e : SUCCESS in 26m 01s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e3c04da72d34ef08331dc1424008b55 : SUCCESS in 21m 30s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c8bd999d15e44de5af14655314cc5b28 : SUCCESS in 21m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fed7c3dabb3844d590f5677edd5889d0 : SUCCESS in 13m 38s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/697f8a0d63bf49988989deb2ba8906ce : SUCCESS in 16m 13s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8acf3b3b88f46c38046bd4315fb6b27 : SUCCESS in 16m 01s","accounts_in_message":[],"_revision_number":1},{"id":"de11074dbcfa2f134a38ab2214966f966767844e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-14 11:40:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb37ec167564ba35fde82a94dee90256026b088e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 13:53:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb0cb6c274c0426faed07d9567552a48 : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f87bbcef8754456d9b93270b66fa53a8 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d0c47515e984e46b4de1bbc1b039738 : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5487780ff41c4605933ba830bc61d21b : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e64edca1227c47dca2e4cc8ebe2fdb4e : SUCCESS in 10m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/492183aa80214768abfc547690e6e8d0 : SUCCESS in 7m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/577fa3f2cfff4cde9d0fa7f33c92bf91 : SUCCESS in 14m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/512afd6da8fc4f0fae5de08f6d5ee6af : SUCCESS in 34m 54s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/871e35a7b1c44614af0f0b29d8d233c1 : SUCCESS in 1h 07m 04s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75818e20cd954990bed459c10e499e2a : SUCCESS in 54m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0487a5c325684714b20ec62d4fa18207 : SUCCESS in 30m 08s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d96b353189bf458b940913badc93e3c7 : SUCCESS in 16m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b6262a811f55448a9e13f0117180c35d : SUCCESS in 2h 06m 51s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd023ea0d03b4f078d57c2946a837642 : SUCCESS in 1h 13m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0d8db2a7187740d8911b006d3fec4859 : SUCCESS in 1h 02m 36s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3404fce7081248ecac739d76652cd855 : FAILURE in 40m 05s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0744de08efaa42429fccbef3a623bc62 : SUCCESS in 37m 06s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d69e1823ef1b4b8dae067170eb23ecd4 : SUCCESS in 34m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7a83e7eeed7d4d2784ac5fe7fc5398f8 : SUCCESS in 18m 58s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96b5d9973a974cd0bbae234609052563 : SUCCESS in 14m 11s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/187a5e0b19d340a591369eed26779bc0 : SUCCESS in 22m 09s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80de8b6190f44ed7be466f80e22c4b2d : SUCCESS in 13m 57s","accounts_in_message":[],"_revision_number":1},{"id":"54f187584904a71af3d35a8abc9e90bc45de6e13","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-14 14:51:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9f9722f0cbc546a46fbee351ed5932f71964eb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 16:53:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0aa3ef22b7344a75830735d9b20b6b93 : SUCCESS in 9m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d926049291e542b7a6405147c90d873a : SUCCESS in 8m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4acbb8f5f87c49ab850ce545599109e9 : SUCCESS in 8m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/463b28bd899b43a99fcf25acfa66034c : SUCCESS in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c107364da9d4308ae6c1a165735d7b6 : SUCCESS in 13m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/314f781ebae54794bb1b6dddbf3274f6 : SUCCESS in 9m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e91c30f99e5349d88fe76cc06b44cfa6 : SUCCESS in 16m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bf837f2eb062469893a2b5b8fac69002 : SUCCESS in 38m 45s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1fb31b85273f49fc978f45e6e959bc95 : SUCCESS in 1h 10m 36s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2871f4ee3d5e4ad9a84a570fa45679a9 : SUCCESS in 59m 01s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/18c23368272e4705a8ee2b5111f07c25 : SUCCESS in 24m 13s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6074cfac3b8c44cab2e534c189cc6c45 : SUCCESS in 16m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/adb60135b3e848b283d6ec487b303b1e : SUCCESS in 1h 53m 57s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7dd0eb52a024839affa1cc32f9a7005 : SUCCESS in 1h 18m 23s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bf94575a7aef4a92bd0619e02cb8c80a : SUCCESS in 48m 59s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5177aa99c1084f93a4e8c2691f5e2bba : SUCCESS in 37m 36s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/52d4015613b444f286d41279fe8aacbf : SUCCESS in 35m 31s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9832ce2eec664261975472cc661a30ac : FAILURE in 19m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/95718d47e47c47fb873467b336b4e6f6 : SUCCESS in 19m 13s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36524905fb3d46d6b80d3ad2078ce544 : SUCCESS in 14m 53s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ae3a87b62c3b467f8d0d86ad101ae7b4 : SUCCESS in 23m 36s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c7c69400d204fd4807f3e807a2aa2e9 : SUCCESS in 16m 05s","accounts_in_message":[],"_revision_number":1},{"id":"40b68ced4e669b097d034de71632dada8419e1d5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-15 22:08:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a05fbc27bbf5cf7017a5b06cfffc5d4332809105","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 23:36:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca7decc23c204874ab733c385e8897d9 : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65cd9b1f04064430b69bb227b2b466a9 : SUCCESS in 8m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d186aac84d964e4ea17e52bed4362044 : SUCCESS in 8m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af7d341ddf974a70b5cc664c8d9e67da : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2a71d8dc2e64f379cd6d9f227092458 : SUCCESS in 11m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/afd7e2c7042b4d00bd3ab3f2f05eb8ce : SUCCESS in 7m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/91cb3e555ebf48d8aba73004ca18301f : SUCCESS in 17m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/527412abc3574c0b832badc06e6b52d8 : SUCCESS in 32m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5f57ca290f5b400dafbe76b7368a72f5 : SUCCESS in 52m 30s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df2d3ea462ba47aca81b0988272114b0 : SUCCESS in 50m 47s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ed045a674c2f40b8ab51dd0628d419f2 : SUCCESS in 25m 28s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebdadacadb4740cdbd77170b006596e1 : SUCCESS in 16m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0f6d67f67b2740f7985b8e87141ceee6 : SUCCESS in 1h 25m 20s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7fcba6ce5b04dd589813c8f4baa40a0 : SUCCESS in 1h 14m 36s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/93c89079068a42429ea6b4796e703e1d : SUCCESS in 42m 38s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6f338ea63124ae6a9e0a9554b385d33 : SUCCESS in 38m 04s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/364b61c9536e4edc95de1e5cb8f20e8d : SUCCESS in 33m 07s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2630e08f5908430e8c69028e5b4cf7bb : SUCCESS in 28m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1f610f3cd53c4a60bb816e2dacfe02ae : SUCCESS in 14m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f3d418355664d5fbd061623bfc72417 : SUCCESS in 13m 37s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/17ec531e09a34bb58548937b7dc21134 : SUCCESS in 18m 47s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b6e04646d7c4b1193463ba5ebeceea8 : SUCCESS in 14m 28s","accounts_in_message":[],"_revision_number":1},{"id":"28868c186d5cde20d5c66c0ff88ab254d200842e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 23:36:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f998584688e62c889dd2306e5d01e8a32b9e3d37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 01:34:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a0798d0ec324c8c9979a58f0889c2ed : SUCCESS in 8m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2b6015a058f4b8c8379bfee1342eb21 : SUCCESS in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/211082b0027e4a2cb3f473b474d86aad : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6211092a03704e459fdd8d3776d35402 : SUCCESS in 11m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b149652afd5845979e3684ebfa2d2544 : SUCCESS in 8m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5e934ab61d224a1ab604f7eea54455e3 : SUCCESS in 14m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7c0bc0de20134cfe810b7cf66e8b99e5 : SUCCESS in 34m 45s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d37c1a81e298443eb5fa836ccde5cc5e : SUCCESS in 1h 04m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d4cd030f8854c0799bc1f87f52aeddf : SUCCESS in 56m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b07df700e39245e6ac02bb6c64f0db8b : SUCCESS in 22m 23s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03b099f1cec54fe59c9767df0bfa1f49 : SUCCESS in 16m 20s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f40d4d2840f14ed4b1c869933e8c24df : SUCCESS in 1h 41m 45s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1dfa76ddf46b449b92399dba9f0a0403 : SUCCESS in 1h 17m 35s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/80a7c379baa141618cbc53bd56f787fc : SUCCESS in 51m 22s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8db80d95b384fc4b73460bc91a18c8c : SUCCESS in 36m 23s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c6518a0331d44287887a10dfe446a9fc : SUCCESS in 31m 27s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1e0588354404188b0933fcef81d3f8c : SUCCESS in 26m 16s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0a1841630355455e8e51429657354afe : SUCCESS in 17m 27s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52943716cc0c4f00be60e23dcc5539db : SUCCESS in 14m 57s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/de2442acd9fa4297b9019b912427fedb : SUCCESS in 21m 39s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ffe3105f63449b590c99c982d8a7333 : SUCCESS in 15m 01s","accounts_in_message":[],"_revision_number":1},{"id":"03b44f3069be3e4cdb942a88b1cdba337acddf65","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 01:34:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a20fc47f0cee0aa398c00651d457cc8f17fdf2bb","revisions":{"a20fc47f0cee0aa398c00651d457cc8f17fdf2bb":{"kind":"REWORK","_number":1,"created":"2022-03-07 13:32:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/25/832225/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/832225/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/832225/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/832225/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/832225/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/832225/1"}}},"commit":{"parents":[{"commit":"678b915aff7b9b7f6286de07c106451537754396","subject":"Merge \"CI: remove qemu-utils installation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/678b915aff7b9b7f6286de07c106451537754396"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-07 13:32:31.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-07 13:32:48.000000000","tz":60},"subject":"CI: Move to pytest-testinfra","message":"CI: Move to pytest-testinfra\n\nChange-Id: I615707976454a91c8f6aecc5eda1852def7197d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a20fc47f0cee0aa398c00651d457cc8f17fdf2bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a20fc47f0cee0aa398c00651d457cc8f17fdf2bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
