)]}'
{"id":"openstack%2Fkayobe~832729","triplet_id":"openstack%2Fkayobe~master~I78e63704cbdb3f63f042a755d369875bca722c3d","project":"openstack/kayobe","branch":"master","topic":"bond-ad-select","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-03-10 08:18:16.000000000","reason":"Michal Nasiadka replied on the change"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-03-10 08:41:07.000000000","reason":"Maksim Malchuk replied on the change"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-10 09:19:39.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I78e63704cbdb3f63f042a755d369875bca722c3d","subject":"Revert \"Add the bonding 802.3ad aggregation selection option\"","status":"ABANDONED","created":"2022-03-09 14:35:19.000000000","updated":"2022-03-10 10:24:02.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"revert_of":830183,"meta_rev_id":"1e65dec096998aa6948e7bdd3576286e84fe6d66","_number":832729,"virtual_id_number":832729,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-03-10 00:08:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-03-10 08:18:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-09 14:35:19.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-03-09 14:35:19.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2022-03-09 14:35:19.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-09 20:32:25.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"}],"messages":[{"id":"5a457aee37b7c93e5aee4bc17605c36eadd1160f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-09 14:35:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2cdc1ae05466edd32828ea408dbd555232efc18f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-09 14:37:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ed5cb288302a68655e2e40fe738ba681281b383","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 16:27:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/980e8efd830e4edb92d79e352ae54975 : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bd24514a8384cbda5e15a3f1affb70e : SUCCESS in 8m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa6f37b1d48e441296529bf8ed4f698d : SUCCESS in 7m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c66d80cb4dfc4249a721020ea9d78bed : SUCCESS in 8m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4411b014374840b886ffddb3610c143e : SUCCESS in 10m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97b016b5ed72488abde42360ed77fdc1 : SUCCESS in 6m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b7224f0f4f8f48219894239b38679c39 : SUCCESS in 7m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ddc0d62d07ef44af878df1b6cd9834ea : SUCCESS in 14m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bff7e9ba110142eb8606099c91f5cd88 : SUCCESS in 33m 04s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/70256c31bbae4a13bec0127e65a573ab : SUCCESS in 1h 06m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e13e3e85598e4bfc977596c4d38972b1 : SUCCESS in 53m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e583effb503844c0a2e8a4ae25f5df67 : FAILURE in 1h 09m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/994c1cb49ac64c14833a76104f4842a9 : SUCCESS in 22m 46s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b286af1afc7d40899115e03e3f1cc614 : SUCCESS in 15m 29s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/edf154f7bbb7482d9b833a4ec717475b : SUCCESS in 1h 47m 45s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b0535c355474cdbb60ed961b3d47111 : SUCCESS in 1h 19m 32s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1da2b356ca7f4ad9a6d962547a03db0f : SUCCESS in 46m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f97cd1d5e280472687b43dc49515fc75 : SUCCESS in 47m 08s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6a76c636b6b640b69d01e2532a0872d5 : SUCCESS in 30m 03s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88e5abefd38343f4b18fd62223ecba4a : SUCCESS in 28m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2bbf7eb4dfc149e0a72286ae37971497 : SUCCESS in 16m 46s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cf3a8d9ce7a4b079df7ee7576868df9 : SUCCESS in 18m 30s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/233fef21818d4a5e86901956db0e64a8 : SUCCESS in 18m 58s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6931d19ec3e5415ea07289cd6cad223f : SUCCESS in 14m 40s","accounts_in_message":[],"_revision_number":1},{"id":"f45fdaec65b9e2e1e13b75b1ca8e4a474e1b23aa","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-09 20:32:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"344b8990e720fdf196f41cb8e1abb76dc3d503b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 22:15:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09b12ace8b7f4ab39b2593a70dee2587 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/044d156d646f4657b4bed7c3fddd75ed : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f605714335224b09bf2ae50ab97fa641 : SUCCESS in 7m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90a58d0f17a44a6bbe2d794d976f2247 : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8c735e5e9334a25950dfda4cfe74384 : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94b5b9a085e24be5ad70de0bb3123672 : SUCCESS in 5m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0a96b60119bf4909b756f9c49a12d545 : SUCCESS in 6m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/26ded17da24e418d94bf3607e66100a2 : SUCCESS in 12m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bb21aa78701e40f28bfde75fa6c3cbb3 : SUCCESS in 26m 45s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/666f7cae597c484d9d15ae3d56b0748d : SUCCESS in 1h 06m 13s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/023191efdd864d1fa641c6995db0c0a4 : SUCCESS in 53m 29s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9a54d8e1592347a5b50f3b9d1129cd9c : FAILURE in 1h 07m 16s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5a88461deed44224b687bb6953999c60 : SUCCESS in 20m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa1e2bf2c84540bf8a90cc5357b204d2 : SUCCESS in 14m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4b474f5f23a24e02823d56ae31918907 : SUCCESS in 1h 35m 45s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0125905bd192467ebb4dd1b1a9d7d536 : SUCCESS in 1h 16m 05s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bf2b308566d04fb8ab89d5026430b70d : SUCCESS in 43m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c48859395f2c4720aa4bc84fcad39bbb : FAILURE in 32m 42s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/65610863a6634f72a2e70102ca9d370a : SUCCESS in 27m 47s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0416e6dc0d34b10b1a10bdf23f674e0 : SUCCESS in 28m 28s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/403d673f174b4d678fb2b4a6bc8375e1 : SUCCESS in 17m 42s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a213f4ed21a741559176f674f9ed69e0 : SUCCESS in 15m 04s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/271b1a95597846b08cb2490c49d7c195 : SUCCESS in 20m 01s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10592136ecca495eade481efe3a4c5b4 : SUCCESS in 15m 45s","accounts_in_message":[],"_revision_number":1},{"id":"eebb9789f4ce55f8c6da52ef4eebed598e9ef655","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-09 22:17:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ffbb8b6137c62b56f43da896c55e06415d93ab2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 00:08:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4aa7d3052ac04e048222c42cad56db5f : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b64519b83614460aad240c44ca7e845 : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6a463bc714a41c7b372ada3ee982545 : SUCCESS in 7m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac39fc62536b41f1a1668c03faac93bd : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6085372abd5148fb91cdc42e5d0b79fe : SUCCESS in 10m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb50dc1a83c04abeb3179421ae86186f : SUCCESS in 4m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/917c48c9774140f5b319000186fce05d : SUCCESS in 7m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9db544a710e24b4790f0602abc776219 : SUCCESS in 15m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0bb640f9321e4a41a983dfcb8cff82b9 : SUCCESS in 31m 06s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a9f483cf300f4e3da4822f2fff2ae463 : SUCCESS in 1h 06m 06s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5a220e62bfb4a1590224e87e1f7282f : SUCCESS in 53m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f810a2ecf6a5420d88deb49e094e4ef2 : SUCCESS in 1h 08m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4057b3939a524df9a94b4944d88a33e7 : SUCCESS in 22m 18s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea56452a3fab433480d2ab300f262e77 : SUCCESS in 15m 19s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/081613e14ce54f85ad3d5faa64bc641a : SUCCESS in 1h 46m 32s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34257772c41b4208a1f71e0d2934c6e3 : SUCCESS in 1h 12m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/16acf5d4ae6848da86dac10b25bbdc1a : SUCCESS in 45m 59s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35a64a9f6a49460bb3920e79182a6c9f : SUCCESS in 35m 49s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7f31735cdb24476396012d261afdda1b : SUCCESS in 29m 18s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8d77ea8bb3d400fa001a55336a0b874 : SUCCESS in 26m 26s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2e0e8ec801d44778a6518e90c5a8ea70 : SUCCESS in 16m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfe7ae60048a4edbad40fc6b1ba0e97d : SUCCESS in 13m 23s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/0cb1c0f22cb2426197e268e178398ff3 : SUCCESS in 19m 34s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be35a61fbe424427a132526aadfadcdd : SUCCESS in 14m 03s","accounts_in_message":[],"_revision_number":1},{"id":"a52d7c4a9bfd3d978856bda07998131e6d9fbc6f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-10 08:18:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"db8694268e8eb7626b7504f65378f0ad416825c7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-10 08:41:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"354f95611090f4e3a65c15d15fb6d22c29d499b6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-10 09:19:39.000000000","message":"Abandoned\n\nBumping the role to v1.13.1 here: https://review.opendev.org/c/openstack/kayobe/+/833007","accounts_in_message":[],"_revision_number":1},{"id":"1e65dec096998aa6948e7bdd3576286e84fe6d66","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-10 10:24:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"666625144b69c9f639e271ab40da1916b85a8fcf","revisions":{"666625144b69c9f639e271ab40da1916b85a8fcf":{"kind":"REWORK","_number":1,"created":"2022-03-09 14:35:19.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/29/832729/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/832729/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/832729/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/832729/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/832729/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/832729/1"}}},"commit":{"parents":[{"commit":"8e36626e3ed41793c79d95951be6000d9b1e480e","subject":"Add the bonding 802.3ad aggregation selection option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e36626e3ed41793c79d95951be6000d9b1e480e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-09 14:35:19.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-09 14:35:19.000000000","tz":0},"subject":"Revert \"Add the bonding 802.3ad aggregation selection option\"","message":"Revert \"Add the bonding 802.3ad aggregation selection option\"\n\nThis reverts commit 8e36626e3ed41793c79d95951be6000d9b1e480e.\n\nReason for revert: The update of MichaelRigart.interfaces to v1.13.0\nintroduced a regression due to\nhttps://github.com/michaelrigart/ansible-role-interfaces/pull/118.\nConfiguration for active interfaces could be removed and not recreated,\nif not present in the kayobe network interface configuration. This\naffects the Kayobe development environment, in which the configuration does not touch the host\u0027s real network interfaces. That \nwould lead to DHCP leases not being renewed, and consequent loss of\naccess to the host.\n\nChange-Id: I78e63704cbdb3f63f042a755d369875bca722c3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/666625144b69c9f639e271ab40da1916b85a8fcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/666625144b69c9f639e271ab40da1916b85a8fcf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
