)]}'
{"id":"openstack%2Fkayobe~833007","triplet_id":"openstack%2Fkayobe~master~I03da729e1d05783af2e290237485f6035d2ddcdc","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-16 01:34:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I03da729e1d05783af2e290237485f6035d2ddcdc","subject":"Bump MichaelRigart.interfaces to v1.13.1","status":"MERGED","created":"2022-03-10 09:19:14.000000000","updated":"2022-03-16 01:36:59.000000000","submitted":"2022-03-16 01:34:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"833007","meta_rev_id":"d9492c4b64b41cbef91951fc247da30a035dd4c1","_number":833007,"virtual_id_number":833007,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-16 01:34:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2022-03-10 09:51:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2022-03-14 09:30:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-14 09:23:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2022-03-14 09:30:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-14 09:23:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-10 09:51:26.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2022-03-10 11:21:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-14 09:23:42.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"ee06bd3cf1d9c5191848166c97ae993e4b733e54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-10 09:19:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"125d510f5830afc5baa89d012b215150b445d823","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-10 09:51:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fd7923f855263ddf028bb47f628e8746b3f3cfc1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-10 10:31:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"40e8a7c0f1c2f031b487a56c1de323a21872e39f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 11:21:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bc2ead775f24ad49d292976062708ec : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7f6940605004dfc8a8d58d5eda08d37 : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2dd69e39b781477d808bfa89cab0a7ec : SUCCESS in 7m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc5e4814cd1146edbf139ed933acb320 : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f14e0f3991764b13a35b070a474154ab : SUCCESS in 10m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/356c5005757b4966b1ca961109003886 : SUCCESS in 7m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1b20a8c085fa4f28803efe1fa6c652e4 : SUCCESS in 13m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b0d6219641cd45beada686ee8380d6d8 : SUCCESS in 32m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/de08d0dd809d4ae4992f4dad749ef55a : SUCCESS in 1h 06m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/523b6e918aee4ab48ccfdb913dbae19d : SUCCESS in 58m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5daf24c1097741e49b08fe98167524a7 : FAILURE in 1h 36m 02s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3647cef49dc8407ab0a7fd9773956f3e : SUCCESS in 21m 26s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79911a405bb645f2aad61f0be99c4fa8 : SUCCESS in 15m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0fd44180863a4e10ba4cc687b26f962a : SUCCESS in 2h 00m 16s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93883965bf264f97bfcfa71b847fb9a0 : FAILURE in 21m 24s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3f039042e4ae4cdf927c2fd4a6a45fb4 : SUCCESS in 51m 12s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f30d3374093f4dc5b466bf52ead09c57 : FAILURE in 36m 04s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/62fc2de23675464dbf2ff11ca1ce7894 : SUCCESS in 30m 26s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13b03c7ce16242a9a6115ddb3a5836e7 : SUCCESS in 27m 32s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/bb939ad959f34e3da02c28adea911102 : SUCCESS in 19m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3695e7fc44f64dca87b4d53be9ad61d3 : SUCCESS in 14m 08s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/7d67c03644d54e6aa22126fd1a73c179 : SUCCESS in 19m 31s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21867592cfee4bbcae6a112ae285a320 : SUCCESS in 12m 33s","accounts_in_message":[],"_revision_number":1},{"id":"a11a0bf63f5b220b0defb8830135b1fc4ef363e9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-10 11:23:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b626f2ccb9565125965f72bf2dcdf09fa39b72ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 13:12:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e372459f5b94c98ad8b41e7fa11b9fb : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b79b1645ac11432383f353bb9ac13654 : SUCCESS in 8m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a59723b1b5774768a2f8e10c9dd2c7b5 : SUCCESS in 8m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/602cf48ad7e049f3ba109473ac353cba : SUCCESS in 8m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/207090b8229e41a7b105e5c369c8947f : SUCCESS in 11m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4451a28112df44658d9650e439330357 : SUCCESS in 7m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fae00ecc73f04434a7c75379ce318a66 : SUCCESS in 14m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/50bfb1f3d59a4f26ab46ec27bb7ca7cb : SUCCESS in 31m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0b5f0864ad8849d4b61c602e64024dbb : SUCCESS in 1h 02m 56s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd885ec9772e45b3be88677bf8e8cf68 : FAILURE in 17m 50s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c14a57d9ce1f409d8f8947c9f41e3a06 : SUCCESS in 1h 09m 11s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/61403b96d7c04635a62118b216f895f5 : SUCCESS in 24m 52s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/768e4e417ab2406e89cb7f7cd968c32c : SUCCESS in 14m 38s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7aac54d2564e414d82959dff2e1b5356 : SUCCESS in 1h 43m 07s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82f4c1b5c8424fbea7c582da9a9bb35f : SUCCESS in 1h 13m 12s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b66ad59f35da40d88a0bfb2b6172c678 : SUCCESS in 49m 34s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77e1a300ce2b466a9e3dad31a83506d4 : SUCCESS in 38m 09s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/25e14260414040cea93112793da29057 : SUCCESS in 26m 18s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7aa4cdef8df440049e3fea0299245234 : SUCCESS in 29m 09s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c4996e28c12e48e5819fdddff873f170 : SUCCESS in 18m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f67c056cd743419aae80cfd3a8b336cd : SUCCESS in 15m 34s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/20b36ef797cc41518965ea2eeb9fba96 : SUCCESS in 20m 58s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3779a3294a84e00a1e04ad0931ae20b : SUCCESS in 14m 34s","accounts_in_message":[],"_revision_number":1},{"id":"1716376a5591a87d75b01ef8390dd4895fab4120","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-10 13:29:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c6656d05bcbada4280899acf3765b3d6aaab1775","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 15:08:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f75927fa7b04abe9919f525a877841c : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a466d463c09d4f5ba03be3f4731d42ca : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79e3157009b24edabcd19aed17806b57 : SUCCESS in 8m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17799a3bb42246c18dfee2993d5a3b60 : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e527046d35124ee48319d0c07833b365 : SUCCESS in 10m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b0d94d7f688049339f8675cce3d42860 : SUCCESS in 8m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4ea6356cae5b422d9a93793bc685496d : SUCCESS in 14m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8d8a0a0789d542169917ccf322679a9c : SUCCESS in 33m 54s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6149291939064f2cb316cfa73409a0f2 : SUCCESS in 1h 16m 12s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/251c8c3ba25a49949b71370f9eb12922 : SUCCESS in 59m 50s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f5c0462b94654031909ee2940cac73b6 : FAILURE in 1h 29m 38s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/84e04a616c0e412db6e3da401ee5be47 : SUCCESS in 25m 01s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d47a147aafcf40e6b7963da8ff5eea6e : SUCCESS in 14m 52s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/275e1fe7fc13464098439fd10774d527 : SUCCESS in 1h 35m 54s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3ca336ca818405f96298ab671527158 : SUCCESS in 1h 03m 01s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8a58975484434f3fa8c562568d4afadf : SUCCESS in 48m 24s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90567c0aa52b4cf985cdac87a4f525ad : FAILURE in 35m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e80a2dfbc57243c0a86df2040d233072 : SUCCESS in 27m 15s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb73356071c04788922c796922f48952 : SUCCESS in 27m 30s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1d5fd3589f844256b46ebc353c3ee527 : SUCCESS in 13m 35s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff06828ef56e420a81d9046fe0d7ebc9 : SUCCESS in 14m 50s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/2af08ea348044db8aacbcca054ed5455 : SUCCESS in 18m 59s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bc54187c6024dbe9c67485ede932855 : SUCCESS in 13m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5271b814aa7baa87dcf32401cea92d27e1cb02bd","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-10 15:12:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"491e081bc658a055742aca93e6bdc5e4f08255a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 17:11:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4ce7a23d9c04116a0f9492a673dd5d4 : SUCCESS in 8m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c251075739fd4e319e409183fd0529f8 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58fba86c2a2b4734a59e85af6e199f5a : SUCCESS in 8m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fc695b857314e39af12a5c3d1527597 : SUCCESS in 8m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a460b796bbb46fea68abaf62a6b8d66 : SUCCESS in 12m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f901ee48355d446299d5fd14486b279d : SUCCESS in 8m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e56eeadbdd604a6bb2399d043f17853e : SUCCESS in 14m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aeecb8daa0a44313b537b98f1077e6ec : SUCCESS in 36m 06s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/300a43be66604bdf97281b1747b697a8 : SUCCESS in 1h 07m 37s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e7e27419ded4fb1bdb49b3661883f41 : SUCCESS in 55m 50s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fc99cbd8fb84432db11016bec6fce13c : SUCCESS in 1h 13m 09s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/778d2f65a8594f309191e9625c1bb57a : SUCCESS in 22m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75dbbc13464c497480eae1bd60133d28 : SUCCESS in 17m 59s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b370d3a789d0411a86c872445fd877c0 : SUCCESS in 1h 52m 05s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/139844d5fe1c4e3bb4ed5d5f85c503cc : SUCCESS in 1h 21m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6d7d4a83bc6b4744af9b3c7db3d4edad : SUCCESS in 53m 03s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af1544d188744cf68c1094b9cd5d31da : FAILURE in 39m 04s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/18604a3db33a422a997f26e4e5555566 : SUCCESS in 32m 34s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97551c19f04d45668c036a474df3a691 : SUCCESS in 29m 38s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/817e70a61cef493da196218caaa25a12 : SUCCESS in 17m 38s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d790a0226ddd4be0a6cb29fd2ebe6f7b : SUCCESS in 16m 59s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ec320fa8a79f4878913fcda5a18e5461 : SUCCESS in 19m 17s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2195883759b46df925ec0b4c49dcec1 : SUCCESS in 17m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a10d49df74d8deed1e521aaf2a00f12c1825b354","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-10 17:53:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4eab1b246800a69883f721735fa343ecc761a2e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 19:53:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f92c89cd5e17472586547910a4bc3e36 : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dde1920aa7f04c02b1e072f294aab2c6 : SUCCESS in 9m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8284b749f3394cec9cd34860d896ac9f : SUCCESS in 7m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bec26bd0f65e47528ae1d6122f948d23 : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7579675d5b704259a1955da4ac9a7f14 : SUCCESS in 12m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cab571f34cc847ac963f2c2c8bf0b9d1 : SUCCESS in 7m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/daa5a77f5c47449ca06f48b0b2acd564 : SUCCESS in 13m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/58b8f9d6c31a4fc3a9c2a751735e3fa6 : SUCCESS in 27m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fb709d79d6614ca98806bf406473bc98 : SUCCESS in 1h 07m 35s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9da649819ea64730a02d31956caae3dd : SUCCESS in 51m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/64b64a4997f34380990e6b682a89a493 : SUCCESS in 1h 11m 34s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b35d7ef5161e461f821fd19149ab5363 : SUCCESS in 21m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcffc893bf404e45bfab181f7df458f7 : SUCCESS in 19m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0fda34244d354a19b8cd7ae4c5da00bb : SUCCESS in 1h 52m 01s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/688aa8d6e0e7493ca99a1d3f21659e16 : SUCCESS in 1h 22m 46s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/700177b560604c1fadcd962ed0e21c8f : SUCCESS in 47m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/102af62a2c7449f9886c174022d9892a : FAILURE in 38m 07s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a4eb20dacdd64004af2dffc0674d746d : SUCCESS in 31m 46s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8346e2858bfc4ad58d1358b7be513e0c : SUCCESS in 24m 59s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8e616be09d1e46e2acd60977dbae6f10 : SUCCESS in 18m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8884bc6a13149d5be5a1fd8d35e8da8 : SUCCESS in 12m 40s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/90ca202697ad47b7b53c9236f708ec5b : SUCCESS in 20m 04s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2b930c1a9b44e80b632a61ed6e99f92 : SUCCESS in 13m 44s","accounts_in_message":[],"_revision_number":1},{"id":"dd4d572db3e6b465cd78eb086ebe95e9b5e2ac98","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-10 20:07:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e4b2354414b71725aa54c97b64345c0ba6ccf535","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-10 20:08:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fdda857e1638af980015ae7cf862029b25c8402c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-10 20:55:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"52e862acff31dfe9cfa2eb4bc8b42a2ef55b69ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 21:58:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d645e646f004513961bf4c54e1e73fb : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e065ec93bb144cd28ceb8225dab7cc0d : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b89d63d08a1429bb6ec8c8548b283f0 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97233ceec7514a6597f01ea0cc8e7683 : SUCCESS in 8m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf6ee2a8460a4f729901ec30632ace36 : SUCCESS in 11m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/23467a8522ea41069173fbe73ea8a5f3 : SUCCESS in 7m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c1b42b00aca04bc4a6d17ba267cdb838 : SUCCESS in 15m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4be0db2304764004943bb9291fb7bd52 : SUCCESS in 32m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/44cd50fbd208456296269f75d6fb8438 : SUCCESS in 1h 05m 37s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c187d1c986945edadbb82e472a4dec4 : SUCCESS in 56m 21s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5f61f04472164bde8eea439629c2f976 : SUCCESS in 1h 11m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/02d81e718c1645e892427450de5a9a59 : SUCCESS in 23m 08s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/741f2fe532f5456c95b22c540fe3df79 : SUCCESS in 16m 54s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e5ae7c93b26a47fb9e2141fc44504407 : SUCCESS in 1h 49m 13s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ec511d84abe4d799edc9cff9b757928 : SUCCESS in 1h 17m 10s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ef856dc3b861464eb611d467504896a2 : SUCCESS in 47m 52s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c303f00e9bd1435cbfcf5ae442ed1b7a : FAILURE in 38m 54s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/da37a7a1b6f24dec9211003de82894f6 : SUCCESS in 34m 34s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86efb02de5434d4684d5a307a4164cf8 : SUCCESS in 32m 45s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/aaa2ad4e14314559be23445bb8e99a51 : SUCCESS in 22m 43s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15635f47fb364883acc84186c40f6bd7 : SUCCESS in 13m 50s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/7c44d326c1594504812c136d11747d5e : SUCCESS in 18m 21s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dde251511164f5dbaac79e9e22df34e : SUCCESS in 15m 09s","accounts_in_message":[],"_revision_number":1},{"id":"88484e7c1f753fc81f0a39d48dd181d1b07eb753","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-10 22:20:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"13dbd5d938fd75a0402a6f13432480f19733f262","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 00:10:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa48b36a85824dcd8dd6084be8ed7f78 : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c70115adaa604f218f0106e83836fc7c : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32564fef90e94d65be86e5b23e638b77 : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1fcd1b752d8410b966034fe7a331819 : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf78245d455f418d849edb24dfdad560 : SUCCESS in 11m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1a979bb02a6246b496157afcfd0bd67d : SUCCESS in 7m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c0f406bb4a7b4437aadd3225c593fd2d : SUCCESS in 13m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f869234b79824f06aca09134651398bf : SUCCESS in 32m 19s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1fb039622b97478fa7d67f30e46187eb : SUCCESS in 1h 07m 48s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ad71d6a6db54c8d8ee8d47456273f1f : SUCCESS in 52m 14s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d0229d97fd654d0fa9cad820382b3cc0 : FAILURE in 1h 26m 38s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c88a1acb3c3744098a845f2b3b4197b5 : SUCCESS in 22m 28s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/685b788acc214a62869f086da215271b : SUCCESS in 15m 33s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fa3cd1beac7341fc9564c69cf20fdbd7 : SUCCESS in 1h 48m 25s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43c9e22478494a259ffee6851c46a299 : SUCCESS in 1h 12m 05s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/515206e9c6784480bc0c1fc61fdce499 : SUCCESS in 48m 58s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9aa66f169db04b64bb121f0e8daa02f6 : FAILURE in 41m 49s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/379e503ef49a4bed99de5dc42860ace4 : SUCCESS in 32m 04s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be958e3fbff5442cb5857c3f9e9f9ceb : SUCCESS in 29m 26s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/cda50b70e2b9464a9238cd3f16aee8f2 : SUCCESS in 19m 12s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3be97ff8954e40b5b837b8951bee9d30 : SUCCESS in 13m 47s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/f371f5eb4bc24987945ea36cedf84e3c : SUCCESS in 19m 05s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9e0eccc58cd4928a3550dfe15cfd96d : SUCCESS in 14m 58s","accounts_in_message":[],"_revision_number":1},{"id":"a20b8f8edead0c9d54843d0fde1bbda15040a566","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-11 07:10:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d87885f15bf2273b8ea36b56afa36be93543f098","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-11 07:55:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"890c95673ff9684748683bfac35b6650b8e3e7c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 09:00:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40050c66c1424eda8a3c667ab2997eb2 : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f0457b88ee84aae98d5c7180a9987ec : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9f1fc393c45494aa68099bd7f5de79c : SUCCESS in 7m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/193c9264415b451f96f000d3423eff2e : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9dc04e048004300b41854d0bb5c09e6 : SUCCESS in 10m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b863cc94c8c24ff181e19234f49ff64c : SUCCESS in 7m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/413b9f2886124bb798f8efdb591b8047 : SUCCESS in 14m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/53a56c6900ee4dba8b3bd19a07191655 : SUCCESS in 34m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fcc7e73fb79a4c2ba6f49497f93c9f81 : SUCCESS in 1h 05m 43s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60630da9b5b942718a456784733b97d1 : SUCCESS in 55m 50s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/873051aa614242fba99628f90b585108 : SUCCESS in 1h 11m 49s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a9c7e658ac7f46e5901ea73458e5e382 : SUCCESS in 21m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63c78dbb15ee423fa94d5fb2f066a720 : SUCCESS in 17m 11s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2db158c40ff04c5cabe4d48c1f3f90c7 : SUCCESS in 1h 48m 10s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e23a812f2cd4466183f1b5a99d3f1a48 : SUCCESS in 1h 19m 32s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6c5ba7be9a314a95af67963de46a2a8f : SUCCESS in 47m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e05aca696a24c8f93062b3e7c91b656 : FAILURE in 39m 28s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a052a47c31a54f2e8c55f8a7fbd2cba3 : SUCCESS in 30m 33s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cab3c8058604f49b7c897b3c4b184b6 : SUCCESS in 28m 50s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/480f3b1a8d794bb6aeea12fb5b572aca : SUCCESS in 17m 03s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/196cf58ab52c481da30cfdb1bff673ac : SUCCESS in 13m 24s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/b4327f76057640a6bf948d64664955bb : SUCCESS in 18m 11s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0db4c66107d646ce98eb49176cfdc2dc : SUCCESS in 15m 30s","accounts_in_message":[],"_revision_number":1},{"id":"db2fd39cf4ecc4315076acba5280c56c4ab75b2f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-11 09:06:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d66c1acb118e002735638269295f60f41b0b3414","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 10:46:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1776e409eb494699bf65c23cf94a0edf : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45806bef343b4234809101100a494f7f : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b2cb6b1204f4f179f8c4a2886ca2d19 : SUCCESS in 7m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2696b61d22941b48e996d32c3aea556 : SUCCESS in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bb97f4f3c9c479dbf33a7c5a7131d1d : SUCCESS in 10m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c0936de1f6974aa1864d25025d2909dc : SUCCESS in 9m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e643208b4a344135a2add5509c050d77 : SUCCESS in 14m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0a118d12007945d3b1dc777a7738c801 : SUCCESS in 31m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e76932017e664fd4b104a4842639c366 : SUCCESS in 1h 21m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8326d723f42640fcb5335f93ec25018d : SUCCESS in 56m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/583ed807d2cc439e87df1150c3fe6ae1 : FAILURE in 1h 11m 30s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d515a44b9c4347de941e35000689e15c : SUCCESS in 26m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d86917f188d24a9e85d28c887a57bebe : SUCCESS in 16m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/afd0ba45a60f498f99e9cb223378ea9a : FAILURE in 14m 07s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed1d6477173b43a88fe73064c66f232d : SUCCESS in 1h 27m 57s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d625d78e420c4635a681ffdcf7858587 : FAILURE in 24m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b91552cfc10149c3893851c0b5f13e14 : FAILURE in 42m 04s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0baabb7a81af4880880f556801c0fab2 : SUCCESS in 35m 15s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f87875643d48440db2c451b8531299b3 : FAILURE in 22m 29s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2fbbc18daee746bcb1c195ae358c8bd7 : SUCCESS in 16m 59s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d892d8ffe598496f94bd60688c647053 : SUCCESS in 15m 08s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c4d634b8d5584bbdb7bbca292f6ed3d9 : SUCCESS in 20m 38s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26d9e5881a1a4664b0ae9cbd83ef3104 : SUCCESS in 14m 15s","accounts_in_message":[],"_revision_number":1},{"id":"96966c6b58aa1568f72a62328a550317ccee417b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-11 10:51:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81504fee1c21ddb073c411d41b984b2a705f9c45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 12:41:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3d3dca5c27c462b9c6e6765001f0563 : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ea3a6008a854805880c13ad81ab86ec : SUCCESS in 8m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07d2889b9b624f9aab6ec6f8ad41a5e8 : SUCCESS in 7m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/865b57e1633a4f9f998d0c384db0dad7 : SUCCESS in 7m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5136d6e0c0b44608b135d0f05c2a9ed : SUCCESS in 11m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0c095857f50c4887b1c364697aca516c : SUCCESS in 8m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ae8af0d1f0f344b18e19234094e65959 : SUCCESS in 16m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/982e4f345f674d3781aa4be95c68b8af : SUCCESS in 29m 44s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/41b3358e5d5243b88d0637e906d09512 : SUCCESS in 1h 00m 02s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abab03690177452c8c1b16b50951888d : SUCCESS in 56m 00s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/fcf2a6e0e8aa45e8b32b1eb32d3740a6 : SUCCESS in 24m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d0e30e99c904bc1b821ef6752f576a0 : SUCCESS in 19m 21s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/144584fc498c4957b4227833cea1eecb : SUCCESS in 1h 46m 41s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fda0ec3fb424d70a4f1164643cbc78c : SUCCESS in 1h 07m 57s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9850f968180d4b03b0521b427b7a2a12 : SUCCESS in 50m 58s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b69d095f6e024203a1f088feb7a0fb78 : SUCCESS in 39m 37s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/35e633028c8849d68d1350834e42555b : SUCCESS in 34m 28s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4256234e792f47c58d852bea344cd49c : SUCCESS in 31m 04s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5b92a05bdb4e4e8db4e81af9d304d41d : FAILURE in 14m 47s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b2ec20efe604668b3cb09089d65ade0 : SUCCESS in 15m 58s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/1d5998b492b8418f96e6ff426aaf2b4e : SUCCESS in 19m 37s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91fa289576624f039248da91023dacde : SUCCESS in 14m 57s","accounts_in_message":[],"_revision_number":1},{"id":"b74238b8afa87558783a9603674cb7e5b7221de8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-11 13:03:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"508edcd577ae1711b9e76c1bc6b27e944ac9d3b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 14:59:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d50cb7292b04e9291c1b127c53d4bb3 : SUCCESS in 10m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/739a8030cc374c9f80c31df5ae198b11 : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28ce880222524cfea49812a8f9bdc60e : SUCCESS in 8m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04f72fceea4c47c296697145cd0a3f0a : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e06b087c96c44df08157469121ced13d : SUCCESS in 10m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0f702e3d522548d5a3b74da47ee48cf1 : SUCCESS in 6m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/63342127255f43f49f67d52cd9eedad0 : SUCCESS in 14m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6f7069f29d1f4a02a9745e9efeb39390 : SUCCESS in 30m 59s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6595f6ab5059483692954b8b07d25375 : SUCCESS in 1h 08m 23s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4811fa612b18404980adb6f1c0153420 : SUCCESS in 49m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4ec75f70c4f7453694edb7eb41b69dc1 : SUCCESS in 21m 02s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0c9d7e44f5943d2a5949abb614c84b4 : SUCCESS in 15m 59s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/efea6f5542eb4d4ba8290fce2f559428 : SUCCESS in 1h 50m 25s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d500dd97508d48a089ed559b71723557 : SUCCESS in 1h 04m 56s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/838eb82efa634059a480d54c39f27214 : SUCCESS in 50m 06s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73cc010f1eee4d2083bbeab5b882ab82 : FAILURE in 40m 25s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6ee2479b6ee44c70a25169842d9c1f62 : SUCCESS in 30m 55s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec566162831a445ebcf7102dc6211548 : SUCCESS in 29m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3e64c5230e4e413fb9788db977ae2af6 : SUCCESS in 17m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e54768aa4c547fba56ec0e1e62c8227 : SUCCESS in 12m 57s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/0901b088165a447c8d3437c509b2c165 : SUCCESS in 19m 57s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ee9b0d4f59847e39373371bea3f74a9 : SUCCESS in 14m 40s","accounts_in_message":[],"_revision_number":1},{"id":"39c4cc057baeab3580899466c534bdacbc1b13b2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-11 16:19:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"790ea041c17641265112f3a24bf6f1097a0b46e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 18:12:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eef4a4ff99174c2697c7cfe50d37c1aa : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/124d733a3a644dfd94c9a52078404113 : SUCCESS in 8m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4697a4e89fdb4eb09fb6d61ead25827a : SUCCESS in 9m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e47e3c9f3c9542ed80b5e27f1e24bfc7 : SUCCESS in 8m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae52180789264860833afb7b47260a1a : SUCCESS in 10m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1aaa9178bacd4700992874bbbc05097a : SUCCESS in 11m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d9ec7e029f6640a78544c406f61aebda : SUCCESS in 15m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d295c65e63e34d6da796b0ed2108b961 : SUCCESS in 32m 31s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9c9b5ac6bd7240beb61e5f648a9e78e6 : SUCCESS in 1h 07m 02s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d33437f9be4c4ddbb78a680a364ec714 : SUCCESS in 54m 37s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8b001b2a51904853a496bea211b0df25 : SUCCESS in 27m 23s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11be18fe5e424cb8b200b89121a0349e : SUCCESS in 19m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/94bc6aeb3d434311962eb286100cec6d : SUCCESS in 1h 46m 02s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14f27f661c3a46dc880d4dce3402fc42 : SUCCESS in 1h 16m 14s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b57311a7a0334545a385b4e077052d3c : SUCCESS in 48m 44s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7994a25ec8f84968ad708c4c91ce7231 : SUCCESS in 37m 26s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5832eac664f440deb5fd22864bc9361d : SUCCESS in 30m 26s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c3f1b78d2934b0c9c7c9fdb8148d74f : SUCCESS in 42m 29s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/eb48c2f873b646d28a5995123e358ded : SUCCESS in 18m 44s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17b95a37d33a49ab9ac113b5d0a49427 : SUCCESS in 15m 21s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/0f224f9be4294e25a0d95021d869e30f : SUCCESS in 20m 38s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bf3e3410acc4f1f821c6146dab86762 : SUCCESS in 15m 36s","accounts_in_message":[],"_revision_number":1},{"id":"9610b9d2e9bdaf9b4682c6075e34645b941e2dc6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-14 09:23:42.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"e28f63a98c271f3c7b76f64ea343dd1b5e2f35dd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-14 09:30:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f94d60aed372e8a31eb324b02aa1f60146499932","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 09:34:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3002e8648c7d709c35f784d64975c3092d0350d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 11:03:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea70248901ec4d2fa8cb66db2ab05d39 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d1aeaa040d74152b0a0347f6d64ef82 : SUCCESS in 7m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5ad60210c5e4cfdb34ba73abdfe22fc : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/446cf7f897d749ffbd04517a59e05542 : SUCCESS in 12m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/33cc8b5e82824eb3be12e5ffac9a2deb : SUCCESS in 6m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7e3a5071087645b8ab736ca7c790a725 : SUCCESS in 12m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a502799f40ed4c3abad3279db5e2abb5 : SUCCESS in 37m 07s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f8397c35c2114efc81473a6d229d0a19 : SUCCESS in 57m 03s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f3c8997d5a84bbeb957a48014a1021a : SUCCESS in 47m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/21f59bced9df43bf9b7b0c2e5f1993f6 : SUCCESS in 26m 47s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/daef731ea8d3427ab5f2459cebb20109 : SUCCESS in 16m 53s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/75257579fff74ac49e5769c9240fd24f : SUCCESS in 1h 23m 33s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f8ee51d2c6249fdbcfd57eb58a2455c : SUCCESS in 1h 12m 10s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/daa5118d9b11404a9bcd99569921fd4b : SUCCESS in 43m 26s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e891f666714849dfb3b4305db5cde72c : FAILURE in 35m 50s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/62b238020e674830bd60b9d5dbe87a79 : SUCCESS in 28m 05s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bb1acbd22254522b2bdce68da904bea : SUCCESS in 25m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ae175c76f044482ca7db202f04ee0686 : SUCCESS in 15m 47s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7393fa425ae4f8988e5326e7fb9dc39 : SUCCESS in 12m 46s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/9e118cb7607342f986454254226035df : SUCCESS in 16m 04s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adc3489c49f24248ac439dc7c1ee8fb8 : SUCCESS in 13m 44s","accounts_in_message":[],"_revision_number":1},{"id":"0b4d847b3269342bc83dca27226d8787bb86a725","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-14 11:11:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a0c21d4c0b855ce34e9e4dc844c406491232babd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 13:08:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/641557b7f5a74579accf850970c1402b : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2d8e72253dc45afb49adf54a97d7cff : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d35ffb69fda5466f9f88bae86c1c04fc : SUCCESS in 7m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a3dfe904db24768ba23961864cf16a1 : SUCCESS in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a52710b5648b495aa06641d95cc62fa4 : SUCCESS in 12m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c9553c9bb48a4080899f274ab2df1821 : SUCCESS in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aa07b8edd4a7450182af3b1f7b2ec97e : SUCCESS in 14m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b48e4bb5690c4860b841a9b4531ec9d9 : SUCCESS in 26m 18s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/696f93034032413486f5de95b7c355d6 : SUCCESS in 1h 07m 11s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d22894902fc842449108f751e62674c3 : SUCCESS in 59m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f51cd3c9a4ac4862bdcf3b5caa107397 : SUCCESS in 18m 49s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7735f6fe7eee472b8bcbcdbbdb18f01f : SUCCESS in 18m 22s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ba80762bf7a8486e9f1ca59a171779a8 : SUCCESS in 1h 35m 30s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/371dd9d7023847248b2f84367002f428 : SUCCESS in 1h 25m 18s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c68638c4db3c44d9b995ee227594d385 : SUCCESS in 46m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6701615bba704fbeb870466dfe59f3ac : SUCCESS in 39m 55s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d6e5094a859b4cbc9bc7081a8498a2e8 : SUCCESS in 28m 19s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c17ee30338034364969a880c63e8db80 : SUCCESS in 39m 41s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/bd93dd89ca60419690e07769f42d0c0b : SUCCESS in 14m 23s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6142e05917740b5b7c36c798fed2e23 : SUCCESS in 10m 51s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/a71e23657e704adf8e1c13ee5e1308fe : SUCCESS in 19m 06s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cd643a38d7849eb88855637351a0eca : SUCCESS in 12m 25s","accounts_in_message":[],"_revision_number":1},{"id":"80dd0bfcc8fb8b5ff8cbc0bc37bad6befaa0c565","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 13:09:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e6cb9173ac4d90c84f7624e504950856099125fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 15:06:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6815e4e541ad44658d05ee3ac389cb34 : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c6d7fe96f5c4cf998e64eb2db786d37 : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6a97d21780c45028f4f0fc96cb2eeed : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cfa20afa32542b98d31e22a7a51dba8 : SUCCESS in 13m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d3e6f6e11f3c4764802660c10498110a : SUCCESS in 8m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4a6335cfdb8f45f48b16083ea26876a6 : SUCCESS in 15m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c71031d6381c45a59b60d45c1d78df3a : SUCCESS in 35m 22s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/45bf525ff23a4501bc8b4dc54d4c4514 : SUCCESS in 1h 07m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8531dcbff40a43288c4e3160cf92a74f : SUCCESS in 1h 02m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/04ad42b647b348a3a76811fc7a0173cf : SUCCESS in 29m 24s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8623102fc75a4710b04fb8376b0f9350 : SUCCESS in 13m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7d59a2826e3c409c9d7f52b8b7c72478 : SUCCESS in 1h 55m 26s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae2484ce2242443aa61e653a80c0423b : SUCCESS in 1h 12m 24s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/222bfbc629554a6584fdf50bd1e46255 : SUCCESS in 39m 40s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40c20557321a4d07a7086b6f15579dee : FAILURE in 34m 53s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d28d84ef99d04ee1aa0c02a07a370564 : SUCCESS in 29m 28s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44172a5425c744b5abc9cdc54b595fe1 : SUCCESS in 26m 49s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6358571e3538402caa7fca7f844f7e5e : SUCCESS in 16m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33a45808576a434d96dcce64a0ab18f0 : SUCCESS in 13m 49s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/f4ebddb8d134451aadc0ef320ce5eb14 : SUCCESS in 28m 05s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f988999e218641eab536c1736ce87c69 : SUCCESS in 14m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a65b1cc8c2201ca155af83f837e54682d19c5fad","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-15 09:53:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a769824f83f9500e7c768ac747ea49bfaa1bcc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 11:59:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2aba36c2b3a24388a0a3125a06f54840 : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca33dcf17e704d2b96ba0445eec53eff : SUCCESS in 9m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de5240229d1f462084baa13810be45cd : SUCCESS in 8m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d61a628189a4eae9624a22c74c5191b : SUCCESS in 9m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3972f73587a4d5e82683f1e2769989e : SUCCESS in 11m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ef5b9cdfdcb046f0b13017ac7ce9042e : SUCCESS in 8m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/700678b99d744f669e08e40b0fbf60dc : SUCCESS in 15m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3d8fb7b993eb4cca9e560ce85384478d : SUCCESS in 35m 16s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/43d7d00819aa48f68e6418cbfda5afda : SUCCESS in 1h 14m 12s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95b901bf22ef4428b0fddc637c55d463 : SUCCESS in 1h 01m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1a867cdfc4324cb28e7f1f96276b93aa : SUCCESS in 21m 08s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb165911d2594d7d81ff65e6d9aa9ba3 : SUCCESS in 19m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f32a2e1ac4f64429b265fdc8c95c18d0 : SUCCESS in 1h 53m 35s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b539cb98f1a4a86ad209d0d9c735316 : SUCCESS in 1h 18m 48s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/74c99d3d0eb4445b886e7875b5c573c1 : SUCCESS in 46m 22s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da8949b7e48648a3a51ac70c3644fd72 : FAILURE in 35m 08s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7d5e90900a4e445c84514406a229e75c : SUCCESS in 27m 11s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/594735f521444975979c567b6877584b : SUCCESS in 31m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/965bfb70fa1042f190de0b33af844552 : SUCCESS in 15m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0e5e3791f8e4b87834976757eaf2278 : SUCCESS in 14m 32s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/9f97e20c0b3c463c8fa5b8e06a98792d : SUCCESS in 20m 51s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4471fccbdde44d76ae6c435f67233b28 : SUCCESS in 12m 21s","accounts_in_message":[],"_revision_number":1},{"id":"feb38bba9bdd609dd4696f96b6f3550614c93fd1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-15 22:07:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6ef76d404298833c0c542ce770b6ea524bf761d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 23:39:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77f3b6297275478692b15aad8b157da2 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a3277a9eddf4fe2a047071e07f2f727 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/591d4132d97c40ef8b99e646e0285840 : SUCCESS in 8m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b2aa2b861c242bd96fa52e7cc8fbbea : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/818d2fe5a89743de8020766b62260ac9 : SUCCESS in 10m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/16a09e21d03c41098ac75922925feaba : SUCCESS in 7m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/32f80f18936f40d3a274048a96483786 : SUCCESS in 12m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eac772d75b164b81acf94eaec50560f6 : SUCCESS in 34m 36s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7249e7c2e43540fba663ca836802ec7f : SUCCESS in 1h 17m 44s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34728e386ee44759ae1307006fa84832 : SUCCESS in 46m 26s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/12701519743446cd9d3ab4ebb5d05c5c : SUCCESS in 23m 41s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc688e71195348af99df4805a0589278 : SUCCESS in 13m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/39cb041324de4d9fb1053ada71703cf4 : SUCCESS in 1h 29m 52s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4417874c2e2e43a1b8fa9b7b599756b9 : SUCCESS in 1h 11m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3ff57a2a1f5a4040b2f31055a31642b4 : SUCCESS in 43m 26s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de9c5de6c30e4582868a2d921e224540 : SUCCESS in 37m 31s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c494a2e913a24085978e2cae7ed9a1d2 : SUCCESS in 30m 29s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba24f4aac1bf46e29c5e53d298067523 : SUCCESS in 24m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e3de9e2bf1d644319af6d49c0f52d704 : SUCCESS in 17m 06s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01ee46def5a94e1caba46ee6cb713028 : SUCCESS in 15m 19s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/1108408a5eeb48cda2ab3b293ba55919 : SUCCESS in 18m 49s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7874cf2a356541aba8f14a09b115f022 : SUCCESS in 14m 05s","accounts_in_message":[],"_revision_number":1},{"id":"372c782b29fdcbd241bf31d3a35d66dfeefba1fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 23:39:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"db27219eae82a9970f15f0873aa197d5fb40381c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 01:34:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e66031efdb54c0c977acd2249f18952 : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31ce5094826046dcadc7a5420f5e0112 : SUCCESS in 7m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b17d63dc731846e7b34db6a9ee09b359 : SUCCESS in 8m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24a68b63d8144c2f8a4c85265c447225 : SUCCESS in 10m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e2fab8e0add54c698d024f99555c514b : SUCCESS in 8m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/03bfa30462c74533a6dbc5b529db00fb : SUCCESS in 15m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0157cd8e8606480595c137fd479e0636 : SUCCESS in 26m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5c312cbd01554de586b81f41e193ec87 : SUCCESS in 1h 14m 40s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92801331f2764f5da71d902f11c58e3e : SUCCESS in 55m 16s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9043445793284d759ce43dfe3db8fa37 : SUCCESS in 24m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc643b1fe1ba4a308df48f642522378e : SUCCESS in 13m 42s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b398225914e6430eab7575774b34d8e5 : SUCCESS in 1h 47m 37s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c9fc3e578c9499a8bd0d0f1c284f690 : SUCCESS in 1h 02m 02s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/233a35d6227c4d55a2e715c3f081f801 : SUCCESS in 48m 05s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8795c58e795d42dabade27ac1e7500ad : SUCCESS in 30m 56s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9746cff459194b8ea7820ea15411ce65 : SUCCESS in 34m 00s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0728329abf004a948bd07704ca4172d4 : SUCCESS in 21m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/750efc10322841b7b6aced3a1885a68b : SUCCESS in 20m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60fdef05b6294993931a66c571179d28 : SUCCESS in 11m 51s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/2dd055f0a2ab40408ccf2a6e74185390 : SUCCESS in 18m 52s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4600dc0a2b1e4b9f9a98ae1df0052ea6 : SUCCESS in 15m 55s","accounts_in_message":[],"_revision_number":1},{"id":"32ec4a51c3c5c09f46fbe1579a766fba1be23b18","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 01:34:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d9492c4b64b41cbef91951fc247da30a035dd4c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 01:36:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4f0759d5ff94e038fcbf9669380c10b : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a786ef20f753125159c7b20aa964fb093283c052","revisions":{"a786ef20f753125159c7b20aa964fb093283c052":{"kind":"REWORK","_number":1,"created":"2022-03-10 09:19:14.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/07/833007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/833007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/833007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/833007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/833007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/833007/1"}}},"commit":{"parents":[{"commit":"678b915aff7b9b7f6286de07c106451537754396","subject":"Merge \"CI: remove qemu-utils installation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/678b915aff7b9b7f6286de07c106451537754396"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-10 09:17:16.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-10 09:18:15.000000000","tz":0},"subject":"Bump MichaelRigart.interfaces to v1.13.1","message":"Bump MichaelRigart.interfaces to v1.13.1\n\nThe update of MichaelRigart.interfaces to v1.13.0\nintroduced a regression due to\nhttps://github.com/michaelrigart/ansible-role-interfaces/pull/118.\n\nConfiguration for active interfaces could be removed and not recreated,\nif not present in the kayobe network interface configuration. This\naffects the Kayobe development environment, in which the configuration\ndoes not touch the host\u0027s real network interfaces. That would lead to\nDHCP leases not being renewed, and consequent loss of access to the\nhost.\n\nThis change fixes the issue by bumping to v1.13.1, which reverts the\nchange.\n\nChange-Id: I03da729e1d05783af2e290237485f6035d2ddcdc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a786ef20f753125159c7b20aa964fb093283c052"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a786ef20f753125159c7b20aa964fb093283c052"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
