)]}'
{"id":"openstack%2Fkayobe~833052","triplet_id":"openstack%2Fkayobe~master~I00e7615581c7a214e0ee8e8f04cacdbb3ea4131a","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2022-03-17 02:01:15.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-03-16 21:51:39.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"I00e7615581c7a214e0ee8e8f04cacdbb3ea4131a","subject":"Use naming convention to infer VLAN tagging","status":"MERGED","created":"2022-03-10 12:33:50.000000000","updated":"2022-03-17 02:05:41.000000000","submitted":"2022-03-17 02:01:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"833052","meta_rev_id":"b63cb827074b249e66111716905271bada3202d4","_number":833052,"virtual_id_number":833052,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-17 02:01:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-03-16 17:08:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-16 15:31:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-03-16 17:08:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-10 14:28:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-10 14:31:50.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-03-11 09:45:38.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-03-16 15:31:24.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"51b73321baef327be8f919fb5cdca2c8357e4773","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-03-10 12:33:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81432457cf4bfe14bc1254af66f1e44e36dac4de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 14:28:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd64540662df4fc89837797c15c924b3 : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af86843babc14845af0273220ea32db0 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f968faa701c4fd48d3fab7d40a4c23e : SUCCESS in 8m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c961c40cb8a40238ddf71f7a723625d : SUCCESS in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81882b0d6c424a558ac8d3901aeb22ee : SUCCESS in 10m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/900cd10838814173813421e7c2aa2f0e : SUCCESS in 5m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/59ebb618453c4709a922cb4ee32b1354 : SUCCESS in 7m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8d776c6f185e4edeaf350f26e65811fa : SUCCESS in 13m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/415dca685dca4d03873c3420567b43e5 : SUCCESS in 28m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/955844f987584b89bb436d8dd14ccbc4 : SUCCESS in 59m 13s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c52244e3ae643eca1dcff01dd2f4ead : SUCCESS in 49m 06s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0f0210b3bf1645cb9d00abdff3f5d612 : FAILURE in 1h 36m 25s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3c70dac4461f43e19b3f67426fd68bed : SUCCESS in 21m 34s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92f1d1354ccd41068e2b3a245a087891 : SUCCESS in 13m 37s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f34e8f3fbda849cbbc7e0af61005c485 : SUCCESS in 1h 36m 32s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9812fd0b7f941faab1a1dd5afadb8e7 : SUCCESS in 1h 06m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/190cb4d58b314e36964397a43b94e920 : SUCCESS in 44m 31s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f17e2282d1f47f0990572c93dc93c16 : SUCCESS in 35m 42s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4b311ceb4c59496ba4dbdfc35fd047c1 : SUCCESS in 29m 18s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e658b0f1492140d98f24718f3ab8f6d0 : SUCCESS in 24m 59s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5276e8dd0fe643f8adac7bad5dd5d834 : SUCCESS in 16m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65c7384420b44b67be9eef8e5badc23d : SUCCESS in 13m 10s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/806f856e173543d2b2a98b08a9cc01b5 : SUCCESS in 18m 41s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/466473eadef44ceab85759dd40b8b869 : SUCCESS in 13m 43s","accounts_in_message":[],"_revision_number":1},{"id":"421874a1a768f5b4f036c271cc024aad95169fb6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-10 14:31:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"402f81e56c141e9c3230e6db2825e50ad7b26b23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 16:25:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a8eae85fc2a4c15864587b7e632fbcd : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed27605aef384e6c9e91e3cff5578162 : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3458a0e4d80d4252886163196fae16fb : SUCCESS in 8m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3a0889852f648579672b4b21f637ad2 : SUCCESS in 8m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b04080b941046e399348db2aaf4d457 : SUCCESS in 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07525ebf7b8e45f1a450e9bfa30725af : SUCCESS in 5m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1e45314602164edd98b40583f599ff22 : SUCCESS in 8m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4ebd77c7b68b4af68718e05090bb61e3 : SUCCESS in 14m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6584a65216584bf795f0081189f6bc0f : SUCCESS in 31m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/814e7d61331f40cb9f459d7a89874d76 : SUCCESS in 1h 10m 50s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09ed3eb02c4444cc979e0ef510d51b04 : SUCCESS in 51m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/578e510472c54247b5e9a256ff6dbcd7 : FAILURE in 1h 15m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/08db0249334047f9af853cc4d2c7adcd : SUCCESS in 22m 18s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1030b09334bd4d66a071c53444ed1356 : SUCCESS in 17m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d01c9776529c4723977cfd106d53ca1d : SUCCESS in 1h 48m 35s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89f91efcd9fe40ec98eafc902c96340c : SUCCESS in 1h 09m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a5ebfd437a634b4f852a8cc5612e89d3 : SUCCESS in 42m 18s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9fd3ad733be4a43b85015319a8595e0 : SUCCESS in 37m 55s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b047553fba8b426280f3bd89513ebffb : SUCCESS in 29m 52s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3979e62170d46e1b0a42013ab9ae410 : SUCCESS in 27m 34s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/57eb44d0e531412c875e2391dcf2818c : SUCCESS in 18m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a139b7d2f2f4217a135384b374234fd : SUCCESS in 11m 26s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/02c5672736294cbcba827c02e2b20aa7 : SUCCESS in 19m 12s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/797e9096a0f34aa3a97c4910b96e1856 : SUCCESS in 16m 03s","accounts_in_message":[],"_revision_number":1},{"id":"2d1adf7ddf14bba4dbdd57487a4ec5ad3fa5a8ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-03-10 17:22:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bb8a7247308770ea19cd27cca24fff6301a34f32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 19:31:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bf0081912694b82a0bc433a4cf0040a : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7212d65728af4bf4b1d1878169e7d7b0 : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe303934e82a4edaa2d4b13c31abf3f4 : SUCCESS in 8m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d553c9ce07104027a56a762a223353c1 : SUCCESS in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dbfb163987a4a6c808f895e260eea0b : SUCCESS in 10m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3def918cac94d2d8d51c99b05399d5f : SUCCESS in 5m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c31fe0b6f70246568652988f15f2a84b : SUCCESS in 9m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/123528298e214f20b7ccb16f7f77a608 : SUCCESS in 16m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4b15b43c294248f1a12defb808fa391a : SUCCESS in 37m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/904e51d338b14f488372a9176094e570 : SUCCESS in 1h 18m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16956303cc334221b558c11a1633fdd5 : SUCCESS in 1h 14m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/98d937e4e1d846409fc0db1b73f33a55 : FAILURE in 1h 43m 39s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/eec7aec37a2c4e5a947d4e4e02eaa02e : SUCCESS in 23m 18s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3072d8341acd49fc8e5a6e62f0effd4c : SUCCESS in 15m 53s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c3dd64f29890445d87db0582a8774cbb : SUCCESS in 2h 02m 17s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7173ab79ecc0437eacd28d4d9452cc05 : SUCCESS in 1h 12m 36s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/457894e443954247b044ef5723a55711 : SUCCESS in 49m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/980aff8b4c52413684757ba0c7a449e0 : SUCCESS in 37m 38s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6744a3ca76904193a95f7c53ce68d247 : SUCCESS in 30m 12s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4590b070d81b4e48b2d671c2ac61e1fa : SUCCESS in 27m 03s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ffbadd9a132d483d8b1e211b216eb683 : SUCCESS in 17m 57s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8155735909144a8b66bdbb7561b6aa4 : SUCCESS in 13m 38s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/4df1e229d42a4f9ca90edf96e52a7440 : SUCCESS in 21m 28s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9298c267593740a5947ca20220fcc01e : SUCCESS in 14m 34s","accounts_in_message":[],"_revision_number":2},{"id":"079ab2facdd9217626aba699bd32231f37e767b7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-11 09:45:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f056795d98c8306773bfe5a08ed1e1575331461d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-16 14:40:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cf35068eb793860385ac61510f9f45d5fba7c9e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-16 15:30:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cabb69be6b2e5f098d6deb6f5c376bb9cc74d9bd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-16 15:31:24.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4a5b674a62131bb0b5f4fa1e653fc9d80afe56ae","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-16 17:08:23.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8eebebe7a220bcf93c78eeec6bc0ddaa0b94a60f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 17:37:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0700733da92456bb9d2e194f73fa170 : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb96f4b6d96c4dbaa7c0d10496bc2b87 : SUCCESS in 8m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d95990c7830a4779a00925c5c9272af8 : SUCCESS in 7m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33ec23b480f74456a790f6df78e287bb : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/697e8bd82d2e4714abb47fb78129da7b : SUCCESS in 10m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78205c255d274c48bd7d6477ea8b7caa : SUCCESS in 5m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e81ff7c95bcb474e85ed009f5b976147 : SUCCESS in 9m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0d8d47ad92fa412ca4d6ad1211d6359e : SUCCESS in 16m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5e5bf69217cb4a02b7312286425fb210 : SUCCESS in 30m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2f941a2c0e094e65a5050bf187911c6a : FAILURE in 1h 11m 56s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0d6fe54cf1344d199f06dcac8731cc5 : SUCCESS in 57m 36s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a5e3b7cb094f46199178777ae7e4f630 : SUCCESS in 21m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/444a4fa7cde94b40b5a6645266fe091f : SUCCESS in 17m 02s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5322e61c30dc4c83896d95dd91793683 : SUCCESS in 2h 02m 26s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13e6510c8b24425bb3471088d9a80dfe : SUCCESS in 1h 18m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e91ad4ed182e4b3a862bcb0e3ed55d52 : SUCCESS in 51m 21s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcf1dd9fac7e4536856bfeaf21e690c2 : SUCCESS in 40m 52s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ea5d5171d64c491a97e51f7a855de821 : SUCCESS in 49m 30s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/242905b4341b4da28fc4d459cd0d1378 : SUCCESS in 34m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b81c0777e50f497a8a8bd175e880a89d : SUCCESS in 23m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbbce193595644bbbeb2fa77ef0f212f : SUCCESS in 13m 48s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/79bdfaa4b7124bf4a494cfe653a165f4 : SUCCESS in 19m 43s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/745e35297bcf46e587092a5db78cf4bb : SUCCESS in 14m 53s","accounts_in_message":[],"_revision_number":3},{"id":"561ccb5339ddb0b257b5791ff167f59c76f90b63","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-16 21:51:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"06e5946c163279c94ac975719c5d5d335fab76e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 23:43:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a26ce5256988414c8ef5c1ce1fa69c60 : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41b6f11b149b4b9dae130f35d07b44e4 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e05bcba0eca04e7d90e8a04b7e11dc29 : SUCCESS in 7m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23c8de50d1cf469997eac365f4547972 : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69f64e78ee2b40b3abda722b8e1dd48b : SUCCESS in 11m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d723e4552b3d4f93ae2a2040f4425b4e : SUCCESS in 5m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/add4168c772e4b93929db899fe94feaf : SUCCESS in 7m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/78f08b05f3ed4300b9606c71f545d2fe : SUCCESS in 15m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8038e346725e4c1aacc5efbf1229c6f1 : SUCCESS in 36m 40s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8d41995c4a534a8b980edc02abc33de2 : SUCCESS in 1h 07m 49s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9e68cbd38bd46aaaa14dbc3f5c503ed : SUCCESS in 57m 12s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1a3a73d90e694aa59b3ae2c1913fb196 : SUCCESS in 1h 09m 59s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3acee38d5fa44f30b5501c47d9f120d5 : SUCCESS in 23m 31s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0dba2d95da642268c04ba9571406523 : SUCCESS in 16m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6bfb54b7fc18445da4502337c5097a60 : SUCCESS in 1h 50m 29s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e2238e4618c48889905be89c62eb597 : SUCCESS in 1h 16m 19s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2700dc86da5d4818925ac7df85f5e622 : SUCCESS in 56m 12s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab2dc8676a294a5a8f1ae4f1c06d2721 : SUCCESS in 35m 07s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a087a6bef7dc45b1b0653ce995dddc62 : SUCCESS in 46m 52s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72d9dfb51bf841caacb2e10fa5ba6ada : SUCCESS in 32m 27s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/43a1f064a72f4f5a85093db2218de1c6 : SUCCESS in 17m 59s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/baaf3971eabf4f50a63e78f5452a05b6 : SUCCESS in 10m 12s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/01cabd3ad41b472bb2c690f40b598506 : SUCCESS in 18m 07s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a5a39ce370f477dbc9d9593ae7987f0 : SUCCESS in 15m 24s","accounts_in_message":[],"_revision_number":3},{"id":"5bb76a01aa5918f86ddc9d8568de399c79f20095","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 23:44:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d469e729538f7cc8ad1452f2f6f1e68cfdacf4c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 02:01:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/466676b4fe9746aa994972c5fa6e2cef : SUCCESS in 8m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5cba9b1478242f789a24cbfd289c7b3 : SUCCESS in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64bfe69f9f5947b185bb5187fe2a85a3 : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9570de93c944a5b971b0d19e14bb05d : SUCCESS in 10m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d4e531fa43c4835b276dd0c771714c7 : SUCCESS in 4m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6f686e7856994b5393302a7ccf082aa5 : SUCCESS in 7m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1c331b75caa8409087d1d77c708e3e62 : SUCCESS in 11m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/84476a4793724d8c8b9ed3b294508fcf : SUCCESS in 34m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c0583647fa7a436a915def5ef9a2516b : SUCCESS in 55m 09s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c18b3dd689f4aa1b893e796417fe98a : SUCCESS in 53m 11s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/29e890d8781449179c99ca0a8addcd81 : SUCCESS in 53m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c2402cfe7ce043fa86b8ad07703161e9 : SUCCESS in 25m 39s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a07cfc696314d48864f42f1970a457d : SUCCESS in 16m 24s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e8e33d3da4274afb81094cd268cf9bfa : SUCCESS in 2h 08m 31s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae8c81dc108047508449e3e6396d5a38 : SUCCESS in 1h 00m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/57ecd8df482a40979d282a074fd3b4b7 : SUCCESS in 46m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/634c275c7f994407b4d4778249134dff : SUCCESS in 34m 36s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9f6b7d6fb63d4d69ac97e9e566b84a1b : SUCCESS in 35m 12s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c881f90cdd314c8bbae62243531715b5 : SUCCESS in 32m 26s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/779b76835ed54710946e917c0f07050e : SUCCESS in 14m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bcf9e47beb894d92b99fd4b88c9ff74b : SUCCESS in 14m 25s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/95f3c3b5751f4c2895bf12725c346512 : SUCCESS in 21m 06s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/228cb7c60e964954b27c55e271ee2c4f : SUCCESS in 13m 08s","accounts_in_message":[],"_revision_number":3},{"id":"575cf313dab38f59dd7e76b1f9ef08e3a00b1257","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 02:01:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b63cb827074b249e66111716905271bada3202d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 02:05:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/978a9d89e1844dd7919c46653dc7480c : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f20fe971db11459984d4584da2dccc2b : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dd956189edc4b031a40453a8af71262d05513935","revisions":{"be622e720b18a70166af694bc2d6c15210cd3ef9":{"kind":"REWORK","_number":1,"created":"2022-03-10 12:33:50.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/52/833052/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/833052/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/833052/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/833052/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/833052/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/833052/1"}}},"commit":{"parents":[{"commit":"678b915aff7b9b7f6286de07c106451537754396","subject":"Merge \"CI: remove qemu-utils installation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/678b915aff7b9b7f6286de07c106451537754396"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-03-10 12:22:48.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-03-10 12:22:48.000000000","tz":0},"subject":"Use naming convention to infer VLAN tagging","message":"Use naming convention to infer VLAN tagging\n\nAny interface not matching the convention: \u003cinterface\u003e.\u003cvlan id\u003e\nwill be assumed to untagged when generating the config drive. The\nlink type will be set to `phy` in `network_data.json` instead of\n`vlan`. This prevents invalid config drive configuration from being\ngenerated.\n\nChange-Id: I00e7615581c7a214e0ee8e8f04cacdbb3ea4131a\nStory: 2009910\nTask: 44736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be622e720b18a70166af694bc2d6c15210cd3ef9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be622e720b18a70166af694bc2d6c15210cd3ef9"}]},"branch":"refs/heads/master"},"88f604333a93dbd32769e4e8b5b20ac73e83b180":{"kind":"REWORK","_number":2,"created":"2022-03-10 17:22:20.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/52/833052/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/833052/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/833052/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/833052/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/833052/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/833052/2"}}},"commit":{"parents":[{"commit":"678b915aff7b9b7f6286de07c106451537754396","subject":"Merge \"CI: remove qemu-utils installation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/678b915aff7b9b7f6286de07c106451537754396"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-03-10 12:22:48.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-03-10 17:22:04.000000000","tz":0},"subject":"Use naming convention to infer VLAN tagging","message":"Use naming convention to infer VLAN tagging\n\nAny interface not matching the convention: \u003cinterface\u003e.\u003cvlan id\u003e\nwill be assumed to untagged when generating the config drive. The\nlink type will be set to `phy` in `network_data.json` instead of\n`vlan`. This prevents invalid config drive configuration from being\ngenerated.\n\nChange-Id: I00e7615581c7a214e0ee8e8f04cacdbb3ea4131a\nStory: 2009910\nTask: 44736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/88f604333a93dbd32769e4e8b5b20ac73e83b180"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/88f604333a93dbd32769e4e8b5b20ac73e83b180"}]},"branch":"refs/heads/master"},"dd956189edc4b031a40453a8af71262d05513935":{"kind":"REWORK","_number":3,"created":"2022-03-16 15:30:56.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/833052/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/833052/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/833052/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/833052/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/833052/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/833052/3"}}},"commit":{"parents":[{"commit":"678b915aff7b9b7f6286de07c106451537754396","subject":"Merge \"CI: remove qemu-utils installation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/678b915aff7b9b7f6286de07c106451537754396"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-03-10 12:22:48.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-16 15:30:19.000000000","tz":60},"subject":"Use naming convention to infer VLAN tagging","message":"Use naming convention to infer VLAN tagging\n\nAny interface not matching the convention: \u003cinterface\u003e.\u003cvlan id\u003e will be\nassumed to untagged when generating the config drive. The link type will\nbe set to `phy` in `network_data.json` instead of `vlan`. This prevents\ninvalid config drive configuration from being generated.\n\nChange-Id: I00e7615581c7a214e0ee8e8f04cacdbb3ea4131a\nStory: 2009910\nTask: 44736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd956189edc4b031a40453a8af71262d05513935"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd956189edc4b031a40453a8af71262d05513935"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
