)]}'
{"id":"openstack%2Fkayobe~833187","triplet_id":"openstack%2Fkayobe~master~I184319a9b1659a675c01240c600491f53e2cc4b7","project":"openstack/kayobe","branch":"master","topic":"tls-job-failures-raw","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-03-11 16:52:24.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I184319a9b1659a675c01240c600491f53e2cc4b7","subject":"[DNM] Try using raw images with ironic","status":"ABANDONED","created":"2022-03-10 22:27:07.000000000","updated":"2022-03-11 16:52:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"561d84643a9adf9413cd4f2ec6fc745cc72b4adb","_number":833187,"virtual_id_number":833187,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-03-11 11:11:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-10 23:41:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ebbfb8158e40aeb7b757e2b40812fbcb78f6f3e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-10 22:27:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c90a60f74328eac845c5994b3077e7baed5c2aab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 23:41:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e1c03546f7747438ade34cf7c4b2513 : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14ef61a207e24a24ba8a0d8f01fe8302 : SUCCESS in 8m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbdaf07ee3434613b074bcb88b96f054 : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a9d22fb09664765bdf0761b065c7860 : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a0f40a34fd74ffdbb105deb4575cd8d : SUCCESS in 10m 28s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/dff16e05acf3412fb3c885b6c6f067ca : FAILURE in 1h 12m 49s\n- kayobe-overcloud-tls-centos8s-1 https://zuul.opendev.org/t/openstack/build/a5b8611d8020452a9ab2abcad9784b30 : FAILURE in 1h 04m 58s\n- kayobe-overcloud-tls-centos8s-2 https://zuul.opendev.org/t/openstack/build/ee60a9d4d00445d1b976f86aae63f2ba : FAILURE in 1h 05m 09s\n- kayobe-overcloud-tls-centos8s-3 https://zuul.opendev.org/t/openstack/build/c482a82034e94617b32d0dde807aefdd : FAILURE in 1h 04m 38s\n- kayobe-overcloud-tls-centos8s-4 https://zuul.opendev.org/t/openstack/build/4c129801e60249f88f49bb3db36872d0 : FAILURE in 1h 05m 00s","accounts_in_message":[],"_revision_number":1},{"id":"3a2fa21a0ec7646170da663289a79a36f2f45ddc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-11 08:22:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d42c0868a621dd7958021b52a38a7ccc9343c6c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 09:26:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75de945bf9a047a69cd91e3bcecec064 : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b53940fbd4494920a3113b74ec5c5b5c : SUCCESS in 8m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30eb2994990b4a52af3b9dc6014d1c0b : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68472ada36a8417195e48c771e29ec6f : SUCCESS in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/257652a0bc194c27b097c52c8eceb0b7 : SUCCESS in 11m 23s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/99089b1c7e974e49ad321ff66841ef92 : FAILURE in 1h 02m 06s","accounts_in_message":[],"_revision_number":2},{"id":"883c0fd58065489500baadd240fdf115e69ba410","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-11 09:33:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"37fafaaf04c33d3bff8a4dab7192ee5a87f87565","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 11:11:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fb8c6ade9f04094b16435e78b1e7225 : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c748049cbf047cfb02d5252821038bb : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6207f55a324f440fa0814ad7ac69f123 : SUCCESS in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f223019e7db345de992aa1647acf665e : SUCCESS in 7m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2e66057626b4d63bebcaa1d4998b170 : SUCCESS in 10m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/de1e20dcced343718af1ba1dbf59008b : FAILURE in 1h 34m 45s","accounts_in_message":[],"_revision_number":3},{"id":"561d84643a9adf9413cd4f2ec6fc745cc72b4adb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-11 16:52:24.000000000","message":"Abandoned\n\nThis doesn\u0027t appear to help.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b1f003bfe6e97b669ffda0082713dbfdc67be495","revisions":{"df725b0fe6da12ea27adeda2966f6e758e5a44e9":{"kind":"REWORK","_number":1,"created":"2022-03-10 22:27:07.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/87/833187/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/87/833187/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/833187/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/833187/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/833187/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/87/833187/1"}}},"commit":{"parents":[{"commit":"678b915aff7b9b7f6286de07c106451537754396","subject":"Merge \"CI: remove qemu-utils installation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/678b915aff7b9b7f6286de07c106451537754396"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-10 22:26:57.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-10 22:26:57.000000000","tz":60},"subject":"[DNM] Try using raw images with ironic","message":"[DNM] Try using raw images with ironic\n\nChange-Id: I184319a9b1659a675c01240c600491f53e2cc4b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/df725b0fe6da12ea27adeda2966f6e758e5a44e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/df725b0fe6da12ea27adeda2966f6e758e5a44e9"}]},"branch":"refs/heads/master"},"efee770abd5af6b25da0c6b659635ea23db46aa6":{"kind":"REWORK","_number":2,"created":"2022-03-11 08:22:38.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/87/833187/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/87/833187/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/833187/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/833187/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/833187/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/87/833187/2"}}},"commit":{"parents":[{"commit":"678b915aff7b9b7f6286de07c106451537754396","subject":"Merge \"CI: remove qemu-utils installation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/678b915aff7b9b7f6286de07c106451537754396"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-11 08:22:15.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-11 08:22:15.000000000","tz":60},"subject":"[DNM] Try using raw images with ironic","message":"[DNM] Try using raw images with ironic\n\nChange-Id: I184319a9b1659a675c01240c600491f53e2cc4b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/efee770abd5af6b25da0c6b659635ea23db46aa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/efee770abd5af6b25da0c6b659635ea23db46aa6"}]},"branch":"refs/heads/master"},"b1f003bfe6e97b669ffda0082713dbfdc67be495":{"kind":"REWORK","_number":3,"created":"2022-03-11 09:33:15.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/87/833187/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/87/833187/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/833187/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/833187/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/833187/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/87/833187/3"}}},"commit":{"parents":[{"commit":"678b915aff7b9b7f6286de07c106451537754396","subject":"Merge \"CI: remove qemu-utils installation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/678b915aff7b9b7f6286de07c106451537754396"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-11 09:33:09.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-11 09:33:09.000000000","tz":60},"subject":"[DNM] Try using raw images with ironic","message":"[DNM] Try using raw images with ironic\n\nChange-Id: I184319a9b1659a675c01240c600491f53e2cc4b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1f003bfe6e97b669ffda0082713dbfdc67be495"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1f003bfe6e97b669ffda0082713dbfdc67be495"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
