)]}'
{"id":"openstack%2Fkayobe~833428","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I8f5f93340642d055cce7ef306d942e75b10c86a9","project":"openstack/kayobe","branch":"stable/xena","topic":"qemu-img-stable/xena","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-03-15 10:05:36.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"I8f5f93340642d055cce7ef306d942e75b10c86a9","subject":"CI: remove qemu-utils installation","status":"MERGED","created":"2022-03-14 09:11:16.000000000","updated":"2022-03-15 13:55:30.000000000","submitted":"2022-03-15 13:53:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"833428-qemu-img-stable/xena","cherry_pick_of_change":831407,"cherry_pick_of_patch_set":3,"meta_rev_id":"0795db8e87fc626021949887e4b02877bb887917","_number":833428,"virtual_id_number":833428,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-15 13:53:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-03-14 09:22:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-14 11:39:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-14 11:39:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-14 09:22:25.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-03-14 11:20:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"75f47a78766038c20873c565457515a46bb41586","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-14 09:11:16.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"29f0d3fc11f836c68aaecc473553d2d4e510a31f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-14 09:22:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a9a6b1149fe9702f05c7a5212021344daa558619","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 11:20:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/471f969c8ac04d4892c03cdfabe90fce : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3fa259ebb5445598764478a89c27633 : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61050272d56c45659524de3c9166c820 : SUCCESS in 8m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d25e810a20742ceb6198d0a499097e5 : SUCCESS in 10m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c5b955b3b9b423490fc0e11ecbd1867 : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/700e2ab6b76a4a1486f263298565da87 : SUCCESS in 11m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/067e7b5732a14d39b95bc24a90733993 : SUCCESS in 8m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6deb1d3a44924728830340919db2d69f : SUCCESS in 14m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2a4fce84785141de84f9c3f5726f07d8 : SUCCESS in 36m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/91477463c30345be956372671d1b1853 : SUCCESS in 1h 13m 38s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb9b624b3f06449485b71a5fe2c2b3ac : SUCCESS in 1h 02m 08s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/25e30dc25c944b51b34f98dc3022aba1 : SUCCESS in 1h 20m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/91b8287ff1694b3782eee393ca1102b5 : SUCCESS in 27m 18s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d87f175a67384b1e84142579a0f35850 : SUCCESS in 16m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a30ee9d959604be2b0b45bc2b6ce1ea0 : SUCCESS in 2h 02m 33s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/441cd57efee9405ba2e3b14aa92e1abf : SUCCESS in 1h 19m 55s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c1ddb18a75ac4dfbb10d5a3f464513f6 : SUCCESS in 43m 11s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a75fbd0dd7b42959a7daefc0cecf7ba : SUCCESS in 31m 34s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0d208375e29342a88a8b6e607ed83689 : SUCCESS in 35m 51s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5d8e0969ccf47d48310b5f82d17bda1 : SUCCESS in 25m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e5abdce905f643719cfe510fa4b66037 : SUCCESS in 20m 58s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ff405408d6e480ea88871f7578e78ea : SUCCESS in 14m 30s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/f3fd2a82cf62415fb2f8d20d25896b72 : SUCCESS in 21m 12s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d498a650e6b24fd883c85983534be52a : SUCCESS in 16m 07s","accounts_in_message":[],"_revision_number":1},{"id":"46f8bdbfb381e164d765691751d1e94d39f624fc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-14 11:39:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2be6dd16106984f27338032c8f429b44f8d714db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 11:40:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"405400a3f1d25ebb7c702729c79b1943ac51c2d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 13:42:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7005da0b78454ff99bfc2d6c056a3159 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1620d7a0f3294e118a8d501ca2fd22be : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c32ce145ff31453e8d0ee309d66a807e : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86e425eef6884487b212e961660307af : SUCCESS in 10m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7563c359bbee4af5a25f7e3753efaa92 : SUCCESS in 7m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/21194dd48a194def87159a29e48b3434 : SUCCESS in 13m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2d9964c9749247faab5755484a6b85df : SUCCESS in 33m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fb1e0ed923f7404e8e9d9223f6e44909 : SUCCESS in 1h 25m 49s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a4b3b9998704a49adb163aa070a7b34 : SUCCESS in 1h 01m 52s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/32b9cfb49e6a4782861bb6f22110b2a2 : SUCCESS in 1h 18m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/35123bc3572048ee9e207686d20c80cc : SUCCESS in 24m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f27ddb6512bf49e4a96d510a7c9ed616 : SUCCESS in 15m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e066f81cd3b545749269cadd2a48315b : SUCCESS in 1h 58m 20s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2fcfbd2ec394e22b29174852d4c8324 : SUCCESS in 1h 27m 55s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/20397d67b08b4246a56757b0d846ed01 : SUCCESS in 38m 04s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad6f3da3d07f4fe899dfc16df67cae0b : FAILURE in 32m 24s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/eb8c1e2f25364fcb8c15e7bd884a2606 : SUCCESS in 44m 18s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acf74c03687345ddba7890943801bf67 : SUCCESS in 37m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e7199a161729459ebc77b68e99f2c210 : SUCCESS in 18m 03s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81ad5aae5c3849feae3a72f9e95c7d50 : SUCCESS in 13m 15s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/fdb44a53946849058042657dd483d6d3 : SUCCESS in 17m 36s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5991615266b4c20a23ee49747b02d8a : SUCCESS in 14m 31s","accounts_in_message":[],"_revision_number":1},{"id":"90fc498a16d4c07b656b4dfd59f18afd2f4006c1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-14 14:12:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"acc03727e7b8e84b0d8a82d7dd059124174d8556","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 16:43:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/744ab33e2ff14a368341c4010c4de41a : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cda49e6b97b469587e1ae2e4e6ea198 : SUCCESS in 8m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b478576bfdea4567b7d835e50a35716f : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f6b6892f448449db2323c09c2e866eb : SUCCESS in 7m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2d31d1057e34d539574b950bba7d285 : SUCCESS in 9m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26cbab7116a54a158f6bd4d64ccf31a7 : SUCCESS in 10m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ddf42ab2c7be492aaac4db56ad0681c9 : SUCCESS in 9m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a6f00c572bb54c8388c8366e7afc5e86 : SUCCESS in 14m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6ccea9b150814b3394dcb6fd4bd99bff : SUCCESS in 37m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6a6435bf476346519c80198e64ba71e8 : SUCCESS in 1h 06m 19s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2844e12b7ab140c684b7c5a11addf75b : SUCCESS in 58m 06s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e974c70269294612a973e959931a59dc : SUCCESS in 1h 59m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ff0a1558139e4a5198ee7c7eafa9996a : SUCCESS in 29m 01s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83c1864ccce04f0eaba85890cff9ca90 : SUCCESS in 17m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/496b091b188b4e4fb01585612cd3f5fd : SUCCESS in 2h 20m 31s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f111ba31dff046cfba27f68359924a4a : SUCCESS in 1h 30m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d55881fa8b2c4395b3cfb26754b54931 : SUCCESS in 47m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96c01058b8dc4ed4af54e772d0f8b46a : SUCCESS in 33m 29s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/52b654ea7e3b4c358420996492d7514b : SUCCESS in 35m 14s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59da6a0cee814f38925c46ce0185498e : SUCCESS in 32m 42s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1c6c92845a6b412dba6406e5565d17e3 : SUCCESS in 19m 13s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/050b665c0bc44b949a57c2eb8fe9571e : SUCCESS in 13m 37s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/93e75cddfb0a4b0dbcb506d49c78564c : SUCCESS in 21m 52s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3e8f74963424656923ff87e69e90527 : SUCCESS in 17m 26s","accounts_in_message":[],"_revision_number":1},{"id":"0accd4a4425511d95b9bb1b59764667755024be4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 16:43:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a326c119771aabab5d7f062d84e2ea182b647883","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 18:35:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65ee6b2efc8640068790b367cae1ce5f : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f87c1a5f0abb41d7bf1ae5030ceab174 : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f45722a89825468d966bedb1206fc7f3 : SUCCESS in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4bfd4755e754397abbd344f66727968 : SUCCESS in 11m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4bb3a3321e1e4eb0b4b363436eb0c94b : SUCCESS in 7m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7ee6764fe03348dda498b5d227bb3158 : SUCCESS in 15m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/548d4d86002b40e6a8e6fceb90ca1831 : SUCCESS in 36m 35s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/70c471df22b14c169ec446595ed6d864 : SUCCESS in 1h 08m 13s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f024cd78d9a34df1989443a46caedeed : SUCCESS in 59m 48s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7e74b069d0b7464bba485bd568c134d2 : SUCCESS in 1h 10m 24s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a59867dcf5354b57a5df366a8c3523b3 : SUCCESS in 30m 01s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab0379a26ee04d0a90ef881bf9dd06b4 : SUCCESS in 17m 20s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/63a8c916585b478e88f08dd07c0943ab : SUCCESS in 1h 50m 27s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03db2a4cae6848e8bdeee9d0e895d7c6 : SUCCESS in 1h 20m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b3e810d1df8f4026b93b6d575a2e8f7b : SUCCESS in 50m 28s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba74d88a6e6948018377c50626a04910 : FAILURE in 44m 41s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/170e988537084339adb89de3b7b1aa36 : SUCCESS in 30m 58s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b32e30ad87604703a0c3f90b80991d84 : SUCCESS in 27m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8667becbe2294482b8a981f4c32c4724 : SUCCESS in 21m 08s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cd9d48969ea4003b4618cfd3a57e8fe : SUCCESS in 19m 39s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/cb1e29ad613d4ce2b6c8c16e94c6cbcb : SUCCESS in 27m 29s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce21cd576b6b4fafb923f0a7ade9d970 : SUCCESS in 15m 00s","accounts_in_message":[],"_revision_number":1},{"id":"73b2f92b27d02e1223bbcac97d3bb795af4ff033","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-15 10:05:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d17464448b6b11c59247bf9449e7c0c706083d54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 12:03:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/497d24aaac4549cea2bbf0dddc25a567 : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37c5e13355d54fdea5d942151f7d9ff4 : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15e930007f904f31a884e4981ee15443 : SUCCESS in 7m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/831fc170f0d9494ab0e6e38dab0663eb : SUCCESS in 7m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/564ec198d7ae4b4f822f67d9bfd3e2cd : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cd3c458e006429fbb76acc54977fc53 : SUCCESS in 11m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/da02102eaa9c4a7cb763d570d3234467 : SUCCESS in 7m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3aea9831dc664236a2f56d6079879ea0 : SUCCESS in 14m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e8fc83bd35bf4f888f834b05caa52146 : SUCCESS in 36m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/93a0f772ad504a3a927bfd4df167f8a6 : SUCCESS in 1h 09m 53s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0b8bc2d7dbe4cbc9a3ece2ac30c4e6a : SUCCESS in 1h 02m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ae82ccaf96f14b53a7329520c847fa9d : SUCCESS in 1h 14m 38s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3f0fcadb94f94601884b36665b92d579 : SUCCESS in 26m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/013857dbf965469b964cfe3a0a5ee95d : SUCCESS in 16m 42s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3d0359c971dd4dba8f95a5132d4382d3 : SUCCESS in 1h 55m 35s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc402632eec34571a9aae6005625584b : SUCCESS in 1h 20m 04s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/63ee409199654a4aaa5bd33957ed671d : SUCCESS in 41m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4eebdf69a4a749a18f3e54b071c1210e : SUCCESS in 35m 45s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5b05d9bb40544c6b81c353645f8ff0b2 : SUCCESS in 32m 47s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8d485650c9d40d289e21fc2836b4d5d : SUCCESS in 27m 49s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7a39715539e7438a843c492e4b48616f : SUCCESS in 19m 50s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c7b54e6fba64f8f94bd8d590cd144d4 : SUCCESS in 13m 16s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/8a1ae9189b63460382721d12020b4b21 : SUCCESS in 20m 51s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aab2763ab9c341a3b4772f6d49a9c74f : SUCCESS in 16m 34s","accounts_in_message":[],"_revision_number":1},{"id":"86da419fc6af22fddefbcb551db9e253144df419","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 12:04:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e7e68a2dcc949e4dbc8e3ef5f8042705da2792d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 13:53:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5719601685e488ca702e1085733cf84 : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c521a7884de14f4683b9a6ecdc341304 : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db8b0115e8124a13963f1bfaf893fa20 : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/264034971fe24630b01974c3fecc3e3d : SUCCESS in 10m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a7c7001bbe9641f8930b9baaafe23112 : SUCCESS in 7m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/985a4b315b9741ff9c31d1b2019d67c8 : SUCCESS in 16m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/35a9e9e56e3c467fb56fdbaaf652116e : SUCCESS in 35m 55s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0340146cb34249bba322988ff578a59d : SUCCESS in 1h 21m 07s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7d082e4de8e47aa95778961ea4b903c : SUCCESS in 1h 04m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/95aba43c961c433da9622e1714c8854e : SUCCESS in 1h 16m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/64abae00fcd642359817017a2022fc56 : SUCCESS in 28m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34753aa846c44f8ea431366267cfb91e : SUCCESS in 16m 23s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2b9a198876d546c3be5f4741a1cc3c52 : SUCCESS in 1h 46m 10s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95185ccb08dd4626b201ff2b2a47916f : SUCCESS in 1h 07m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3c9083f227644347b58cb1475786f2d7 : SUCCESS in 44m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37e5a474b9f44da9abb3c08aeafb1586 : SUCCESS in 33m 42s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9e928b86af304727a4953349de0911f5 : SUCCESS in 33m 07s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39e4d4a502734cb9bfddcb684ab7e566 : SUCCESS in 30m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ed8417bc7ba04c10aa9ec51fdb79f0d1 : SUCCESS in 18m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc87f370bbe54d21bcb1278c77f50058 : SUCCESS in 14m 49s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/2991eac0f90940dfb5ab15048e3da7af : SUCCESS in 16m 42s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f9656e6314b4ab49ca9d9ede382a366 : SUCCESS in 15m 02s","accounts_in_message":[],"_revision_number":1},{"id":"48746d2663546fdc4c1a40d8b20baddbbf8eff49","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 13:53:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0795db8e87fc626021949887e4b02877bb887917","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 13:55:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd5a1310fa364f838f873266aec4e228 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15790c983e9f88f8bd735f7e7cb2b6db605adee2","revisions":{"15790c983e9f88f8bd735f7e7cb2b6db605adee2":{"kind":"REWORK","_number":1,"created":"2022-03-14 09:11:16.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/28/833428/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/28/833428/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/833428/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/833428/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/833428/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/28/833428/1"}}},"commit":{"parents":[{"commit":"d5006cc65b739ec1d230fe02363a8d8e048bb2c7","subject":"CI: stop using zuul as kayobe_ansible_user in TLS jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d5006cc65b739ec1d230fe02363a8d8e048bb2c7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-03 09:01:21.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-14 09:11:16.000000000","tz":0},"subject":"CI: remove qemu-utils installation","message":"CI: remove qemu-utils installation\n\nnova-libvirt images now include qemu-utils on master [1], xena [2] and\nwallaby [3].\n\n[1] https://review.opendev.org/c/openstack/kolla/+/830401\n[2] https://review.opendev.org/c/openstack/kolla/+/831411\n[3] https://review.opendev.org/c/openstack/kolla/+/832892\n\nChange-Id: I8f5f93340642d055cce7ef306d942e75b10c86a9\n(cherry picked from commit e453f3c57053bf52b4b70d0a1c1776d360170157)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15790c983e9f88f8bd735f7e7cb2b6db605adee2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15790c983e9f88f8bd735f7e7cb2b6db605adee2"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
