)]}'
{"id":"openstack%2Fkayobe~833429","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I8f5f93340642d055cce7ef306d942e75b10c86a9","project":"openstack/kayobe","branch":"stable/wallaby","topic":"qemu-img-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-18 17:22:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8f5f93340642d055cce7ef306d942e75b10c86a9","subject":"CI: remove qemu-utils installation","status":"MERGED","created":"2022-03-14 09:22:35.000000000","updated":"2022-03-18 17:24:12.000000000","submitted":"2022-03-18 17:22:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"833429-qemu-img-stable/xena-stable/wallaby","cherry_pick_of_change":833428,"cherry_pick_of_patch_set":1,"meta_rev_id":"e39c17351f2d598b0ec9509ca56635bdd5aa1a23","_number":833429,"virtual_id_number":833429,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-18 17:22:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-03-15 10:22:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-03-15 09:55:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-15 10:05:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-03-15 10:22:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-15 10:05:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-14 09:22:35.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-03-14 11:11:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-15 10:05:41.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-03-15 10:22:01.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"e52ade26733cf0159ec98d11f33da1bc62cd8ec5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-14 09:22:35.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"3c43c26ea3c2d9f13f24ef377a66aa803f22e536","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 11:11:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b68419435b240c3af8e3ac351ef8024 : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb828261358b40158d5aad051f2e7780 : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8279885ab914527a468091fbbd46f74 : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdd2c9c33c224fa29a597422aaca2669 : SUCCESS in 7m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f31e026af034cc4a933776feebad0af : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daf2085faef04953bb62da5c6b148b78 : SUCCESS in 10m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/761ac155d0b84479b6b823bac0ce469b : SUCCESS in 7m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/05c316ec7acc437cbe541c28285a0143 : SUCCESS in 12m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b879b3895a5445c8987f43d336061ac2 : SUCCESS in 18m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7b5e0882b5ec4c8d8a2a2476a60b7f92 : SUCCESS in 1h 09m 22s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/efafea7cd5a846a884ac8a2ad04762b8 : SUCCESS in 1h 00m 14s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/807faa976cf249ad94cb5682a12b35db : SUCCESS in 1h 15m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a98dd74281e74f09b4696a1fa4ea2bec : SUCCESS in 23m 01s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c13197b98c9345179f36139e4ec26539 : SUCCESS in 14m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d1470989847542f29ed4b52a80dc3691 : SUCCESS in 1h 46m 43s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4ebcaf05fa3a4829980d1e5d9be2d246 : FAILURE in 23m 32s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1885e8c7581648eb99161f0d07569406 : SUCCESS in 30m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/157fbd21590d477f8ef34c5656495513 : SUCCESS in 29m 41s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/94b5c499d74d4dcfa21cd13080ea1e1b : SUCCESS in 19m 27s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b80e3b3c2014fa8aefc957a41bd52c0 : SUCCESS in 12m 44s","accounts_in_message":[],"_revision_number":1},{"id":"e6be04dc6f05cc1ea1ef5350bc0ed19b6362d501","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-14 11:40:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d67d11878c360afcd913c8af38334aaf161c50b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 13:57:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b84525e6ff1540558f49375505ea48eb : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5afa871599d8486fab3b68879e4de7eb : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22c6625d8d694624b68b633a61cd3558 : SUCCESS in 7m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8300475ab0cb4dc08ff08e927acd3241 : SUCCESS in 7m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88c4887184cc4739bf0738113e3c4dc3 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09784c1a0efb40bbb87ade5611447c77 : SUCCESS in 10m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bf404403e1214461be78461dc492fa91 : SUCCESS in 7m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9113bf3e548743bc8541f88a7cde113c : SUCCESS in 11m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ac7f1c6e523a44e7ab240f27ef16fe45 : SUCCESS in 20m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6a3f8d7976c44e00af7187419e4a1884 : SUCCESS in 1h 25m 25s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/856d3e7c82a44aa090fc5d15743aa9ff : SUCCESS in 57m 57s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b9d8bead690d4241a05161e204a7d53d : SUCCESS in 1h 21m 09s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2f368687446d43238c8b086096ebf182 : SUCCESS in 24m 07s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84b399664ef24682b9c243d2931b7b55 : SUCCESS in 14m 51s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/da63f34d3c6f4f078688aed2d7d44967 : SUCCESS in 2h 10m 10s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e723633221bf4ae7baa7197db86ab242 : SUCCESS in 43m 43s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01617d1868b44603a2a5d6e37b030cf7 : SUCCESS in 33m 35s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/749cf0f165ef488bb6718f750f393a66 : SUCCESS in 32m 38s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/cf1e2d874f404dc48bd1968f263f91c8 : SUCCESS in 18m 26s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c82236840084e2aac27aafa116a6fd2 : SUCCESS in 12m 01s","accounts_in_message":[],"_revision_number":1},{"id":"5cd16f49544eca5e650ddd1292cf1367048fedbc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-14 14:51:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2ed96494daa61ffa515b7f7cec4d290680144d9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 17:00:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fd9faf94b3447eab8dc83260d437328 : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f72c150c213e47e98b52f860204e0d3a : SUCCESS in 8m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7b9fbef0ee74dbda69ecd233eee64ef : SUCCESS in 8m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8cc9d745ab4a4a80ac609129f002d111 : SUCCESS in 8m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f887e746a7354ff38c15d048411ef7b2 : SUCCESS in 8m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef0c6d6da7334deb8e332d3a3cb0771c : SUCCESS in 11m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f13874d5564a4d27a9dd9e4286f818a4 : SUCCESS in 8m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b3dc68d608f545ba98e7bbbc0cd0e623 : SUCCESS in 14m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e00f2391eb324eb98a9cbd05956e01c7 : SUCCESS in 20m 48s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6582febbc2f049cfaab8d51c53f79e98 : SUCCESS in 1h 13m 05s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f60b89c3a6d444b49bd6483b7a76eee6 : SUCCESS in 57m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/70bbcdf7f9e84d009f9837e16ac698c3 : SUCCESS in 1h 05m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/db42e882d18d4147a14e541b100208ef : SUCCESS in 24m 15s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21298eba4bc34612bb8e31a5437342a9 : SUCCESS in 16m 13s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4c27890d1ff8443ab15d9586191255a8 : SUCCESS in 2h 05m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/448c0d6815e0498c8678c5fb18276edb : SUCCESS in 45m 53s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d9567537aa54258945893dcea369e3e : SUCCESS in 31m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bbaefeb59c7d49ceb115be5226475321 : SUCCESS in 28m 29s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/eeb8040cd32e48859a7a28bff402f225 : SUCCESS in 17m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5365f02504641bebc196ddd0c9050f2 : SUCCESS in 15m 19s","accounts_in_message":[],"_revision_number":1},{"id":"df564ab8a25d7270c55981419a1543e4791f44ed","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-15 09:55:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"048d56afa14a037f728d6a379a7bd2640b1afcc6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-15 10:05:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aa46dbac73d7dbea6d8c00fe91e999e190d97abc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 10:06:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"97b2dddcff5e309edf8bc91c731cf37c7a397496","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-15 10:22:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"21caaeb9858bb96a0bef3c849cbf65e7f0cc5243","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 11:49:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5c3b26cd04a48ecb46fe47fec98aa3e : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f587b8803d944a01997b491e84d55894 : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69045b9ec1db43b9a43f345ceaad6d3a : SUCCESS in 7m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27a0dd442a4c48b3811f134581eef938 : SUCCESS in 7m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44b9378341cf46b6956acbb88028a3cd : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef1fa99600284a43a0409c7640cc6e80 : SUCCESS in 10m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/603e0a1efb4b4c0298450328da2ee40c : FAILURE in 8m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4f77efa0cbd44220aeee4a74aa8f9e8a : FAILURE in 8m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3ad52ee4e1e94046a689a46317fa19a3 : FAILURE in 9m 22s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f925cf87249d4c92a1b756e7d76f9524 : SUCCESS in 1h 25m 55s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afd2d20f65b647bfaa081200e4d0fb3a : SUCCESS in 47m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7a586ceb9c42461d944e6706acaefc85 : SUCCESS in 1h 15m 05s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5c7c43743982490388ac187cefc0f476 : SUCCESS in 19m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aeb13779938b4e0c811af12a6fc6c253 : SUCCESS in 15m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1e276d4f2d7349d4accc79d7b0c40c47 : SUCCESS in 1h 40m 45s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ac65e584526b4f9d8010b0581c621947 : SUCCESS in 41m 31s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f18ea145cb814a56b02c0af61a34614b : FAILURE in 31m 07s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7eac8a769e8341aaa412177f94f27502 : SUCCESS in 30m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/47fce7baa2e24f38a09847eb5e322470 : SUCCESS in 14m 44s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/feea057ea9f44b19aba62f8bf4f9c0f8 : SUCCESS in 14m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f8c3fbce36dc4850c6d45eef159fa112362294da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 12:19:14.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/502dab4e70d34f6bbeed0ba376418f71 : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/927ddc780067482f913247aa1b438588 : SUCCESS in 8m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/814a8c8fd8bb453bb4a21bbef0337f52 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4d6adcc6f7f44529933c8f5afece544 : SUCCESS in 10m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2b41eee9556242a6a91af2644f4d3db4 : SUCCESS in 7m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4737935506b34537841b4fb0504bed06 : FAILURE in 8m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2a6c0e83fe5c4d54bf28a03ea54fdf50 : SUCCESS in 21m 21s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6997ef25867e46f6b074489943c9dda8 : SUCCESS in 1h 21m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c27d332273624a68b682cd7f35ede2e5 : SUCCESS in 1h 00m 21s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a8e65122323a4570b2e33dc1d6e16980 : SUCCESS in 1h 15m 06s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/830771f0027540fc9eec749909d4da7d : SUCCESS in 23m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/594396b149f34e00a4c846d9675cfb0c : SUCCESS in 14m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c53b61210c28452a96e8d7ec5f5859df : SUCCESS in 2h 07m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e98e0536dd974b8da8f6ab7635d86aea : SUCCESS in 39m 33s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83bf5df7f2334b948bcc92b6e4f9831a : SUCCESS in 34m 59s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3dc6e0eb13274f7580aa5f32e5932c6f : SUCCESS in 33m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/51b050ef76dd46258c4edfea73856847 : SUCCESS in 18m 17s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f933c6902a1549b790ec5ba845081ed8 : SUCCESS in 12m 58s","accounts_in_message":[],"_revision_number":1},{"id":"ed1f58a657fd8a6558d2c33bb51b2e6f4a84ac85","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-18 13:20:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f81e04021ed533b328c7c397db41882f5e10cc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 15:13:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92f633be2bde47ce9c078cf21b7501bd : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dd3e885466c49c1be35f64a81fb4f26 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e26ac4dab6fa4b16ba16c324765f906f : SUCCESS in 7m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a66b209f20842f8874336bae744d835 : SUCCESS in 7m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea2b019d94b340d492696cdaf6479470 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f2290e76aaf4503a3aa14f1bc350015 : SUCCESS in 9m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/23b2afe57a524ef4ae9e5e8578590aa3 : SUCCESS in 7m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0aebf533c525451487529db5121753a0 : SUCCESS in 12m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/659d8a6aa63946d1b6428b56363706a6 : SUCCESS in 19m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e97fc2db15e54c3f9423f8d46b5dfab0 : SUCCESS in 1h 15m 08s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd68feb63b884fc19cd7ab874101b412 : SUCCESS in 55m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/91e25444d7644350b6b3051ae3cf6805 : SUCCESS in 1h 03m 05s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/92dc7d58f497424eb8ce5e571f588eea : SUCCESS in 19m 02s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/497666d880ae40a49bb27b881d5af9ed : SUCCESS in 14m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fa007f61b2334066994889e7b537014c : SUCCESS in 1h 48m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/29ef3bff79554abe88c4619df3dce06c : SUCCESS in 37m 45s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f85de7362704e2fb97e8c0b044a92a0 : SUCCESS in 31m 30s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a3067b81d7da4a94a1d1823936f734d3 : SUCCESS in 29m 47s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4fe21a7d5fd94323b9ca9d82fe76d6fd : SUCCESS in 15m 06s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1868ca8ebdf2435b9e2e978d69fedfe6 : SUCCESS in 12m 15s","accounts_in_message":[],"_revision_number":1},{"id":"8f9bd7fe085eff2f951c2e0643ceded2c1149209","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 15:14:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"52e1b012d520aba2e966f77d1c62eb7a3a5aa47f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 17:22:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46d6076488ac45f9820fea6ad8965899 : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c118daeb228548db921cca07f7b0d181 : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94588b6620a8433b831b0b5f1561b202 : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fafdf19c1b04122b7f1c40cef1433d9 : SUCCESS in 9m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bef83ca405bc4e978a66e8c7ba2cfff2 : SUCCESS in 8m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d19040f115b140cfbc275cf038bb7655 : SUCCESS in 13m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/480a3aff5aa04e56904def07bbacb6f5 : SUCCESS in 21m 14s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9d0bad34e10a495da8940de2823edd35 : SUCCESS in 1h 11m 55s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58263f5d5c1a40918b3732b33a3a04af : SUCCESS in 59m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c79389a647e64872b64db4f6a47409aa : SUCCESS in 1h 16m 28s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d773210a6542484a84e614c5ba88e683 : SUCCESS in 23m 59s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4c901fe86ab4bff8244cab67056eb63 : SUCCESS in 15m 58s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/99d5fe2f14b14e1fbc617aee10e7a979 : SUCCESS in 2h 01m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/04f551c5ecf34cdfa0d2796b30ce72d4 : SUCCESS in 38m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2a1e3f9fe434322ae9f6024c9c732b3 : SUCCESS in 33m 22s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/152428ccc65a4350af19d778b068f781 : SUCCESS in 28m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/30c7f36abb3947c6953fc46f56cd2c86 : SUCCESS in 19m 13s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/107902896cbf4db1957c1c6672efb297 : SUCCESS in 12m 14s","accounts_in_message":[],"_revision_number":1},{"id":"ca1c49823f68c1cde8402c678041a8d67eb05985","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 17:22:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e39c17351f2d598b0ec9509ca56635bdd5aa1a23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 17:24:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e92a40b787d4deea4cf1b51c0c75947 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"00d6ad653ef47d8c4ff71745c81bec9789fbef58","revisions":{"00d6ad653ef47d8c4ff71745c81bec9789fbef58":{"kind":"REWORK","_number":1,"created":"2022-03-14 09:22:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/29/833429/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/833429/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/833429/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/833429/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/833429/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/833429/1"}}},"commit":{"parents":[{"commit":"ee6d4614f58be025a0af7c08de45831fd5c523ec","subject":"CI: stop using zuul as kayobe_ansible_user in TLS jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee6d4614f58be025a0af7c08de45831fd5c523ec"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-03 09:01:21.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-14 09:22:35.000000000","tz":0},"subject":"CI: remove qemu-utils installation","message":"CI: remove qemu-utils installation\n\nnova-libvirt images now include qemu-utils on master [1], xena [2] and\nwallaby [3].\n\n[1] https://review.opendev.org/c/openstack/kolla/+/830401\n[2] https://review.opendev.org/c/openstack/kolla/+/831411\n[3] https://review.opendev.org/c/openstack/kolla/+/832892\n\nChange-Id: I8f5f93340642d055cce7ef306d942e75b10c86a9\n(cherry picked from commit e453f3c57053bf52b4b70d0a1c1776d360170157)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/00d6ad653ef47d8c4ff71745c81bec9789fbef58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/00d6ad653ef47d8c4ff71745c81bec9789fbef58"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
