)]}'
{"id":"openstack%2Fkayobe~834018","triplet_id":"openstack%2Fkayobe~stable%2Fxena~Ib59c8bc2d8938eca18c943bb2e66ed185152a739","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-20 20:15:18.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-03-19 21:37:40.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"Ib59c8bc2d8938eca18c943bb2e66ed185152a739","subject":"CI: Don\u0027t download Cirros or IPA in seed jobs","status":"MERGED","created":"2022-03-16 17:01:36.000000000","updated":"2022-03-21 09:53:30.000000000","submitted":"2022-03-20 21:40:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834018","cherry_pick_of_change":833759,"cherry_pick_of_patch_set":2,"meta_rev_id":"1b286303342aeb5ea966aa5fdb741403c6dfc916","_number":834018,"virtual_id_number":834018,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-20 21:40:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-03-20 20:15:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-19 21:37:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-03-20 20:15:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-16 18:55:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-16 21:49:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-03-18 13:28:22.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"3ebc9da4b4a896c32b3acc4a1d955cb531ba2a66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-16 17:01:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"33cd0a05a1409c09c6b1478a9ffad9fc50c77ade","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 18:55:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1a954e4e4cd49fba83dd6df067fec4f : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9f0688df21f49ce9d3d7e854fefc6c1 : SUCCESS in 8m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0fdbda44e2a409591d978c4d2e3cca2 : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47a4c2e3758d4b119f42937f3e215069 : SUCCESS in 8m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3ae87a01ac44085926c06e9078766ae : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55661058748c4bdf82deaf234926c5ab : SUCCESS in 11m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/85e54e091ad947d78fcf94c4a9b6066d : SUCCESS in 7m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1d2aa1d5a06c4c389fa223368f062986 : SUCCESS in 16m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/02d5a0e990674f098eafea14fca9f918 : SUCCESS in 34m 25s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b351f53600d94f7f9af2c565b1a0f319 : SUCCESS in 1h 21m 30s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6abcbc5639534789a9e0605f27646eb4 : SUCCESS in 1h 00m 03s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4f104522e3bb464b92a5feb1e0250b84 : SUCCESS in 1h 02m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/485087057db14e29bc141d5edf075278 : SUCCESS in 26m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d9b6fa536fe4b07ab0e89a9463ba578 : SUCCESS in 16m 52s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/025e771d8f1244919540d27f7742cfd8 : SUCCESS in 1h 51m 16s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96efdacaa1a94f23b2f9f07b107c6c14 : SUCCESS in 1h 22m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/eeb203bd36a842859c6f4a35ecb476c5 : SUCCESS in 58m 37s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e62366f525a4a47b412347078dcde7d : FAILURE in 35m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d14662ded670456082b4e7b57a2d3fb7 : SUCCESS in 48m 34s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13e66e4759ae4202a59fd0a51a77c115 : SUCCESS in 30m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f1b5aa61bf034212a9056e28aa231459 : SUCCESS in 19m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/918b18513d344ac489a6e5e0bb553031 : SUCCESS in 17m 32s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/43e9d8673fca4dea9f64ab6df86fe2e8 : SUCCESS in 21m 01s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b0a0c7b0fc14803a7c54e7f239deeec : SUCCESS in 17m 31s","accounts_in_message":[],"_revision_number":1},{"id":"72b490a8ff8e2bae3ed888be403f49f8ba7a8069","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-16 21:49:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9f15e0f0ab3bd7516e8e32eec8bd43698e3e699","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-17 09:40:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e3d5e413583cdfc34fcbda97accfa5294ffa0754","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 11:37:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac52c25e64284ee69c1017fc974cc137 : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7c89cbbec734132bcbd5072a638a1ed : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/058937c95d7c435b88b453be2549a1c8 : SUCCESS in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ee976f4a1d24565ac8fe8f88c673f5b : SUCCESS in 8m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2ffd7dbed3348a68c71bd6216a309e7 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cdfeee994934cac86d223190cc2b214 : SUCCESS in 10m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5e68670316434b31a74c5ca9fdc1105f : SUCCESS in 6m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f449dd8269914f85ab3b64eb19f5d884 : SUCCESS in 15m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b1ddae697a8043839c927c1f08bd2844 : SUCCESS in 37m 06s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/22243c118b524eadbb9f037f66dce106 : SUCCESS in 1h 01m 06s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b844ae8aa8c48488495fe6c740895b5 : SUCCESS in 56m 13s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5e109931d2054087a0847298ea37cc01 : SUCCESS in 1h 02m 47s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4d29b33e8a5f405cbc06e2097a5149b2 : SUCCESS in 26m 44s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65a7d7019df54c7ebfe39e8eefc667b0 : SUCCESS in 18m 50s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/186e6c5d32de409fb9fac5f44b31f9a9 : SUCCESS in 1h 55m 26s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12edd9fdec444a4dabf11189f66e37be : SUCCESS in 1h 07m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d9ffcda7c2ae425380f207d644c60ab5 : SUCCESS in 56m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5903a9e55e44ce8b1643ca4aeba5cdb : FAILURE in 34m 56s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9b6eae94a03540089681eb6bdbcedcce : SUCCESS in 42m 56s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0f29810af734a2d8e6a9f1dbf736f9f : SUCCESS in 38m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9b3a0eae4a6346a6a5842da5af8f0a84 : SUCCESS in 17m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/410339280d0644ff81cec16855a813fc : SUCCESS in 21m 08s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e2704a875e814194bbcc6ccbc552d4b1 : SUCCESS in 20m 05s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52fe8de1a7da46609035d3b5e5583f14 : SUCCESS in 17m 14s","accounts_in_message":[],"_revision_number":2},{"id":"2c926776ba6341aa21c11cb528a7f09b930d00a3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-18 13:28:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a962ed47c2c041d0b663e69ee98670a93c40fc9b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-18 13:28:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"45bba3e7537cd08ba6dccab411801b06139bc0d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 15:06:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c858d31bdd9a4540bc493147bbfd5c37 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdfb4868c1dd4e0eb0a0cbfd711a6d98 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b725393d262a482292f15976df75b751 : SUCCESS in 7m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc3105c79d9349938c6937b7df7cd473 : SUCCESS in 7m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc70ceb4f36b4ca7ae280119b032422e : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60cabed60d2042fc8a14f7c05998b43c : SUCCESS in 10m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/78e1d84f2a164896a3e25da5d4ed3de1 : SUCCESS in 7m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2b194dc72cbd40eda32f81c01a8e52a4 : SUCCESS in 15m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c81b500c7d6a4743b03a8cfb9793c1cf : SUCCESS in 33m 53s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f0938e46b66144f9a2a3357375912b02 : SUCCESS in 1h 05m 12s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc262ec91ecf4752800124ad330c7a72 : SUCCESS in 1h 00m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0bca0399cb8846689671176319fcd3df : SUCCESS in 1h 07m 45s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4bfda1b6ba544c189cfe8562faa6fbc4 : SUCCESS in 26m 01s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c13ef79cfa7a45de9d7cf07ac2070daf : SUCCESS in 16m 24s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3fc3dcae12e44e958a4b246970259963 : SUCCESS in 1h 36m 49s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7d334009ef640d2a06fd16ca0a07f5b : SUCCESS in 1h 22m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/392af1dd6e024f3aab1c7ccebad2efc0 : SUCCESS in 1h 05m 11s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee1b204bfb0b4d36a6c4fce845c81526 : SUCCESS in 37m 26s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1311af6b36684d55995561da61dca07b : SUCCESS in 54m 00s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bc81161657847a583ba281f97372639 : SUCCESS in 28m 36s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2a17b6aa4af84e91966fe9f6d891c188 : SUCCESS in 18m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8a455db5ede4ec78b8ab443a50034fa : SUCCESS in 13m 54s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/d83eaf509f5e44b7bac6dd2e71a7c4f3 : SUCCESS in 27m 12s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c66d2ed5e6b464fadd1505968c4fe61 : SUCCESS in 15m 24s","accounts_in_message":[],"_revision_number":2},{"id":"b82bfcadc551ec2af3e38e9a08a7c4d8f9104d80","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-18 17:09:38.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"b88091f2e4cf311e2a35e6be623c91229bba0d42","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-18 20:29:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6dce3547a47ad89f244e8f60e973b3b987bf698e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 21:58:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97d8e0a4048644ffb2410b592e490f29 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe345e97c1494706bd6192deace9d43e : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9a3831475dc4f5098b1f2ca2152af59 : SUCCESS in 8m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c91f9be4ea84b4ba66bc179aea19155 : SUCCESS in 8m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63509fa6f6144cfa902591cb7ccc1dbd : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ca8917364334f2495d66bf2179f862c : SUCCESS in 10m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/af152add6a914fd3a3f5f342426d13c1 : SUCCESS in 7m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4bd898da9e144b69ae3c3f883621ccd5 : SUCCESS in 13m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7d7e49fcf1634825a955ec69aafaa94e : SUCCESS in 34m 51s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5590e99df40746ab8abdfafbe92f06ac : SUCCESS in 1h 09m 53s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dcf6dec85c74a09ba22765d7d73c8f6 : SUCCESS in 57m 21s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c8e5e40cf7534476b4871372df349ffb : SUCCESS in 1h 14m 34s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ef92ea7b26a444428aed7612beca5372 : SUCCESS in 26m 52s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a4559f559934ec297d35a2caca9ed80 : SUCCESS in 16m 58s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3f2dd7eb40ea44ee8e1b3a7155680d5e : SUCCESS in 1h 28m 09s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b70b1aa94e24cc1921f0467ea07513b : SUCCESS in 1h 20m 29s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/98631fa9fa8c4b1c8048605b11aa7a47 : SUCCESS in 40m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f15d6eceb014c4085b1c7806fb61ada : SUCCESS in 25m 53s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/889ca07ffac94743964ad7b420865277 : SUCCESS in 40m 27s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1050eba97564702b20e93495857f852 : SUCCESS in 31m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/630c1ca862284c1bad437e84d47e7675 : SUCCESS in 18m 04s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/452d7a6b44654c54b954205c9c739099 : SUCCESS in 12m 56s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/14fb6125519a467ca114cea23f86f75f : SUCCESS in 20m 52s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5924946c6b54387aee5e180d6553ccb : SUCCESS in 14m 20s","accounts_in_message":[],"_revision_number":3},{"id":"1bbebb79b3c0443d33902f201124d06f06d60c8b","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-19 14:47:47.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"9ac6b8d6a1e48f3802c2279cb97f7c5cdd119509","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-19 21:37:40.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5088537f1eec4d695d9192faf65aa38d22a1618b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-20 20:15:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"50cd553c726687fbda9d1536acd9bcc26f914f3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-20 20:15:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9a06bfed13187ce472c2c5a6126aee033f81d7fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-20 21:40:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3a77d36ac8d473ebcb3884418fe8b83 : SUCCESS in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cdb9fdacfaa47ba859092f3d320567c : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8b4b334a9a54081a6f0fb6ab01b2c03 : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8dd23c173724a7c8f7c829cd7cd9ec7 : SUCCESS in 10m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9df5777b4c054e5dac5243f3561cc7cf : SUCCESS in 7m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a4079e298a1b4e0187d0e233bb5ef3e2 : SUCCESS in 14m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0e7486c7526d415c88b5b2fd871c2791 : SUCCESS in 35m 04s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d22717e13287415dac45b60781a2901c : SUCCESS in 1h 00m 22s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a92cb401cde34eeda8d9ca7dcfc91f17 : SUCCESS in 44m 36s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9cb57f5d4af746b6bd93fef0513871e7 : SUCCESS in 56m 07s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/45478f3f12564dd4969b5af8608e5395 : SUCCESS in 20m 13s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b67d95f27a54519bab024aa78cd4f43 : SUCCESS in 14m 46s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f7b564863b3d4db9bb04589b5892bdcc : SUCCESS in 1h 24m 21s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd1715a5c9614b36ac14dcb0a13d6492 : SUCCESS in 1h 03m 04s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ea3c67e09072493094e564b825ddb416 : SUCCESS in 35m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0cd14fc2b6524ef4b4af3187ef69bd42 : SUCCESS in 28m 52s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d3412c63a72b40128039a83c59f8d16d : SUCCESS in 37m 38s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee51f9e051204519a31e02441263d6c0 : SUCCESS in 27m 17s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b9b8cb2e8a324524b17214f1e65ddf5c : SUCCESS in 15m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aaea76f011b64a7c9cd50f3c497b3817 : SUCCESS in 11m 25s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/9848a0a1858f4d9e8e0a26930797d039 : SUCCESS in 20m 26s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e20f3cae9d1448f8617bf498e77f04c : SUCCESS in 15m 56s","accounts_in_message":[],"_revision_number":3},{"id":"7e6cb3a6bfedb4c13aa168cea02fe6a742e43acb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-20 21:40:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"91921b2792474dd90443f6e9a73d02552f8abd18","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-20 21:41:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e93bb0fe478f486dae91bb2fc688e6ab : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3},{"id":"a4f9985cb83aff4c9d6b73a735b866d8df925f07","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-21 09:46:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1b286303342aeb5ea966aa5fdb741403c6dfc916","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-21 09:53:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1ff569ac7ba4b8db5cfe8d588a59b89a51d3f387","revisions":{"b603b21d7eadba93719418a7df758ad69226f395":{"kind":"REWORK","_number":1,"created":"2022-03-16 17:01:36.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/18/834018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/834018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/834018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/834018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/834018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/834018/1"}}},"commit":{"parents":[{"commit":"15790c983e9f88f8bd735f7e7cb2b6db605adee2","subject":"CI: remove qemu-utils installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15790c983e9f88f8bd735f7e7cb2b6db605adee2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-15 10:12:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-16 17:01:36.000000000","tz":0},"subject":"CI: Don\u0027t download Cirros or IPA in seed jobs","message":"CI: Don\u0027t download Cirros or IPA in seed jobs\n\nWe build IPA images and a deployment image in the seed jobs, so we don\u0027t\nneed to download Cirros or IPA images. Also, these downloads depend on\nexternal resources which may make jobs less reliable.\n\nFor seed upgrade jobs, disable IPA and deployment image downloads.\n\nChange-Id: Ib59c8bc2d8938eca18c943bb2e66ed185152a739\n(cherry picked from commit d77a30db36c7bc936e80f87a1b366431f42116a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b603b21d7eadba93719418a7df758ad69226f395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b603b21d7eadba93719418a7df758ad69226f395"}]},"branch":"refs/heads/stable/xena"},"5ef1493cafe4579674450447431073ab95872fe8":{"kind":"REWORK","_number":2,"created":"2022-03-17 09:40:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/18/834018/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/834018/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/834018/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/834018/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/834018/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/834018/2"}}},"commit":{"parents":[{"commit":"15790c983e9f88f8bd735f7e7cb2b6db605adee2","subject":"CI: remove qemu-utils installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15790c983e9f88f8bd735f7e7cb2b6db605adee2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-15 10:12:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-17 09:40:26.000000000","tz":0},"subject":"CI: Don\u0027t download Cirros or IPA in seed jobs","message":"CI: Don\u0027t download Cirros or IPA in seed jobs\n\nWe build IPA images in the seed jobs, so we don\u0027t need to download IPA\nimages. We don\u0027t use the deployment image, so we don\u0027t need to download\nCirros. Also, these downloads depend on external resources which may\nmake jobs less reliable.\n\nFor seed upgrade jobs, disable IPA and deployment image downloads.\n\nChange-Id: Ib59c8bc2d8938eca18c943bb2e66ed185152a739\n(cherry picked from commit d77a30db36c7bc936e80f87a1b366431f42116a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ef1493cafe4579674450447431073ab95872fe8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ef1493cafe4579674450447431073ab95872fe8"}]},"branch":"refs/heads/stable/xena"},"1ff569ac7ba4b8db5cfe8d588a59b89a51d3f387":{"kind":"REWORK","_number":3,"created":"2022-03-18 20:29:52.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/18/834018/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/834018/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/834018/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/834018/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/834018/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/834018/3"}}},"commit":{"parents":[{"commit":"15790c983e9f88f8bd735f7e7cb2b6db605adee2","subject":"CI: remove qemu-utils installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15790c983e9f88f8bd735f7e7cb2b6db605adee2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-15 10:12:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-18 20:28:38.000000000","tz":0},"subject":"CI: Don\u0027t download Cirros or IPA in seed jobs","message":"CI: Don\u0027t download Cirros or IPA in seed jobs\n\nWe build IPA images in the seed jobs, so we don\u0027t need to download IPA\nimages. We don\u0027t use the deployment image, so we don\u0027t need to download\nCirros. Also, these downloads depend on external resources which may\nmake jobs less reliable.\n\nFor seed upgrade jobs, disable IPA and deployment image downloads.\n\nThis backport also includes the follow up change,\nIbc46a18d00c4c51b67ea18f24a4dd138ef59a4ff:\n\nCI: use dib.yml for bifrost overrides\n\nFollow up to Ib59c8bc2d8938eca18c943bb2e66ed185152a739, which stopped\ndownloads of the Cirros image in the seed jobs, but inadvertently\nenabled DIB builds in the seed upgrade job. This is because\nkolla-ansible sets create_image_via_dib to true in dib.yml, and that\nfile takes precedence over bifrost.yml.\n\nThis change fixes the issue by providing Bifrost overrides in CI using\ndib.yml.\n\nChange-Id: Ib59c8bc2d8938eca18c943bb2e66ed185152a739\n(cherry picked from commit d77a30db36c7bc936e80f87a1b366431f42116a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ff569ac7ba4b8db5cfe8d588a59b89a51d3f387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ff569ac7ba4b8db5cfe8d588a59b89a51d3f387"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
