)]}'
{"id":"openstack%2Fkayobe~834019","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~Ib59c8bc2d8938eca18c943bb2e66ed185152a739","project":"openstack/kayobe","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-21 09:44:49.000000000","reason":"Mark Goddard replied on the change"}},"hashtags":[],"change_id":"Ib59c8bc2d8938eca18c943bb2e66ed185152a739","subject":"CI: Don\u0027t download Cirros or IPA in seed jobs","status":"ABANDONED","created":"2022-03-16 17:01:44.000000000","updated":"2022-03-30 09:22:07.000000000","total_comment_count":3,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"cherry_pick_of_change":834018,"cherry_pick_of_patch_set":3,"meta_rev_id":"a5765f6adb1d67364e46fa5183ebf9aa029ea5a6","_number":834019,"virtual_id_number":834019,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-03-20 22:05:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-16 18:29:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-18 13:20:20.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"}],"messages":[{"id":"a2f3f4e4685b8b3561a11f07117c712d373871a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-16 17:01:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f2fc9e3a5cb23bfcc4b0c254da354e47c0a00c04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 18:29:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/603f64a9208346f7a2f44e298f085f5f : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb49bf5a4e5b4fd7888c717d82f6ee9a : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c87d58d2d2a148448e490a5124c3de43 : SUCCESS in 8m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fefceaa23284998a445f750fafb5c78 : SUCCESS in 7m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f2172ff980a451c9f95b4d8e8b9c30b : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d8b11fd255548fb812c77f2fbb757bb : SUCCESS in 9m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0dab92b9d4424925b93e5528cdd82b0e : FAILURE in 8m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/542f9664299e41e5a180d9c47428afe4 : FAILURE in 7m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3ea0a599d13248d78b2131a419e7243c : FAILURE in 8m 50s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8f76106b114a4203aa5b37d7744d6dd4 : SUCCESS in 57m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d3a8e01ad6f483dba071f2cad34bd64 : FAILURE in 32m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/582ceeeaea2345099b8746a72b6372bd : SUCCESS in 1h 02m 47s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/efd8385f567c42b7bf41b6cf0a1b4a75 : SUCCESS in 20m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d851fb12eae7456cb258adb51bed788e : SUCCESS in 14m 06s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9b10123bfe2c4f47b63324e6d00118e1 : SUCCESS in 1h 21m 20s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/314327327b504eefb187865adb1228a9 : SUCCESS in 45m 31s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3975c73f7ea54a1e9ff3c504f9af428a : FAILURE in 36m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ddd683e477dc4412b3dbe8738514c920 : FAILURE in 19m 18s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/05bc7ebcc6a244d0a94869900ccdab4f : SUCCESS in 16m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89c34d91041b420b8520dafbb74057d3 : SUCCESS in 11m 08s","accounts_in_message":[],"_revision_number":1},{"id":"52da1beac77ef88eb9e5792c8544076e079c2af9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-18 13:20:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"371c54c8a3a76171fee74f4e0f2291d8989cf2a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 15:20:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ada21074df7049d6a2a30c934838b95c : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/845c0b083f1142c2ba97b727426ec527 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bfd196e7521428ba09c0fb9386468f0 : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a82470e2384414eaf7bde2f2d8cc5f2 : SUCCESS in 7m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9384c37dc71413581592856a10ef18e : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07c782eb8de64f71962648f65e633f50 : SUCCESS in 9m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c72d2235a0864583b727ef7271981050 : SUCCESS in 7m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f32cc3c90db94efcac6a2301c7782026 : SUCCESS in 12m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b4df3830a39d4eb48dc6d82d05e97a53 : SUCCESS in 22m 14s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/53357be887fd42ca9cd16f6f826cf64f : SUCCESS in 1h 09m 33s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ecf344906124a3c813cd434f31cc444 : SUCCESS in 1h 01m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/70ef987beb0f4bbfb08c9fdd15e2c5bd : SUCCESS in 1h 27m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a3a9ae02d75f4b97b5d2912cfbb8d212 : SUCCESS in 21m 32s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/793d60ae099f4cb0b13ba758b4f2894c : SUCCESS in 18m 09s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/045d3e2c76694183a22b7e6e903cb45c : SUCCESS in 1h 55m 00s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/635e571bb0224e27a1bab6963be33b81 : SUCCESS in 58m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f71d6b400525452ba451843d1e17b028 : SUCCESS in 40m 12s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0130c52e76ea4fa88b0f4719b152c962 : FAILURE in 19m 12s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8a977c729005405bbef0310ab56ece6e : SUCCESS in 15m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6384f5732dab49cd8962dc416e924d27 : SUCCESS in 12m 27s","accounts_in_message":[],"_revision_number":1},{"id":"57df929c5010b1a1d209b43962b6b8fc4f0b46e9","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-18 17:09:42.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"a79203e51bc2c30589d47485c84060f167b8f52f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-19 14:47:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"42317f167b164aedecc44dfc250010c05414ab94","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-19 14:47:58.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"22c8a82d99ca352022ac1fb28162ce8152727efc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 16:44:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6843763391a6493daa0637bce65eec40 : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed2db9f609d247f98aec96620853daa9 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b0ef0be2eaa4fc4b2f90c91b468c0e8 : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f8b70ed7fb044d691036b47cf8bb400 : SUCCESS in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4a5a3bc914b4c918e4d8b62dd26bf93 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8846e8174304c32aff410fcb9f047d3 : SUCCESS in 9m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b43fab772f10414e8711938f66915c6e : SUCCESS in 7m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9708e0e5437243c0a3767b115390a6d5 : SUCCESS in 12m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/370f312bec344939a796fe75fe9bcbdd : SUCCESS in 18m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c7472bb417b146e2bcbd3561b255ce70 : SUCCESS in 1h 01m 55s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6bfa38252d542eea42789e66c7b7488 : SUCCESS in 1h 05m 19s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c2b7d56690404e8dad20f91e4c19caec : SUCCESS in 1h 07m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/cddcc3629faf4b61840f03b06e94fd82 : SUCCESS in 20m 09s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17bbe55c41864f3595a180a5c824ab63 : SUCCESS in 16m 11s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0717a8eb241b49ddba2be020bb31cccc : SUCCESS in 1h 54m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c9a22e91f26041f19439484dcd4c7745 : SUCCESS in 57m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d8de4699bb54cef8ae9a131b8b02aea : SUCCESS in 33m 12s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/adee240e69294cce869b8d6ee6c98bd1 : FAILURE in 20m 21s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ae1e8290462f41589c129267b29c9e4d : SUCCESS in 17m 24s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41a1b7cbd4df41c9b91aea7ca1cb35f8 : SUCCESS in 12m 24s","accounts_in_message":[],"_revision_number":2},{"id":"c4cc17b2e1144b56bcf12a7afc13451708e44679","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-19 21:38:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9b4125db1a856c146c54f04a53be545a152e2be7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-20 20:15:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c606d4dd8a3805104edee944392486cbf2f11eb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-20 22:05:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45013d480d0d4ef480a4eb6e5420fe8b : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/132e5ca1db394c2aa692b92c00f64a2a : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f359ecad3b348d7a5360bb5cc5339e7 : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a92e82168e24326906e2785ad90831f : SUCCESS in 7m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51d1f924994d4aeaa6e48b11007b3eaa : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7232e1f408548789e7a6c30baa97fb2 : SUCCESS in 11m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/656d8e536e4445c6a1905767724442f0 : SUCCESS in 7m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3af2f288645f4a1bafb7ed5304ac3a90 : SUCCESS in 15m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9f610e3964c446fa87a63bf48dd1d7f5 : SUCCESS in 19m 27s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/82dc5f29b0524ae4af1d872fb2b30ca8 : SUCCESS in 1h 10m 36s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f648a2d44064ddc928a8f9b6a8ac20a : SUCCESS in 54m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a9595f03655042c2ab76820085d56e64 : SUCCESS in 1h 11m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ef9a14ae6d664a91a6c3455e33600a89 : SUCCESS in 20m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e28d544dc769453686ac0145340ccecd : SUCCESS in 14m 54s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f5912acc19fc41bb907771e881462499 : SUCCESS in 1h 45m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a2c80b0859c0406abd84da4551d27562 : SUCCESS in 50m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8aca5e4669704bf886b51b8f2869911e : SUCCESS in 31m 48s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fefedd71edcf41af9c718d8d2223361a : FAILURE in 19m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4cbaa3553a484120a28a1ccd07a78fb8 : SUCCESS in 15m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/124fd999e44d4e5c951f74add80fa1df : SUCCESS in 12m 02s","accounts_in_message":[],"_revision_number":3},{"id":"be13edf35ee27be80214a9527ebfba416af684dc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-21 09:44:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"50e4d6672c9a7f3832de27eb48fe13478deb786f","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-21 09:44:52.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"a5765f6adb1d67364e46fa5183ebf9aa029ea5a6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-30 09:22:07.000000000","message":"Abandoned\n\nGo with this: https://review.opendev.org/c/openstack/kayobe/+/835811","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b632f5dfba5df8c259f2812038a9c40d834da5bb","revisions":{"d5d343002ef2ca88b1722dc2f7653e7d4c9a7c47":{"kind":"REWORK","_number":1,"created":"2022-03-16 17:01:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/19/834019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/19/834019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/834019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/834019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/834019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/19/834019/1"}}},"commit":{"parents":[{"commit":"ee6d4614f58be025a0af7c08de45831fd5c523ec","subject":"CI: stop using zuul as kayobe_ansible_user in TLS jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee6d4614f58be025a0af7c08de45831fd5c523ec"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-15 10:12:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-16 17:01:44.000000000","tz":0},"subject":"CI: Don\u0027t download Cirros or IPA in seed jobs","message":"CI: Don\u0027t download Cirros or IPA in seed jobs\n\nWe build IPA images and a deployment image in the seed jobs, so we don\u0027t\nneed to download Cirros or IPA images. Also, these downloads depend on\nexternal resources which may make jobs less reliable.\n\nFor seed upgrade jobs, disable IPA and deployment image downloads.\n\nChange-Id: Ib59c8bc2d8938eca18c943bb2e66ed185152a739\n(cherry picked from commit d77a30db36c7bc936e80f87a1b366431f42116a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d5d343002ef2ca88b1722dc2f7653e7d4c9a7c47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d5d343002ef2ca88b1722dc2f7653e7d4c9a7c47"}]},"branch":"refs/heads/stable/wallaby"},"b6113bbfb9a085c454a9d1f798d1beaacc5690d8":{"kind":"REWORK","_number":2,"created":"2022-03-19 14:47:53.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/19/834019/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/19/834019/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/834019/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/834019/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/834019/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/19/834019/2"}}},"commit":{"parents":[{"commit":"c9945c8afcc885de7e6483a6a76f0c1f19a8c20d","subject":"Merge \"Only create patch links on overcloud hosts\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9945c8afcc885de7e6483a6a76f0c1f19a8c20d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-15 10:12:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-19 14:47:53.000000000","tz":0},"subject":"CI: Don\u0027t download Cirros or IPA in seed jobs","message":"CI: Don\u0027t download Cirros or IPA in seed jobs\n\nWe build IPA images in the seed jobs, so we don\u0027t need to download IPA\nimages. We don\u0027t use the deployment image, so we don\u0027t need to download\nCirros. Also, these downloads depend on external resources which may\nmake jobs less reliable.\n\nFor seed upgrade jobs, disable IPA and deployment image downloads.\n\nThis backport also includes the follow up change,\nIbc46a18d00c4c51b67ea18f24a4dd138ef59a4ff:\n\nCI: use dib.yml for bifrost overrides\n\nFollow up to Ib59c8bc2d8938eca18c943bb2e66ed185152a739, which stopped\ndownloads of the Cirros image in the seed jobs, but inadvertently\nenabled DIB builds in the seed upgrade job. This is because\nkolla-ansible sets create_image_via_dib to true in dib.yml, and that\nfile takes precedence over bifrost.yml.\n\nThis change fixes the issue by providing Bifrost overrides in CI using\ndib.yml.\n\nChange-Id: Ib59c8bc2d8938eca18c943bb2e66ed185152a739\n(cherry picked from commit d77a30db36c7bc936e80f87a1b366431f42116a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b6113bbfb9a085c454a9d1f798d1beaacc5690d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b6113bbfb9a085c454a9d1f798d1beaacc5690d8"}]},"branch":"refs/heads/stable/wallaby"},"b632f5dfba5df8c259f2812038a9c40d834da5bb":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-03-20 20:15:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/19/834019/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/19/834019/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/834019/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/834019/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/834019/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/19/834019/3"}}},"commit":{"parents":[{"commit":"ba53ee383a2df89106de3cf1b73a2dabdd46c894","subject":"Merge \"Use naming convention to infer VLAN tagging\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ba53ee383a2df89106de3cf1b73a2dabdd46c894"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-15 10:12:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-20 20:15:35.000000000","tz":0},"subject":"CI: Don\u0027t download Cirros or IPA in seed jobs","message":"CI: Don\u0027t download Cirros or IPA in seed jobs\n\nWe build IPA images in the seed jobs, so we don\u0027t need to download IPA\nimages. We don\u0027t use the deployment image, so we don\u0027t need to download\nCirros. Also, these downloads depend on external resources which may\nmake jobs less reliable.\n\nFor seed upgrade jobs, disable IPA and deployment image downloads.\n\nThis backport also includes the follow up change,\nIbc46a18d00c4c51b67ea18f24a4dd138ef59a4ff:\n\nCI: use dib.yml for bifrost overrides\n\nFollow up to Ib59c8bc2d8938eca18c943bb2e66ed185152a739, which stopped\ndownloads of the Cirros image in the seed jobs, but inadvertently\nenabled DIB builds in the seed upgrade job. This is because\nkolla-ansible sets create_image_via_dib to true in dib.yml, and that\nfile takes precedence over bifrost.yml.\n\nThis change fixes the issue by providing Bifrost overrides in CI using\ndib.yml.\n\nChange-Id: Ib59c8bc2d8938eca18c943bb2e66ed185152a739\n(cherry picked from commit d77a30db36c7bc936e80f87a1b366431f42116a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b632f5dfba5df8c259f2812038a9c40d834da5bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b632f5dfba5df8c259f2812038a9c40d834da5bb"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
