)]}'
{"id":"openstack%2Fkayobe~834020","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~Ib59c8bc2d8938eca18c943bb2e66ed185152a739","project":"openstack/kayobe","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-21 09:44:55.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"Ib59c8bc2d8938eca18c943bb2e66ed185152a739","subject":"CI: Don\u0027t download Cirros or IPA in seed jobs","status":"ABANDONED","created":"2022-03-16 17:01:54.000000000","updated":"2022-04-12 10:00:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"cherry_pick_of_change":834019,"cherry_pick_of_patch_set":3,"meta_rev_id":"1d92e211019326365d7927c2a2d881627b863202","_number":834020,"virtual_id_number":834020,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-03-20 21:23:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-16 18:18:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2bd11378cf07364ad0c341bda92864ff1ba549fd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-16 17:01:54.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* playbooks/kayobe-seed-base/pre.yml","accounts_in_message":[],"_revision_number":1},{"id":"e347f41fe06fa546b429f0d25849d6d908b55dd8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-16 17:02:40.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9e5beae221a58c68e1c87e9df35ea4fae2f194e8","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-16 17:02:44.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"2dec9f0aaa9858dbe182ddb186a5ec080d1352d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 18:18:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5843b30d26924ec39c58f2c4e65868ff : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cad0a2d8da064b508a71cc27a6f5edb8 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/815efb29a12f4d0cab665a6f00be0e94 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd1d24f067d347d8bf4bcbc338ab71ce : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e2b39792b96474aaf3d9f45ef4446ef : SUCCESS in 7m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ca07fcbcbf3a47088e1681766bd26c53 : SUCCESS in 6m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/00524b135b6340d5b12a5d4607440ece : SUCCESS in 10m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b5bab23dac1241aaa3970d71cbc25e01 : SUCCESS in 19m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c2d7c84b87ac4ad6bba11772429e8a9d : SUCCESS in 1h 10m 22s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/01088f53aecf48e0b8c8323de41f04cd : SUCCESS in 1h 07m 07s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bb94b4763d0643f9981203db076d9169 : SUCCESS in 18m 50s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/dfa5084323934297b2c44df2532ab271 : FAILURE in 20m 48s (non-voting)\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2bc6586b3d574c91a3a8de363f3e6dad : FAILURE in 19m 03s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c4eb1073498d43f7afa51b2128ed03f4 : FAILURE in 19m 43s (non-voting)\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/77836bb1450d4e009108fbe354d26c92 : SUCCESS in 13m 42s","accounts_in_message":[],"_revision_number":2},{"id":"4e013ee9bc0d3920166a458292e49836c4c3eb7d","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-18 17:09:47.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"708a11e6a0e733976b89214777ed445bcf73c438","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-19 14:48:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"63eadbf82005d0f425a369979b6fb750c7913799","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-19 14:49:02.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"ac438b009d49350559fe2a00c449170084f07dd3","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-19 14:49:14.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"893c722fdf39051ee490387c094375a00bb2e66d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 15:54:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d21b1a7c41a94b0995e873cefb4f67b3 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9547940a3f4a4baa983f8d4c93e84b34 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/420e62daf88a441ca06a2da15262ef1c : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd931e2af6014c6b96a256bf4741d496 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/305d8e4f5d7a4ccf88fad44e3baebf60 : SUCCESS in 9m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a8e01836837345b3886da811ea8bf1c8 : SUCCESS in 6m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2f5a0091fc194c4095a09c0ba92271b4 : SUCCESS in 10m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7cbf8974c8a64980869aefa9d09fabd5 : SUCCESS in 19m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/03d14bb236ca4a75a426dfc96edefe3c : SUCCESS in 1h 01m 23s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/64708f80fb5e442abaaf204c681dabdb : SUCCESS in 1h 04m 25s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6d12d3a84b07441ea824d59e2e1716c6 : SUCCESS in 17m 13s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/110ef26482f44d51ab4547269d789cfa : FAILURE in 18m 11s (non-voting)\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a119aa660c33469cb94c0ff9474eeacc : FAILURE in 18m 41s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/aa08d49505d64d0eb85eb691b8216c8c : FAILURE in 20m 17s (non-voting)\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f969a452e6b1496aaf27493cc25b5f1f : SUCCESS in 11m 45s","accounts_in_message":[],"_revision_number":4},{"id":"1922d75096ec71ea8c0891c10d3701f49c109c82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-20 20:15:44.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"81365addd62d0a55ecd83269ead2e74a0e601e85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-20 21:23:01.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e7becf1522249d3a9e711dab271cc3d : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e15bdd18c888421d95bcb714088bf73c : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d806962143244e3f9f85b5b288a752ce : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf9e588b49a44679ba2f76a156fb663f : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a09184c4c6584b16a4c5ed4c5c2f95f7 : SUCCESS in 7m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c8397d60502649609381da20f3221636 : SUCCESS in 6m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/81a2c05ee1ed4a5dbfbe325e13ed3557 : SUCCESS in 10m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/507c79f7831b483eb0bca1faedf6d8e4 : SUCCESS in 19m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7ae1b797a87e4c88a3b879d28d80526b : SUCCESS in 1h 03m 13s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/861bcda1bce747f59b5eadc6613f2e59 : SUCCESS in 55m 55s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3da59d72d66c4c9dbc2f7e72d6caebf0 : SUCCESS in 16m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/885548b43c3e404b9f12b0b382a636e1 : FAILURE in 16m 33s (non-voting)\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8be839e0e15f4efb96f4af55421b5f14 : FAILURE in 18m 48s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/54f319717aba40919a43549ca4f9c540 : RETRY_LIMIT in 6m 02s (non-voting)\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/564af4911cc44e04bc8bcffc34083ffc : SUCCESS in 12m 32s","accounts_in_message":[],"_revision_number":5},{"id":"95dc811257af5bd5f2962067cdf9d5ca8797ab28","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-21 09:44:55.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":5},{"id":"1d92e211019326365d7927c2a2d881627b863202","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-12 10:00:37.000000000","message":"Abandoned\n\nGo with https://review.opendev.org/c/openstack/kayobe/+/835801","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d5a76f47d62b91ce1de4cf59afa48a8437215373","revisions":{"3f263550f141eb9ec07002ff3581a6f63394db83":{"kind":"REWORK","_number":1,"created":"2022-03-16 17:01:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/20/834020/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/20/834020/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/834020/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/834020/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/834020/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/20/834020/1"}}},"commit":{"parents":[{"commit":"7cd08b030bbd53e6faed13586d631281c9492edf","subject":"CI: stop using zuul as kayobe_ansible_user in TLS jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7cd08b030bbd53e6faed13586d631281c9492edf"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-15 10:12:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-16 17:01:54.000000000","tz":0},"subject":"CI: Don\u0027t download Cirros or IPA in seed jobs","message":"CI: Don\u0027t download Cirros or IPA in seed jobs\n\nWe build IPA images and a deployment image in the seed jobs, so we don\u0027t\nneed to download Cirros or IPA images. Also, these downloads depend on\nexternal resources which may make jobs less reliable.\n\nFor seed upgrade jobs, disable IPA and deployment image downloads.\n\nChange-Id: Ib59c8bc2d8938eca18c943bb2e66ed185152a739\n(cherry picked from commit d77a30db36c7bc936e80f87a1b366431f42116a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f263550f141eb9ec07002ff3581a6f63394db83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f263550f141eb9ec07002ff3581a6f63394db83"}]},"branch":"refs/heads/stable/victoria"},"77bb5f5a6a390e74cb1448c8612210d53097a2ca":{"kind":"REWORK","_number":2,"created":"2022-03-16 17:02:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/20/834020/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/20/834020/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/834020/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/834020/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/834020/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/20/834020/2"}}},"commit":{"parents":[{"commit":"7cd08b030bbd53e6faed13586d631281c9492edf","subject":"CI: stop using zuul as kayobe_ansible_user in TLS jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7cd08b030bbd53e6faed13586d631281c9492edf"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-15 10:12:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-16 17:02:40.000000000","tz":0},"subject":"CI: Don\u0027t download Cirros or IPA in seed jobs","message":"CI: Don\u0027t download Cirros or IPA in seed jobs\n\nWe build IPA images and a deployment image in the seed jobs, so we don\u0027t\nneed to download Cirros or IPA images. Also, these downloads depend on\nexternal resources which may make jobs less reliable.\n\nFor seed upgrade jobs, disable IPA and deployment image downloads.\n\nChange-Id: Ib59c8bc2d8938eca18c943bb2e66ed185152a739\n(cherry picked from commit d77a30db36c7bc936e80f87a1b366431f42116a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/77bb5f5a6a390e74cb1448c8612210d53097a2ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/77bb5f5a6a390e74cb1448c8612210d53097a2ca"}]},"branch":"refs/heads/stable/victoria"},"f13975412451751de932d1d266e614f756d408c8":{"kind":"REWORK","_number":3,"created":"2022-03-19 14:48:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/20/834020/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/20/834020/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/834020/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/834020/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/834020/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/20/834020/3"}}},"commit":{"parents":[{"commit":"0356f7d5384fdebcf04efd0905adce05156be8ff","subject":"Only create patch links on overcloud hosts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0356f7d5384fdebcf04efd0905adce05156be8ff"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-15 10:12:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-19 14:48:07.000000000","tz":0},"subject":"CI: Don\u0027t download Cirros or IPA in seed jobs","message":"CI: Don\u0027t download Cirros or IPA in seed jobs\n\nWe build IPA images in the seed jobs, so we don\u0027t need to download IPA\nimages. We don\u0027t use the deployment image, so we don\u0027t need to download\nCirros. Also, these downloads depend on external resources which may\nmake jobs less reliable.\n\nFor seed upgrade jobs, disable IPA and deployment image downloads.\n\nThis backport also includes the follow up change,\nIbc46a18d00c4c51b67ea18f24a4dd138ef59a4ff:\n\nCI: use dib.yml for bifrost overrides\n\nFollow up to Ib59c8bc2d8938eca18c943bb2e66ed185152a739, which stopped\ndownloads of the Cirros image in the seed jobs, but inadvertently\nenabled DIB builds in the seed upgrade job. This is because\nkolla-ansible sets create_image_via_dib to true in dib.yml, and that\nfile takes precedence over bifrost.yml.\n\nThis change fixes the issue by providing Bifrost overrides in CI using\ndib.yml.\n\nChange-Id: Ib59c8bc2d8938eca18c943bb2e66ed185152a739\n(cherry picked from commit d77a30db36c7bc936e80f87a1b366431f42116a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f13975412451751de932d1d266e614f756d408c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f13975412451751de932d1d266e614f756d408c8"}]},"branch":"refs/heads/stable/victoria"},"12aad67cbf6395201f82f970e5acc4f95d3e03fd":{"kind":"REWORK","_number":4,"created":"2022-03-19 14:49:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/20/834020/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/20/834020/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/834020/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/834020/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/834020/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/20/834020/4"}}},"commit":{"parents":[{"commit":"0356f7d5384fdebcf04efd0905adce05156be8ff","subject":"Only create patch links on overcloud hosts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0356f7d5384fdebcf04efd0905adce05156be8ff"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-15 10:12:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-19 14:49:02.000000000","tz":0},"subject":"CI: Don\u0027t download Cirros or IPA in seed jobs","message":"CI: Don\u0027t download Cirros or IPA in seed jobs\n\nWe build IPA images in the seed jobs, so we don\u0027t need to download IPA\nimages. We don\u0027t use the deployment image, so we don\u0027t need to download\nCirros. Also, these downloads depend on external resources which may\nmake jobs less reliable.\n\nFor seed upgrade jobs, disable IPA and deployment image downloads.\n\nThis backport also includes the follow up change,\nIbc46a18d00c4c51b67ea18f24a4dd138ef59a4ff:\n\nCI: use dib.yml for bifrost overrides\n\nFollow up to Ib59c8bc2d8938eca18c943bb2e66ed185152a739, which stopped\ndownloads of the Cirros image in the seed jobs, but inadvertently\nenabled DIB builds in the seed upgrade job. This is because\nkolla-ansible sets create_image_via_dib to true in dib.yml, and that\nfile takes precedence over bifrost.yml.\n\nThis change fixes the issue by providing Bifrost overrides in CI using\ndib.yml.\n\nChange-Id: Ib59c8bc2d8938eca18c943bb2e66ed185152a739\n(cherry picked from commit d77a30db36c7bc936e80f87a1b366431f42116a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12aad67cbf6395201f82f970e5acc4f95d3e03fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12aad67cbf6395201f82f970e5acc4f95d3e03fd"}]},"branch":"refs/heads/stable/victoria"},"d5a76f47d62b91ce1de4cf59afa48a8437215373":{"kind":"REWORK","_number":5,"created":"2022-03-20 20:15:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/20/834020/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/20/834020/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/834020/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/834020/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/834020/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/20/834020/5"}}},"commit":{"parents":[{"commit":"0ea814e0b9f85dce90079191fb284b91d4aaf774","subject":"Merge \"Use naming convention to infer VLAN tagging\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ea814e0b9f85dce90079191fb284b91d4aaf774"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-15 10:12:22.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-20 20:15:44.000000000","tz":0},"subject":"CI: Don\u0027t download Cirros or IPA in seed jobs","message":"CI: Don\u0027t download Cirros or IPA in seed jobs\n\nWe build IPA images in the seed jobs, so we don\u0027t need to download IPA\nimages. We don\u0027t use the deployment image, so we don\u0027t need to download\nCirros. Also, these downloads depend on external resources which may\nmake jobs less reliable.\n\nFor seed upgrade jobs, disable IPA and deployment image downloads.\n\nThis backport also includes the follow up change,\nIbc46a18d00c4c51b67ea18f24a4dd138ef59a4ff:\n\nCI: use dib.yml for bifrost overrides\n\nFollow up to Ib59c8bc2d8938eca18c943bb2e66ed185152a739, which stopped\ndownloads of the Cirros image in the seed jobs, but inadvertently\nenabled DIB builds in the seed upgrade job. This is because\nkolla-ansible sets create_image_via_dib to true in dib.yml, and that\nfile takes precedence over bifrost.yml.\n\nThis change fixes the issue by providing Bifrost overrides in CI using\ndib.yml.\n\nChange-Id: Ib59c8bc2d8938eca18c943bb2e66ed185152a739\n(cherry picked from commit d77a30db36c7bc936e80f87a1b366431f42116a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d5a76f47d62b91ce1de4cf59afa48a8437215373"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d5a76f47d62b91ce1de4cf59afa48a8437215373"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
