)]}'
{"id":"openstack%2Fkayobe~834130","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I00e7615581c7a214e0ee8e8f04cacdbb3ea4131a","project":"openstack/kayobe","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-19 23:50:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I00e7615581c7a214e0ee8e8f04cacdbb3ea4131a","subject":"Use naming convention to infer VLAN tagging","status":"MERGED","created":"2022-03-17 09:05:30.000000000","updated":"2022-03-19 23:51:45.000000000","submitted":"2022-03-19 23:50:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834130","cherry_pick_of_change":833052,"cherry_pick_of_patch_set":3,"meta_rev_id":"c96189fb1c151401d70dbe1ef4c165558493cf62","_number":834130,"virtual_id_number":834130,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-19 23:50:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-03-18 20:24:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-19 21:39:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-19 21:39:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-17 09:05:30.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2022-03-17 11:23:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-18 13:20:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-03-19 21:39:43.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"f821905176aa05a6fd790db8be74f21254d7f643","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-17 09:05:30.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"077d65a82df8d6c7d46d9e58b69ce3d7d9afb074","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 11:23:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a59b0450bab41919cb10fc64573c059 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6b0381713724dfba53a7842c45ab54d : SUCCESS in 8m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36ee809f04b4442fb4666f51195e58e0 : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf2b9064ab2940698323623cc581f8ce : SUCCESS in 8m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3b5b1d4e7e6435d98ddf22701a7ec2c : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60e8b4c7385c42c19c80f90fb29fc82a : SUCCESS in 10m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68043a012fd54c0b9468f732a4f508af : SUCCESS in 5m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/facfb531d5a5481197b0b33835fc7a33 : FAILURE in 8m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8249e31bf999470a9e6c5ffa76f7d1a6 : FAILURE in 8m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/afef92c4154c49989c3476d367d3e37d : FAILURE in 9m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/76fa17541ec247e7becb57e9d18ff305 : SUCCESS in 1h 11m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d748601c48f843f4b61ba6e78801018f : SUCCESS in 57m 14s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9bce1c26e82344fcb4449ef01b36d367 : SUCCESS in 1h 15m 40s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/fc8444fdc3924dd69d226e0bd54a0756 : SUCCESS in 21m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91f572c0b96441e28f2373c9d0d1da90 : SUCCESS in 16m 17s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/63ddd7c8c54b45dfb15e86edc9107e05 : SUCCESS in 2h 13m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9b3b9a4bfabd4ef68c00cc9921112bca : SUCCESS in 40m 49s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b940b7891b3f48448cb8dc0f86779085 : FAILURE in 32m 35s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/584b9b22824047018828a08bb0cba936 : SUCCESS in 28m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b089850424b0440d8a84bf76fb19b189 : SUCCESS in 16m 17s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7227404f84c94b6d9526ba1433454575 : SUCCESS in 13m 03s","accounts_in_message":[],"_revision_number":1},{"id":"68bb107fafc96945eb14e0803e20024462a5f6c3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-18 13:20:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af2df87a542d8530480f64ee1083e7cb390e793b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 15:06:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55b1c4e2d1f74b6da5e6a60c942fcc7d : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed434f3bd31a460099dd3877b14a62dc : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4806bd06ac884fec84cb0cc5d5cba614 : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ec225079c4a439e990b9f4f4abef5d5 : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e76eea14feb842ddab7609adf6267faf : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3892011255d4040821e6e8db3bb9712 : SUCCESS in 8m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac1bf42e3da9480592c89f5b6d3cba6e : SUCCESS in 5m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b5446e27d4ca44a0a0e8de245ce8ee6e : SUCCESS in 7m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eb02fd4f011d48aea1f57a5ad151db10 : SUCCESS in 11m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ab3be097482b484abe5481f9f43cf578 : SUCCESS in 21m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/67af27dc8ff4444a8b2f39f6f3ec0c28 : SUCCESS in 1h 02m 48s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a438ef3564ed46d7b6a3ac693f644b8c : FAILURE in 35m 34s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/08c9debcbd484995bd7c6bfee8264555 : SUCCESS in 1h 19m 45s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c83435b551794c5790d0981302444463 : SUCCESS in 20m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a95304f238a448781e1000cedf717f5 : SUCCESS in 16m 58s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ceb2a17c8cfa440a9b0dcdb3afdbbfef : SUCCESS in 1h 41m 14s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/edb32dd0b661492c8687b36de9706c53 : SUCCESS in 34m 29s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aeca8bd51b1d440f91f469014f91c767 : SUCCESS in 29m 36s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/13d3cf0f6f434cba9390d1f39dcbced6 : SUCCESS in 29m 26s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7c3f1fc447e5404aa5208be54bfd295f : SUCCESS in 15m 13s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19db32ae044749bbb725aceb1e4699f9 : SUCCESS in 12m 14s","accounts_in_message":[],"_revision_number":1},{"id":"77294ede2b4d656ae7bce5c995ac53f4b0fc328c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-18 17:13:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae9b971cdb5cafe0b670a636cca2398a7f56ee6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 19:08:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed4d5dffe15c4c0397d414e2a42964c8 : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4bbebae05994156a49886a443904c6b : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27adaa467620471d8fc7b884d84ce438 : SUCCESS in 9m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d481240698d4bf38d20d6dbe4686235 : SUCCESS in 10m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b8ac1512de74ed88cbcee7892d6950a : SUCCESS in 8m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b99c57074c64859981b3c314159f4c9 : SUCCESS in 12m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a50d9f4fa724c9794f51a93aaf16c21 : SUCCESS in 5m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fab3acf353eb4177aacf2ef10ac50dd9 : SUCCESS in 7m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/47cdfc308f584bd6a33adc6d7723c8b6 : SUCCESS in 12m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/35f3744e733e40e3b011fdb4b8aaaa62 : SUCCESS in 19m 59s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b19ba315fd91441c9382952a14dbaa63 : SUCCESS in 1h 30m 31s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85996fb5803d4978bbf11f88d7e3a0a7 : SUCCESS in 56m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e659eea516ec457484fe48c4301b233d : SUCCESS in 1h 19m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/98266296fbab4407afa249c2e7776cca : SUCCESS in 22m 28s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02845538ef6f4893a743554e9d495bba : SUCCESS in 16m 18s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/744b8f0491c84de488b07defea467077 : SUCCESS in 1h 50m 32s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/002cdddfdd0c4e6e928719d10f00fe9a : SUCCESS in 38m 09s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfcd54223ca245d881b82ddcccbef38e : SUCCESS in 34m 04s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/70bee7ba66684c47b021173e0a535257 : SUCCESS in 31m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/40283dcd0dcf466f8b70bc91061b65b0 : SUCCESS in 17m 34s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d889366ef2e44b6f927193e24c6e0488 : SUCCESS in 14m 11s","accounts_in_message":[],"_revision_number":1},{"id":"54bec747da876987f831fe30e02650a0d1622ed3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-18 20:24:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"96e694e1b494b84a211677f0843cf5bde95e54c4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-19 21:39:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"813d29d5f6ce989ac0a71cd022abad24bf708957","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 21:40:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7d2eb5ee824e17de7b1d3cc1f3655b04882607e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 23:50:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6458802f7c624bebbd6b7cc5852cbb91 : SUCCESS in 9m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38b2c142b99f4ceeb79ba41ef4ad61ef : SUCCESS in 10m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/421cee27257846cba50dd82b05c19b42 : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4545f71685774d5e91477db1a0c0727b : SUCCESS in 9m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/545174f1011145cba6080cc2fa51d321 : SUCCESS in 6m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4ac9fc5945904f01bec925636b178295 : SUCCESS in 8m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/73fae22fe11649ff9bf28b96ca9d22aa : SUCCESS in 11m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8ec54f119b32432fa124e96603c9bb6c : SUCCESS in 20m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c29cefcba75643fd8421b4914291fb50 : SUCCESS in 1h 04m 22s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0dbfa557d52247db81a254d4977f3c76 : SUCCESS in 53m 18s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9e5c0a4e5f6f438e90c8b0db5e52feed : SUCCESS in 1h 07m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1898669d83a34020b18e84d9c753aafc : SUCCESS in 21m 26s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4d23d360fa34a08856dd86da18a444b : SUCCESS in 15m 53s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a98f822f6c9a4652914c595ad359043a : SUCCESS in 1h 37m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/eee8314e7ee149759fb2c5f08fa28831 : SUCCESS in 37m 04s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7df4ff2ef8484ed6a8b1d64087f75cbb : SUCCESS in 31m 07s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/43235a0cc5e44303b63366dc5ff8c41e : SUCCESS in 28m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c2e382966865435eba586721ec1c7e02 : SUCCESS in 17m 22s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac49300edf344c549ffbbd11b821a74e : SUCCESS in 12m 17s","accounts_in_message":[],"_revision_number":1},{"id":"035aa5553be8bd921f075a2f90b5dfb54e9aa3c4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 23:50:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c96189fb1c151401d70dbe1ef4c165558493cf62","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 23:51:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aafd459b1e6a4602a204405fd82390ce : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4aaf5911a56847258d05af8f53efb343 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f006150bb7e4997add2e4829f8d3c0ea9bf9b7b","revisions":{"1f006150bb7e4997add2e4829f8d3c0ea9bf9b7b":{"kind":"REWORK","_number":1,"created":"2022-03-17 09:05:30.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/30/834130/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/30/834130/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/834130/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/834130/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/834130/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/30/834130/1"}}},"commit":{"parents":[{"commit":"ee6d4614f58be025a0af7c08de45831fd5c523ec","subject":"CI: stop using zuul as kayobe_ansible_user in TLS jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee6d4614f58be025a0af7c08de45831fd5c523ec"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-03-10 12:22:48.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-17 09:05:30.000000000","tz":0},"subject":"Use naming convention to infer VLAN tagging","message":"Use naming convention to infer VLAN tagging\n\nAny interface not matching the convention: \u003cinterface\u003e.\u003cvlan id\u003e will be\nassumed to untagged when generating the config drive. The link type will\nbe set to `phy` in `network_data.json` instead of `vlan`. This prevents\ninvalid config drive configuration from being generated.\n\nChange-Id: I00e7615581c7a214e0ee8e8f04cacdbb3ea4131a\nStory: 2009910\nTask: 44736\n(cherry picked from commit dd956189edc4b031a40453a8af71262d05513935)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1f006150bb7e4997add2e4829f8d3c0ea9bf9b7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1f006150bb7e4997add2e4829f8d3c0ea9bf9b7b"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
