)]}'
{"id":"openstack%2Fkayobe~834134","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I648fa6423ad9ff43120c7808f080b0359ad8621c","project":"openstack/kayobe","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-18 16:55:16.000000000","reason":"Mark Goddard replied on the change"}},"hashtags":[],"change_id":"I648fa6423ad9ff43120c7808f080b0359ad8621c","subject":"CI: Disable container image builds on Ubuntu","status":"MERGED","created":"2022-03-17 09:41:19.000000000","updated":"2022-03-18 18:59:51.000000000","submitted":"2022-03-18 18:58:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834134","cherry_pick_of_change":833756,"cherry_pick_of_patch_set":1,"meta_rev_id":"fd8695b67de9466a7a6d1941aad1bd394e307500","_number":834134,"virtual_id_number":834134,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-18 18:58:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-03-18 16:55:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-17 09:51:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-03-18 16:55:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-17 09:51:05.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-03-17 11:49:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6ad1397e704797fe0abb440c67bc6704e234ba1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-17 09:41:19.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0052c11633894a536f6347f417c727fce848ffc2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-17 09:51:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0e3486a0ef8ee305e0975af54436c25d40282454","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 11:49:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc0da8c6b922471485593bff8a2e6ae2 : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b77a533ce9e4acfbf6e440e74517d25 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7444ae03278f4f699991c90a124c8d1c : SUCCESS in 8m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3115b911159148bd9e7f1fd42d0f192c : SUCCESS in 8m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ee66f94eb064efda6bf566687b86233 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26adc03206fd45f6b3ac42e14f311b4e : SUCCESS in 10m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/48cef17862854163b7060101f5b63d10 : FAILURE in 8m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/de2e5edd2fa349d08c4e32ba2cbf0b43 : FAILURE in 8m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cff89208de1c433897ea6514f618bea6 : FAILURE in 9m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e74cd38d0d6f41849bccf208545a1cbf : SUCCESS in 1h 09m 59s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d04fa8c6394f48809f6089d840ec4fd4 : SUCCESS in 1h 01m 36s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0ea3bb2180274c2d946606ac3d42fcdf : SUCCESS in 1h 23m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e648766882ab495ab759e2824af8b11a : SUCCESS in 21m 56s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09565d7817734d75882b5959869394a5 : SUCCESS in 16m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/911e2389b9c4482a9bfe49ea5d6e177d : SUCCESS in 2h 01m 01s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/846550dee306444791f989e8abbb52f9 : SUCCESS in 33m 06s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5025d343ca23442bbed071745f2130c0 : SUCCESS in 30m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fbc3faff3de9477f8531075127ff8ca7 : SUCCESS in 38m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/edadb1c38a5f4e8eb25f70bad41fbeaa : SUCCESS in 16m 03s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/732facac77b541b8924902589b8822c0 : SUCCESS in 13m 54s","accounts_in_message":[],"_revision_number":1},{"id":"0606d1810235a4e3af0adfb258f3c5f1a887e13d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-18 13:20:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"31420cfc5d5f96245212d979ee97a3a725b3fa69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 15:20:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/206783728a164cb7965fff524c0cfa6e : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5979d4350ad4794a125435d79a2c7df : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/684364238e844decbc0c61662e177299 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bde3d2e5c7614d88bbe568f7a00d2d21 : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3d121b06a8e48909e18eeca352785ab : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4bd7b442e9e4666834423affbcccedc : SUCCESS in 10m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/473faf52760c4262b7c2e2686c4a3fa5 : SUCCESS in 7m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/272482854830436b9b17cabf86a01de8 : SUCCESS in 11m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2d72094d96164ef5bfe2dbb5e5828d09 : SUCCESS in 21m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1e8c137eb1bf4a3fbda26c726a345b25 : SUCCESS in 1h 02m 49s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc64236e5ddd4714aff04fc723ca16dc : SUCCESS in 1h 19m 28s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/152bc0e968a842afaf7747b4ab3f1ad7 : SUCCESS in 1h 07m 28s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/786038bee41c475d877eeaf3e5858046 : SUCCESS in 20m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/680c033b33234983a521befd3b2fb470 : SUCCESS in 13m 46s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/edb09edb7f554e48b5846e5ef3e9d1dd : SUCCESS in 1h 57m 12s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/991faa537a9f44d8903af55c928b910f : SUCCESS in 59m 12s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ac78309a4894cf1a717f4cecc5a8325 : SUCCESS in 30m 44s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/896e604f162f433d99df02dd4e484e76 : SUCCESS in 25m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6eae56f66d7c439dbb15e20df4429350 : SUCCESS in 17m 10s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ece5e7ec7a3943f7ac78d36407468718 : SUCCESS in 12m 43s","accounts_in_message":[],"_revision_number":1},{"id":"aff30d7fde432ff3af4e91262695af3d24d3cabe","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-18 16:55:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c078e0c04a5b082bea62c0446af7b536ffc09b9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 16:56:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3b8000697d48b9cd3bcb9ec54d1f8ec66b01236e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 18:58:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a20e27498a14dd38ffdbfb2c0cfb760 : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d475929df1b74245ac17ad0bd3054e51 : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16d7cff91de0475aa0a8e9b4ab049734 : SUCCESS in 7m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4cd2a9a595045838f8522333aead59b : SUCCESS in 9m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/55ce86aae6d34caebeb9b2e49961147f : SUCCESS in 7m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/94dced9996e745af9d17faea646224e9 : SUCCESS in 11m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4734abe793a34b75ad8e158f5dbdc7c5 : SUCCESS in 20m 32s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9743525b4476443c807cb410c824d407 : SUCCESS in 1h 18m 37s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f79dd6e098f5463e893fa8e9fc510c65 : SUCCESS in 1h 13m 50s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/75e9ff8f6f044f4e9a2a073602f0cec0 : SUCCESS in 1h 18m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6e7a07609921400984253f2664a12355 : SUCCESS in 24m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/845b4cd901ac44fe993fd36bca232d05 : SUCCESS in 16m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/50c7736bcb32439aa41154556f009c01 : SUCCESS in 1h 58m 19s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ae6185a2f3d04a638113c3239eb87074 : SUCCESS in 36m 55s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/786f7b97c0f94350a854f8435d09de7a : SUCCESS in 33m 55s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/54aba4254c3649649ea626254e2ceae2 : SUCCESS in 29m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/82015096f379429cbb899eae69fa2a8f : SUCCESS in 18m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ede229c512d47f2b649341c7d50f189 : SUCCESS in 12m 45s","accounts_in_message":[],"_revision_number":1},{"id":"9594438bf425dc2eda8a9eedff4b79801dc17630","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 18:58:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fd8695b67de9466a7a6d1941aad1bd394e307500","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 18:59:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d455e963e51495386c79d6429d8274c : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c129808cd573bf33a0afb5c548971cb81c63602b","revisions":{"c129808cd573bf33a0afb5c548971cb81c63602b":{"kind":"REWORK","_number":1,"created":"2022-03-17 09:41:19.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/34/834134/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/834134/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/834134/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/834134/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/834134/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/834134/1"}}},"commit":{"parents":[{"commit":"ee6d4614f58be025a0af7c08de45831fd5c523ec","subject":"CI: stop using zuul as kayobe_ansible_user in TLS jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee6d4614f58be025a0af7c08de45831fd5c523ec"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-15 09:52:05.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-17 09:41:19.000000000","tz":0},"subject":"CI: Disable container image builds on Ubuntu","message":"CI: Disable container image builds on Ubuntu\n\nThe kayobe-seed-ubuntu-focal job is currently fairly unreliable, often\nfailing to build the base container image.\n\nWe are not using the mirrors provided by OpenDev infra, which may be\nmaking these builds less reliable.\n\nThis change disables container image builds in CI on Ubuntu. It should\nbe reverted if they are made more reliable.\n\nChange-Id: I648fa6423ad9ff43120c7808f080b0359ad8621c\n(cherry picked from commit acf6d0d51fe54588e5083b19b3824041e1fcb49d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c129808cd573bf33a0afb5c548971cb81c63602b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c129808cd573bf33a0afb5c548971cb81c63602b"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
