)]}'
{"id":"openstack%2Fkayobe~834135","triplet_id":"openstack%2Fkayobe~stable%2Fxena~Id5af3969da63150e892201f7518f50a3da73e852","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-18 15:41:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id5af3969da63150e892201f7518f50a3da73e852","subject":"Only create patch links on overcloud hosts","status":"MERGED","created":"2022-03-17 11:08:37.000000000","updated":"2022-03-18 15:43:48.000000000","submitted":"2022-03-18 15:41:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834135","cherry_pick_of_change":833125,"cherry_pick_of_patch_set":1,"meta_rev_id":"5841b2a8eb61b446dacfc6e6ad530fbce48a0fc4","_number":834135,"virtual_id_number":834135,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-18 15:41:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-03-17 13:03:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-18 13:20:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-18 13:20:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-17 11:08:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2022-03-17 13:02:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-18 13:20:54.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"a1cfc2771b0ac7830eb2360f3dd651e4909136fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-17 11:08:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6690a0ef0e8822c9046ff894637addad06ba184b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 13:02:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da213eaf30154ef384818eab3d119c57 : SUCCESS in 9m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3f2db7959da4ed8970b9f23cd5e602b : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b24e345a56f449ca808580f43967b933 : SUCCESS in 7m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ef43d60f4ce4514a8b47e905d82809c : SUCCESS in 7m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f44304e89785409ab2d9c279f7911e66 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/640e8479062642209d5b32804d4f9312 : SUCCESS in 11m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1e9d591006449cfb252578ff84900f6 : SUCCESS in 5m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5b73271e4c964cb48f63583f5137c05d : SUCCESS in 8m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e03dc60a2e8c4792b5353c7c3a82113f : SUCCESS in 14m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/94e40d17a29645c8a91fc39a2513fbbe : SUCCESS in 33m 28s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2fe30e1f4e0841559c401d420b61dba0 : SUCCESS in 1h 08m 17s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66ba8499e7e94f3d80ef5d5f89230c21 : SUCCESS in 54m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0bfa20ccf5ac4258ba86440bb7abf2b7 : SUCCESS in 1h 08m 16s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/71a14c86291d4607aca178444c5c0e61 : SUCCESS in 24m 28s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/240949c9e2b64c7d9cc462d864dc580e : SUCCESS in 16m 11s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9a467e28d2f14b6e92b850f45514310c : SUCCESS in 1h 48m 33s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0286eb13c66430ebdae5681295bd4ff : SUCCESS in 1h 15m 56s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a5754dd03bd344d182ffad81c753cc05 : SUCCESS in 40m 34s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7357e0d32b8c4472b989ec4b5d99a69a : SUCCESS in 32m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cbd30304e8bd4944b138c8e5dd842d42 : SUCCESS in 32m 19s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c759c57885646dd97c4f14ec69072ea : SUCCESS in 27m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/eaee4626950c47aa987b5a66f8f9a66a : SUCCESS in 17m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77d718c4629c4322af07a97b810c2593 : SUCCESS in 13m 27s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/80663c7ba77649f28be9f8ee74999f2e : SUCCESS in 19m 33s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f77acd9cae14a03afaaf5b4b88499de : SUCCESS in 15m 27s","accounts_in_message":[],"_revision_number":1},{"id":"08c289c96db2c53f1b1c81b2dd971575b80b34af","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-17 13:03:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a35a0b3850ccf9e2632f355a09131df9d53ebf0f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-18 13:20:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3761ee34453d8ecb2d63d64f9b07b15d6657fe45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 13:21:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"04ecdcc0b589ad89d8498fd1e38fd19a97d37389","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 15:41:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84326d9bbf0948348caaa9c409dcf596 : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e30aa8af8aef4e9c8738071f8a2bc246 : SUCCESS in 8m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b439ab6d2ff84771bf608b729ee2ac13 : SUCCESS in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5083395213ba47e39feaa3049f52e2b7 : SUCCESS in 10m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bef8f91acfb4b2fa6dd12aec82cba08 : SUCCESS in 5m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8abfeb3f63614812bc36d2e253035ad8 : SUCCESS in 7m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fefe1649949e42fd82448c158d28f8c0 : SUCCESS in 14m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f68357d6bf1246e2aab74f8d51dec4ed : SUCCESS in 39m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4c98ef30e0774843983dc989ab928432 : SUCCESS in 1h 10m 02s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6018cebdfc1460c9c8d34abc7d3d306 : SUCCESS in 1h 00m 29s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5e705b6eb9c849718e4e27be8df69209 : SUCCESS in 1h 02m 48s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/62cc9fd4dc5445cda4e098319e499c31 : SUCCESS in 25m 39s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a213cd56f7e04ba8a6275e1dae1f1111 : SUCCESS in 16m 01s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3446b8bc0e324df19e1b57cd1a375a4e : SUCCESS in 2h 13m 14s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5d811125c704e4a95b2a2f33b09d7f6 : SUCCESS in 1h 09m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9b84707d83e54eda81e26547b2825a48 : SUCCESS in 38m 36s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb0389886fc84f9fb47e2d92e501eb52 : SUCCESS in 32m 13s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bbd17dba9b134eae8bac811de2fd852f : SUCCESS in 30m 47s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a4c591a43914f5fa145facd33abcae8 : SUCCESS in 26m 50s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b14095f1eead494e9a104b62485a3d4e : SUCCESS in 17m 56s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e88477889fe448ca3017d607e7e7414 : SUCCESS in 14m 34s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/6e30cb10599742dc856b83bef7666d28 : SUCCESS in 19m 30s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43c70ea6c9e64df5b4b66331924d84b8 : SUCCESS in 15m 55s","accounts_in_message":[],"_revision_number":1},{"id":"f155fd18a86a6ac8fc6fa2aebb1be785bca16711","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 15:41:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5841b2a8eb61b446dacfc6e6ad530fbce48a0fc4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 15:43:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a76f5c605fc54b8496c6c1237bf2baec : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7e83459a84e4e708c05b09f145af2a1 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"512f4c1e9e8a37c7ca3dd4925ff57ccdf1ec3159","revisions":{"512f4c1e9e8a37c7ca3dd4925ff57ccdf1ec3159":{"kind":"REWORK","_number":1,"created":"2022-03-17 11:08:37.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/35/834135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/35/834135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/834135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/834135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/834135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/35/834135/1"}}},"commit":{"parents":[{"commit":"15790c983e9f88f8bd735f7e7cb2b6db605adee2","subject":"CI: remove qemu-utils installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15790c983e9f88f8bd735f7e7cb2b6db605adee2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-03-10 16:51:38.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-17 11:08:37.000000000","tz":0},"subject":"Only create patch links on overcloud hosts","message":"Only create patch links on overcloud hosts\n\nThese are only neccesary on network and compute nodes.\n\nCo-authored-by: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: Id5af3969da63150e892201f7518f50a3da73e852\nStory: 2009911\nTask: 44740\n(cherry picked from commit bb9da9fb2428149b3f213268c7cfd6575edf787d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/512f4c1e9e8a37c7ca3dd4925ff57ccdf1ec3159"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/512f4c1e9e8a37c7ca3dd4925ff57ccdf1ec3159"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
