)]}'
{"id":"openstack%2Fkayobe~834188","triplet_id":"openstack%2Fkayobe~master~I88bdf1e3d0c08271ac8938ae2f9ac3f9fee1efa5","project":"openstack/kayobe","branch":"master","topic":"story/2009927","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-04-07 10:31:06.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-04-07 12:25:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I88bdf1e3d0c08271ac8938ae2f9ac3f9fee1efa5","subject":"Fix Ansible inventory generation when reusing group names","status":"MERGED","created":"2022-03-17 15:35:22.000000000","updated":"2022-04-07 12:27:05.000000000","submitted":"2022-04-07 12:25:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"834188-story/2009927","meta_rev_id":"f13848f9ce06e2d34f069330d4d1460533dd5be8","_number":834188,"virtual_id_number":834188,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-07 12:25:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2022-04-07 12:27:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2022-04-07 10:31:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2022-04-07 10:31:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-17 17:33:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-18 09:51:28.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-04-06 16:45:57.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"35aac302d7169890695a245ae21e9d5d41810795","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-17 15:35:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b514d0c6a4feeff68452d56228041e9fa5a7b292","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 17:33:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fe7d83254aa4ea3a0e02859cbb26e8e : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c31a16a80a0945b8851b9a0d317f3987 : SUCCESS in 8m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/658bc44b1b3f40f6b476056a19a92c42 : SUCCESS in 7m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4cd750b1ee42483dbba7cd5b33c2f919 : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0edf3eefcaa244b88e49cb5f62df2ff5 : SUCCESS in 11m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d53c6e86475249179f1ad42c9c072a1a : SUCCESS in 5m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ae32f247c26244579f0914f820f52117 : SUCCESS in 8m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/540ab81f6c1a471893c17e6b39d81f2f : SUCCESS in 14m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fdc62b95f59c4e8488f2cc90861be2e4 : SUCCESS in 32m 50s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f9592bd472fa43b785310bd40f593ec8 : SUCCESS in 1h 07m 04s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6127cc5db7e4fd5a43cfb309dee6fb5 : SUCCESS in 54m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/67e65dbb3d9a4c3d960217b1bb1255da : SUCCESS in 1h 07m 27s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6d4cfd1fe91b4f4381a767e2563ddba9 : SUCCESS in 22m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db3bc46d3098427eacc947721cc1d8ac : SUCCESS in 15m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4716efea9b3b43fd8f25010eab77a919 : SUCCESS in 1h 52m 32s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/951d63a4872945ed8ebcbfe655280ce5 : SUCCESS in 1h 21m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6c34a13d6e2045dc8b37bc8f527322ce : SUCCESS in 54m 27s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf31c26ec60e46d2a3a3fb7b78efb215 : SUCCESS in 39m 39s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8f15e1b37f944cf2b0ca55e0efb942e4 : SUCCESS in 40m 49s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b0c7782edc1419ca04b21b4f22bd8f2 : SUCCESS in 35m 57s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/910049c1ecd945bf8cc7b0b56d24af41 : SUCCESS in 17m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d59cc4cbd654c4eb9ed1d97c55c3209 : SUCCESS in 14m 31s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/6469cae13bf545988eab125cde37305e : SUCCESS in 18m 17s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c24f4fbac6cb4ae593307edcacb4c4e7 : SUCCESS in 15m 49s","accounts_in_message":[],"_revision_number":1},{"id":"b1ff258691c42c73e8e0f760578e456c2905a70b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-18 09:51:28.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d101c1383a56afa2128fa88e61b61e9d30d6216","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-18 23:12:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2646f1046101f9bdd3a2eb794f1c2dd32589257","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-04-06 16:45:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"12980a34edcea1c132324b6203bed4964267e70f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 16:46:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e877316276acb45d50819720d1146ede4e1b90e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 18:47:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e266a91049b46589c6bd38fcf4b8790 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e5d70264a0046d98a0454bb58652d5f : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35d9c1fb9ac64411bba47890f2a6c076 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc3d581b54aa48c9ae4d1cf6fd29878e : SUCCESS in 9m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38b23d60b2984bf680c7a85cd9755d61 : SUCCESS in 3m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/83e5a8e60edb40e7b6f7aabde91e958e : SUCCESS in 6m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f07df8ea5a86463a9c8cb2607bb65e56 : SUCCESS in 13m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/122325d049bd4c09a1c6fdfc213172fe : SUCCESS in 29m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5ed65f71bff74856b29fdddaddfeffca : SUCCESS in 1h 05m 42s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/3508c63cb8c049b599dda55c0b353806 : SUCCESS in 1h 03m 12s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c092fd61a47460c8954dc820712a336 : SUCCESS in 52m 04s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e41196ad29ab4dc5b983176308355f03 : SUCCESS in 1h 05m 27s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/737ba86cd0e4461bae27c8d9a73a749b : SUCCESS in 25m 50s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/176d91c8aa3e4034b9f24e6bd9392e2d : SUCCESS in 21m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/448e475576644c8f9b52a74ace10d75d : SUCCESS in 14m 21s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1c01651f1f5e4c4f9095bda26c8b7e17 : SUCCESS in 2h 00m 19s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a8182df15f048cfb9baa29f8c25caac : SUCCESS in 1h 14m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/16bd410579434e22b50b750c84d32171 : SUCCESS in 19m 48s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/37f0ebcd3c854c3b9d39ee1f4e6fdb3e : SUCCESS in 18m 32s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/313fca3044ed48c4975f3f35bb3d4fc7 : SUCCESS in 15m 27s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c79efbc4b1424207ad0ef9e9a3257e24 : FAILURE in 21m 27s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3009e67eb24b486ea5c401ef2e4ce9b6 : SUCCESS in 28m 42s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d2d19947cd3a40e3ac2fbc87a238533e : SUCCESS in 15m 49s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f770ddb640c34d2c8c9aff64c8c25836 : SUCCESS in 13m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99c86d7179fd41d2a12befa43e054917 : SUCCESS in 12m 34s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/d8fc197e041d46a899fb094aab227b92 : SUCCESS in 17m 15s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c0f347c0bfd441e5acbc285a31c9dda3 : SUCCESS in 17m 32s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77adce1328854d6abba22c17285fa7ca : SUCCESS in 13m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f8a4fd264e5b06b411ebb83c580eda605bd0e900","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-07 08:02:02.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"9eea7985f208f5a82dfe3f1b63b53569f3d2edbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 09:59:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/044b95f773a1424e8cb9efa3da2123c2 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42a699d4980a4238aa2053186a2e8644 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd5641f1a10c4561abf320efdf9b05db : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f033bde31d90497193d73b38468a9789 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc9d4cd6de62442a9cf5dfdf9216909d : SUCCESS in 8m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e32ca5b22f9406597fdbdb13bb2a599 : SUCCESS in 3m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a585e8fb177c4fa0bb35b0dde0ae6157 : SUCCESS in 7m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5a348f9887f443a2b4a84bd1a0e42cdc : SUCCESS in 13m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d6399f42266340de92671ac9ce8b52a6 : SUCCESS in 32m 36s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b887ed26e085411e8b15a31ed08dd9e2 : SUCCESS in 1h 07m 23s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/3449a95083ea43f5aa084d8e33624ab1 : SUCCESS in 1h 06m 51s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac08e9213592449f8872655205ae7b0a : SUCCESS in 47m 56s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/048de3cc3874444e87038539e77f0852 : SUCCESS in 57m 55s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/044bc424f9f248cca72d29bd3beba005 : SUCCESS in 20m 44s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/3b5e1ff3b7b44f5988d075aeb3aec468 : SUCCESS in 21m 45s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc4684c50c564ce4967e6e0fc82404bd : SUCCESS in 13m 53s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d98421f0c09049eaa3125048af78766a : SUCCESS in 1h 35m 55s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1df463d0b8ac4a81948266feb63cf074 : SUCCESS in 1h 02m 44s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b2867d2d62ac4e848ea2275d460b8894 : SUCCESS in 17m 04s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/cd798d90814340b1b796504fa6c9812f : SUCCESS in 21m 00s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9bd38e005974b6db5fd4ff36c4b4a29 : SUCCESS in 14m 36s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/6828142ce0724c248bf45de192c67bf2 : SUCCESS in 53m 53s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1203a335bff34a75833a2e58fc16154f : SUCCESS in 38m 45s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5f17e13c67f4d53a45b61a94839ec2b : SUCCESS in 26m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4b0950aab78e4029beb3eb6cb307e661 : SUCCESS in 16m 32s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/65222055481340c0bf6a3458e59348e8 : SUCCESS in 17m 12s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b48549c447440439e4d06e9659c4bf4 : SUCCESS in 10m 47s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/830e6f173122461c90299bee67217d6e : SUCCESS in 14m 36s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f8189811dd5b4ab4bcc3df8658215358 : SUCCESS in 19m 13s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e87f871a1b3044708c00a9687369a722 : SUCCESS in 13m 29s","accounts_in_message":[],"_revision_number":2},{"id":"3e887e2fea7aad8f67eca5d72d48b220eb857b67","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-07 10:31:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9d3593b42efe210b27f51237e7a7f881cbf03852","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 10:31:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a3f8c3c29f2a68f51d16451058ea8f21836b8b42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 12:25:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae18aef37a73457f889ead15e9314414 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f77d92e110d4838bbbabaf988ba2e0e : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bc2275d13b84f42bae1e4cb2185d2b7 : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92ddd6e703dc4efaacd4f9fc50920e10 : SUCCESS in 9m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ca028178df642b4a1222dcff95cce43 : SUCCESS in 3m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8f4e54ff8e67480e8014b09932a2fb04 : SUCCESS in 6m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0e730d20677c4fd0aaeeea34f46ad8f3 : SUCCESS in 13m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8eab50470df8498ea324806e0c05da4c : SUCCESS in 30m 45s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/88c3b9852f5c4f259de700965a57cea8 : SUCCESS in 1h 07m 54s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/b4691b87ae0a4b44ab30935347360243 : SUCCESS in 1h 07m 12s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/956f6f62a69742cca515f93c7bd5c9a2 : SUCCESS in 54m 53s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e2d35debecf44ea88724d2dad69bbf82 : SUCCESS in 1h 10m 12s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5a498554a9564490b1f8794a296669c1 : SUCCESS in 25m 17s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/8f1fa631ad184e1cb8b7586f43250674 : SUCCESS in 22m 44s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fe0f6c87c3e4d9db97f1e558a507952 : SUCCESS in 15m 11s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4e118383ea114eeebb44288cde1634bb : SUCCESS in 1h 50m 28s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc7f5e7776574734b5f8694c09537e5f : SUCCESS in 1h 17m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d59766bd0b874394ac5304be3bee01f6 : SUCCESS in 20m 21s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/004ff7cb1cf14ff9b20243bcc4e1b2a5 : SUCCESS in 23m 09s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13491fc9f79a40e09603524c9a3cb7a4 : SUCCESS in 17m 26s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cb5bdd13ec904efbac068e40ec6fa2fd : SUCCESS in 44m 32s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed4e771a0ce24a9283dc760b9c98464e : SUCCESS in 31m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c70b21cfc818444fb51605a665b4f998 : SUCCESS in 17m 28s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5bede8d7dea84676b8302687603eafd3 : SUCCESS in 16m 32s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78c2d55b72bb4bdca94fd9085e145372 : SUCCESS in 12m 40s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/1ddbc4282d9b4389b0a78a2f34c5bcaf : SUCCESS in 15m 53s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/df09dbd9677546f89e7c6f6790aaf730 : SUCCESS in 17m 04s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db8d9b9ea56a447f8383c6e8fe6d16b8 : SUCCESS in 13m 48s","accounts_in_message":[],"_revision_number":2},{"id":"312bc1e96f49586d4d0522b812fd89d8ec82b8f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 12:25:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"56caada6c63ad1893ee875629834b49a57511548","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 12:27:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/849e4c12ba0d43afb66737f8ad517b90 : SUCCESS in 1m 31s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bbef5c5f23946e693142c216eab7638 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0c190a22f0f50385f598fa92d6962842bd9b96eb","revisions":{"b8150d0c17cd651c50f0dace9c7296f5502f48b6":{"kind":"REWORK","_number":1,"created":"2022-03-17 15:35:22.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/88/834188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/834188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/834188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/834188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/834188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/834188/1"}}},"commit":{"parents":[{"commit":"a1ed4a05bb51e89abf7841715dee8251f9fbf53a","subject":"Merge \"Use naming convention to infer VLAN tagging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a1ed4a05bb51e89abf7841715dee8251f9fbf53a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-17 15:34:54.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-17 15:34:54.000000000","tz":60},"subject":"Fix Ansible inventory generation when reusing group names","message":"Fix Ansible inventory generation when reusing group names\n\nWhen a kolla-ansible group is composed of a kayobe group with the same\nname and of at least one other kayobe group, kayobe would generate an\ninvalid Ansible inventory such as:\n\n[compute:children]\ncontrollers\ncompute\n\nBecause the top-level group should already be defined, we only need to\nadd as children the kayobe groups using different names.\n\nChange-Id: I88bdf1e3d0c08271ac8938ae2f9ac3f9fee1efa5\nStory: 2009927\nTask: 44798\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8150d0c17cd651c50f0dace9c7296f5502f48b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8150d0c17cd651c50f0dace9c7296f5502f48b6"}]},"branch":"refs/heads/master"},"0c190a22f0f50385f598fa92d6962842bd9b96eb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-04-07 08:02:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/88/834188/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/834188/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/834188/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/834188/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/834188/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/834188/2"}}},"commit":{"parents":[{"commit":"60d74df8159b14e35ea28051ff0b66b8b16cb703","subject":"Merge \"CI: separate image builds into a non-voting job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/60d74df8159b14e35ea28051ff0b66b8b16cb703"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-17 15:34:54.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-04-07 08:02:02.000000000","tz":0},"subject":"Fix Ansible inventory generation when reusing group names","message":"Fix Ansible inventory generation when reusing group names\n\nWhen a kolla-ansible group is composed of a kayobe group with the same\nname and of at least one other kayobe group, kayobe would generate an\ninvalid Ansible inventory such as:\n\n[compute:children]\ncontrollers\ncompute\n\nBecause the top-level group should already be defined, we only need to\nadd as children the kayobe groups using different names.\n\nChange-Id: I88bdf1e3d0c08271ac8938ae2f9ac3f9fee1efa5\nStory: 2009927\nTask: 44798\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c190a22f0f50385f598fa92d6962842bd9b96eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c190a22f0f50385f598fa92d6962842bd9b96eb"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
