)]}'
{"id":"openstack%2Fkayobe~834309","triplet_id":"openstack%2Fkayobe~stable%2Fxena~If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-03-23 11:27:06.000000000","reason":"Michal Nasiadka replied on the change"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-03-23 13:33:12.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-22 16:25:15.000000000","reason":"Mark Goddard replied on the change"}},"hashtags":[],"change_id":"If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad","subject":"Add support for Rocky Linux 8","status":"MERGED","created":"2022-03-21 11:24:13.000000000","updated":"2022-03-23 13:41:27.000000000","submitted":"2022-03-23 13:33:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834309","cherry_pick_of_change":831652,"cherry_pick_of_patch_set":53,"meta_rev_id":"7fc41d023b3865271e18b49a7d39e0868e80129b","_number":834309,"virtual_id_number":834309,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-23 13:33:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-03-23 10:06:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2022-03-23 13:41:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-23 13:41:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-03-23 11:27:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-21 12:12:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-22 12:43:59.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-03-23 08:27:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-03-23 10:06:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"9ac6788aa430e709e552db20102f4d5d8dee9708","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-21 11:24:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* ansible/group_vars/all/infra-vms\n* ansible/group_vars/all/overcloud-dib\n* ansible/group_vars/all/seed-vm\n* doc/source/configuration/reference/overcloud-dib.rst\n* doc/source/deployment.rst\n* etc/kayobe/infra-vms.yml\n* etc/kayobe/overcloud-dib.yml\n* etc/kayobe/seed-vm.yml\n* roles/kayobe-ci-prep/tasks/main.yml","accounts_in_message":[],"_revision_number":1},{"id":"f78cc59929f7afcf83a4cc6e20a66b3e7246fa88","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-21 11:26:35.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"65c3dff2dbbf66db72cd284b7f80a6550bb794ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 12:12:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6fbed4873724e138a6507095e728b7a : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63b44f560c4248638b74434d90df9ca3 : FAILURE in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac18cc861ce24a37bb1dfcbb88bbc97c : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ce68426648447ab8f842f3e37b0d843 : SUCCESS in 7m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc7e91a6f1694373aa98db45cad53366 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a17e3a2667d496da7c56ae6336a174a : FAILURE in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a29d626c49b546989bef9d1881b61f7e : SUCCESS in 5m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/abb25bccb4394146adbbacd044d516d5 : SUCCESS in 7m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/32117b37ee2e4c23844dc0e186017bf4 : SUCCESS in 12m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cc7576dc2d6f40f18b9bdfe72da7d993 : SUCCESS in 28m 13s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/72d8735ee6d6423aa8d595f59e3b4584 : RETRY_LIMIT in 5m 27s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/964908f1a7ea48dca8c9b717d3e84fc8 : RETRY_LIMIT in 3m 49s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be9ad2082d0e49878db209cc54c1fe3b : RETRY_LIMIT in 4m 45s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c22e5262873d48bd87fe579ba36a915a : RETRY_LIMIT in 5m 30s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7e863cdfb75a40cba9dc5bf1e07bdea8 : RETRY_LIMIT in 4m 22s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/e568edbac9f74b6ba1ac923a457099be : RETRY_LIMIT in 4m 13s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b473a3754b241faac9f9e914790065d : RETRY_LIMIT in 4m 51s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c36ac344e840402c9799d2db67bac990 : RETRY_LIMIT in 4m 25s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dbfb68a6b804f378498cd5c7ff9fb52 : RETRY_LIMIT in 3m 50s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ae97f6fb040442f0b1d63d485599c709 : RETRY_LIMIT in 4m 12s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/b7e564028d0f475480ed9d7bccf929b6 : RETRY_LIMIT in 3m 38s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64f3224d66ec456a9264c9e249f0dc6e : RETRY_LIMIT in 3m 52s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6a3b93d106444834ab66201bcd724980 : RETRY_LIMIT in 4m 20s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b61e916da924697a0290ffc074c8ec9 : RETRY_LIMIT in 3m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2901264586264dad92a3edc33ec1373f : RETRY_LIMIT in 4m 20s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/66a572bd128d4a27b1a85b05b4a3cc3b : RETRY_LIMIT in 4m 12s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/320ed6c1771d437ba3c351b1d71502ca : RETRY_LIMIT in 4m 04s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/073e98302b0a4d8288a8e0fd8fea1e48 : RETRY_LIMIT in 4m 22s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/32326bf4b6384179b8b41835bd6c4b3d : RETRY_LIMIT in 3m 46s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3083148698354397b422e1fac7e1dc08 : RETRY_LIMIT in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"d83d46857f0f49e0f86cfbb73a8673b5cb26d397","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-22 10:51:37.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3c5bda7ae75e90b879f9edfccd5daa344193dc83","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-22 10:52:10.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f6d1f6e67a2358a85aeb4f07efa0e5b6e5d67d6d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-22 10:52:42.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"8b8bc46c4fbaca4d0576ff19a7d6096dd466aa53","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-22 10:54:22.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"993937e1639267755d7eda8821d64a1cdbadaaac","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-22 10:55:06.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"ddc1a5902533ff19ec43f349b9937d62ba3fdd39","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-22 10:55:51.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"222270c9de436db723bea848dad5bdaef7c4ea90","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-22 11:41:39.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"fa546271a3724002e75eb2a77c99439306c952ce","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-22 11:43:04.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"9635d13bc5773e227bb80a17a7c089a0ed95214c","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-22 11:43:27.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":9},{"id":"d86db984b7860bde39a1b9e5a1c5f113ce9afab1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-22 12:43:59.000000000","message":"Patch Set 9: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"9d9b2c3a37ca42251b4909ba03648011a7e1065f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 13:39:18.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ac43d087d194b348948c709e9e8e21a : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5104bbd43e4a4dd59a76bf6c4ae2081b : SUCCESS in 8m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f33e0779920f43629ebd2f643e0722bd : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8e01d42a6364ad2af89abedf3f9c92d : SUCCESS in 7m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f45516c7172e49ffb139b4654372ba95 : SUCCESS in 11m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebaee21c57094ec290d9a7fdab731925 : SUCCESS in 11m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d38e808d076472893c0833b7e2a570f : SUCCESS in 5m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5428e772373e46bbb5013ba3030a2740 : SUCCESS in 8m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e6627b251fe24ff287e89b2f06f9e718 : SUCCESS in 15m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/66601ef69a9e44b39c3fb474c5fd1fa0 : SUCCESS in 34m 50s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3d106f29e681473eb46e077af2a341a0 : SUCCESS in 1h 22m 57s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/f24aea6f40a9408abea2a16191fe9db4 : SUCCESS in 1h 17m 19s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11fff0247ee54a16a228857ce813b869 : SUCCESS in 59m 14s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1a3ee543a6944ba29a9cc199318f1d04 : SUCCESS in 1h 12m 33s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a89587b93377463dba32b5cd736f7394 : SUCCESS in 25m 29s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/64ed4044e63e40db9a08b3b81e015634 : SUCCESS in 22m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7b7bd7f4f1d46bab6e2e79a2b1230fd : SUCCESS in 24m 52s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4cb3fe40fa2b4b1e9181f5a47f12a3b7 : SUCCESS in 1h 50m 38s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a013dbc6485d4b72aae81f3d092560ee : FAILURE in 1h 51m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8f136cab65584934a53985aa44a461e2 : SUCCESS in 51m 30s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/52164da65481486dbfbaee4da0922646 : SUCCESS in 37m 30s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d5ae14e53844a7e92a26fd4f3abc23f : SUCCESS in 56m 05s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2fe277741d8b4561856065e8e111c7b7 : SUCCESS in 40m 31s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d4d5ac235734d7ea5f3e5025043a957 : SUCCESS in 55m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a0316801b13f4b138b762537a556c110 : SUCCESS in 15m 23s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/8a4a667aaac14daba594935a311e9e8d : SUCCESS in 15m 50s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5409a111b924e1ab4b8cf479b6b2ae9 : SUCCESS in 11m 23s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/18ddcd3810d2466ab4fbf776a4fc8956 : SUCCESS in 19m 49s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c3e21a131ed44d12b09606bfa06e5be5 : SUCCESS in 17m 33s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39fd77d4bb4846b4a8e966dadaf84c65 : SUCCESS in 15m 23s","accounts_in_message":[],"_revision_number":9},{"id":"0b240d73a10394800d1d086b42ffbed96b183eee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-22 13:47:20.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"84ca1b41474c8784176b593db6d4cb3548d9dcc7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-22 13:47:25.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"48cfc614cb849e2f8b1ea26d96fd7e6a05e4ec55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 15:50:14.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa3585f0580c4a86a1729fc5189dd780 : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e5faeda324645e9aee66ee58525897d : SUCCESS in 10m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4a49bf1948c4953929cf6e1c0ac7b50 : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c88b5885532c4c36b57446706f08ea77 : SUCCESS in 8m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d79256e28ce4d998f0bc2c5b76a9850 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55272f879a514af2992528aa66441faa : SUCCESS in 11m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/effbaa0c82b742c8ae82c3f843b31650 : SUCCESS in 5m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c467c798acc440659f1f58dd7fbd0a42 : SUCCESS in 7m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c4ce34192bbb4217914859b08db56399 : SUCCESS in 16m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ac79092f551c4c34b37bc367b6ad034f : SUCCESS in 36m 14s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0fcf38d9d425472688adb7d1182809e8 : SUCCESS in 1h 07m 16s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/cd4014fd82ee42b6839957050d592638 : SUCCESS in 1h 11m 57s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5240de00d8cd412b811df3107b625e90 : SUCCESS in 57m 36s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4582cd207d034c8cab2159a7a64f21b7 : SUCCESS in 1h 28m 22s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/859a1d3697644f53a2500373275c1b6d : SUCCESS in 25m 58s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/6ba4b3ac5f1f42598d352e622dd62178 : SUCCESS in 25m 04s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d0896dd03194b3ebce29be30c9d952b : SUCCESS in 16m 17s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7de75d4c74da45d3a8b526c3350fc33d : SUCCESS in 2h 00m 49s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef9a1ac8cd614b2fb168ae1203bae84f : SUCCESS in 1h 47m 24s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/35eb4d76e92e4dc4892c6becdfe621b8 : SUCCESS in 50m 14s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/4d03c07365324ad0b18be0e9ce0b6e66 : SUCCESS in 35m 46s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b6a4123906d400f9f364ed0d2a88f0b : SUCCESS in 41m 02s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2d18f6b64af34152bf900a67c03035bb : SUCCESS in 47m 52s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0695fc8905c4c7b8b19932153dff6dc : SUCCESS in 31m 30s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/86612515790647da940a6fe079ec5721 : SUCCESS in 21m 18s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/27506e5224c74d3298a04c23aac72894 : SUCCESS in 16m 13s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/869f16423b4f4408a805c408c5d7d25a : SUCCESS in 11m 16s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/827b5801c66f4691b975c72aee26f941 : SUCCESS in 19m 35s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/49526d826cb7435aa83933b963f1dfa3 : SUCCESS in 17m 08s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1534efa6a4d4940a0903215bb9db3fb : SUCCESS in 14m 50s","accounts_in_message":[],"_revision_number":10},{"id":"6d2336a65c2d3bb84248db7559ddbdc6ed5e6204","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-22 16:25:15.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"7124e3e1246c3a30a451d1a3cc9be2827756246b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-23 08:27:35.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"f39b5c350d7f694e3dcd4e5388de8f499f15a307","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 08:28:20.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"bae8197da50f19fdc66fc256b16288fa2044a741","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-23 09:25:23.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"bf0d4719eb82c711af34fb400665e78b2e889cb2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-23 10:06:39.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"51a0df6de18ef3991dac20c73e0567abb4aa9088","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 11:26:52.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7b441bf31794e0faf29f2d6d480168f : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e45590abbc24d40b0876f51fdd39bf5 : SUCCESS in 9m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d122c7f6e482478e8b907f3d28daddb7 : SUCCESS in 7m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3175ea4d79ad49a19b8d2f0b2dca6918 : SUCCESS in 9m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abb646462ae34b7a831d864e6c3467ec : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e69332048bc04f54a75864da9144efe2 : SUCCESS in 10m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c630923a1ed4c8b99c2e351ba394a59 : SUCCESS in 5m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/94c4550cde6347f7b90bf027d5a14c53 : SUCCESS in 7m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/578c095f88dc4827b39e7698b599c4d3 : SUCCESS in 13m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5799c8e4dc9e41c49b8ac145eab7c581 : SUCCESS in 32m 26s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/276fc290540c4de9b94d67984d65696b : SUCCESS in 1h 07m 53s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/0f2e57b9e7d7429ab67b1df380404064 : SUCCESS in 1h 06m 58s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a427e699cf1d493da1fa33e2e05935e5 : SUCCESS in 59m 45s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c1fc557038694615b834c581137ae94f : SUCCESS in 1h 07m 00s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/27cfc7fa8c504f569c450a719c2be348 : SUCCESS in 25m 15s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/999136ada5a94fb68e14a3202775d593 : SUCCESS in 22m 53s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81c134745fd849c1b5139ebd2e758591 : SUCCESS in 15m 46s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9ddcb3b21a4f4940869e143fef13b95b : SUCCESS in 1h 59m 15s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/271703780aee4cfd979f619ebe19fb56 : SUCCESS in 1h 12m 38s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a200a5e4fbab4007a309c5e80d713eaa : SUCCESS in 40m 28s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/f1585cfbe44c49aa85a852e1e9602cdf : SUCCESS in 32m 58s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e83b0701699f4b049c118e41b346e7f0 : SUCCESS in 29m 05s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4f82f47150294c8397af56ee4aca6ca6 : SUCCESS in 43m 55s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee438825da1a4c48a228d7fbc88981d0 : SUCCESS in 36m 14s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d85b7894c90746e59875c1c19373d09e : SUCCESS in 18m 44s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/37cdb000a2764d67a68d5051f8c22590 : SUCCESS in 17m 26s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6946ff11693d4c28b1029dac86448e08 : SUCCESS in 14m 00s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/40cdd0c78d754425b832a66266434e4e : SUCCESS in 21m 49s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/bf8774b3734745e2b89de6e47d8a3b98 : SUCCESS in 19m 49s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/796c3bc318b54ac0bc2039f6c0141a0a : SUCCESS in 17m 14s","accounts_in_message":[],"_revision_number":11},{"id":"56cb822bc6e396c2c0b4af9239c24cfaa36535ce","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-23 11:27:06.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"f242cd0513ef430c1c2d57e74ccd9941be7dc280","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 11:27:55.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"86279e4b970434f845fb2d9806e03f2a0eb43bee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 13:33:12.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a52244f18700479999c4efc8cdb1316b : SUCCESS in 8m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b60f4e8b9124724bf480c1f181fd4dc : SUCCESS in 9m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/628f043af6af4b78bb3a50b1dea73c3b : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd058c30038142bcba13c75f18726b9b : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29f56285d88a4ad0bc72093950832eea : SUCCESS in 6m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/95029e7b92ae41219525e51b7ddea529 : SUCCESS in 8m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6b0ed5dee0704a6a9e640481f9a5b922 : SUCCESS in 14m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/167042d83be14d4488421f7837586d22 : SUCCESS in 34m 28s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ef4f27de111249c0a92206ad06eff247 : SUCCESS in 1h 14m 27s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/73de2d8c992c4b4389b9d5768b308a94 : SUCCESS in 1h 09m 14s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/501af1cacdb14ed6ba476e961b27b194 : SUCCESS in 1h 40m 17s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2c9dea3ac4014350b780fc3da6415622 : SUCCESS in 1h 15m 38s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/cbc45746d83f4642b271ded57a2465ca : SUCCESS in 26m 34s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/de96e55298fc47279041535e1e8cb2ec : SUCCESS in 25m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e871b1e4a9ae478289c9ca202b72cc79 : SUCCESS in 16m 59s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/369538ab4275428193718075a86f2221 : SUCCESS in 1h 50m 08s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c991b5ccefc94bd0854e5dc28b93861d : SUCCESS in 1h 25m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c627b5a01ee042e8acb1e39a0c6753f9 : SUCCESS in 42m 51s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/3b61817137bd4a7ca57fed15a3db4e9b : SUCCESS in 32m 30s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4445002cd73d4604a38072fff0930f45 : SUCCESS in 31m 07s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/26f113af4e174c9f9b4b5ad405b26366 : SUCCESS in 40m 18s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d19be8a4d8d34b1fbd9829a7168a2e4a : SUCCESS in 31m 49s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2710992881c44731b8407dad1471dbf5 : SUCCESS in 20m 15s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/146f8ecf7954406086fce7f6a4308e88 : SUCCESS in 16m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/458f26bea9534c779ae681108ab9de14 : SUCCESS in 13m 59s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/b577582c27d14cc499e9cb19b8fbed1c : SUCCESS in 21m 06s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/917b5a4c2df542d98cea9c3f8a8d9921 : SUCCESS in 20m 01s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc4590b44bf34d549f96d5defa162b93 : SUCCESS in 15m 25s","accounts_in_message":[],"_revision_number":11},{"id":"24f6baf9888c72274303b9ea0eefe15237a2126f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 13:33:12.000000000","message":"Change has been successfully merged\n\n10 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":11},{"id":"7bb32c2c795bd4a8ab5dab8bff09bfaf583b6643","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 13:41:27.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/135d4433b6fe40d39c6ff96a71473f81 : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcd909c167c940dfbdf37699ff546102 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"c083073ca26ea9e06a4c95da6cca81eebb155f73","revisions":{"dbaca9a7f5f9333c25d43786738d1d42ee10ce1e":{"kind":"REWORK","_number":1,"created":"2022-03-21 11:24:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/834309/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/834309/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/834309/1"}}},"commit":{"parents":[{"commit":"55850d24968048d4d2a0ec134cf352fe31ce228a","subject":"Merge \"CI: Don\u0027t download Cirros or IPA in seed jobs\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55850d24968048d4d2a0ec134cf352fe31ce228a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-21 11:24:13.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n(cherry picked from commit 8e55ea08a474d3138fed160129acb8322f9eeae9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dbaca9a7f5f9333c25d43786738d1d42ee10ce1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dbaca9a7f5f9333c25d43786738d1d42ee10ce1e"}]},"branch":"refs/heads/stable/xena"},"84a97a83a3b6c5b582226a58f6abc7c4322e261e":{"kind":"REWORK","_number":2,"created":"2022-03-22 10:51:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/834309/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/834309/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/834309/2"}}},"commit":{"parents":[{"commit":"55850d24968048d4d2a0ec134cf352fe31ce228a","subject":"Merge \"CI: Don\u0027t download Cirros or IPA in seed jobs\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55850d24968048d4d2a0ec134cf352fe31ce228a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-22 10:51:35.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n(cherry picked from commit 8e55ea08a474d3138fed160129acb8322f9eeae9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/84a97a83a3b6c5b582226a58f6abc7c4322e261e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/84a97a83a3b6c5b582226a58f6abc7c4322e261e"}]},"branch":"refs/heads/stable/xena"},"3be95bd0d4d8f7e042afeefdf22463f3f3a2d844":{"kind":"REWORK","_number":3,"created":"2022-03-22 10:52:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/834309/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/834309/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/834309/3"}}},"commit":{"parents":[{"commit":"55850d24968048d4d2a0ec134cf352fe31ce228a","subject":"Merge \"CI: Don\u0027t download Cirros or IPA in seed jobs\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55850d24968048d4d2a0ec134cf352fe31ce228a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-22 10:52:05.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n(cherry picked from commit 8e55ea08a474d3138fed160129acb8322f9eeae9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3be95bd0d4d8f7e042afeefdf22463f3f3a2d844"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3be95bd0d4d8f7e042afeefdf22463f3f3a2d844"}]},"branch":"refs/heads/stable/xena"},"65a74159aaf36e0bc931447268723c74b3ebe9e9":{"kind":"REWORK","_number":4,"created":"2022-03-22 10:52:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/834309/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/834309/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/834309/4"}}},"commit":{"parents":[{"commit":"55850d24968048d4d2a0ec134cf352fe31ce228a","subject":"Merge \"CI: Don\u0027t download Cirros or IPA in seed jobs\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55850d24968048d4d2a0ec134cf352fe31ce228a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-22 10:52:42.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n(cherry picked from commit 8e55ea08a474d3138fed160129acb8322f9eeae9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65a74159aaf36e0bc931447268723c74b3ebe9e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65a74159aaf36e0bc931447268723c74b3ebe9e9"}]},"branch":"refs/heads/stable/xena"},"d7688ebbc3ce03bb183db2b6bf74820ec149950e":{"kind":"REWORK","_number":5,"created":"2022-03-22 10:54:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/834309/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/834309/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/834309/5"}}},"commit":{"parents":[{"commit":"55850d24968048d4d2a0ec134cf352fe31ce228a","subject":"Merge \"CI: Don\u0027t download Cirros or IPA in seed jobs\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55850d24968048d4d2a0ec134cf352fe31ce228a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-22 10:54:21.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n(cherry picked from commit 8e55ea08a474d3138fed160129acb8322f9eeae9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7688ebbc3ce03bb183db2b6bf74820ec149950e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7688ebbc3ce03bb183db2b6bf74820ec149950e"}]},"branch":"refs/heads/stable/xena"},"610d656242f77e60b2811960c9a6f8e38d8de950":{"kind":"REWORK","_number":6,"created":"2022-03-22 10:55:06.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/834309/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/834309/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/834309/6"}}},"commit":{"parents":[{"commit":"55850d24968048d4d2a0ec134cf352fe31ce228a","subject":"Merge \"CI: Don\u0027t download Cirros or IPA in seed jobs\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55850d24968048d4d2a0ec134cf352fe31ce228a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-22 10:55:06.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n(cherry picked from commit 8e55ea08a474d3138fed160129acb8322f9eeae9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/610d656242f77e60b2811960c9a6f8e38d8de950"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/610d656242f77e60b2811960c9a6f8e38d8de950"}]},"branch":"refs/heads/stable/xena"},"905b4d7ed44223303210201965a99296ba60b228":{"kind":"REWORK","_number":7,"created":"2022-03-22 10:55:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/834309/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/834309/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/834309/7"}}},"commit":{"parents":[{"commit":"55850d24968048d4d2a0ec134cf352fe31ce228a","subject":"Merge \"CI: Don\u0027t download Cirros or IPA in seed jobs\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55850d24968048d4d2a0ec134cf352fe31ce228a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-22 10:55:51.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n(cherry picked from commit 8e55ea08a474d3138fed160129acb8322f9eeae9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/905b4d7ed44223303210201965a99296ba60b228"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/905b4d7ed44223303210201965a99296ba60b228"}]},"branch":"refs/heads/stable/xena"},"da958005a4dc05a8eec8d3ccad1b709c8eef9782":{"kind":"REWORK","_number":8,"created":"2022-03-22 11:41:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/834309/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/834309/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/834309/8"}}},"commit":{"parents":[{"commit":"55850d24968048d4d2a0ec134cf352fe31ce228a","subject":"Merge \"CI: Don\u0027t download Cirros or IPA in seed jobs\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55850d24968048d4d2a0ec134cf352fe31ce228a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-22 11:41:39.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n(cherry picked from commit 8e55ea08a474d3138fed160129acb8322f9eeae9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da958005a4dc05a8eec8d3ccad1b709c8eef9782"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da958005a4dc05a8eec8d3ccad1b709c8eef9782"}]},"branch":"refs/heads/stable/xena"},"19f359c7dc8ee775a026c7a425104df79362173b":{"kind":"REWORK","_number":9,"created":"2022-03-22 11:43:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/834309/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/834309/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/834309/9"}}},"commit":{"parents":[{"commit":"55850d24968048d4d2a0ec134cf352fe31ce228a","subject":"Merge \"CI: Don\u0027t download Cirros or IPA in seed jobs\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55850d24968048d4d2a0ec134cf352fe31ce228a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-22 11:43:04.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n(cherry picked from commit 8e55ea08a474d3138fed160129acb8322f9eeae9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/19f359c7dc8ee775a026c7a425104df79362173b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/19f359c7dc8ee775a026c7a425104df79362173b"}]},"branch":"refs/heads/stable/xena"},"81acf54fb7437dc48327fec3b18f1efd747055ee":{"kind":"REWORK","_number":10,"created":"2022-03-22 13:47:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/834309/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/834309/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/834309/10"}}},"commit":{"parents":[{"commit":"55850d24968048d4d2a0ec134cf352fe31ce228a","subject":"Merge \"CI: Don\u0027t download Cirros or IPA in seed jobs\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55850d24968048d4d2a0ec134cf352fe31ce228a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-22 13:17:05.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n(cherry picked from commit 8e55ea08a474d3138fed160129acb8322f9eeae9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/81acf54fb7437dc48327fec3b18f1efd747055ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/81acf54fb7437dc48327fec3b18f1efd747055ee"}]},"branch":"refs/heads/stable/xena"},"c083073ca26ea9e06a4c95da6cca81eebb155f73":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2022-03-23 09:25:23.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/09/834309/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/834309/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/834309/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/834309/11"}}},"commit":{"parents":[{"commit":"efa8209c97b15f507a5b9c67b3833d3a65bf61f3","subject":"CI: pin pytest-metadata\u003c2 for molecule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/efa8209c97b15f507a5b9c67b3833d3a65bf61f3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-03-03 10:27:11.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-23 09:25:23.000000000","tz":0},"subject":"Add support for Rocky Linux 8","message":"Add support for Rocky Linux 8\n\nChange-Id: If7d6e58b19f98ccb7cc4c209e458cb6f4f4765ad\n(cherry picked from commit 8e55ea08a474d3138fed160129acb8322f9eeae9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c083073ca26ea9e06a4c95da6cca81eebb155f73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c083073ca26ea9e06a4c95da6cca81eebb155f73"}]},"branch":"refs/heads/stable/xena","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
