)]}'
{"id":"openstack%2Fkayobe~834357","triplet_id":"openstack%2Fkayobe~master~Id7e4bd7f199ad9dcb21d5db082e7a187cb310df9","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-19 07:55:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id7e4bd7f199ad9dcb21d5db082e7a187cb310df9","subject":"ironic: switch to kernel_append_params","status":"MERGED","created":"2022-03-18 20:42:37.000000000","updated":"2022-03-19 07:56:42.000000000","submitted":"2022-03-19 07:55:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834357","meta_rev_id":"1298c41043e410a52bb11a3b4d003e37dfacb45a","_number":834357,"virtual_id_number":834357,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-19 07:55:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-03-19 06:03:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-18 21:49:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-03-19 06:03:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-18 21:49:47.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-03-18 22:27:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-19 06:03:48.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"a10e1c36790315c745e0824e325f217f0c054052","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-18 20:42:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"818881395fdc91e94a203490133f09216e0c9554","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-18 21:49:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2a9a42c3f5d818bcbf4f29a27231e5844469e1c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 22:27:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2dc9d480ff6f457dbb0dfe662dee99a1 : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc0f4e190fd944bd8b4d1761e697e832 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81cca3702cb840ffaccf582e3109bb78 : SUCCESS in 10m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06819931874443339e3c5be02d7b8879 : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74d9df60d2994f1497b7bf150eccef29 : SUCCESS in 13m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8b567010cc60495ca70c64df5d116b64 : SUCCESS in 8m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cd86f64ee54446259ebe46093574c655 : SUCCESS in 13m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/46673aa4fd7945c999c5d5536f082711 : SUCCESS in 26m 48s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f3df28e533f44ea5836697490202ecd3 : FAILURE in 55m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1cf3f711a444423b847902921d78c3c : SUCCESS in 54m 08s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4908d76240e74ed1a89e54bf92310173 : SUCCESS in 1h 05m 50s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/103e135268fa44208a07fa02912334c7 : SUCCESS in 23m 42s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/564a9b642d5744b9bbffff93ca48a1c9 : SUCCESS in 16m 44s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f92f7e3090b747bea18c8d93ec8086c4 : SUCCESS in 1h 41m 36s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee0e162e48214e17b0098b884b27746e : SUCCESS in 1h 00m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/918f4aa43b3e419193d784eb11fb02e2 : SUCCESS in 42m 09s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c72a90a566144ae1931c6866d0a50ddf : SUCCESS in 34m 07s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d854163291b34278ad843bbe3ccd8cf2 : SUCCESS in 33m 11s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b30836c3c1db4defbebdaa1f0cb6597c : SUCCESS in 32m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/435a17db0ed74c0dbc5697f07f4378ff : SUCCESS in 15m 04s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f75a4f3793c41a6a74ad1d3084c4ae8 : SUCCESS in 12m 35s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/0e77d59e18a74786b32c31b44288915e : SUCCESS in 19m 27s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5aa7c7093334480d9dab78aab58d2469 : SUCCESS in 13m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e4cdcfd1ba65a96f44ffe30407859f227f705f55","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-18 22:38:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf54d8c37cf77f559404895193495018c5a74aec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 00:32:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/926a688f2a984c80a5b9653a8ddaf96f : SUCCESS in 9m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8662725d8d8433d837801f44cb6c77f : SUCCESS in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b940f19dc34447868d72fe9609628e63 : SUCCESS in 8m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f0e9b764bd443319097f6ba8d22eaf2 : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24b5edea2c0f4324bd0bf912564955d2 : SUCCESS in 11m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e2ef22efc772464896ee680bf63c46a2 : SUCCESS in 8m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5a73c9dd6ff14ab2aef8eb0894391ffd : SUCCESS in 16m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dadaa2d1a11a48f88f11a02f71fad8f8 : SUCCESS in 33m 34s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7ab78b578e0a43ca8ecd6eaad9b87000 : SUCCESS in 1h 08m 56s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a62e02a2460a48a6a71c9b0391deb163 : SUCCESS in 56m 52s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/28a1b2836a604af4a30e4c93718b9c70 : SUCCESS in 1h 11m 08s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/45d1ff56eb964774ba2e7cab742e9c91 : SUCCESS in 26m 58s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3c4075e7b994dd993791514a22d7f88 : SUCCESS in 19m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/09b5bdf60b954cb9a35174e7a2954ab6 : SUCCESS in 1h 51m 27s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77f8227305334b83bcbc93795e4c1978 : SUCCESS in 1h 22m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/87dc9a9989e141cd8151be4d9fceecca : SUCCESS in 53m 59s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a5691ff77ac4394b6620a04d3339d56 : SUCCESS in 35m 02s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6eab1fe691ee4d2b9fb31c8bbe0808d8 : SUCCESS in 43m 13s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df4a27ed46d04f1db3ed65f9a2f20919 : SUCCESS in 32m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/96e63fdade4c47e58afd14b70df6c1b9 : SUCCESS in 18m 27s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86bdeb4e9fd74fb19c032e87c6aed82d : SUCCESS in 14m 37s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/b4e13b0d81704fd98ba0d5869026dec6 : SUCCESS in 23m 40s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8205ee4298c48fe810c6284d9a886fd : SUCCESS in 16m 25s","accounts_in_message":[],"_revision_number":1},{"id":"1af544b9e47165bf0cd14b089c5074715d6c3918","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-19 06:03:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1977bfcd80ba673a936f046f901dd1b209f03d82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 06:04:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d63e7e0b3391944776ed8f29ffe368d02fb2a196","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 07:55:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37582e66442340299b0ca3b1a25c79b7 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d568af463d1f4e24b55bdecd506689e4 : SUCCESS in 9m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b10646c35a4149508f0c1489def5e5a5 : SUCCESS in 8m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51bcf9bf0ea545ccb687d46a687505eb : SUCCESS in 10m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/533d9a55117a4136a3e05a1802815f12 : SUCCESS in 6m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8ad49b02970743c890fc7819b48f8e20 : SUCCESS in 13m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/264fa56d7b314e198a4137d6dc4f83e1 : SUCCESS in 27m 13s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a0aa7829224845db806761201935b42c : SUCCESS in 1h 08m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35c171be49fd4e8ab00636ed05f9647e : SUCCESS in 45m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/20cc896237f04d0b9c2e7e724299628b : SUCCESS in 1h 03m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b5f8c74cbd9b4285acb4286b6b124a1e : SUCCESS in 25m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bce9f4229c164d89be03b196eae432b6 : SUCCESS in 11m 46s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bf88bc23389d4ebeab83fc8a6fbbc234 : SUCCESS in 1h 50m 12s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/820607554cef4ad4b75555ffef5f757a : SUCCESS in 1h 18m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a9b4f6d87e9b4245a1dc344335aae2b1 : SUCCESS in 47m 27s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d507baa7468e4cf89a93bc5eb310b705 : SUCCESS in 34m 08s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d8afffda6b5b4002a2a503027bd6e79b : SUCCESS in 39m 48s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f1c9cd7bb2d49e3bbde51d9fcb2ae9e : SUCCESS in 31m 32s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ddd29ce44b38459882a6b99e8dca0e74 : SUCCESS in 17m 57s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47f56c197fbc43f5a731c1404cbd6798 : SUCCESS in 13m 03s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/2290646acaf641709d2b37eb14cf4d85 : SUCCESS in 18m 38s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1c3e7cfc6c0401d8fde4a71b7e5c738 : SUCCESS in 15m 25s","accounts_in_message":[],"_revision_number":1},{"id":"e731a7525df84178a3ac3f0685581806bdda7747","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 07:55:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1298c41043e410a52bb11a3b4d003e37dfacb45a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 07:56:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1196464fef44462183a47d52e5ff3b76 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"994af643e67b1efc9368e38997aaa7c412eabf3a","revisions":{"994af643e67b1efc9368e38997aaa7c412eabf3a":{"kind":"REWORK","_number":1,"created":"2022-03-18 20:42:37.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/57/834357/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/57/834357/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/834357/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/834357/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/834357/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/57/834357/1"}}},"commit":{"parents":[{"commit":"1de4f2a4a3f6ef16399d812a544c6da622609e24","subject":"Skip IP address allocation and configuration if needed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1de4f2a4a3f6ef16399d812a544c6da622609e24"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-18 20:41:23.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-18 20:42:25.000000000","tz":0},"subject":"ironic: switch to kernel_append_params","message":"ironic: switch to kernel_append_params\n\nSince I4fb42d376636dc363cd86950ed37de4a3d28df73, kolla-ansible sets\nkernel_append_params in ironic.conf, instead of the deprecated\npxe_append_params. Make the same change in the Kayobe ironic.conf\ntemplate, to avoid getting shadowed by the kolla-ansible default.\n\nThis also fixes the overcloud TLS job, which started failing because it\nlost the ipa-insecure kernel option, making the Ironic API callback\ninvalid.\n\nChange-Id: Id7e4bd7f199ad9dcb21d5db082e7a187cb310df9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/994af643e67b1efc9368e38997aaa7c412eabf3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/994af643e67b1efc9368e38997aaa7c412eabf3a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
