)]}'
{"id":"openstack%2Fkayobe~834458","triplet_id":"openstack%2Fkayobe~master~I174623bbcfeb35c59cd4c155f36078800d4c35cd","project":"openstack/kayobe","branch":"master","topic":"yoga-feature-flags","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-03-31 15:46:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I174623bbcfeb35c59cd4c155f36078800d4c35cd","subject":"[release] Sync Kolla Ansible feature flags for Yoga","status":"MERGED","created":"2022-03-21 12:20:31.000000000","updated":"2022-03-31 15:48:24.000000000","submitted":"2022-03-31 15:46:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834458-yoga-feature-flags","meta_rev_id":"a7805cc7a4c18566244854fbd4f1ae59cfbdf074","_number":834458,"virtual_id_number":834458,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-31 15:46:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-03-31 14:07:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-03-31 08:53:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-03-31 14:07:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-21 14:21:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-23 08:24:20.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-03-31 14:07:34.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"e4cafb204f68521b6e9e9f01d725e900c46198db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-21 12:20:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b06759bc19f6bfe467f13ed1d7d03dfddaddfe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 14:21:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ba8d168048f470ca0b5af3b96131e6f : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98b096563e184e0fbea840357fd040a4 : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2c8caec37044c3ea11e69e337a10fe4 : SUCCESS in 8m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3eb2bcca8b44d439062f4be2284148f : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df6cc2669a6145e2a1f0815132ed6f05 : SUCCESS in 11m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1829ae674a3748fcbf261d669ad4c66c : SUCCESS in 8m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3be67da1e1544bc297168b7ba25fe84e : SUCCESS in 15m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fa5ef4402795458cb25153ff2f4539ce : SUCCESS in 34m 32s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0ab1aa92f5f5496ba48687e36ba7de2b : SUCCESS in 1h 15m 50s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/f0c0e34178974dc291343932c7045c94 : SUCCESS in 1h 06m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb5cf0c53dd440e4a6d53eecd11ee4e5 : SUCCESS in 56m 10s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/55a2ca906e594a129d9e0651caf792a3 : SUCCESS in 1h 11m 11s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b9da56d8f15e4c7f8847fc1637c5ce68 : SUCCESS in 27m 48s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/969bae485ee3477bb001d96416fbd31e : SUCCESS in 22m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/548cefe3f3ad4d6c9f8eae887ad0aec0 : SUCCESS in 17m 42s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/52f71207a00c4a1b9603b4531b43d12d : SUCCESS in 1h 59m 15s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e53fb917a0324806aab0822cf4a9318d : SUCCESS in 1h 20m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9ad75767f20244d79d837598190a58de : SUCCESS in 52m 52s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/4b1223dab67d4441a456cc1ba6972aac : SUCCESS in 43m 50s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0871d18580047108182b845ec9b601d : SUCCESS in 39m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5b6f97a845f74ff9a4a671f6766948af : SUCCESS in 41m 31s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64e1bf50354e49fb84aa068dfd7e0cfe : SUCCESS in 34m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/650b11cfd9f341c3ba8b15cb9f601098 : SUCCESS in 24m 48s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/b683b51cd657471fb50dea99843c7e3b : SUCCESS in 17m 22s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c60828ee0e58449eb86c25fc40cca6ee : SUCCESS in 14m 05s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/4c89c62d31794629aa30936b3a3d4296 : SUCCESS in 20m 29s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/43efa139d442428da574f87fb88347c9 : SUCCESS in 18m 54s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17e99affb323438bbb06437cf6bf59fc : SUCCESS in 15m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b923ad9bc101bd59b4c8f72b363d98e1a6a6339d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-23 08:24:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"49c2c5c2a997f68cac527fed9ee15f9b5bf2c80a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-31 05:30:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6154ea20cb8a4d3232ad2de084a2e9fb62e4b7b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 07:22:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/181de5f9e3c04adbb97e745fc57243c4 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2714c0494fc40c78d1a873615cdae1d : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf1030e8ad854c78b3163981046c345e : SUCCESS in 6m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58c755be90f345be95719c2533d89f5d : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/663b88e798ff4c4da8497457daf8870e : SUCCESS in 9m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b3d07b88ff644fa19a86e54dedfd4467 : SUCCESS in 5m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7bd532fc321249e59fe25946e1258fb2 : SUCCESS in 14m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8f0d0f3180f5490a840a3e9acff5bfd0 : SUCCESS in 34m 05s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3d88b24e6509420d87bc5cefecb2ac21 : SUCCESS in 1h 10m 38s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/d833751e1fbb4767b1bda808b596ad72 : SUCCESS in 1h 13m 40s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8a13abf044b4acd9b02fbecbc9be79f : SUCCESS in 56m 49s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/df36844e49574bd496ea28a90dbd662c : SUCCESS in 1h 26m 46s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d785c013e2a54d2db9891d4c5e819a77 : SUCCESS in 24m 12s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/f225677669294afe9314d9f7090eac98 : SUCCESS in 21m 56s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/089e39ec2b54405f8b75666a0ea9c611 : SUCCESS in 14m 28s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/154b5d0e30dd46918c6839ee073e9c23 : SUCCESS in 1h 46m 55s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31449087fc664ca8ac1ed95495af7d91 : SUCCESS in 1h 30m 17s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bb3f4d340d74448c8f4a75d25d610a5d : SUCCESS in 24m 52s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/9fa32ee96572476ab2548e4a6c5f43de : SUCCESS in 23m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fec177e64be49548981c7a00d83bc80 : SUCCESS in 17m 21s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2ec09a4015f34b55bb66b1483438a3a0 : SUCCESS in 39m 02s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8660fec578f47e9826e68d3c973ea35 : SUCCESS in 30m 20s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/520d7f08011b46e5b4071cd3290fb9ab : SUCCESS in 16m 40s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/045a164f956b4ab3bd3aa95ba8135f67 : SUCCESS in 14m 36s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c32a21ac6944bff806ebe5c318d0b47 : SUCCESS in 13m 20s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/73b5ff199990464180ec3239823d6a27 : SUCCESS in 18m 48s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6878468b40804697acee18c150f588b6 : SUCCESS in 14m 54s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfec379271ea4129862f6cbc26609437 : SUCCESS in 14m 00s","accounts_in_message":[],"_revision_number":2},{"id":"553c81c892df9043b69c5274d9b18b092d0e2187","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-31 08:53:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"433cd5a5979f428fbd91edb6dc478758058cfb04","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-31 14:07:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9a5292ec6b404ef505f4b49cd1aa21d34a04ac7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 14:08:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8ef7ea3e4a50e47d8f380002a33861e99eb7ef58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 15:46:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bb67233d127435d8c6bc8bd599c580a : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/295eed10c5644a51925f27abea3ec2ca : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5409eb99e904983962576eac3d6d14e : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7075f3bc18114c3b9837b982e34d9172 : SUCCESS in 10m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1ea3eadeb61e49cc88356f6257580bb4 : SUCCESS in 6m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/34134b1c1cfe4a919433b5879a398ac8 : SUCCESS in 11m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ad44c5ad82424b2cbf899e5e8c730b1b : SUCCESS in 29m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/986846998e184546b94f0989a9358983 : SUCCESS in 1h 06m 44s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/23943ffec6f244408586632d89cc109a : SUCCESS in 1h 05m 08s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2779f823a4d41e2b156be335bf2ca39 : SUCCESS in 53m 38s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5564ca5f581c4dd886d05268eb5a2550 : SUCCESS in 1h 05m 39s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7d2dc133eb63405a8f809e8b052397a2 : SUCCESS in 25m 33s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/e0d5889705c64396ae9e336df3e6b58b : SUCCESS in 21m 34s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/413c0a00c91a45a5a320210ec905e733 : SUCCESS in 14m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/72db6c13785e464799ec119ba83bc9d4 : SUCCESS in 1h 35m 32s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6147d0f63ae464ea3f8f9da78ec8500 : SUCCESS in 1h 15m 43s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/85304b9f93da49098096709a0f0691bd : SUCCESS in 25m 29s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/ba670fd642ba448098d317ee263cb405 : SUCCESS in 23m 44s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f61b862172034202af48cecbbdaa01d0 : SUCCESS in 19m 58s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d4dc23769ea94b00b08d0cbe456ebbba : SUCCESS in 42m 22s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2243d9a92f5548c38022770e7ffae6fe : SUCCESS in 31m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/aaf20c6394254f03a06aad358c238b8c : SUCCESS in 19m 40s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c9e74aabfe8d434b85a02b40d08d75fa : SUCCESS in 15m 32s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/024e78cd135141c290c67e3bd006dcc4 : SUCCESS in 14m 07s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e85d07cd84224f828f03bb9afa5ac9bd : SUCCESS in 18m 42s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e380e4b133c0469e87efa83cd9d34a49 : SUCCESS in 17m 39s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/380e08f6077142d7a9ba9399093dfc8c : SUCCESS in 13m 12s","accounts_in_message":[],"_revision_number":2},{"id":"0a88650f14e82d854453ed07d7e3e63a391addde","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 15:46:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a7805cc7a4c18566244854fbd4f1ae59cfbdf074","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 15:48:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aad7b1761b0437cba6ced577df5d343 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c2186b1bd3da12a04a298d48e5bd442899440113","revisions":{"523f71ad2e12c0087e052e89366e3d4be071a9de":{"kind":"REWORK","_number":1,"created":"2022-03-21 12:20:31.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/58/834458/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/58/834458/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/834458/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/834458/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/834458/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/58/834458/1"}}},"commit":{"parents":[{"commit":"4616c87010fe87caea302dc3c67d0b9b0b0f287d","subject":"Merge \"Add support for Rocky Linux 8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4616c87010fe87caea302dc3c67d0b9b0b0f287d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-21 12:15:54.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-21 12:15:54.000000000","tz":60},"subject":"[release] Sync Kolla Ansible feature flags for Yoga","message":"[release] Sync Kolla Ansible feature flags for Yoga\n\nChange-Id: I174623bbcfeb35c59cd4c155f36078800d4c35cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/523f71ad2e12c0087e052e89366e3d4be071a9de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/523f71ad2e12c0087e052e89366e3d4be071a9de"}]},"branch":"refs/heads/master"},"c2186b1bd3da12a04a298d48e5bd442899440113":{"kind":"REWORK","_number":2,"created":"2022-03-31 05:30:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/58/834458/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/58/834458/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/834458/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/834458/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/834458/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/58/834458/2"}}},"commit":{"parents":[{"commit":"91f2d9da2daa7f9e51c1cb734f4e6d8488dbce0b","subject":"Merge \"Fix custom config idempotence\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/91f2d9da2daa7f9e51c1cb734f4e6d8488dbce0b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-31 05:29:35.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-31 05:29:35.000000000","tz":120},"subject":"[release] Sync Kolla Ansible feature flags for Yoga","message":"[release] Sync Kolla Ansible feature flags for Yoga\n\nChange-Id: I174623bbcfeb35c59cd4c155f36078800d4c35cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2186b1bd3da12a04a298d48e5bd442899440113"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2186b1bd3da12a04a298d48e5bd442899440113"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
