)]}'
{"id":"openstack%2Fkayobe~834723","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I60cf89ce2f8c236b7febb92d437abf7e21bb7acd","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-03-23 09:52:51.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"I60cf89ce2f8c236b7febb92d437abf7e21bb7acd","subject":"CI: pin pytest-metadata\u003c2 for molecule","status":"MERGED","created":"2022-03-23 08:50:48.000000000","updated":"2022-03-23 12:51:46.000000000","submitted":"2022-03-23 12:50:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834723","cherry_pick_of_change":834790,"cherry_pick_of_patch_set":1,"meta_rev_id":"4e0b60ab6956f55f854e1e1db1c92fa080d64665","_number":834723,"virtual_id_number":834723,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-23 12:50:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2022-03-23 09:23:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2022-03-23 09:47:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-23 09:52:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-23 09:52:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-23 08:50:48.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-03-23 09:23:47.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2022-03-23 09:47:57.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-03-23 10:54:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e2f88d99545a6035787066657ade1d790bea976","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-23 08:50:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"872cc2350fccadcaad8a2b00c22886d50af6ac95","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-23 09:23:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"47b423143497081940fa4e6d4be5ec8d7bb52ea2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-23 09:47:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6e4e7ad43d3cd4e9e2300ab12525f394c64a0b73","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-23 09:52:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0ad770eddf0109366254705600630bd90e1c3131","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 10:54:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5bfadaa4d954e079013dda3c8b3118a : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67d2f3574f4945a9b3c0fb733078fe93 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f998edc6af44bc7869cdfce4ae76a1f : SUCCESS in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d508a82b9b84751b2cb3cfb16ded242 : SUCCESS in 8m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ecce82907e6444fbbc62f571b402ea1 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44134a5c7b974226beb66c6f38443bdb : SUCCESS in 10m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9a4acc6b1a5d4dd684ed4807f61ca5ae : SUCCESS in 7m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ce6a8a56f9fa41a3976c92be872329f8 : SUCCESS in 14m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cdc094a057154da8bf3565210ec93595 : SUCCESS in 32m 57s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2d9a79265ab64b3ca729d37069a4bf9b : SUCCESS in 1h 19m 05s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9df7f5cfe1884d1c931a0a92d8f072d6 : SUCCESS in 1h 02m 21s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8da4ce153c3f4ff3bb9b1ebab8f9d66e : SUCCESS in 1h 19m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/42e755c2c9fe4df4bdfed800f359f5bb : SUCCESS in 26m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff757276977c4af8bf2f9c5679fdb9cb : SUCCESS in 17m 06s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d326f7eb6a67430cac53f0e287d46eaf : SUCCESS in 1h 56m 51s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de8f4e4e23dc4b15a72653a98df078a1 : SUCCESS in 1h 24m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8398757ea9c040a4a4c44514537bc625 : SUCCESS in 43m 21s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e6612fa586d420b81d1c4c41e609b91 : SUCCESS in 31m 40s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/919f73fd05f64a9c8817eae47002f225 : SUCCESS in 46m 49s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cd34826793e4bc4a6fdbce7dfc09367 : SUCCESS in 31m 53s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6b58f3522a194917944defd51f98b962 : SUCCESS in 18m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01fc3ed6a4334fc2b8203726d52b96fc : SUCCESS in 13m 43s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/9e855805a01f40f998a5573fad851d15 : SUCCESS in 22m 04s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6014370cc1d4027954f782fe19171b8 : SUCCESS in 16m 26s","accounts_in_message":[],"_revision_number":1},{"id":"1441342f24e71a9eab889bd3e48ff3bdef1a82c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 10:55:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"17ad969c6d968140436525f516995057072079e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 12:50:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2729d99caa84fe19a83c8600e481d5c : SUCCESS in 8m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88990102bfd443a1a81f1c91975dbdf5 : SUCCESS in 8m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da4d0d03672d435e9b988226138cb99d : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2554f47c7557417f85387fc6abc8508a : SUCCESS in 12m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a3e99f008fec4bfd8aa35b5f1f30664f : SUCCESS in 8m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2b83c45d0f85466aa08cd6f031d183b1 : SUCCESS in 14m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bb029a17771e4de28f73531b7e041eb5 : SUCCESS in 34m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4e2bc246fb3c4a5f8c69c56c9e55c688 : SUCCESS in 1h 03m 24s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8579fd825464e9bba4962be423f8160 : SUCCESS in 59m 17s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/39ad69423d974e3d9cf508e8407e767e : SUCCESS in 1h 22m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bc42e0c5f5de47c5a195f33caab8579d : SUCCESS in 25m 56s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b593bd118434071bdead15cf4856005 : SUCCESS in 25m 24s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1d696f38419c435aa8f03dae2b2cdb96 : SUCCESS in 1h 50m 46s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0cd9f1fa8ed84c358e9e367632da9873 : SUCCESS in 1h 22m 05s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c49eddc4226a4fdba09fc0c67606c88a : SUCCESS in 41m 08s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1eca50b0dd5d4e118f13aad501209d58 : SUCCESS in 29m 42s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0caf0a54ad334910ae33ddab182afff9 : SUCCESS in 42m 09s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbbd2496a2b949378f50a3e914992aab : SUCCESS in 30m 09s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/331c70f06e354ec8a06ceb32086944fd : SUCCESS in 18m 10s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c205cf18732454f922231bab8c4069a : SUCCESS in 14m 28s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/026af2f40107405284ff3502ab772826 : SUCCESS in 21m 57s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c751265e827460dab5feadfb97e69d6 : SUCCESS in 19m 24s","accounts_in_message":[],"_revision_number":1},{"id":"594556fd02ab45c8f64f73c1249270ca9f4ceb2c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 12:50:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4e0b60ab6956f55f854e1e1db1c92fa080d64665","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 12:51:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49f7d2a828f64162aa2ac26d4760fe8e : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"efa8209c97b15f507a5b9c67b3833d3a65bf61f3","revisions":{"efa8209c97b15f507a5b9c67b3833d3a65bf61f3":{"kind":"REWORK","_number":1,"created":"2022-03-23 08:50:48.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/23/834723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/834723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/834723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/834723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/834723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/834723/1"}}},"commit":{"parents":[{"commit":"55850d24968048d4d2a0ec134cf352fe31ce228a","subject":"Merge \"CI: Don\u0027t download Cirros or IPA in seed jobs\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55850d24968048d4d2a0ec134cf352fe31ce228a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-22 20:09:21.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-23 08:50:48.000000000","tz":0},"subject":"CI: pin pytest-metadata\u003c2 for molecule","message":"CI: pin pytest-metadata\u003c2 for molecule\n\nThe release of pytest-metadata 2.0.0 appears to have broken the molecule tests.\n\n  AttributeError: \u0027Config\u0027 object has no attribute \u0027_metadata\u0027\n\nChange-Id: I60cf89ce2f8c236b7febb92d437abf7e21bb7acd\n(cherry picked from commit f3bca3e0f69982ce6c4eac72d1ab95f8732f4feb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/efa8209c97b15f507a5b9c67b3833d3a65bf61f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/efa8209c97b15f507a5b9c67b3833d3a65bf61f3"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
