)]}'
{"id":"openstack%2Fkayobe~834847","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I3f821937b0930a0ac9341178de7ae5123d82b957","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-03-24 11:34:56.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-24 09:36:23.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-03-24 08:37:55.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"I3f821937b0930a0ac9341178de7ae5123d82b957","subject":"Ubuntu: add support for Apt repository configuration","status":"MERGED","created":"2022-03-23 11:04:07.000000000","updated":"2022-03-24 11:36:14.000000000","submitted":"2022-03-24 11:34:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834847","meta_rev_id":"d2c8a775a7f33e6427a8b3b895d75b3a7a579dc5","_number":834847,"virtual_id_number":834847,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-24 11:34:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-03-24 09:36:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-24 08:37:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-03-24 09:37:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-23 11:04:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-03-23 13:00:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-23 13:52:33.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-03-23 13:52:33.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"8a4fb79d74956d8785a0be95c18386bd7db51439","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-23 11:04:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cae622bf3e6ec29f9e0eb08f9824e11f664615b5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-23 11:06:36.000000000","message":"Topic vtb/xena-mark-goddard-816364 removed","accounts_in_message":[],"_revision_number":1},{"id":"a18f9dc0ee74b0e7f428d7d9f24698dc884db731","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 13:00:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe5034bf451545d68b8596e099757ec9 : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75abf757b30e4a638f468c3dce71d2b9 : SUCCESS in 8m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0909fab6b744697988ca07078d3f3ba : SUCCESS in 8m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ce5fa75710342bdaba68d3f93237df0 : SUCCESS in 9m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e71a76977add43e4bacbc29d2f56dfa0 : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d79f1367fe984e4b804537ec776e43a2 : SUCCESS in 12m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b25a5b7b25b34a238fdba08e13465163 : SUCCESS in 5m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b95a85f1177047df8c497624971a048d : SUCCESS in 8m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/af807d796a344d23a937e2d28ae214df : SUCCESS in 14m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/173451a3ba274b65b3462cee2782be33 : SUCCESS in 35m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5c4660df8a464f51808dbac624df416e : SUCCESS in 1h 15m 40s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/a511cf9d4535406ab6184e148f352688 : SUCCESS in 1h 12m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5caf258c776a4703a2586afe69fc1ded : SUCCESS in 54m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ba45e0a3ce0d4365935c6587cae712b7 : SUCCESS in 1h 11m 24s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/803984b0ee2b42e49e49fb896bc98843 : SUCCESS in 25m 59s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/ddabd26346cb4a26867cd9ba7f33b5b8 : SUCCESS in 24m 45s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07b29d11a68749a9aa5d482548c857f5 : SUCCESS in 16m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7f3a801f19ec432bbde9cf8f2cef6e70 : SUCCESS in 1h 49m 02s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2945cf23852c4a309d7998f37026fb9c : SUCCESS in 1h 14m 41s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c6a51cd615c44127b15c333ade09a505 : SUCCESS in 37m 02s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/a769b777b5324fb0a146e17cdec42a64 : SUCCESS in 32m 43s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b671dd6064f4276bb846ba14c5f30d7 : SUCCESS in 27m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6214e31bcc52433fba7b438954b79842 : SUCCESS in 39m 49s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33d1506835f14a27b7caba95075959fb : SUCCESS in 28m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d5777bf3090e461a85ec9b114c5f7eae : SUCCESS in 18m 40s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/fff42f466eab4b71adbd674e8cf4982f : SUCCESS in 17m 26s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fba6912b80742f49fc41953e80d54a2 : SUCCESS in 14m 34s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/2edfe5c7a0da41d08529dcf91ff621ee : SUCCESS in 22m 21s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/20d1454336f642869d932903e156b00c : SUCCESS in 17m 38s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e548b135269f47c6b1290e703581712a : SUCCESS in 16m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8d956a9a69f5ef9520cc978a718881a2a5486b01","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-24 08:37:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"77106220ed7a9a6d7beee26f4d2df0c519232270","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-24 09:36:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"59c09ac0a88ff5172ce9b144099a5a273500446e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-24 09:37:19.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"93c34bba8d024a652a2e56385fc320122e8a773b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 09:37:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9309d18f9d9aa88472520201c8165a0aaaee561e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 11:34:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98de6f2fb0af493f8b4524ec83790c9f : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/703721ea2eb14b0eac1cbaa296ed692c : SUCCESS in 8m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ecc491a76434d7d856b3fa6a6989c23 : SUCCESS in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1b18fbe71df4f6f8035735ef3eb9fe2 : SUCCESS in 11m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40e210dfe9fd4b14af6209ddbfc3c283 : SUCCESS in 5m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/865f1627a6804758ac875556405e11f2 : SUCCESS in 7m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/68e1d70cc1884f198e4d66b239b89bb9 : SUCCESS in 15m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a90202e1fe6d48f7bd2d0219fbbb62dc : SUCCESS in 35m 08s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f8ae3ba859874e488204287a54e872b8 : SUCCESS in 1h 12m 38s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/1a279bb6e57c4ad59ef8b1377f3980a1 : SUCCESS in 1h 14m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ad23bb96c0c460e9a199a09e1ae564a : SUCCESS in 1h 19m 05s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bf4b4e7efb8f4f7e95d8472da17975fa : SUCCESS in 1h 16m 42s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a78f1ecc1e114aa6aac11c69a920ab0b : SUCCESS in 25m 09s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/96a48bc1d582425a88e3660abf2f9798 : SUCCESS in 24m 00s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e45320ff2b4d4b649eb2a5d2cdc0f4be : SUCCESS in 17m 02s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bd65f5125ae144f7b6f0456260166145 : SUCCESS in 1h 56m 05s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/378b8807a6d4400284c8262ea3b6e988 : SUCCESS in 1h 12m 09s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9e7a7cd008fd436594db06a492df72f5 : SUCCESS in 37m 49s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/e5022a837a964511b8265dd437a33cc2 : SUCCESS in 35m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b4148e7b0144b0d9d8b8c41aa4d3545 : SUCCESS in 33m 41s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0329134beeae49178befcea28fd5aa62 : SUCCESS in 42m 10s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5740b87cd29348b091a937e3a1d060e7 : SUCCESS in 27m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/db3c2c2840bb454a8e881042fef3988a : SUCCESS in 17m 58s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d6def2514dbf4880b924912c50a51d3f : SUCCESS in 16m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2aab269f836c4330acd4a3db65fb8357 : SUCCESS in 12m 45s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/00e3dd7fad634055a1ad335ac507f1c8 : SUCCESS in 20m 47s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/869bb0c515d4431dbd23ec49bc29e849 : SUCCESS in 17m 39s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e28a2ca75dd74b1f93b30b5fb9eed3ff : SUCCESS in 15m 03s","accounts_in_message":[],"_revision_number":1},{"id":"6974b1e81b55d267efcf0735c2f39f41cff7a25a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 11:34:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d2c8a775a7f33e6427a8b3b895d75b3a7a579dc5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 11:36:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b74ba65298dc4f90b0525ef7201e8ed0 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/491177d7c64d4a768af79a5f415e2887 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"98d7cc138170a22101121467324b00466b7fd1d9","revisions":{"98d7cc138170a22101121467324b00466b7fd1d9":{"kind":"REWORK","_number":1,"created":"2022-03-23 11:04:07.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/47/834847/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/834847/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/834847/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/834847/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/834847/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/834847/1"}}},"commit":{"parents":[{"commit":"c083073ca26ea9e06a4c95da6cca81eebb155f73","subject":"Add support for Rocky Linux 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c083073ca26ea9e06a4c95da6cca81eebb155f73"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-11-02 10:40:56.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-23 11:00:23.000000000","tz":180},"subject":"Ubuntu: add support for Apt repository configuration","message":"Ubuntu: add support for Apt repository configuration\n\nThis change adds support for configuring Apt repositories on Ubuntu\nhosts during host configuration.\n\nRepositories are configured in a single file\n(/etc/apt/sources.list.d/kayobe.sources), using the modern deb822\nformat [1]. This format is more flexible and readable than the original\nsingle-line format, particularly if multiple options are used.\n\nUsing a single file allows us to more easily keep the set of\nrepositories in sync, since Ansible doesn\u0027t make it easy to clean things\nup.\n\nSupport is added for marking repositories as signed by a particular GPG\nkey. This approach is now preferred over the deprecated [2] apt-key\ntool, which resulted in a set of globally trusted keys.\n\nIt is also possible to disable the repositories in\n/etc/apt/sources.list via apt_disable_sources_list. This allows for\nreplacing the standard repositories with a local mirror.\n\nCI tests and documentation are provided.\n\n[1] https://manpages.ubuntu.com/manpages/focal/en/man5/sources.list.5.html\n[2] https://manpages.ubuntu.com/manpages/groovy/man8/apt-key.8.html\n\nStory: 2009655\nTask: 43818\n\nChange-Id: I3f821937b0930a0ac9341178de7ae5123d82b957\n(cherry picked from commit c603be253641aab2199330c1afb2c83051be3e3a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/98d7cc138170a22101121467324b00466b7fd1d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/98d7cc138170a22101121467324b00466b7fd1d9"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
