)]}'
{"id":"openstack%2Fkayobe~834929","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I9d7d18851359e97cd01b4c2287bf79110796b25a","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-03-29 16:13:13.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-24 09:37:14.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-03-24 08:38:07.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"I9d7d18851359e97cd01b4c2287bf79110796b25a","subject":"Ubuntu: add support for Apt configuration","status":"MERGED","created":"2022-03-23 17:20:36.000000000","updated":"2022-03-29 16:34:12.000000000","submitted":"2022-03-29 16:13:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834929","meta_rev_id":"774abcb9af271b8479e756ae3ce67f63079e1a31","_number":834929,"virtual_id_number":834929,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-29 16:13:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-03-29 13:44:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-29 16:34:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-03-29 13:44:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-23 17:20:36.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-03-23 19:24:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-23 20:20:52.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-03-23 20:20:52.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"8429355f401968ca6036251df48391024bcfe8e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-23 17:20:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ea3d5967d5c3f16983dc39c1d1454a2f3775f1d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-23 17:20:52.000000000","message":"Topic xena-mark-goddard-818068 removed","accounts_in_message":[],"_revision_number":1},{"id":"3745f3e00b867ec186747cd62b8bd5cf00ab7428","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 19:24:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/343bc127ecff4705bd768ef46e9b90f9 : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19ed52f4e35c48b8b7c79690fb7671ee : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38209fbff79b4780a2144cae447c98be : SUCCESS in 8m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34bfeacfe76e4c12bbe1915cfcfeb8da : SUCCESS in 8m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dfe6745fe07d4456b6176442de707737 : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10aeacb12c1848a7bfab7798e93d5e9c : SUCCESS in 11m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/432d57abd72c4841886f06b8da39d068 : SUCCESS in 5m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e4c830f85b2f442a8c6d32d30a7c04ee : SUCCESS in 8m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4f8b4e11d84b4bb3ab70e8b4d0a2c7d9 : SUCCESS in 13m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c7fad5be1d6f443cb63c95990a4af523 : SUCCESS in 36m 06s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4aa56b4d8d714e4abf822becacee3143 : SUCCESS in 1h 05m 11s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/184cd5e58ee94ed5977c86a6d8bab291 : SUCCESS in 1h 12m 24s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9432bcc821a48e49cdc7ff7aa08f131 : SUCCESS in 53m 18s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/142e6b401228441ea38a1fe590b2ccfd : SUCCESS in 1h 22m 49s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/dc46871697a242af964a1972128ccc49 : SUCCESS in 28m 29s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/9cba4768ddde4e74bb34bffef71959b8 : SUCCESS in 25m 56s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8b68920bbd8445a984d4a757e224fc4 : SUCCESS in 16m 45s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e73912ac32b94e5a89cc18e8d39b0c26 : SUCCESS in 2h 01m 43s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06a7bbba4920400d90e68661ca3eeb80 : SUCCESS in 1h 22m 02s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0f48374e18b94f2898f8e7a88676dad3 : SUCCESS in 38m 44s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/39f5cfd7278845ee98f6c37b7a4929f3 : SUCCESS in 33m 28s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/548eaf19d2d14aad99a0e1fee163fa6e : SUCCESS in 31m 54s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/818fe238014846a799a35c5fbd65b687 : SUCCESS in 38m 44s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ca5ec1105ce4d31bf895df0b8fc51b1 : SUCCESS in 31m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b627b9acb4e142c8b3ca982ed4225d24 : SUCCESS in 18m 44s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/4e3a96c507114d5598a3897e4e877abc : SUCCESS in 17m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6dd0861323eb4328b3470ae331195836 : SUCCESS in 15m 18s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/9f213e2ef92c4d72abc7b3094e321e93 : SUCCESS in 19m 03s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/51f78d57530d4a708622e3a50408e9cd : SUCCESS in 19m 17s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb351dff84e14ed18053e08933111fe0 : SUCCESS in 14m 51s","accounts_in_message":[],"_revision_number":1},{"id":"26c31b8e742c60993e6a7c820e67a1379f93351d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-24 08:38:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d2be7100af608c7b5028f6663ff1740d3749f820","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-24 09:37:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0e0e3c21ad99f44a0114d41f03e46b16efda2627","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 09:37:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c703132a654c92a03fb7abdc564bb8aec29cc8a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 11:56:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2a3160275fe42298246795c6de8c459 : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/702ce34d311040e3be059f982a61228f : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2625136ff35d4c4c82653b4ee54160e0 : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe8e5fedcc5f41bf8685c38db958a36d : SUCCESS in 12m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a8eec0b206b4d22bca6ee1ce36c0dd0 : SUCCESS in 5m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8e513c8cf79d411e9769ccf2b8edb424 : SUCCESS in 7m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4eeb48dba1da47c7893bfee699ed04f0 : SUCCESS in 14m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fa59cda1fed9404bb4646e203e05028e : SUCCESS in 37m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ef03ea5935b149d18e333dbd90e02f2f : SUCCESS in 1h 26m 37s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/5db697772db140a2bc1fc5a5b2028290 : SUCCESS in 1h 12m 35s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea6d3bb871f74a2dadf4e02a1e294818 : SUCCESS in 56m 32s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6e60d6559b314c7fa8a8ae3d6ece0c7c : FAILURE in 1h 31m 26s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e6f9cb5d88b4466dbe50ce3bcbff7af2 : SUCCESS in 31m 25s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/29be4dc6d19644c6b6feef203e72276e : SUCCESS in 25m 06s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3dbe51b84e243fcafd1fcd72b5848b5 : SUCCESS in 21m 22s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e0a947504fc447b3be504aae842296f5 : SUCCESS in 2h 16m 10s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b56768fb01d346d7a5043b3a3ad998d5 : SUCCESS in 1h 23m 06s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/da31269719304f768d9098139cf91681 : SUCCESS in 42m 22s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/8a87164ded60410fa812029f9d50bf16 : SUCCESS in 38m 48s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/496fae710eb24fd389a545bf0d1096e6 : SUCCESS in 33m 31s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cb14de655e9a4bcfb3946b861a58e9bb : SUCCESS in 45m 18s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4239921dd7084e34955f98d70ebda759 : SUCCESS in 31m 47s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b60572c6d2974b0f85402706eca59eaf : SUCCESS in 19m 15s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a09cad7dbab248ef99497b1adc35a62e : SUCCESS in 17m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/127df5ee65d149929a66bb3ce26d9016 : SUCCESS in 17m 25s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/33bacae48f15472487fee0d67767aeab : SUCCESS in 20m 54s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/00e89a82ac344dd18d253a2f5520b138 : SUCCESS in 20m 02s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61e6864d21914e34a4249d82f3981b77 : SUCCESS in 14m 40s","accounts_in_message":[],"_revision_number":1},{"id":"5bcc53e67b17ff95735562d19451b585c9cb9fd2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-24 12:04:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2f078883d901a662a64ace2e48377b1b6f9dcfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 14:13:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5568171cb92c494f999135010b57a750 : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18afb62a98884db1b08353d3fd82fb76 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/676dfc65a62c468a9ecec849fd0efb0d : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2d7f73487774cdcb5c1a9cab08c80ff : SUCCESS in 8m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/655b29dbde8b45d49dfaa111bd2dc574 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/334f36e479894158a35058c1cc38dea2 : SUCCESS in 9m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a94c979df21d421f9fcf88770769d361 : SUCCESS in 5m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d738a0dd8b6346249fcf5c2fe8086455 : SUCCESS in 7m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/beb594ab2aa6488f8ae8fa49e8b2faf2 : SUCCESS in 12m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/82980c233cb940dab3a5c0ca124cc584 : SUCCESS in 36m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9035c92138a74826a0921c8bf730e4b7 : SUCCESS in 1h 09m 21s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/a287af20660e4dbda73bfef53eddf929 : SUCCESS in 1h 18m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ef6b2ef5029445dbf5cc81ee538066e : SUCCESS in 58m 25s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bbbc9b718341475c8d2a458bae6f1dc4 : SUCCESS in 1h 13m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a9a7d76890194866b9229e65bd1d34e3 : SUCCESS in 26m 22s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/64c56a0b9c5e472bbf69ca0cf877ad6e : SUCCESS in 25m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f51a6a817e1944acb0cd97c6f0422cdc : SUCCESS in 17m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b152df76a8cb451dbb8fbc7f980f8e42 : SUCCESS in 2h 07m 51s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e02ce957fec4197b97bd41e7ad1acd8 : SUCCESS in 1h 18m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/087bdbd341244900bc3e13dbe9e76fd8 : SUCCESS in 45m 18s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/1c0b7c8dcde24e4eb76ff5cc8ce306a4 : SUCCESS in 32m 37s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f13e0100e66043a180d93b0eaf9cff5c : SUCCESS in 29m 12s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/90e3f3011b9447bfadef8b488cceec40 : SUCCESS in 46m 12s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7113d395afe54169a08327a7a3d1f4aa : SUCCESS in 30m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5c84fad22b9f46a29755e6abf972ffb0 : SUCCESS in 21m 51s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/8a547f961ac74861a626eefb166740f5 : SUCCESS in 16m 54s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d07494c83924f5db6409e2978edacfd : SUCCESS in 15m 27s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/195d279a8a7744cdb1037354f028c2b9 : SUCCESS in 20m 36s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2a6e500a1311424dbce14475bfb2ceef : SUCCESS in 17m 50s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14563f822de04bf7a75eeeb5b053f771 : SUCCESS in 16m 05s","accounts_in_message":[],"_revision_number":1},{"id":"7f29c32a834c2f2ed701cdd211ec4f5c348aaea8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 14:13:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0f4ad082824a6af3ebcd672a9b537079bcc98f44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 16:24:31.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/914417aced8e42ec9c0644eceb206fc7 : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f0473e5adb84fc6a892564bb3815fdd : SUCCESS in 8m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f57874fd7ffa40ddaf630b4483149aa9 : SUCCESS in 9m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/914927e9d8d74f28a6b601370b7a64f8 : SUCCESS in 10m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89d2cfa1c1a749969389668e0ad5d61c : SUCCESS in 5m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8f4ed70630aa459780863c6153250cae : SUCCESS in 8m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/75bc809d1af448d1a139a1f2e4c5be32 : SUCCESS in 15m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/067a2139bfbf4e51b5f2be90eafae62e : SUCCESS in 37m 42s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b58d2bf22edf49739b91f8b7d79758be : SUCCESS in 1h 13m 52s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/77e4d48024154cd2a84f4265f68397a5 : SUCCESS in 1h 14m 34s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa2c4c564b0b4715a0ecfef63fd3bc65 : FAILURE in 10m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2fbaa3643dc5489bb23bb06c3da060a5 : SUCCESS in 1h 22m 07s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ab6ef830c8fc46b68f3bd2fc5f182874 : SUCCESS in 27m 35s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/7d78b994900c429fb6010a23d8a1f5a5 : SUCCESS in 24m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de373b143668494597224a343a1e5523 : FAILURE in 10m 27s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/463dcb8f290647349dddb444bcfa47b2 : SUCCESS in 2h 04m 35s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8acf34e8270495e84ed5b57b59ca131 : FAILURE in 9m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/878fd9620cba450ba84d38a1f226cdfe : SUCCESS in 37m 17s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/24b517453b4843908dc8c4f42310fa01 : SUCCESS in 31m 15s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c34e169052c04d2eb37315426f9c920b : SUCCESS in 27m 51s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/91833948d7a04905b7c3c95023887136 : SUCCESS in 43m 21s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c46a5430977b48818ebd6b3a4df265b8 : SUCCESS in 52m 29s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5ec9af229bac4ad6b667d157f84f3e4e : SUCCESS in 18m 45s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/b7f3a92e581a41889ef9aed846d08f0d : SUCCESS in 16m 46s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36297f69d46e4eea87562f0645a49adb : FAILURE in 10m 31s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ef873e7e9a8847a1a08d3e4bb1b47000 : SUCCESS in 20m 50s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/1f2a82a1216b422d8c990acdd88c6837 : SUCCESS in 18m 59s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74e11148be1c4e36b7259f8f92e12351 : FAILURE in 9m 52s","accounts_in_message":[],"_revision_number":1},{"id":"43a05e5c29db88def5582d50d28bf8a61d67eeb9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-24 17:27:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29bbab78f41e7f92e92eafbb436a42e36cb2e081","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 19:58:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fd7b0b265cd496dac951a0054c7a75f : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a927ef78e78421b99c02fa624ade575 : SUCCESS in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61d75ec4365b4815835023eb29ad3d3e : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b2f5b01f3694fd0a53bc3e1ca235ce0 : SUCCESS in 8m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8fe4e626a454e7b8751032ee95529ad : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0492bc9327c34150961cf3c8c1c90435 : SUCCESS in 10m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73af4e4fa4f8479990d632919ada6b93 : SUCCESS in 5m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f59d697b47c24ae492469edc6574fc0c : SUCCESS in 7m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/787833e4d6af4635a35c2c97e6efb741 : SUCCESS in 14m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fe5582cc6d124cf681138572bd52b413 : SUCCESS in 36m 44s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a4203608a8b64a688ad184c69b496af2 : SUCCESS in 1h 19m 29s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/27c111166b8848959a74326fdb881876 : SUCCESS in 1h 11m 12s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76328d06180344c28fee5533ff0f9b14 : FAILURE in 11m 02s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/afa9d02d672f434b9e5728989c423612 : SUCCESS in 1h 13m 05s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/12c30c7c99724f8e9d214af8e9020201 : SUCCESS in 27m 42s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/54db3d29a9cf4f089ae53fc8bd7331a1 : SUCCESS in 26m 18s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9277175828a46b6b90b31e4d907025c : FAILURE in 11m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/07c35f011e8d4e5f847bd55aa54a5323 : SUCCESS in 2h 24m 05s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e10c5a67dc04f3e9f9452e94a5c47c6 : FAILURE in 9m 30s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5fd1f3db590e4fef9a852ea96c8c0011 : SUCCESS in 37m 50s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/63deb1ed24824293bc5044667159a6d9 : SUCCESS in 31m 56s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f2b1da1097e47e2a7f244af229a88de : FAILURE in 14m 14s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d11beeb52ca9445a8a8d95c834ce874c : SUCCESS in 46m 58s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7cc672e4273341dda845e3e77123eb7e : FAILURE in 10m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/458ec4786d3c49f7aef16287c8815a84 : POST_FAILURE in 17m 46s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/aaee884d49304c099a08e58750f4917a : SUCCESS in 16m 10s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/065bf296b25f4d668fb769a46fe0d6e7 : FAILURE in 9m 48s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/413ba0ea813a44e2a04c232d1689805a : SUCCESS in 30m 13s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e06c2b3cbd4141ff945c494de615e45a : SUCCESS in 17m 23s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4de7361fbb4f4d20a75d455c4eb6fea4 : FAILURE in 13m 54s","accounts_in_message":[],"_revision_number":1},{"id":"d2b8092d530d333672372190077f3fb2f4db09aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-25 17:38:43.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1529e7b024e30bccb38ac009de1f5ca933c0a2dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 19:36:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0691cd1e0f724a069da3f6b5f5c60333 : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e232aa6f7e445a4818b782410d26066 : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a0da1cc64d046028f7c20fbf8bceade : SUCCESS in 8m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce35a2c4cfbf461eb201efccd53afd12 : SUCCESS in 7m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7fc85b590d634819b615f7e66b8064ab : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29cc562f114c4d8ea53ca9416f4e1d2d : SUCCESS in 10m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/840c66bad88e457d81eed94ec4e0d384 : SUCCESS in 4m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5b437369e9eb451f980d35010d1b87c1 : SUCCESS in 9m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9ff96f4a351b4afab0b5c5073fbc93ee : SUCCESS in 15m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/541db45e5df447a08df23532b7432bdc : SUCCESS in 34m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/24b28e5c1d2645fb87ebc33fcfd996dd : SUCCESS in 1h 19m 52s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/9f11d9c524514ea88ad099fbb6694379 : SUCCESS in 1h 07m 38s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d41f91498609409d81d03f1b784eefc0 : SUCCESS in 55m 51s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4ff9a6e2cab641029c5b956452a36966 : FAILURE in 1h 16m 11s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2520a09e4a404db39480ad19fe258fee : SUCCESS in 25m 14s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/e7ccb3e9ece14afa95ac991741361dc6 : SUCCESS in 23m 45s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3217bacc1183493397e2b7361bfcf6cc : SUCCESS in 16m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c4c166a7ac8e436987823f06a6033d01 : SUCCESS in 1h 55m 19s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5488e06a384476b96c6ed5484dfbf8e : SUCCESS in 1h 12m 37s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/10242a0e0a2a4e63a35d5af45db44622 : SUCCESS in 36m 31s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/0f50545d5058491b955552c86a654707 : SUCCESS in 37m 53s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bbe8d34f4e64dd1bd76bb1b3925e656 : SUCCESS in 32m 49s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a9eae4c0ac3f4aa3ae6899497d39fcbb : SUCCESS in 45m 03s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bea60cacc28749ba8da878db13f09c53 : SUCCESS in 27m 18s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9bdbc9799ca944068a5954ae3def7358 : SUCCESS in 17m 46s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6d5a62ff4ac84561b0830cf912ddcd30 : SUCCESS in 16m 55s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6696562b0584863b61806afd7379161 : SUCCESS in 13m 02s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/8e5f84baef584a1d81ce9eb1e22a9100 : SUCCESS in 17m 21s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/0103b60da7844636ad15215e0e4bcea6 : SUCCESS in 17m 37s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b68aaa1fec364c3ea80c38c62d8cce2e : SUCCESS in 16m 41s","accounts_in_message":[],"_revision_number":2},{"id":"7e435af5c8ae4aabc849949f1b46df01cb0eaefe","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-25 20:07:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7097a9198dab3ab2e2f3bbcc0312ba13ce2839ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 22:04:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6ca3a82f8644bd9bd575ce0865695f4 : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7efc72b61ed14246a00eeb5372064cd2 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b1d407874624f3a9b84b3d34e6914e6 : SUCCESS in 9m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fe52cc9ad0f445384ce66ae38f162cf : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2954fe9041994ac5befbc14493100716 : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf6d0b8a198e44bda52e19d68400d981 : SUCCESS in 10m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/140339e122b241139a3385740f2edf61 : SUCCESS in 5m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cc07acc5d9ef4df09debe36c9a3500f7 : SUCCESS in 7m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4541f52d930f4f2589d7860ffcc85eb9 : SUCCESS in 16m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f8e4a424a4e443459c808ed13054c4b0 : SUCCESS in 29m 49s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c59157440ba24e89859e127933abdd0a : SUCCESS in 1h 20m 11s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/df24f09f81d2410ea8f7b928383c7f82 : SUCCESS in 1h 11m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5b2ed779497413ca0649312ca36568b : SUCCESS in 49m 48s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9b205e667d384154920b47efde886b68 : SUCCESS in 1h 15m 16s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2c747007a37c436cae4502dbdcb9ea3a : SUCCESS in 27m 09s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/3c2ea2fbd3a344bc9c90f48a28c8c634 : SUCCESS in 24m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69245033e01a43b79c307aa72467bc3a : SUCCESS in 18m 44s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9ee999e8a44948cdb482af5903c682dd : SUCCESS in 1h 55m 23s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7ef212ad36b448da3c237398834e853 : POST_FAILURE in 1h 22m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a6ec37ba212647768877a2d25d405878 : SUCCESS in 44m 13s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/8e238cb91c034b4287feac5708bbd102 : SUCCESS in 32m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/675f99a20ff34ab186036de9cceb9570 : SUCCESS in 29m 42s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ea95a8ad1adc4f7ba515f8f1d3af7c5d : SUCCESS in 47m 36s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8462de676c94d0db4b27ce42e3c0235 : SUCCESS in 26m 14s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/371004f0d92240cb8daba7c76b046c46 : SUCCESS in 21m 26s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/110b7c3d5e7447ae95aa86d1a0b737ee : SUCCESS in 18m 24s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ec46f12fe3249d3bab76926ffbd4411 : SUCCESS in 15m 18s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/f18bbdaac74a46b3b4110b63b9f9f138 : SUCCESS in 23m 40s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/393fda347f56406e9ad65f902d505fa0 : SUCCESS in 19m 28s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfd6a4497cc8453ea7a5f4106768bb34 : SUCCESS in 17m 04s","accounts_in_message":[],"_revision_number":2},{"id":"2582517169a7bbd66a71789ec4c8bc36007fb97c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-25 23:28:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"163ba9f1a61131a0d2ad6a4d3dc4b49d03038074","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-26 01:16:22.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b780531e89b34e9dbbb919d356701581 : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e79f30590b74de7ae428414e0c3a545 : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9dba0e1d4854207866b5391ff7189ac : SUCCESS in 7m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a42e60c305542328e0f8a7b33aa0075 : SUCCESS in 7m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97482caf2fe44a55934fdd48a9f9968c : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/806d464ad4cb4af99217fe407306b33a : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a192e7705324b5ebeab14ae0b61edf8 : SUCCESS in 5m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/df2fd958a2144f53832dc34a153c43bd : SUCCESS in 7m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/58b6dbee0c284aeda67e5eb8a0af660d : SUCCESS in 14m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/355a8bb12a43471c80ed25fccc345ff5 : SUCCESS in 33m 20s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a89b0e6c6f0d448d92a705b7656fcfdc : SUCCESS in 1h 11m 19s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/575cff71cda04a84ae516803d5aa6254 : SUCCESS in 1h 10m 20s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8895dfe19f05400890c460c3644fcef1 : SUCCESS in 55m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ec5056b0f8fe40f8aa2ce861ae0c49d2 : FAILURE in 1h 11m 33s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0dde853763af4404ad4805b9b6832ea6 : SUCCESS in 27m 20s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/23c37b6392464a3e9ce2d2ee67f3e491 : SUCCESS in 24m 00s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a44e5c580cd4b579729b691a0c66563 : SUCCESS in 17m 52s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3e62eb36ca414c169f76ab935b93a10c : SUCCESS in 1h 45m 49s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a1ff2a6bb11435d9125228c167136c3 : SUCCESS in 1h 13m 53s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5e294b60026e45fea7e2cb82da561b65 : SUCCESS in 39m 35s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/69c2e533c29c456798584789216dc5b3 : SUCCESS in 31m 28s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef89e550729c4a419d8ccb5262536f6e : SUCCESS in 27m 35s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/18d9192d458a4252a3cfe3f40b741052 : SUCCESS in 39m 00s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/854f1ddde43b4304aa817cff1fb0f964 : SUCCESS in 28m 18s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4fd2b60a479b49bbbec9ace920caa816 : SUCCESS in 20m 08s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/1721289db95d4439bcc1ffffa7efe10a : SUCCESS in 16m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86ce32c03bf04d739b5c51dbee52be5a : SUCCESS in 12m 51s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/2360a6bd6fd64f069ceaa4d96783669d : SUCCESS in 20m 55s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/09224d1c36ce43b88e0982e17a13cb10 : SUCCESS in 17m 31s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d56f70d09cd7473c92c99b147dd5a66f : SUCCESS in 15m 35s","accounts_in_message":[],"_revision_number":2},{"id":"848fa21bd34fd9e878ae6a193f8f964e9168e420","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-26 09:17:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a75970d7f52839afbc53040fde05a062c833950d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-26 11:14:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb4b9d933b224fed9ec2af37fdc66683 : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94f4a722cdc84e1c86a5c56d6fe414e4 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b726c0233c29436d8532f6509e5c431a : SUCCESS in 7m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6db77285df5a4d19815491209de56fc4 : SUCCESS in 7m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95bd2a9761104bbcbb2569bbd34e2103 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6af52f61aa66419481f1944108e201ae : SUCCESS in 10m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de40a5eadadc4bfdaeebcc0a36585a15 : SUCCESS in 5m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a05fb05d041549df8dcade8a83aa537a : SUCCESS in 7m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e650890c459c4eac81a8965fca91b166 : SUCCESS in 16m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a3a244cf065c40cabfb2922191a3a6a0 : SUCCESS in 33m 13s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/cfc9f857d2cc4788abff0f4345cb445b : SUCCESS in 1h 19m 58s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/89137e2588654d6a85d5e4efcbb9554e : SUCCESS in 1h 09m 34s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/622537aa0b754194bbf0548c4e3fc56c : SUCCESS in 58m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d25dfd1a79b74a0e8261691a1b031459 : SUCCESS in 1h 13m 12s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/fe279285d4c04990a7ebee9c697a973f : SUCCESS in 25m 42s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/9966315a59514eed9014d1ce8ae69154 : SUCCESS in 22m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3a38fef0253420092c7c89a99e91efb : SUCCESS in 16m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/85485b8358bf4056be3c444885bf60a2 : SUCCESS in 1h 55m 16s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb70c108b7e44613942479cd79850ad3 : SUCCESS in 1h 18m 35s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/41094b3dedea44fe9d70f8043181fa61 : SUCCESS in 39m 50s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/4d33c85a9f894f359eec9ed369959570 : SUCCESS in 36m 16s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55f4eef30b514055b79a3476817c46ec : SUCCESS in 31m 32s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f6a5bb72fe044ea0b1344d9019dc38ff : SUCCESS in 41m 27s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea91f7a5f64a4e248d02e4116c22cc99 : SUCCESS in 29m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/634998287a694120b5f49f0422a9d001 : SUCCESS in 17m 56s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/bac4e5efe45c44fb973c3a6459c4c4b2 : SUCCESS in 15m 58s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfe70eb9bb954a9498de0cc671d96ef3 : SUCCESS in 13m 38s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/dd48316f8c2b486db6c30e54db04bc5c : SUCCESS in 23m 27s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/9d63212f4af44a62b16321e618505e8a : SUCCESS in 21m 47s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d72148880504e7295892f292c7c092c : SUCCESS in 15m 45s","accounts_in_message":[],"_revision_number":2},{"id":"cd83c853a13075836e1d36fccaa88a799de7c985","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-28 14:18:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"730f873234661aefce3117d47ade97968af8e5e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 14:45:47.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"8ef86fb4f79863518098c0a607a7aa261883e051","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-29 11:30:15.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ba25117328e5be10a6bd1644e04d3f1ec8de75da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 13:33:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9fc8ac34bd94b0487fe71315c34bb08 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09989c1371bc472ab987b5ccb4356b0b : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c53a92496234476d97674e083c6bff9f : SUCCESS in 8m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0494294655f342e4bbc2ddc7b9c8a7dc : SUCCESS in 8m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/077acbaf07a24c958745ab7815eca300 : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fc2480745334631a2cd5093436d1fcd : SUCCESS in 11m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0310351ecc534d4a9345c09eb17df6f8 : SUCCESS in 5m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d6762e3e2afc479cb784dc930c73cf64 : SUCCESS in 6m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e63dacc2bbd84317b12c819b65bfcd50 : SUCCESS in 15m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a7514fc3547b4523bb4f684dce53bbbe : SUCCESS in 34m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b726c4eb0d9241c2bc8b1b658bd3b4b3 : SUCCESS in 1h 11m 46s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/a17f1ae612b14e17a7eb29ea3c668c75 : SUCCESS in 1h 14m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a88b4db8604b45128babec323099d46d : SUCCESS in 56m 36s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/33dfd9b0862942ab9539875e0af410f1 : SUCCESS in 1h 10m 44s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c7900052abea4fe4afdaa9f356c4921b : SUCCESS in 25m 49s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/610b58e57d2f49fb9258a606f9f4a26d : SUCCESS in 23m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12726fea0afd446ba903df9edfdffe5b : SUCCESS in 17m 19s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/25cca4fbdf054369a0058694e6af363b : SUCCESS in 1h 57m 40s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9be94bbaa7764773b6d4444e0966c987 : SUCCESS in 1h 14m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/57fa96450c6c48258b730b5dfd575061 : SUCCESS in 29m 57s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/03549bcb1ef845deb709f9a33dc9f689 : SUCCESS in 20m 46s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9f3c75f7ce9431a8cfe83c1d2df5e4d : SUCCESS in 17m 00s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/94a23978e04e4ef0b119cfcbbb92dfcd : SUCCESS in 33m 36s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf3ebf620fd0428185eb40e9fe6a35d8 : SUCCESS in 26m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/145cd877c739438d8e52172814146ffe : SUCCESS in 18m 03s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a4a4521414f341b5a75485dcd7b839ea : SUCCESS in 16m 03s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/461a9e15b5834cf88336004b8687fe81 : SUCCESS in 14m 10s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/0be2f235fb1648dcba532f273fe66b2a : SUCCESS in 21m 44s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e4bdc398a0f044768fb4403091a79966 : SUCCESS in 19m 14s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8afe01e3c0504c60b15e4efd9336772a : SUCCESS in 16m 41s","accounts_in_message":[],"_revision_number":3},{"id":"1328929c082ec2c99495ac27691b3ac286684aac","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-29 13:44:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e094ee4f7ba70ffbbcdfe6f5d7c4e8036733fdc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 13:45:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9c20fb427928ea1e84b467fd270b8ce837992b38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 16:13:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8da27db66164bea83627bd251c8844b : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ae6bf21efac4f02a29453a051055f9c : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b6c57df5e6e4669af5314fb844dcd58 : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22964fd7279e4150827d31dc15272ce7 : SUCCESS in 9m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0a3f4f5084940bd8350eed9ead12ebc : SUCCESS in 5m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/67fad7f0f16c477ca069afbc5ffa6d67 : SUCCESS in 8m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a67e95fbbd6944fdb82a0e42f41689de : SUCCESS in 15m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cd40aa62ea6f42e7a842c617e90afe3f : SUCCESS in 36m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4020e64024d84315838e3c064295547e : SUCCESS in 1h 08m 32s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/05efe05d4a2646408f32ca48482d78cf : SUCCESS in 1h 08m 01s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/694623426810414db3f39189f82dc8d4 : SUCCESS in 56m 32s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c89324033c97469883c4c3cd7e372a69 : SUCCESS in 1h 12m 46s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/69a3f435e0e94d91abb5c7ffebc8f1b6 : SUCCESS in 23m 46s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/e6f0f10ec01c4ef0b8666312d36d90ff : SUCCESS in 22m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d543b73c42af4d8e8e56b6d795c2164b : SUCCESS in 17m 31s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2271d194652444c1ae644dcbc7cd2acf : SUCCESS in 2h 14m 20s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d020a539291c456b8cc81d69f785f782 : SUCCESS in 1h 25m 18s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/481fcca507644ff2a6a708deb1b2c9e8 : SUCCESS in 28m 22s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/84c157b1be914d1a8be34d32182dd2ba : SUCCESS in 20m 51s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4e1f98527bf4c46a10f6f0dbcf9e52c : SUCCESS in 18m 04s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6a8809f107ec49e695493739e2eb4514 : SUCCESS in 34m 47s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/966ed8820218495a8c75175ea66b73f6 : SUCCESS in 27m 41s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/097d5f484d284b04857d810adffc165c : SUCCESS in 27m 05s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/7676d15072d843e9934b0420eb7f3ef8 : SUCCESS in 17m 13s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e0465391add44f49f452c676f6701c9 : SUCCESS in 13m 09s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/a46e7a323aaa4d51b631bdfd9972f151 : SUCCESS in 28m 04s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/877909411bd346b999b22642c1737bd7 : SUCCESS in 19m 09s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfad86a1be344a018fb32097a96a8c3f : SUCCESS in 14m 25s","accounts_in_message":[],"_revision_number":3},{"id":"6b700fc5f520974f2e889309a0a3c7e5e79bc449","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 16:13:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"77c1dcfb88988e9d158845a2d726802a86ff2d74","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 16:34:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95a734f420004054986f04ff8b9f9ab8 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28f8f5219d54466d94d89e04dd8ffd80 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4efb80a10d60268e42e64741da09317e42d5e67e","revisions":{"294e07309d71b00e86c052b108b39c2b39cad3f6":{"kind":"REWORK","_number":1,"created":"2022-03-23 17:20:36.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/29/834929/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/834929/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/834929/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/834929/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/834929/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/834929/1"}}},"commit":{"parents":[{"commit":"98d7cc138170a22101121467324b00466b7fd1d9","subject":"Ubuntu: add support for Apt repository configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/98d7cc138170a22101121467324b00466b7fd1d9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-11-16 11:27:37.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-23 11:01:19.000000000","tz":180},"subject":"Ubuntu: add support for Apt configuration","message":"Ubuntu: add support for Apt configuration\n\nThis change adds support for configuration of Apt package manager in\n/etc/apt/apt.conf.d/. This allows adding arbitrary global configuration\noptions for Apt. Options can be added in different files, allowing for\ndifferent filename-based priorities.\n\nCI tests and documentation are provided.\n\nStory: 2009655\nTask: 43987\n\nChange-Id: I9d7d18851359e97cd01b4c2287bf79110796b25a\n(cherry picked from commit 5c661b888e821bcadf5e93fc0b09f86eb9ba901a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/294e07309d71b00e86c052b108b39c2b39cad3f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/294e07309d71b00e86c052b108b39c2b39cad3f6"}]},"branch":"refs/heads/stable/xena"},"0b125c69109736f42d1ab900a62ac86420b75bfb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-03-25 17:38:43.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/29/834929/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/834929/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/834929/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/834929/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/834929/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/834929/2"}}},"commit":{"parents":[{"commit":"b648057d45b4b7ed73c259a0129053703d0e8a69","subject":"Use jinja2.pass_context instead of contextfilter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b648057d45b4b7ed73c259a0129053703d0e8a69"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-11-16 11:27:37.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-25 17:38:43.000000000","tz":0},"subject":"Ubuntu: add support for Apt configuration","message":"Ubuntu: add support for Apt configuration\n\nThis change adds support for configuration of Apt package manager in\n/etc/apt/apt.conf.d/. This allows adding arbitrary global configuration\noptions for Apt. Options can be added in different files, allowing for\ndifferent filename-based priorities.\n\nCI tests and documentation are provided.\n\nStory: 2009655\nTask: 43987\n\nChange-Id: I9d7d18851359e97cd01b4c2287bf79110796b25a\n(cherry picked from commit 5c661b888e821bcadf5e93fc0b09f86eb9ba901a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b125c69109736f42d1ab900a62ac86420b75bfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b125c69109736f42d1ab900a62ac86420b75bfb"}]},"branch":"refs/heads/stable/xena","description":"Rebase"},"4efb80a10d60268e42e64741da09317e42d5e67e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-03-29 11:30:15.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/29/834929/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/834929/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/834929/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/834929/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/834929/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/834929/3"}}},"commit":{"parents":[{"commit":"5b78b3757577107c6e38ff9e8b18872571fff4fc","subject":"Use jinja2.pass_context instead of contextfilter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b78b3757577107c6e38ff9e8b18872571fff4fc"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-11-16 11:27:37.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-29 11:30:15.000000000","tz":0},"subject":"Ubuntu: add support for Apt configuration","message":"Ubuntu: add support for Apt configuration\n\nThis change adds support for configuration of Apt package manager in\n/etc/apt/apt.conf.d/. This allows adding arbitrary global configuration\noptions for Apt. Options can be added in different files, allowing for\ndifferent filename-based priorities.\n\nCI tests and documentation are provided.\n\nStory: 2009655\nTask: 43987\n\nChange-Id: I9d7d18851359e97cd01b4c2287bf79110796b25a\n(cherry picked from commit 5c661b888e821bcadf5e93fc0b09f86eb9ba901a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4efb80a10d60268e42e64741da09317e42d5e67e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4efb80a10d60268e42e64741da09317e42d5e67e"}]},"branch":"refs/heads/stable/xena","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
