)]}'
{"id":"openstack%2Fkayobe~835093","triplet_id":"openstack%2Fkayobe~master~If657bf5b0117812d3c53942464cc41cf86cc8ad5","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-03-28 21:10:05.000000000","reason":"Maksim Malchuk replied on the change"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-29 14:20:00.000000000","reason":"Mark Goddard replied on the change"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-03-29 12:43:02.000000000","reason":"Maksim Malchuk replied on the change"}},"hashtags":[],"change_id":"If657bf5b0117812d3c53942464cc41cf86cc8ad5","subject":"Use jinja2.pass_context instead of contextfilter","status":"MERGED","created":"2022-03-24 16:34:54.000000000","updated":"2022-03-29 16:43:01.000000000","submitted":"2022-03-29 16:40:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"835093","meta_rev_id":"4723a4b0c0c0e59fd10a33b8277e110f5cabc2ee","_number":835093,"virtual_id_number":835093,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-29 16:40:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2022-03-29 13:46:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2022-03-29 14:20:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-29 13:50:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2022-03-29 14:20:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"CC":[{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-24 19:07:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-25 10:15:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2022-03-28 08:20:46.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2022-03-28 10:49:49.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-03-28 13:05:33.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-03-29 09:46:30.000000000","updated_by":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"reviewer":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"state":"CC"},{"updated":"2022-03-29 13:50:30.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"9ef20ead700e07cc0ce336b0befb8574989893be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-24 16:34:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e32d58c302578a96bbc9e5dfe9febf83add3f90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-24 17:09:15.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d0b29656592b06a1bb12bc0f7209bbebdd0c5f77","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-24 17:09:19.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"046e1beef20c47fd784bdccef8f0d55e6015efa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 19:07:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b685ea244be4b15ac9b4d61ec61483d : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23fd791b1a104b218c657b5a3f56b74e : SUCCESS in 8m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84253bd98bb44e2184586f1ddc3d0c95 : SUCCESS in 8m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be4e9aada2184f33ba350c7e99f9238a : SUCCESS in 8m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1321cd4f62d74522a26356868e6b7e47 : SUCCESS in 12m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc73a05fdc6944a6bb0fc667479bd11c : SUCCESS in 6m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ab297021121742098913da02d79a69cc : SUCCESS in 8m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/67691f1e3b9547e1bbfc511ed2e71ebe : SUCCESS in 15m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b9963b6b93034725af2974118b677265 : SUCCESS in 48m 51s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/afb04f4c06194e13854858ade1963122 : SUCCESS in 1h 06m 29s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/ac14d7742f0f4c1dbf838fb051d3a199 : SUCCESS in 1h 08m 19s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ea4ffca691c4ee79ea681a70faa0245 : SUCCESS in 57m 30s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ce970d946360490eabcd26b43bb0c090 : SUCCESS in 1h 06m 40s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c2ec99bbe38e4b61b1696b1ceba811b6 : FAILURE in 20m 13s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/4261cff697024abcb37db205fd5938c9 : FAILURE in 15m 49s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80f6aec8041049b2aa7ecdc4751167fe : FAILURE in 13m 52s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5a9693b703124fb3a727aa682b23f511 : SUCCESS in 1h 55m 24s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ee4cdc4cc394057b407de502a666c7e : FAILURE in 11m 54s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1f4bc92d77554371bc61a7b3cdf5a6b3 : SUCCESS in 51m 18s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/ea164509624b489d96267a839e37ee64 : TIMED_OUT in 1h 34m 33s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a57e464c952a4af1bfedf2af8d257c7b : SUCCESS in 33m 57s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ac6a992e769542a5a9eddddda844c028 : SUCCESS in 40m 45s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b5b30eda46f40b2a8fbe4d5d1e0d647 : FAILURE in 10m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3f1af757724446288fdb6db56a495391 : SUCCESS in 16m 31s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/4e6c657475674bf094b6ddd9b0e66f59 : SUCCESS in 17m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb0662aaaf074dcd997873bf71c9c8cb : SUCCESS in 16m 15s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/265611cf9a01496fba8bfea770a2a732 : SUCCESS in 20m 02s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/38af397dbd844506aef2ed28dd3e27cc : SUCCESS in 18m 08s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8347268f7781462abe5bc541063e2b49 : SUCCESS in 15m 17s","accounts_in_message":[],"_revision_number":2},{"id":"8fb1397d44a493bfd761bc443db89fb5c18f2260","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-24 20:12:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8732d4d7add5873a87d013a1bc8d091adecb1add","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 22:13:20.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45b2734a25f4425d8e90696cb786a4e9 : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b60b65ad39c44e9a913a2666bb575374 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c361de10f84542fe8e8d787aa110f011 : SUCCESS in 7m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fbea5d8322d24c2ea4d0f05733b6aa94 : SUCCESS in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86bfad28184d4509be105ebb1ac23a2d : SUCCESS in 10m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5b6f9b64ae1473c9b60701f9a51e153 : SUCCESS in 5m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c0d62d359d4844769ca493928bec4958 : SUCCESS in 8m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0db2188142364b5895e38e450818aa3e : SUCCESS in 14m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e744249851134e1a8624a30d34d908ba : SUCCESS in 34m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3ebc2120a8a3439e8db1be7207311a12 : SUCCESS in 1h 12m 12s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/ca0d0ebcea4e4bf6930e416a7ad1b74e : SUCCESS in 1h 06m 20s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/055c62cea4134d6ea6abf3f9aec61c47 : SUCCESS in 57m 49s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5ebb2bddc166447891da4f64a0d61c48 : SUCCESS in 1h 07m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7e995dceea22420cac110d4d33241197 : SUCCESS in 26m 11s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/bf06541066ae40f8bd650d1192f3995f : SUCCESS in 22m 24s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e7fedcecc4343a2b6e30e691a8d9687 : SUCCESS in 17m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bcd5ee8e9a7341d1a0df16a724b4e979 : SUCCESS in 1h 55m 59s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f2003adfd4049d7bcfe0e48ee2ae86f : FAILURE in 9m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/976b71bfa32544c8bbbe2c59c48ed83e : SUCCESS in 1h 19m 50s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/83109af4e82643a6916f523a09895500 : SUCCESS in 39m 51s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77503a8cc0014f4a821adea4f1d2ca98 : SUCCESS in 35m 30s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d0a62e2ba2f349ba9ef86bc7802c3f68 : SUCCESS in 39m 26s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0c8d3799249420c873dea7b7f2ec6b7 : FAILURE in 10m 49s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3999636f83d24fb08e10057ed5d99b37 : SUCCESS in 21m 11s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/bd115ebc053242e4828248eb31bdb31e : SUCCESS in 16m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e05893d64df485895933210ad40f279 : SUCCESS in 13m 42s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/b097cc8015134c0792a143d2a34c93b2 : SUCCESS in 22m 33s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/590baa6b2bfe4952bb533b359a5a3711 : SUCCESS in 19m 06s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8943812d79374292999e6353b503f324 : SUCCESS in 15m 36s","accounts_in_message":[],"_revision_number":2},{"id":"3e36bae7c0b17b03f84b101c489ab17d77840362","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-25 19:08:30.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"dd629c48fd6fa8cee7ee91b57e28edf687e57930","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-25 19:32:44.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"3ac8d566b7fab241ae29a296c6992739c27bf974","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-25 19:43:07.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"9642e45be5dc14574e61cce7c4419e689e0eda94","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-25 20:05:27.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"8780745e6ed4c21b0ae8662f5fc596177478b6a0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-25 20:15:35.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"c73f9229affbabdbbb62089e6aa39f3425223403","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-25 20:44:46.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a0c14347a1bf0e0688893198b1ec0c96aeddd3e3","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-25 20:44:57.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":8},{"id":"e1e7d21260c9d9b3c1afc663e001f69963b94cb4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-25 21:19:10.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3d33cce2d119de7b991692f058cd36a240c34de6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 22:39:28.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da99884f961b4289803a3320317bf72e : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70b0727f06504a4785d18ca0c55c0bb5 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef9ef7ba3f3d4572b69425b917346cbe : SUCCESS in 8m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed7a1c56d3134327bf4460f98fc00cd3 : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27d2466ce827495b834696b7e80ff465 : SUCCESS in 10m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d7a5810b5fb4ad3ba504e7d59a01bd1 : SUCCESS in 5m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8f19f58539bc4359a05db9189d6ce174 : SUCCESS in 7m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/823b95a1af884f80aa1ecc1951133fb5 : SUCCESS in 14m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c9ae081375c44e5686a004001e42846a : SUCCESS in 35m 07s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e05ffb998c6643f68c698f30fc2055d8 : SUCCESS in 1h 11m 48s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/7a2e2a1aa4bf49fdbe3e01c741be6fe9 : SUCCESS in 1h 06m 35s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e71233708288421580eff900069873d8 : SUCCESS in 55m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/dbbe787288844828a1c7721e781e2fdc : SUCCESS in 1h 03m 54s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f1110f85b5f64e79bc1777622f35398c : SUCCESS in 28m 41s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/928c46ca95d340e3931655d9d8ed1196 : SUCCESS in 21m 31s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c79d03fa7e94f9d92e84bdeedde9832 : SUCCESS in 17m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0d2ef2d3893d41078134cca8f08fa577 : SUCCESS in 1h 52m 10s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72852a21e98f4f2384b4ab1c07d7552f : SUCCESS in 1h 15m 08s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e6ad748f272b406cbf790779fb324d34 : SUCCESS in 1h 00m 04s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/9470cd42292a4f339f23e673db5f0614 : SUCCESS in 39m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b60b56dcc2854b78a1822e513c7b2a5a : SUCCESS in 36m 20s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c8e7e8c3689c47639cfd64e00cc83459 : SUCCESS in 47m 40s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/422c5e9769dc41fe9e54e26277b7baa6 : SUCCESS in 33m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/23ce6e6a35bf455fae87fe85d1365806 : SUCCESS in 22m 04s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e94d4f58e4004e11ab9535d007e521ae : SUCCESS in 16m 04s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9efd98966144cae8ae798106c38f08c : SUCCESS in 14m 55s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/8f8d259d44be456ca202461fe59eca2e : SUCCESS in 21m 14s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3d4cc716f43d41c09ae71e7d78443295 : SUCCESS in 17m 22s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/323d2eaa253d49eaa7b06840a82b0408 : SUCCESS in 15m 13s","accounts_in_message":[],"_revision_number":8},{"id":"b05fda4ece7d539b128dd5d399bdb75a115e836d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-25 23:26:51.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"8eaac05f573f88327735dd8f9b72e5fe749bdbd0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-03-28 08:20:46.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"1b80d1358ef4e6dd42d427ce66e9ba8e6f1e976a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-03-28 08:22:13.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5a18fd0b6427bef16f3aff81a951b74871b4a5d2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-28 10:25:30.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"50a4e8ff68725a587bf4b89c589bf0e1352ca479","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 10:49:16.000000000","message":"Patch Set 8: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"354ec68745225cdeb1580cfd3eae18172d9cbc1c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-28 10:49:49.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e3e468b986a9b5ba171622668a0d8f3c23220be3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 12:54:13.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57d5ff8473b446809a66a134b3b1053b : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/251e4f9e3dd44a11aca58399b337b43d : SUCCESS in 7m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1ea1040776d4f3e8857a7be09c7d20c : SUCCESS in 8m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54a925a0eb5c452bab2a3c93d31e1ff4 : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/302ba45372da4a26b2aa89415d06d45b : SUCCESS in 11m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10e09e4e63ba4d1ab881f0fbda54f5fc : SUCCESS in 5m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e084c4bd49d1427eaab3e83f8d7a8436 : SUCCESS in 8m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/881b4d50f33d4a53bb473f5e88e1e0fb : SUCCESS in 16m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/490356f31a0042459f309991ad69978e : SUCCESS in 40m 02s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f9b024d6621f4e69bdb3aca5b6a19572 : FAILURE in 1h 11m 28s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/89974143f364465da6210c3a1726d04a : SUCCESS in 1h 22m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/697b607e5e1c4b0f957930ee61e1d14c : SUCCESS in 53m 40s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0cd01053d25c4f59b2bff81f3b9ae216 : SUCCESS in 1h 10m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1fe93b7c812a447aae1e8bdde9ad7439 : SUCCESS in 26m 57s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/51f9e211f7cd4a50afd79616609b283f : SUCCESS in 22m 58s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fa75a1472fd45778c4e0845fd1d25d5 : SUCCESS in 17m 58s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/08c1f3472f0b469d827c0f44b01f2ddd : SUCCESS in 2h 00m 30s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a568a11033244bc814a8a1b7184e2df : SUCCESS in 1h 21m 19s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e9994004f77b4151bcbb37fdda9c5050 : SUCCESS in 52m 31s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/a89269c4b86341c5a5aa6fd6c2a083a1 : SUCCESS in 37m 51s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09f3d70700074a5db604ccbb3cea2250 : SUCCESS in 39m 59s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b0a9327db6544b17b29cf0d0c13c3368 : SUCCESS in 38m 34s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/473e6e0bc6c74a868a7ed0eea08ef762 : SUCCESS in 31m 39s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/bd5670e0c37f4056a7fb01993f7f3df7 : SUCCESS in 18m 04s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a025aae97a0f4377b72d8b988443e088 : SUCCESS in 17m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8795a78f7ca4feeb5529970c38a767a : SUCCESS in 13m 47s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c2173a1379a145db889f01aaf9308408 : SUCCESS in 20m 43s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/b6de111814ea4ae9afd6830cf240beb7 : SUCCESS in 20m 36s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a73c12073d2b483986149656834fbd9d : SUCCESS in 15m 19s","accounts_in_message":[],"_revision_number":8},{"id":"9ab27eb74cbd2ff01fd9448e46763605c7485048","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-28 13:05:33.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"00d52680bf2a54d07daafc90ff5be47aaae239f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 15:46:19.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26c0a095765b4e09a3836f068227fae9 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2310ea9ffe0a40cb802b4b5b3552be4a : SUCCESS in 8m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f342a41f005f4066ab9a267009c80d2a : SUCCESS in 7m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b7ecb4fc83b4cb98a641cd3e76a4fb9 : SUCCESS in 7m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4465a70c69e4a9d9197bb14435aa51d : SUCCESS in 10m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4c0bb0a35ad47ebb3b25e28bbe9bf31 : SUCCESS in 4m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/75f7e21534524df6a1a0ffaf8af5e9fb : SUCCESS in 8m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/caa43526aa00486aba0bd57892232cc9 : SUCCESS in 14m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/36e4fb3871d34dffbfb0bd2f6a8b9d8e : SUCCESS in 34m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2b37cf704d754599af34ee5b40990ae1 : FAILURE in 1h 29m 41s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/ce32b2204c034121990e0afafbe90d11 : SUCCESS in 1h 05m 58s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afedcf85be8148dbba1febd8ae01bc20 : SUCCESS in 47m 45s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ec74f8e8ff1d422da94bd30e3c35451b : SUCCESS in 1h 11m 11s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/71032e3f52be466e9e5c2b7a80871927 : SUCCESS in 24m 55s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/f68d0cba06c6472ba0b4cfea9d19038c : SUCCESS in 22m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cf1d7f08f8d4cdebda586b46b9008ef : SUCCESS in 15m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ce1077fb09b1498c9882f2cd61304e31 : SUCCESS in 2h 31m 33s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81379566bc634519a17694479464bc7d : SUCCESS in 1h 38m 36s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2e2e621b1ed84207952c31f07d4b6c8f : SUCCESS in 49m 08s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/a2d469aee1a743a2bd147ecca4a10e4f : SUCCESS in 40m 06s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/630d8a7ee30945cabc81e976a07399a9 : SUCCESS in 59m 58s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6b00062575dc4bc1be783c6ad0efb4cc : SUCCESS in 39m 42s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d968429e02c4444993e65dda4590a344 : SUCCESS in 32m 27s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6baf39910cd0479eaf0a3f0e99044b43 : SUCCESS in 18m 28s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c9e1d9607fed45c8aec6ee1441e7f309 : SUCCESS in 16m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1a051151cd34f219a600ea24043f435 : SUCCESS in 13m 10s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/b01009e0262a44a2bb01f369eaf83516 : SUCCESS in 20m 07s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/763fe32fa96f448fadbf320be9212be4 : SUCCESS in 18m 32s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fbdd5c305d848d29150bbd03c62c64a : SUCCESS in 15m 19s","accounts_in_message":[],"_revision_number":8},{"id":"da823f2e225631484986e9ce1bbe408297bafeb3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-28 16:37:08.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"37c8982a11a7fececd69023cbda2307e6f2142e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 18:28:23.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f178d0350774ef2b3baaf17e85d8c58 : SUCCESS in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc405ea5a2eb439d9a749b639439f989 : SUCCESS in 8m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e81a209e54e9415098a3828f679563c8 : SUCCESS in 7m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3830e0b900124b01bd4b2fe3ca0138ac : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6b1cf30fda84a8b9bc3d68358cad0c8 : SUCCESS in 11m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aab0137511214ff99aa70da6c904b442 : SUCCESS in 4m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b2563928ed4d4088a89baca6cac237c3 : SUCCESS in 7m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4b9174be4a634dd78482d9496bea7e70 : SUCCESS in 15m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6fddeae67a264079bf6c83ecdbb2aeba : SUCCESS in 37m 54s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/bbb96561538049c28f7da09d1a566131 : SUCCESS in 1h 06m 16s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/8010582146a946d18e195078e5457903 : SUCCESS in 1h 05m 10s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f0aef0c96054c5ebdaede788fe0ea00 : SUCCESS in 59m 56s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/61d65643d5754884806e537701782784 : SUCCESS in 1h 11m 39s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a0478079db63440e99990380425f24bb : SUCCESS in 23m 39s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/aa891f3c0e9c4140a8b6d86a0abdd0fe : SUCCESS in 23m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bba5414d5a9943dbace3d47953c333e4 : SUCCESS in 15m 09s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fcba3bc208d94adda779cf520a7f10a5 : SUCCESS in 1h 46m 00s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d14d8d2349f241058f758e87e0c6d6f7 : SUCCESS in 1h 31m 49s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/78ed2b28bb074fe389f096831a7c3c6c : FAILURE in 30m 29s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/7d79dfc7c42d4b488a64f1b2e4b6c38c : FAILURE in 26m 35s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96e6bbd73de94851adcacc6450456f79 : FAILURE in 27m 55s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f4b56ddcf5d24b79a9aff783fe7e5555 : FAILURE in 24m 17s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d726c1756813439b80b6895a5afb0ce7 : SUCCESS in 30m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/fbe86ef97f054d83af6a341436d136cc : SUCCESS in 18m 13s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/af4fd382141c42e8b2f0494cfada928f : SUCCESS in 15m 52s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/587140f25cca4b609d3f56d040be2607 : SUCCESS in 13m 11s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/4dac60efac1a4f70b1be615a0ed4ca0e : SUCCESS in 19m 51s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/29271c5376914ba3bb680b6b76bc7683 : SUCCESS in 17m 46s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35fe2a6341a24cd7837f80bbb97a91a2 : SUCCESS in 23m 23s","accounts_in_message":[],"_revision_number":8},{"id":"cef513d3f2718eafaceb71f531d1d6f24c5f7318","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-28 21:10:05.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8bf9a3d8a26300485a7579f365567e7de43195b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 00:12:42.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e4fe41255f24ebd8fa873fba7fdbcc1 : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fad8231946ac474e80186c7d57a8440a : SUCCESS in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73e0bd26eddb4e3aa80572e9ee85d95b : SUCCESS in 7m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7353d578d8894ceebf5722109381a5b7 : SUCCESS in 8m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c34c5abdacba4090b2818522a316146c : SUCCESS in 15m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a80dc68ec71b4dc5b50c4c87f1c48d10 : SUCCESS in 7m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7ab3ff54c83a41ed9488966f1285c98c : SUCCESS in 8m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9765f099a2184089b80f04c876834be6 : SUCCESS in 13m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a7127c53d13a462a9c1605712b89e6ab : SUCCESS in 28m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7bee419b39c348aaaaefadff36658368 : SUCCESS in 1h 07m 15s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/e21dc1b746db459eaf56c15edef1e8c9 : SUCCESS in 1h 14m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a21c2da84a7d46528cfee9f97fc9cae6 : SUCCESS in 57m 04s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ee4eb7c35bf944b6a4e57f54e63cab13 : SUCCESS in 1h 07m 07s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/43c5076f13674c30a08350962bd77b95 : SUCCESS in 23m 47s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/e6cc539ecb16478d80afb43d4808d9fb : SUCCESS in 22m 36s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8557dba2e71f4515859b94dfd1969905 : SUCCESS in 16m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8920871b6817482180c76ce4d96181a2 : SUCCESS in 1h 45m 10s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f04ba8e32bb4ad493715b7082679130 : SUCCESS in 1h 19m 23s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c4d5d5a8ab2c4b31a9bd45c8ca632454 : SUCCESS in 46m 42s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/1651e06aefd6408f84c15fcd19e37f49 : FAILURE in 25m 37s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad994e0f3db34aeaba2db2204244081f : FAILURE in 22m 41s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5aec584ae96444fba723000b97b6a7b5 : FAILURE in 27m 28s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/461abe50f8af4b1c9c88b6bab0dc96af : SUCCESS in 32m 50s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2318d8f7d8dd4224a92879fbc73a95dd : SUCCESS in 18m 27s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/fc2695dc73794ea89093f07e0fa15bf3 : SUCCESS in 16m 52s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9954e4de57e42d4b6ab66a92b1841af : SUCCESS in 14m 08s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/74cc0d1655be41109cda7c45dc3515e9 : SUCCESS in 21m 59s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d7e673ecb1c341f898ad7ef5327e27a7 : SUCCESS in 19m 33s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a1f165ffadc4147848368a157e5a674 : SUCCESS in 15m 28s","accounts_in_message":[],"_revision_number":8},{"id":"572627a4a42cfdc2536530cfdbc7f09110f201c2","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2022-03-29 09:46:30.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5729ac79b0a3e47ca1a544e84851161c0f99625d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 11:19:52.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"58e16b5b758b0bf158d1a829cb73ec24ef75842c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-29 11:29:31.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"da86121a3a23b5303cbe989b427637b963617515","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-29 12:43:02.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d1a9233ba6207b894196c08ce646af81fcd5f32f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-29 12:43:17.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"2d6dd2c131a757049c20f76ecbe96745b59a1844","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-29 13:00:03.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"330728fee8f6c7d5a6a287ab262a5b443035825d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-29 13:46:10.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"123ad904c2567937d4464d5c513d96e96bf0fdaf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-29 13:50:30.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"59bb11a8b3106563ab52922f06dfb989dbbc69e0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-29 14:20:00.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"aa75df6586e52d9711b12efb98a88b3c0172db17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 14:47:21.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9eaa60f2fe34f02b49f8a02c29f8f3a : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20124c00630d4a0fa21eebb4699ace4a : SUCCESS in 8m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c9d5f080bdc4230a239e220c32496f5 : SUCCESS in 7m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1230466a731d4471ae378f3d02c372ee : SUCCESS in 8m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dcb49fdb0c44c2c8c8a05d8985438d6 : SUCCESS in 10m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f115004c8dee421da8f4132b38fe0ebd : SUCCESS in 6m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d91d21a19c204843a167103165945377 : SUCCESS in 8m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/34cc01cdca1b466db687966ab4221ad3 : SUCCESS in 15m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c59ba4dbbc4e465aa8e225b2bcb8a112 : SUCCESS in 35m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fa95d2c3c8fc42e395d7083f5ed23c27 : SUCCESS in 1h 06m 40s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/4427e6e9002b475d84e3eabdd7c96388 : SUCCESS in 1h 05m 51s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61e41f82b0db43d4a1723dbd7d8b3af3 : SUCCESS in 1h 00m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e09a5e09fbd7445d96aa6deb5f5e1d3e : SUCCESS in 1h 05m 22s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bc45c5107e8c40688dfb4bb2bae13005 : SUCCESS in 22m 51s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/c87e09c25f294bf6bba6b99600d097a5 : SUCCESS in 22m 32s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/220fb2266bd84855808fc4ade618c7a6 : SUCCESS in 16m 44s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/295e7f6e837b48cd8291e97c0b857a87 : SUCCESS in 1h 45m 06s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98879650da8244ab9f3581ea5f5e15c3 : SUCCESS in 1h 14m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/904e74538fbd40608a9d88a4e407de95 : SUCCESS in 25m 20s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/b7e36aec7ca241b28544660bfa42fb62 : SUCCESS in 26m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dcfa9d1a00f74d3f81c8d417d30bd861 : SUCCESS in 16m 46s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cee80381494f459aa9739199311a782b : SUCCESS in 39m 47s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/171efd4f3fb543558c0f04a378495c45 : SUCCESS in 27m 36s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4462581bb88d4ee8acded46f14389874 : SUCCESS in 18m 17s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/fb21f7d808f8432ab57fc156c3250af8 : SUCCESS in 17m 30s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b93158d6f6814026b94c6f019d3b5013 : SUCCESS in 13m 35s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/aa743ca5a6704a05b8d08c3341f7d66a : SUCCESS in 19m 22s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/613cd7549045407884689505a30683af : SUCCESS in 19m 38s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7dca0649c984321a22b0bfc789bc4bb : SUCCESS in 16m 19s","accounts_in_message":[],"_revision_number":10},{"id":"1494d78d4c4c3928b1420da7436a87777c815ae1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 14:48:12.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"454cd7056e93217a516e2565e186e8a69ad5c010","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 16:40:12.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b9f3a2a7e8e44da8b816955f322e8e0 : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1be2dfb182a8402eac93816f6e8f7e66 : SUCCESS in 8m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55aa4237211f4ffda942c8ba91a3cb65 : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab5789f1ac2a413cb50a07bc0036b50e : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc218b0c69994e06bd0295448f3966ff : SUCCESS in 5m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b33166d83d3047a4a2ba9b961da41ca0 : SUCCESS in 9m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2e25898e4f334971ab161dc1057eebc8 : SUCCESS in 12m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/342466d079314674812ac3be185201a0 : SUCCESS in 29m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/625021b1fdff44078258604acec6882c : SUCCESS in 1h 09m 12s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/afe117a2b0414533958e08ab1125d3b3 : SUCCESS in 1h 13m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70bb8ab9262244ae8b6cc525ba608009 : SUCCESS in 55m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1c738402e1cb4cddab84550dba8ca079 : SUCCESS in 1h 21m 36s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/82f22b1b4b5545319f1176eb031c4f0d : SUCCESS in 24m 58s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/a7b8c033b3734df5a659673ebf2fdf61 : SUCCESS in 21m 28s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2569fd02c57a4ab69e19404932bd96a8 : SUCCESS in 17m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/248358b077df4229a2102e8bce81221e : SUCCESS in 1h 47m 11s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f75d13f219e240a1b769f9d9305485fe : SUCCESS in 1h 16m 38s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9752090c3ef84d9aaeb1b8d8711743a7 : SUCCESS in 41m 20s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/500f5546d4f844c699edc1b4aa8ec0e8 : SUCCESS in 27m 37s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62d72889cedc44e982741bb9e3fd903e : SUCCESS in 18m 32s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/adcfc015a397407db15d5889487d40da : SUCCESS in 37m 58s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e59ab19f290442f1b70b88ba40ffa508 : SUCCESS in 29m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6036798682914dfb9a9598113f7d656d : SUCCESS in 19m 31s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/bcf9fd9f02444ee89b48ca72fc48bd1f : SUCCESS in 16m 10s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a61f7279d8ec4a28b8f2ca9647927571 : SUCCESS in 14m 21s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/61e65b130cb1475abd3bce0b40f2ef59 : SUCCESS in 19m 17s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e91b610bf90141738b90c48d14c9d9fb : SUCCESS in 17m 32s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab5721d4d025489287631688c64f0054 : SUCCESS in 14m 53s","accounts_in_message":[],"_revision_number":10},{"id":"adec6e7821e17b2a58a75d71855149365f6b492e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 16:40:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"4723a4b0c0c0e59fd10a33b8277e110f5cabc2ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 16:43:01.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0f5f95e57fd40208a6a0fe3e427e8e5 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/016c5de3bf814773a1efbb2104d86f8c : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"c9c0019d7e66b21be82974278311925300acd594","revisions":{"6403035e4726c815bf4c7bd761af6c01e8c825d9":{"kind":"REWORK","_number":1,"created":"2022-03-24 16:34:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/93/835093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/835093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/835093/1"}}},"commit":{"parents":[{"commit":"24b2da7cbc6157299a46c9fa5773514d0ac91187","subject":"Merge \"Ubuntu: add support for Apt configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24b2da7cbc6157299a46c9fa5773514d0ac91187"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-24 16:34:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-24 16:34:01.000000000","tz":0},"subject":"Use jinja2.pass_context instead of contextfilter","message":"Use jinja2.pass_context instead of contextfilter\n\nThe contextfilter decorator was deprecated in jinja2 3.0.0, and has been\ndropped in 3.1.0. This results in the following warning, and failed\nattempts to use filters:\n\n    [WARNING]: Skipping plugin (networks.py) as it seems to be invalid:\n    module \u0027jinja2\u0027 has no attribute \u0027contextfilter\u0027\n\nThis change switches to use the pass_context decorator. The minimum\nversion of Jinja2 is raised to 3 to ensure pass_context is present.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/835092\nChange-Id: If657bf5b0117812d3c53942464cc41cf86cc8ad5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6403035e4726c815bf4c7bd761af6c01e8c825d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6403035e4726c815bf4c7bd761af6c01e8c825d9"}]},"branch":"refs/heads/master"},"51a740af19572f7dbf8980ec8827ddf47f9350c8":{"kind":"REWORK","_number":2,"created":"2022-03-24 17:09:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/93/835093/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/835093/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/835093/2"}}},"commit":{"parents":[{"commit":"24b2da7cbc6157299a46c9fa5773514d0ac91187","subject":"Merge \"Ubuntu: add support for Apt configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24b2da7cbc6157299a46c9fa5773514d0ac91187"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-24 16:34:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-24 17:09:15.000000000","tz":0},"subject":"Use jinja2.pass_context instead of contextfilter","message":"Use jinja2.pass_context instead of contextfilter\n\nThe contextfilter decorator was deprecated in jinja2 3.0.0, and has been\ndropped in 3.1.0. This results in the following warning, and failed\nattempts to use filters:\n\n    [WARNING]: Skipping plugin (networks.py) as it seems to be invalid:\n    module \u0027jinja2\u0027 has no attribute \u0027contextfilter\u0027\n\nThis change switches to use the pass_context decorator. The minimum\nversion of Jinja2 is raised to 3 to ensure pass_context is present.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/835092\nChange-Id: If657bf5b0117812d3c53942464cc41cf86cc8ad5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51a740af19572f7dbf8980ec8827ddf47f9350c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51a740af19572f7dbf8980ec8827ddf47f9350c8"}]},"branch":"refs/heads/master"},"fa4377b5a5231e486a5018bae4cf187bae092d07":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-03-25 19:08:30.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/93/835093/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/835093/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/835093/3"}}},"commit":{"parents":[{"commit":"24b2da7cbc6157299a46c9fa5773514d0ac91187","subject":"Merge \"Ubuntu: add support for Apt configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24b2da7cbc6157299a46c9fa5773514d0ac91187"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-24 16:34:01.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-25 19:08:30.000000000","tz":0},"subject":"Use jinja2.pass_context instead of contextfilter","message":"Use jinja2.pass_context instead of contextfilter\n\nThe contextfilter decorator was deprecated in jinja2 3.0.0, and has been\ndropped in 3.1.0. This results in the following warning, and failed\nattempts to use filters:\n\n    [WARNING]: Skipping plugin (networks.py) as it seems to be invalid:\n    module \u0027jinja2\u0027 has no attribute \u0027contextfilter\u0027\n\nThis change switches to use the pass_context decorator. The minimum\nversion of Jinja2 is raised to 3 to ensure pass_context is present.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/835092\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/835279\nChange-Id: If657bf5b0117812d3c53942464cc41cf86cc8ad5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa4377b5a5231e486a5018bae4cf187bae092d07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa4377b5a5231e486a5018bae4cf187bae092d07"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b2858b614a48f27b9e2445c38d3e860a691279ce":{"kind":"REWORK","_number":4,"created":"2022-03-25 19:32:44.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/93/835093/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/835093/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/835093/4"}}},"commit":{"parents":[{"commit":"24b2da7cbc6157299a46c9fa5773514d0ac91187","subject":"Merge \"Ubuntu: add support for Apt configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24b2da7cbc6157299a46c9fa5773514d0ac91187"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-24 16:34:01.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-25 19:32:44.000000000","tz":0},"subject":"Use jinja2.pass_context instead of contextfilter","message":"Use jinja2.pass_context instead of contextfilter\n\nThe contextfilter decorator was deprecated in jinja2 3.0.0, and has been\ndropped in 3.1.0. This results in the following warning, and failed\nattempts to use filters:\n\n    [WARNING]: Skipping plugin (networks.py) as it seems to be invalid:\n    module \u0027jinja2\u0027 has no attribute \u0027contextfilter\u0027\n\nThis change switches to use the pass_context decorator. The minimum\nversion of Jinja2 is raised to 3 to ensure pass_context is present.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/835092\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/835279\nChange-Id: If657bf5b0117812d3c53942464cc41cf86cc8ad5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2858b614a48f27b9e2445c38d3e860a691279ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b2858b614a48f27b9e2445c38d3e860a691279ce"}]},"branch":"refs/heads/master"},"4b3169d38dfea7196c0f9b9db0ef31c681ac2f43":{"kind":"REWORK","_number":5,"created":"2022-03-25 19:43:07.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/93/835093/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/835093/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/835093/5"}}},"commit":{"parents":[{"commit":"24b2da7cbc6157299a46c9fa5773514d0ac91187","subject":"Merge \"Ubuntu: add support for Apt configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24b2da7cbc6157299a46c9fa5773514d0ac91187"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-24 16:34:01.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-25 19:42:58.000000000","tz":0},"subject":"Use jinja2.pass_context instead of contextfilter","message":"Use jinja2.pass_context instead of contextfilter\n\nThe contextfilter decorator was deprecated in jinja2 3.0.0, and has been\ndropped in 3.1.0. This results in the following warning, and failed\nattempts to use filters:\n\n    [WARNING]: Skipping plugin (networks.py) as it seems to be invalid:\n    module \u0027jinja2\u0027 has no attribute \u0027contextfilter\u0027\n\nThis change switches to use the pass_context decorator. The minimum\nversion of Jinja2 is raised to 3 to ensure pass_context is present.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/835092\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/835279\nChange-Id: If657bf5b0117812d3c53942464cc41cf86cc8ad5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b3169d38dfea7196c0f9b9db0ef31c681ac2f43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b3169d38dfea7196c0f9b9db0ef31c681ac2f43"}]},"branch":"refs/heads/master"},"7e4b98a26ae89175e6f4ba028a322cf1ddf190a8":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-03-25 20:05:27.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/93/835093/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/835093/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/835093/6"}}},"commit":{"parents":[{"commit":"24b2da7cbc6157299a46c9fa5773514d0ac91187","subject":"Merge \"Ubuntu: add support for Apt configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24b2da7cbc6157299a46c9fa5773514d0ac91187"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-24 16:34:01.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-25 20:05:27.000000000","tz":0},"subject":"Use jinja2.pass_context instead of contextfilter","message":"Use jinja2.pass_context instead of contextfilter\n\nThe contextfilter decorator was deprecated in jinja2 3.0.0, and has been\ndropped in 3.1.0. This results in the following warning, and failed\nattempts to use filters:\n\n    [WARNING]: Skipping plugin (networks.py) as it seems to be invalid:\n    module \u0027jinja2\u0027 has no attribute \u0027contextfilter\u0027\n\nThis change switches to use the pass_context decorator. The minimum\nversion of Jinja2 is raised to 3 to ensure pass_context is present.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/835092\nChange-Id: If657bf5b0117812d3c53942464cc41cf86cc8ad5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7e4b98a26ae89175e6f4ba028a322cf1ddf190a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7e4b98a26ae89175e6f4ba028a322cf1ddf190a8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a956eb084b4b7780cfa3d36f55b67a182846b050":{"kind":"REWORK","_number":7,"created":"2022-03-25 20:15:35.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/93/835093/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/835093/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/835093/7"}}},"commit":{"parents":[{"commit":"24b2da7cbc6157299a46c9fa5773514d0ac91187","subject":"Merge \"Ubuntu: add support for Apt configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24b2da7cbc6157299a46c9fa5773514d0ac91187"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-24 16:34:01.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-25 20:15:35.000000000","tz":0},"subject":"Use jinja2.pass_context instead of contextfilter","message":"Use jinja2.pass_context instead of contextfilter\n\nThe contextfilter decorator was deprecated in jinja2 3.0.0, and has been\ndropped in 3.1.0. This results in the following warning, and failed\nattempts to use filters:\n\n    [WARNING]: Skipping plugin (networks.py) as it seems to be invalid:\n    module \u0027jinja2\u0027 has no attribute \u0027contextfilter\u0027\n\nThis change switches to use the pass_context decorator. The minimum\nversion of Jinja2 is raised to 3 to ensure pass_context is present.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/835092\nChange-Id: If657bf5b0117812d3c53942464cc41cf86cc8ad5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a956eb084b4b7780cfa3d36f55b67a182846b050"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a956eb084b4b7780cfa3d36f55b67a182846b050"}]},"branch":"refs/heads/master"},"6196efcebb6fdfbd091b417f1c3e2c461e225822":{"kind":"REWORK","_number":8,"created":"2022-03-25 20:44:46.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/93/835093/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/835093/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/835093/8"}}},"commit":{"parents":[{"commit":"24b2da7cbc6157299a46c9fa5773514d0ac91187","subject":"Merge \"Ubuntu: add support for Apt configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24b2da7cbc6157299a46c9fa5773514d0ac91187"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-24 16:34:01.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-25 20:44:07.000000000","tz":60},"subject":"Use jinja2.pass_context instead of contextfilter","message":"Use jinja2.pass_context instead of contextfilter\n\nThe contextfilter decorator was deprecated in jinja2 3.0.0, and has been\ndropped in 3.1.0. This results in the following warning, and failed\nattempts to use filters:\n\n    [WARNING]: Skipping plugin (networks.py) as it seems to be invalid:\n    module \u0027jinja2\u0027 has no attribute \u0027contextfilter\u0027\n\nThis change switches to use the pass_context decorator. The minimum\nversion of Jinja2 is raised to 3 to ensure pass_context is present.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/835279\nChange-Id: If657bf5b0117812d3c53942464cc41cf86cc8ad5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6196efcebb6fdfbd091b417f1c3e2c461e225822"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6196efcebb6fdfbd091b417f1c3e2c461e225822"}]},"branch":"refs/heads/master"},"f270271c8337c8a54c3f56b2c27687e1e3d39662":{"kind":"REWORK","_number":9,"created":"2022-03-29 11:29:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/93/835093/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/835093/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/835093/9"}}},"commit":{"parents":[{"commit":"24b2da7cbc6157299a46c9fa5773514d0ac91187","subject":"Merge \"Ubuntu: add support for Apt configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24b2da7cbc6157299a46c9fa5773514d0ac91187"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-24 16:34:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 11:29:02.000000000","tz":60},"subject":"Use jinja2.pass_context instead of contextfilter","message":"Use jinja2.pass_context instead of contextfilter\n\nThe contextfilter decorator was deprecated in jinja2 3.0.0, and has been\ndropped in 3.1.0. This results in the following warning, and failed\nattempts to use filters:\n\n    [WARNING]: Skipping plugin (networks.py) as it seems to be invalid:\n    module \u0027jinja2\u0027 has no attribute \u0027contextfilter\u0027\n\nThis change switches to use the pass_context decorator. The minimum\nversion of Jinja2 is raised to 3 to ensure pass_context is present.\n\nThis change also includes some changes to address issues with image\nbuilds in CI, caused by CentOS Scream.\n\n1. disable IPA image builds in seed deploy jobs\n\nIPA image builds will be split out into a separate job. For now, disable\nthem.\n\n2. disable overcloud host image builds in seed deploy jobs\n\nOvercloud host image builds will be split out into a separate job. For\nnow, disable them.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/835279\nChange-Id: If657bf5b0117812d3c53942464cc41cf86cc8ad5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f270271c8337c8a54c3f56b2c27687e1e3d39662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f270271c8337c8a54c3f56b2c27687e1e3d39662"}]},"branch":"refs/heads/master"},"c9c0019d7e66b21be82974278311925300acd594":{"kind":"REWORK","_number":10,"created":"2022-03-29 13:00:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/93/835093/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/835093/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/835093/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/835093/10"}}},"commit":{"parents":[{"commit":"24b2da7cbc6157299a46c9fa5773514d0ac91187","subject":"Merge \"Ubuntu: add support for Apt configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24b2da7cbc6157299a46c9fa5773514d0ac91187"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-24 16:34:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-29 12:59:56.000000000","tz":60},"subject":"Use jinja2.pass_context instead of contextfilter","message":"Use jinja2.pass_context instead of contextfilter\n\nThe contextfilter decorator was deprecated in jinja2 3.0.0, and has been\ndropped in 3.1.0. This results in the following warning, and failed\nattempts to use filters:\n\n    [WARNING]: Skipping plugin (networks.py) as it seems to be invalid:\n    module \u0027jinja2\u0027 has no attribute \u0027contextfilter\u0027\n\nThis change switches to use the pass_context decorator. The minimum\nversion of Jinja2 is raised to 3 to ensure pass_context is present.\n\nThis change also includes some changes to address issues with image\nbuilds in CI, caused by CentOS Scream.\n\n1. disable IPA image builds in seed deploy jobs\n\nIPA image builds will be split out into a separate job. For now, disable\nthem.\n\n2. disable overcloud host image builds in seed deploy jobs\n\nOvercloud host image builds will be split out into a separate job. For\nnow, disable them.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/835279\nChange-Id: If657bf5b0117812d3c53942464cc41cf86cc8ad5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9c0019d7e66b21be82974278311925300acd594"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9c0019d7e66b21be82974278311925300acd594"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
